)]}'
{"id":"openstack%2Fswift~868144","triplet_id":"openstack%2Fswift~stable%2Fussuri~I35cade2c46eb6acb66c064cde75d78173f46864c","project":"openstack/swift","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-01-05 23:31:37.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I35cade2c46eb6acb66c064cde75d78173f46864c","subject":"Extract SwiftHttpProtocol to its own module","status":"MERGED","created":"2022-12-20 00:18:07.000000000","updated":"2023-01-06 18:54:54.000000000","submitted":"2023-01-06 18:53:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868144","meta_rev_id":"2e13f936dcff03aed2179cc1aef4be9754273dee","_number":868144,"virtual_id_number":868144,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-06 18:53:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-06 17:36:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-06 17:36:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-20 02:11:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"afce63fd74aaf576e8f7ff985f28859219819aec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-20 00:18:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71953bff6d79acf04302e5ed9650f84d7acbf6a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 02:11:31.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b15567774814de78ef82019805da90a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f52ee9c7193409386c36e265c641fb8 : SUCCESS in 6m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/9b16f5e705ae4d14a012517194199f32 : SUCCESS in 1h 18m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/898f1ddf5de3480287c7e345bba7fdd9 : SUCCESS in 47m 35s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/948f2f2ed46a4b9480b47a5f973de820 : FAILURE in 2m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/eb12024208444055973a0d4ce503ed1d : FAILURE in 3m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d05f2e5150de4f27927b4ae56bbf1852 : FAILURE in 12m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cafc196c6795466780c92ffa47d13e21 : FAILURE in 13m 27s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/052fbaed237448be9d3bf3031aa93ffa : FAILURE in 11m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/027703d5357a4501ba62a6d9f7df2005 : FAILURE in 13m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/113bd41c2f6541a4983ef2d9c2a7c169 : SUCCESS in 8m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0c8220a1d8594e0db40a911578f3caa6 : SUCCESS in 9m 31s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/114c0dc52e3a4ae28fe55c9db7836700 : SUCCESS in 8m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3604401ccc014fc7977e5105a6ddbf2d : SUCCESS in 8m 42s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/8495511417a542388351e36323825a1b : FAILURE in 14m 04s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d04f85ca87a34690825c10cb46c3f3d1 : SUCCESS in 7m 28s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/de226acc65da47e69ba470bfaede140d : SUCCESS in 10m 36s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/40fdecfcc9934bbba8558ad709b66b3e : SUCCESS in 7m 29s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4ad6d3d119da40ee9f48c6e17fee466f : SUCCESS in 9m 01s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/33144fd6c18e4e84b8b5879efa8e9c73 : SUCCESS in 12m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d9210c0f47e749d1bcb54698fcd2e8fd : RETRY_LIMIT in 3m 37s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e3deb464be20462aa2606924f2741195 : RETRY_LIMIT in 3m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/419219f01d094e1aaa766e5d8cead64d : SUCCESS in 25m 23s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/555abbd79c8c45779e3fcc5cfb306c8f : SUCCESS in 24m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bee3b0d24c0947f496fbf4bd3c19a3da : SUCCESS in 25m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7795abd31af241aaac6cff35b1d4c2b2 : FAILURE in 10m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4a16dd99e1349c5b53a326b03abe532 : FAILURE in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0df05905f8cd476997a2ffc8b3a416ac : RETRY_LIMIT in 5m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d8bb2c806c34531902d1c28ea00ee98 : SUCCESS in 1h 04m 17s","accounts_in_message":[],"_revision_number":1},{"id":"6be8e57a1d31abce007c5830166121d7dfe2902e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-21 04:41:42.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ce312ec38f4ffe7276476baf40a2514f270abbb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 05:39:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad6e90998e734c199b571fba68e87cc2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/984f7eebc1a9425ea36347d22f43a4fe : SUCCESS in 6m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/8f6af931b89d409392fb44290cdee4b6 : SUCCESS in 54m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f341ecfe38ec4dd8a2f4349b11b0a1bf : SUCCESS in 33m 38s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3b7061a1f1384de79ec261b7d4c2bb00 : FAILURE in 3m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2637e0028d524a32a5799a9009cef122 : FAILURE in 3m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2026ddaa8b034c1f956314cbbffad414 : SUCCESS in 11m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/55b3ca6cbec64f3ea92d66a43205de96 : SUCCESS in 12m 22s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8cf709f72f374eb3bba388c548b2c729 : SUCCESS in 13m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/06bc75ac3d1b4473934d301248c3b18b : SUCCESS in 12m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/29ee06c8f540458d8b21c0170a45ed06 : SUCCESS in 9m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cf519683ddd445dc9c7da166aa5b071a : SUCCESS in 11m 14s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/74a2e6cc11744e76a835da943032d2e8 : SUCCESS in 9m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/24871c8efdcd4bcaba3a53471560830f : SUCCESS in 9m 27s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/2440a5c9f16744e5990cdc588e6afa50 : SUCCESS in 14m 32s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/08389190a7b740a1a15edf9a03d6c146 : SUCCESS in 9m 06s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d191028eb91f477d839d88a4aab80c84 : SUCCESS in 11m 29s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/18ffed234bb843dfbdd5cb71cf7d4dc4 : SUCCESS in 9m 58s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/745ab27837084e549587566e54f8dd0c : SUCCESS in 9m 49s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/4d54dc31860246718a5ee4599233f0ed : SUCCESS in 14m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/22360d8b74874c288bbaa33d30056c3f : RETRY_LIMIT in 3m 58s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b3a8e11833a34fdabbbda99a8c3661fa : RETRY_LIMIT in 3m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/00e1e12c7b0945e68396327fd4f8e761 : SUCCESS in 27m 26s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/22aac32c53ef4435b11d44412caed674 : SUCCESS in 30m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9a2c1adb2d154bbd8a1631417df23bde : SUCCESS in 26m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/637ddfa8ef76464a9088e9450bf8f830 : SUCCESS in 13m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2366ab0692d54b54930fffb32064b702 : SUCCESS in 8m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5d57234bce9f4c20af457525bfb4a756 : RETRY_LIMIT in 5m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/60bced2275d247599c28d4c63e5c21a2 : SUCCESS in 48m 01s","accounts_in_message":[],"_revision_number":2},{"id":"98db1e97550e8fef96c86d301502f366a655f4d0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-05 23:31:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"020452266278258d6920204a3d6dcd29e1d181fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 00:31:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/672c9362bebb451db8350f4b7f1ed9a4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e033d3288224c758593a92214a8c2f3 : SUCCESS in 8m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/320016e2f1854643b80041db5b2e8ff4 : SUCCESS in 47m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d5e676910085492a96a7cd61dba1c5ba : SUCCESS in 51m 54s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c3679e614e534eb1b6b48f176a03a61c : FAILURE in 3m 21s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/52374e73da4e4b8caee252ff8b5cb935 : FAILURE in 4m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ef6c2a00977648dba02ef068a65b8f07 : SUCCESS in 12m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0873dab8530a426ea3508cfacbb50f41 : SUCCESS in 14m 13s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/937ecbf34887450dbad1b6e734dec433 : SUCCESS in 12m 18s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0fc8f0b610bf4a9e895eb34b26f2ba1a : SUCCESS in 12m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/19a7bd16045742a2a1f5bc5f4faf46b0 : SUCCESS in 8m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/621d791232e2416c9b9af87fb05deb50 : SUCCESS in 12m 58s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/32bdf9154a2b408594b257855380b50c : SUCCESS in 9m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/565edb7d7b124e1ea1566cf34de04180 : SUCCESS in 10m 58s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/9e7899a9d68b4daca3c7b4cdce87e968 : SUCCESS in 13m 26s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/01971a2cbab142d08170a2782c0bb028 : SUCCESS in 10m 14s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/15290fc7f2674f6f9301002883837466 : SUCCESS in 12m 30s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/f91edfdcce514cc3b65a65a6e5c0c7fb : SUCCESS in 10m 25s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/34eb124f7307438691cea55aaa86a45c : SUCCESS in 9m 47s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/e32749d10218494eb3d4cd4e8a3f9a6f : SUCCESS in 14m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bb9ea6dc8014429ca6056d5291b5ad6a : FAILURE in 7m 00s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a5ea97f860984521abe46225aaafc3d4 : SUCCESS in 52m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c7d70a9be97c4fa9bfa1d150687d543c : SUCCESS in 21m 14s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/f3ff7865fa98489ea8b4e574117d4292 : SUCCESS in 20m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3baf42f61b9547679fba4e5afc03919a : SUCCESS in 21m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1cdc30830a5d41259b56165b2d29a544 : SUCCESS in 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1551127f18ae47ebb237966d13e112e1 : SUCCESS in 5m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/50a7223c43f849578f116531ae08ef30 : RETRY_LIMIT in 5m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aef2883960844210903d65d02d8d027a : SUCCESS in 45m 36s","accounts_in_message":[],"_revision_number":2},{"id":"b8f472e1dc5a3a44d79b9e54d9ed41279bee4027","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-06 17:36:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"894703937444ab3738b760f598bcd899c44412c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 17:37:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c77e1ecac9185b08b0d1eb5c0a8f3b677ffbd76b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 18:53:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2446b24763ea46f39c002269deddc168\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2e7d1ced5754f07827cc32b7e353033 : SUCCESS in 6m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/b040d88a9d9347dfa835189652220752 : SUCCESS in 1h 09m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8a2e8dc10c624c919c5caea54b5a786e : SUCCESS in 33m 25s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/02846f6fb90248ad96aa5688bfc93462 : POST_FAILURE in 4m 44s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/84a60d637b0941cca918a5fb850d48d7 : POST_FAILURE in 4m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f6995e0373eb4f3b943c8c2910e5632d : SUCCESS in 13m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5d2bbb1fef8f4e2282eb9674ac0c02c9 : SUCCESS in 12m 55s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/41d113ebcb0e46588d6b50b6aebd8404 : SUCCESS in 13m 33s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b243d591b610432c8f799d9b34fca4cd : SUCCESS in 13m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4c3412e1f7c44addabb53e1a1286181b : SUCCESS in 9m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a67a474f39354971893713e3ebd2d2c8 : SUCCESS in 13m 33s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/ca27d4d8e90347618a57052d7259191f : SUCCESS in 9m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/65bbc2501e43435eb65ddfad3cf31b6c : SUCCESS in 10m 51s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/fef107b137c942d3ac49c4e0af988a3d : SUCCESS in 15m 35s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0731553e94004cfc98780776ba8f5000 : SUCCESS in 7m 54s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/cab0bcc43aca40338acf523403ed00b6 : SUCCESS in 11m 13s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/034d5f5fab9b4087bae961660265e683 : SUCCESS in 10m 25s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/521385e2243b407cb12da2f36f822d75 : SUCCESS in 16m 12s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/28ce650e4edb40ba8a4d7a00579b8e14 : SUCCESS in 40m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/59273917bc6049558d58109f825eb520 : SUCCESS in 23m 10s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/a32a6514a7324d9f87f52f9934c6654d : SUCCESS in 34m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b05a788de14e428e935e9f51d47e4f9c : SUCCESS in 22m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dfaf159f92f3433ab831a2f3ce418849 : SUCCESS in 13m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e574426e0be54fc4bd719bce249c3f0c : SUCCESS in 4m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b594e74389f6400fa9122d3cd345cd25 : SUCCESS in 56m 03s","accounts_in_message":[],"_revision_number":2},{"id":"3de9dcdf30d77ec415e39af8555e33b57492447f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 18:53:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2e13f936dcff03aed2179cc1aef4be9754273dee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 18:54:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4accb54ca9c43478ac07c480e416df0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/793a94e7db9147faa0f17716f33d5c8b : SUCCESS in 51s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/8dcd6cb2dd9a4a9fb13be89ee3743e62 : FAILURE in 26s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/209dcb9972614a9aa4dfc6d3c443d04c : FAILURE in 31s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3d6bd3fca960763ab74ce677e1edbec578afe966","revisions":{"3798eff03e7c942177af6391cec5826e48e40353":{"kind":"REWORK","_number":1,"created":"2022-12-20 00:18:07.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/44/868144/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/44/868144/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/868144/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/868144/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/868144/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/44/868144/1"}}},"commit":{"parents":[{"commit":"430bb7f5681f80ad96c7d7aca2b40023508d6843","subject":"s3api: Use constant-time string comparisons in check_signature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/430bb7f5681f80ad96c7d7aca2b40023508d6843"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-12-06 19:15:53.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-12-20 00:17:18.000000000","tz":-480},"subject":"Extract SwiftHttpProtocol to its own module","message":"Extract SwiftHttpProtocol to its own module\n\nChange-Id: I35cade2c46eb6acb66c064cde75d78173f46864c\n(cherry picked from commit 597887dedcf1f2c855edfd1c591d4f30222be580)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3798eff03e7c942177af6391cec5826e48e40353"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3798eff03e7c942177af6391cec5826e48e40353"}]},"branch":"refs/heads/stable/ussuri"},"3d6bd3fca960763ab74ce677e1edbec578afe966":{"kind":"REWORK","_number":2,"created":"2022-12-21 04:41:42.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/44/868144/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/44/868144/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/868144/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/868144/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/868144/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/44/868144/2"}}},"commit":{"parents":[{"commit":"430bb7f5681f80ad96c7d7aca2b40023508d6843","subject":"s3api: Use constant-time string comparisons in check_signature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/430bb7f5681f80ad96c7d7aca2b40023508d6843"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-12-06 19:15:53.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2022-12-21 04:41:42.000000000","tz":0},"subject":"Extract SwiftHttpProtocol to its own module","message":"Extract SwiftHttpProtocol to its own module\n\nChange-Id: I35cade2c46eb6acb66c064cde75d78173f46864c\n(cherry picked from commit 597887dedcf1f2c855edfd1c591d4f30222be580)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d6bd3fca960763ab74ce677e1edbec578afe966"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d6bd3fca960763ab74ce677e1edbec578afe966"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
