)]}'
{"id":"openstack%2Fswift~868145","triplet_id":"openstack%2Fswift~stable%2Fussuri~Iae28097668213aa0734837ff21aef83251167d19","project":"openstack/swift","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-01-07 05:45:29.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Iae28097668213aa0734837ff21aef83251167d19","subject":"Inline parse_request from cpython","status":"MERGED","created":"2022-12-20 00:18:07.000000000","updated":"2023-01-07 07:53:17.000000000","submitted":"2023-01-07 07:52:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868145","meta_rev_id":"d21768ca03146821d23b70d0bb9e63b64106f17d","_number":868145,"virtual_id_number":868145,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-07 07:52:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-07 02:43:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-07 02:43:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-20 01:49:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6e68f0da978b0ae255d65d116d083d913009cdc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-20 00:18:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3170198bd156a35e38186f7d5e7ab9c4ee3d54da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 01:49:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/abe8b813d93642668b283b6c0326b97d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4772b60de57a4c398ce2397b8d65ce17 : SUCCESS in 8m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/16e2b732f3ec4f9eb3d2ba5daab2b524 : SUCCESS in 1h 25m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/aba0503194df467fae148be367d226f0 : SUCCESS in 36m 42s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8ecfc9382c96400ab9f3594ac08fa2fc : FAILURE in 3m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/31294305c5604b809e42862c9763a81c : FAILURE in 3m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/75172c36d97c4718b3b7aed21084e1bb : FAILURE in 12m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3d1bea6c0df143f887658810c9c09ec7 : FAILURE in 12m 20s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/caca8b1ef0c744f2bc82e19fae05acc2 : FAILURE in 14m 16s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/364bfcaa6d4248bd8ad5eff05f8b7137 : FAILURE in 10m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4be9d982aa03467390f11c44b3f78afa : SUCCESS in 9m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/554e67078927499ea7f7769a0cf7173d : SUCCESS in 12m 36s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/cd211ead77b0416e8cf0c9be156397bf : SUCCESS in 8m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/58ad7c8542264b878fe01ebb404ee884 : SUCCESS in 11m 26s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/4d4d76dd0d734fd389aac8fca3a42fee : SUCCESS in 14m 22s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ee50a0dde5b744af8d9bab97cebd23b6 : SUCCESS in 8m 22s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/3aededc1c3e94477ba714b35dbe06496 : SUCCESS in 12m 31s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/1c1904eba5624fd298d42dace6bcdd4c : SUCCESS in 9m 15s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0adf2a549cfd4ee388edf6c752cef78b : SUCCESS in 9m 45s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/7f617b76c3e346b5abbeb8dcc9e18fa0 : SUCCESS in 14m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ab84ee3d98044ffc9a84eefe53912811 : RETRY_LIMIT in 4m 05s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ff1647cc143e48c4ac2e376baa7ee1d4 : RETRY_LIMIT in 4m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/561123b4d4104aaab288476787f3e1b2 : SUCCESS in 37m 38s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/67c6651cdb4045a9bbcd10e04385f7c2 : SUCCESS in 31m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/85ac998c4d0444b9ac011e68f4c6e799 : SUCCESS in 26m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1df78f942244437c97d5038dd56bcfdd : FAILURE in 11m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c11ef12f82344bfe82828d1591bbf4bb : FAILURE in 4m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9527e81f639545d0bc6df91aad374a8a : RETRY_LIMIT in 6m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/02cfe104599f4f5bb10d1e50fde21b53 : SUCCESS in 54m 55s\n\nWarning:\n  Comments left for invalid file test/unit/common/test_wsgi.py","accounts_in_message":[],"_revision_number":1},{"id":"195e3257ee6ba8056cda1be39cbe886a4a7b28fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-06 20:13:00.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"29d5f6d0207b6df23f2661a307eff10ec659c3c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 21:33:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c3ca94d23ad4d7dbc43712deac0ed2d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73ffb1c27a86478081bfb8f45e3023b7 : SUCCESS in 7m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/1db0c02eab3847f683e2b730715115a5 : SUCCESS in 1h 19m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c998c7e5cf49435aa6059755fe494c35 : SUCCESS in 48m 50s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e9bb927bd23d498da54e703ad033a811 : FAILURE in 3m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3b4c338a5b094b138fb6e3536dc1c9ca : FAILURE in 3m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6609df794fe34b9084a29b3fdfb3b9f2 : SUCCESS in 13m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/18e08ee2b0cb45e19dfba5142f30e2ad : SUCCESS in 15m 29s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2887e1847cd84dc3b115071004c17511 : SUCCESS in 13m 33s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1050003b49264eff87efbb7041fd129b : SUCCESS in 15m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9edbdd3889e94c798dbce87c66b1888b : SUCCESS in 10m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c62ae6cd6ad4453089d0b58f490c5fc4 : SUCCESS in 11m 51s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/d9135f0100a149f9aea1e2fa60a24428 : SUCCESS in 10m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e184009eefe047b3abd40d6adc9e1b47 : SUCCESS in 11m 14s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/f4a4592608f7454593000be1f3307253 : SUCCESS in 14m 49s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/938f44709dd443a0a9f2f26741e1f219 : SUCCESS in 11m 58s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/baca63bb8e7d4dc4b59cdee678773c71 : SUCCESS in 15m 01s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/888f90de3c1243f1a12a3a9b1cd41760 : SUCCESS in 9m 57s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e1fd8f08741a4a3997bcc5d573903917 : SUCCESS in 10m 53s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/8f9000eb2f6749e8a064e79878bcc89d : SUCCESS in 14m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/11eb44d45a694c66ba4560bf7a3714c6 : FAILURE in 7m 45s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9df4f79cada6469f8d85ad8f426bd7ae : FAILURE in 42m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c671da7e2d7e4f3e83c7953b5735d117 : SUCCESS in 27m 07s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/293def399f3c4c889bf620519f919b24 : SUCCESS in 38m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/31fb5fbe3d984336a9f25a9bfe9325ea : SUCCESS in 38m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4ddd4aa00194b4c9c17eb9372b95d2c : SUCCESS in 12m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34b6138c09ce44d383009d0aabd8a870 : SUCCESS in 4m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/be9701b2cfdf4ddcb3ec01da7c6823a2 : RETRY_LIMIT in 7m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/382986d1884f4ace807f2a1733f8e603 : SUCCESS in 48m 42s","accounts_in_message":[],"_revision_number":2},{"id":"9cb59d7c9de0143def0ec0e47a80c555a7119143","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-06 21:39:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2c25f747fbfbf297cd00a82aacecd3f63748a12c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 22:36:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dcc193e725a4c2293dc785e2193ff3a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4607196459f4fbeb676dacc54e5d061 : SUCCESS in 8m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/3a61e49bd06b4fda97265eae090d75c3 : SUCCESS in 52m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f4ff465a5dca4eccbf22f3f9b1bbf3b6 : SUCCESS in 36m 42s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6d0336d8c4274d8295f977f557f0f508 : FAILURE in 3m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7ed8b8705c5b4925a2a106b3215df554 : FAILURE in 4m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f19dd7efa4594afcbbb027e8c55967b4 : SUCCESS in 17m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d58eec9278e64b4f8cd9f96e88049e22 : SUCCESS in 13m 27s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/38e085e64874475aaf1b8184604ddf8f : SUCCESS in 13m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5a7348790cd34784b20267a35696537a : SUCCESS in 12m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e5e26a8260f34ba8bd340f1ff43c3273 : SUCCESS in 14m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bbb424a6be654ceba45389bc712d4b35 : SUCCESS in 11m 28s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/f7686d62efb74a5cbad7553b2b48f0ca : SUCCESS in 10m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9e4c280c385147af8a64c3567d1ea5c5 : SUCCESS in 9m 43s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/e72990d2658745528c22f991a4ab1e03 : SUCCESS in 20m 48s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d995f0b6583c4b46b8a25ff49b501877 : SUCCESS in 9m 51s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/0b5d8d2b29f04e3f86a52f17b497eec5 : SUCCESS in 12m 46s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/9606ad17235d4a1aabefbd90f137cdb3 : SUCCESS in 14m 25s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8983caeff22d4736bbcff9a93e77cef3 : SUCCESS in 9m 52s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/847b51534f5d4a8095cad29706eda396 : SUCCESS in 15m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8eff32bd7d17405a882f3828a92d7cc2 : FAILURE in 7m 40s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/90021523d7d2469cb507637aa6c52b03 : SUCCESS in 39m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a76b7cd83dae43febf6596e88104bf16 : SUCCESS in 31m 59s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/b2fc9390e2f64620ace0543c9bb8d833 : SUCCESS in 31m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b866de25161d493f82f7ca02d939f259 : SUCCESS in 27m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8bdece35b03f416d93099858694c82fc : SUCCESS in 13m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/749388e66d584da6a289aa686f7685e7 : SUCCESS in 5m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/afb8e63142a842769bea913d3550c586 : RETRY_LIMIT in 5m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ae2774bae3645869589f4476ee03288 : SUCCESS in 55m 33s","accounts_in_message":[],"_revision_number":2},{"id":"07240bfc0ec8a0151724c75fead9c4cf12a758a0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-07 02:43:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a21318c0170cabc21ea04a8991a208b05324b9be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 02:44:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6d8ca31c6e06fe22e401a675f16b8cbb8df81740","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 03:48:07.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/14bebc10c1b9492484e09d4c3083efcd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab3cc77c19f44fcb933afc39555929d2 : SUCCESS in 8m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/47db7910398a4d89acf04f04636c84fd : SUCCESS in 1h 00m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6c45373642174e19b0138bad4ef68501 : SUCCESS in 35m 26s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/cb6757e9b4dd4be8af6e0a3da76a122d : POST_FAILURE in 5m 33s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/f46d147ce6b944cfa7934b5a1615a9b2 : POST_FAILURE in 5m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6ed4e0025d1242e0a34ff3bb020fec4b : SUCCESS in 9m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/02481984f9bf44e5ab9792243b539127 : SUCCESS in 11m 40s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/eac990baf2b741a3bd1e860c9103dc49 : SUCCESS in 13m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/432df35f54144e53a1c2b4eaf2ed850b : SUCCESS in 10m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/444329e0d7eb45de8fc5fb45ef131d2f : SUCCESS in 9m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7db151fcc5da4565ae314f3421caf6ca : SUCCESS in 12m 11s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/767a03992b6f4e69a2df668449332ec3 : SUCCESS in 10m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8236a9c3f6c84413a7139fbeb1eec2ce : SUCCESS in 10m 18s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/35635bd8c6684397a3cf3aa3dd42aa79 : SUCCESS in 15m 24s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3fc1685acf804c5baad92eb504d88539 : SUCCESS in 10m 53s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/0a28159ec46c44b3a9cd97a164efe8da : SUCCESS in 10m 48s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c256564f6aef4746b187a8592320d78c : POST_FAILURE in 11m 38s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/16d13d466f1a4f1c9639f1fc14be0c6e : SUCCESS in 13m 19s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/063b5d33fc60421abb3aadba4eee5d52 : SUCCESS in 45m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/34373be4e2ae4b9da697311a7b0a8e0c : SUCCESS in 32m 39s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/326585f92ff144dc83cc92b944ac5560 : SUCCESS in 32m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/832fd897c02b4c0f930b733010d67597 : SUCCESS in 31m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a2faa521d804a0893857a3e8fcd6a53 : SUCCESS in 12m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8affa552a3f49168dfee3d107db865f : SUCCESS in 4m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b773528733445118cbc05149a52b78a : SUCCESS in 56m 16s","accounts_in_message":[],"_revision_number":2},{"id":"466dbfb466c07141d6c7fdfc2f8d54ebf7c9a698","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-07 05:45:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"35ec51c5c3824f417f24b1a8a7f4e9f4f6c58d51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 06:47:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20d1ada3c4354d349aca9b1bec2c2744\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8669498033b846b8827fbcd3664627ca : SUCCESS in 7m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/af9d5ae153934cd19213603db2a21ab2 : SUCCESS in 53m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/35f7ce8a991c4eaa8fafad96ec529fc8 : SUCCESS in 54m 10s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/084520173b84474ea966c6ab3904e395 : FAILURE in 3m 06s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/11d8d27f794b4c968526909b7c72101a : FAILURE in 3m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8c9d975b397d495190d5f8a58a1569ba : SUCCESS in 10m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/69b202466ac046c9b5b53900c66246b6 : SUCCESS in 13m 33s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/74eb567d6410450bab3910c477fada12 : SUCCESS in 13m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/19e3f2364a8e4a239c88628c10dd3368 : SUCCESS in 14m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c91cd4e1765f4bb6ad656c469c5f214b : SUCCESS in 7m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3fa84b109e834271b0b72f36f9c34ee6 : SUCCESS in 11m 53s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/fd63760e244b49a6bb694a35b3c8b5c7 : SUCCESS in 8m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a437c2d051c74394bcb257932c680f78 : SUCCESS in 9m 50s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/1d107d5f6baa4a8081477e62f3faa810 : SUCCESS in 16m 10s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7d67877e5db6436bb18f0bbdb548b699 : SUCCESS in 10m 23s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/aad7a50c09a24ab78cc94f694c83bd40 : SUCCESS in 9m 52s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/4270bac4e37442b8a3d707243a3f5e4a : SUCCESS in 9m 45s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/996af7d42f014d0eac91f898ea3fd2bd : SUCCESS in 10m 29s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/c2afdc5a1640453d964532271d23ee96 : SUCCESS in 12m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fa5163a2e04e428084dc8a80d7528fa9 : FAILURE in 5m 50s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8d4470aa003048348b24523177c76dd6 : SUCCESS in 34m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b482683f8fde4e779bbf6703d9f6e003 : SUCCESS in 22m 54s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/7b7a1aaa6b5a44c18e4427a62dc83f2e : SUCCESS in 20m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/89c098aa671a40b3aa2810994fd4f689 : SUCCESS in 22m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3924a04b1eee4d509a3b40a9d90ed3cd : SUCCESS in 13m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd0d207a30d1476bb974b39a8c2bd708 : SUCCESS in 5m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/28825a0feecf431fbe2d9017f44e2e97 : RETRY_LIMIT in 6m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96db4f5fe27840298128d19ef1d68310 : SUCCESS in 53m 56s","accounts_in_message":[],"_revision_number":2},{"id":"0428dfc4976418b6d5b3215e77a4107b90aaf9df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 06:48:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"522a8db758df2ce2fa7e7929d84da6a383010b9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 07:52:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/101c1b90cab74429a6ff8d60f03b007b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16b994c5a807477f8271cd5dca074d45 : SUCCESS in 7m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/5b22f3dcc7d34ff9893f9003f6c0486b : SUCCESS in 48m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/05e2dd93232d45199e10e875532e4413 : SUCCESS in 30m 10s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/05bafb708e6b49d0a3d38045b88a972e : POST_FAILURE in 4m 57s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/27829ea4f4b24c34a79f17864482dcd3 : POST_FAILURE in 5m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f791dd556b694638ab85536732990b0b : SUCCESS in 11m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/434e4bed2e1b420298a88703490bbd0c : SUCCESS in 11m 19s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a6e5f4f26bf9431cafce4c5363993215 : SUCCESS in 11m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3f8e9992c5d64f81b7151378f2336f0d : SUCCESS in 11m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3f877ea31ec9433eb7675475f6545f91 : SUCCESS in 8m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/05a7790a693b4660902ded5288f2b21d : SUCCESS in 11m 21s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/92a39914f2eb45d89c789282634a82d1 : SUCCESS in 8m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/98026a9ede254f17b7da50d88b9d2fda : SUCCESS in 9m 48s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/79a8c0b362d54473b776f9ae1c0f2da4 : SUCCESS in 13m 35s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ee86d48d1c6f4be5b519fbe61522614e : SUCCESS in 9m 36s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/6dd5024d044d48de89532a3bca32c566 : SUCCESS in 10m 36s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/bf4022d9391d49e99bf72728b80caba9 : SUCCESS in 11m 35s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/f4cdd626ae9b4066a070234197f53c83 : SUCCESS in 16m 14s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bec7c9db4b404e44bdcb0dd73e700a32 : SUCCESS in 37m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f67b6efc561c416785ec82ffd60f51e8 : SUCCESS in 36m 28s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/216a5d8e98574c6b8853d5a83f00f801 : SUCCESS in 36m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e3ef26db4500456a95bdd1b9c2020910 : SUCCESS in 20m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/acf7136e8d554d8e81ddbbba0d5931c8 : SUCCESS in 13m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/830a2d8c769341beb2b1bdcc5f9b5808 : SUCCESS in 4m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c9f9d2aed2064ff0828cb78ffb4ca9c2 : SUCCESS in 59m 46s","accounts_in_message":[],"_revision_number":2},{"id":"c498f087f686a3d700fc6c2eec6c7e87957b902f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 07:52:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d21768ca03146821d23b70d0bb9e63b64106f17d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 07:53:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66e8d092e398407b83371a0296fbf6c3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5c7c4077547498d8c175571a6c6957f : SUCCESS in 42s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/099dfd6c1e7d45bbba6db34cd161bde6 : FAILURE in 32s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/d8870421e9f24ca18c85db358209627e : FAILURE in 28s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"28d6cf1ba8808b34e8b80728238a61f0b7000054","revisions":{"b49fd32b649a938b11080d5644c6c3090bc7a614":{"kind":"REWORK","_number":1,"created":"2022-12-20 00:18:07.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/45/868145/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/45/868145/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/868145/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/868145/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/868145/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/45/868145/1"}}},"commit":{"parents":[{"commit":"3798eff03e7c942177af6391cec5826e48e40353","subject":"Extract SwiftHttpProtocol to its own module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3798eff03e7c942177af6391cec5826e48e40353"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-12-06 19:21:37.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-12-20 00:17:22.000000000","tz":-480},"subject":"Inline parse_request from cpython","message":"Inline parse_request from cpython\n\nApplied deltas:\n\n- Fix http.client references\n- Inline HTTPStatus codes\n- Address request line splitting (https://bugs.python.org/issue33973)\n- Special-case py2 header-parsing\n- Address multiple leading slashes in request path\n  (https://github.com/python/cpython/issues/99220)\n\nCloses-Bug: #1999278\nChange-Id: Iae28097668213aa0734837ff21aef83251167d19\n(cherry picked from commit 884f5538f8fb187b6ff18316249f1bd4b97b0952)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b49fd32b649a938b11080d5644c6c3090bc7a614"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b49fd32b649a938b11080d5644c6c3090bc7a614"}]},"branch":"refs/heads/stable/ussuri"},"28d6cf1ba8808b34e8b80728238a61f0b7000054":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-01-06 20:13:00.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/45/868145/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/45/868145/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/868145/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/868145/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/868145/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/45/868145/2"}}},"commit":{"parents":[{"commit":"76cdb4d3e9048ffcc127df79ccef7964fd3e6dd8","subject":"Merge \"Extract SwiftHttpProtocol to its own module\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/76cdb4d3e9048ffcc127df79ccef7964fd3e6dd8"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-12-06 19:21:37.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-06 20:12:37.000000000","tz":-480},"subject":"Inline parse_request from cpython","message":"Inline parse_request from cpython\n\nApplied deltas:\n\n- Fix http.client references\n- Inline HTTPStatus codes\n- Address request line splitting (https://bugs.python.org/issue33973)\n- Special-case py2 header-parsing\n- Address multiple leading slashes in request path\n  (https://github.com/python/cpython/issues/99220)\n\nCloses-Bug: #1999278\nChange-Id: Iae28097668213aa0734837ff21aef83251167d19\n(cherry picked from commit 884f5538f8fb187b6ff18316249f1bd4b97b0952)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28d6cf1ba8808b34e8b80728238a61f0b7000054"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28d6cf1ba8808b34e8b80728238a61f0b7000054"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
