)]}'
{"id":"openstack%2Fswift~868934","triplet_id":"openstack%2Fswift~master~I0e76809586f05a85c11fc4f3078263bbc2fe7cc3","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-03-10 21:09:44.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I0e76809586f05a85c11fc4f3078263bbc2fe7cc3","subject":"Let zuul use new tox if it wants","status":"ABANDONED","created":"2022-12-30 22:15:08.000000000","updated":"2026-03-10 21:09:44.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ca84a70813ae227d4271efd9e0e0554becc48e4f","_number":868934,"virtual_id_number":868934,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-12-30 23:36:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-30 23:24:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-30 23:36:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-31 06:15:45.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"ba49fa86a638acdb1c38661754f8a3b688b6ad6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-12-30 22:15:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8dc89e598627cf684381e83f52bd69ce526b0775","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-30 23:24:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a072ceec9664961a697fb14b3fbebf1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/502d225f86084d8489b89a72c450a679 : SUCCESS in 1h 06m 51s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3fbfa7bc0d81425ba480b60aaeda00c1 : SUCCESS in 41m 20s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7454866cf19f4ed888206c0dbc80bc5d : SUCCESS in 1h 02m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d66e7ec505554f6bbf853a9a9c3315ed : SUCCESS in 57m 41s","accounts_in_message":[],"_revision_number":1},{"id":"e76fcf95e9101443e6be508810bf3486b112d3ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-30 23:36:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e952283d3978413c9d059c73f60cfeb9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2086ad2af6664faf9aa735c287c1e1f1 : SUCCESS in 8m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/a91ae3b5491b40348e8d121cb13f0d9a : SUCCESS in 1h 00m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e40b86b7f57f4e68ac80cf3c25abc937 : SUCCESS in 48m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f89e987668504133b8cbcb5f7fef88af : SUCCESS in 43m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b239f338980849afa526b201e571514e : SUCCESS in 16m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c44b75047f4e49cc98c7698da33aaa48 : SUCCESS in 19m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d8dab012efe4e3eb6ed984ef669e406 : SUCCESS in 17m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1efe136263754973a17aecb0c70ce7ae : SUCCESS in 7m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9adc72dd01844ad69b869a44e3dd5bc4 : SUCCESS in 4m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/81bc9df4ae884d27846ddc784b46b327 : SUCCESS in 11m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ef080a10ce674334af372084ee640689 : SUCCESS in 16m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4955288b2926444db0ef85d44fc7f8da : SUCCESS in 10m 33s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1bfb0b517a1d4834bbb1ab6b1dc46727 : SUCCESS in 16m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/76c7fd3f3efd408a9e83f12c0835bf96 : SUCCESS in 14m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0f3198be7e79421ebf22575eec5a7af8 : SUCCESS in 12m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cf6080e0a00f4598bdd826222a2b87fc : SUCCESS in 19m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/541653a0e167406ba9766d9eab65432c : SUCCESS in 13m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3736084b3f074a40a36fc1c29c5edd47 : SUCCESS in 12m 09s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c90dcc4fc375420c8a64e6277b99f1df : SUCCESS in 16m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7b90f586d9b049b49a90d8332272a49a : SUCCESS in 6m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c64248e012fe495cb541df81429d0614 : SUCCESS in 17m 14s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b898581e7d6243168e839066c240cab2 : SUCCESS in 6m 21s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/766c0c38f9404535bad29683a1cb5766 : SUCCESS in 1h 16m 17s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f3fe92449e7347a2b8463ded723989ea : SUCCESS in 1h 12m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b159e70bea6d456fabb0a65b5c2ad9b4 : SUCCESS in 35m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cdf8bc894acc4264a788b4f58be9dc2f : SUCCESS in 27m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d7bc4a783d54ed790b6865c58a57539 : TIMED_OUT in 40m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10135cbeb8b3494184a48611eda0be6c : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/172118cc54f24dba9d2f6c3554de021f : FAILURE in 22m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0067903134a84e79871e628878ce2bdf : SUCCESS in 31m 15s","accounts_in_message":[],"_revision_number":1},{"id":"dfee4d6c78c97d4c5ab3983f79beba68ee3fd91f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-31 06:15:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ca84a70813ae227d4271efd9e0e0554becc48e4f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-10 21:09:44.000000000","message":"Abandoned\n\nThis pin got removed in https://review.opendev.org/c/openstack/swift/+/881035 and is likely moot since at least https://review.opendev.org/c/openstack/swift/+/949684.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3fd0a8f6dd0ac5905f1fc946c1082a63dc27e79d","revisions":{"3fd0a8f6dd0ac5905f1fc946c1082a63dc27e79d":{"kind":"REWORK","_number":1,"created":"2022-12-30 22:15:08.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/34/868934/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/34/868934/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/868934/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/868934/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/868934/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/34/868934/1"}}},"commit":{"parents":[{"commit":"2d7c1dc6dd8a672c1f89a46b2423898d33ff63d2","subject":"CI: Fix our usage of tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2d7c1dc6dd8a672c1f89a46b2423898d33ff63d2"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-12-30 22:14:14.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-12-30 22:14:14.000000000","tz":-360},"subject":"Let zuul use new tox if it wants","message":"Let zuul use new tox if it wants\n\nChange-Id: I0e76809586f05a85c11fc4f3078263bbc2fe7cc3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3fd0a8f6dd0ac5905f1fc946c1082a63dc27e79d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3fd0a8f6dd0ac5905f1fc946c1082a63dc27e79d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
