)]}'
{"id":"openstack%2Fswift~869132","triplet_id":"openstack%2Fswift~master~I84dc4dd72d88b70c0dc641767527eb4fadb2403e","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-01-04 20:29:02.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I84dc4dd72d88b70c0dc641767527eb4fadb2403e","subject":"Simplify our warning suppressions","status":"ABANDONED","created":"2023-01-03 23:22:28.000000000","updated":"2023-01-04 20:29:02.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e4d20baaf6da76f34d65df314048a74130beaf6d","_number":869132,"virtual_id_number":869132,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-01-04 01:01:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2023-01-04 00:19:32.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-03 23:30:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-04 00:19:32.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2023-01-04 01:01:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a5c6da9eb04ae9b6d84c478a67fdd54d6554edab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-03 23:22:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9195d6ac03e7f610395be21849874caee72bf6a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 23:30:52.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37d8eb0f4db3457fb8897c6ba2e5a231\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4cc04fb076914643b340aba3661f60cb : FAILURE in 6m 14s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/32a2102b055940349a507540f05b1de2 : FAILURE in 6m 26s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/917ba0a5cf01437b8190212ae9670826 : FAILURE in 6m 03s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/061d32cf438a4d03be1d9a7e1e45b51c : FAILURE in 6m 12s","accounts_in_message":[],"_revision_number":1},{"id":"32e191956a3427fe0d4a59f37fd354c49ee8b7de","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-01-04 00:19:32.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a9968745c3189aba0600736be60c0bbdf6ce31cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 01:01:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f22c0de8762b4b089c134bf6450220f4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96b9c56dca2648d09af662afa99ee7e6 : FAILURE in 4m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/75b8c570f40643679daa97ba15ce7de3 : SUCCESS in 1h 04m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/70639f36080248e18c65aefe609a3689 : SUCCESS in 38m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d822573d96d54877860c2b8da12e2a1b : SUCCESS in 43m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e21b6ad6374148658b331fc1cffc4e6d : FAILURE in 5m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/900e5770e5994d1b951a7e885bb96944 : FAILURE in 5m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/17865568bdeb45978bea3917cec08da4 : FAILURE in 5m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/37c53beba8fa40baa4e54320941ff022 : SUCCESS in 9m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cc8c4fb960f04c06a32179d749fb44a8 : SUCCESS in 4m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f0d8dc21eeb54055bfa2f268ba40ad59 : SUCCESS in 11m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/94eb360cc67945638c3d6cb4b013d3e0 : SUCCESS in 11m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/59e5dece5fed45c49e0c10bd0bf5c3c9 : FAILURE in 2m 55s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/67ff5f2853b6430f9f5f15a2b747b930 : FAILURE in 2m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/06c12d19bd0b44649689ec268d285868 : SUCCESS in 13m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/29862da82a1d451faea3be9a463553b0 : SUCCESS in 18m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cfc1152b40ba4027b2bc49cffe537d1f : SUCCESS in 17m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/15e7a0a98e0f431ba70c8b276fd5c1a4 : FAILURE in 3m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/67239e34d30a485d9c1b64a29eb434b3 : FAILURE in 2m 42s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6f198fafcaeb4a3eb795099e0d725e5f : FAILURE in 2m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dd27fa588db04736bb3c993e8bf44643 : SUCCESS in 6m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/118956f81ff344fd924f3fcc669eccd2 : SUCCESS in 17m 55s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1245f1f6ceb14493ad86028b1cd4baf8 : SUCCESS in 7m 32s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/98ea0849716f453380639d441b678d7d : SUCCESS in 1h 27m 49s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7ce80e87311f4f4198e7a78bb252e1f3 : SUCCESS in 1h 31m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2d755276a9bd44b593f8eee7ac575d83 : FAILURE in 20m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8477c15deda4446ba0b9c6053abaa451 : FAILURE in 23m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/376366cf4009469690048fd6119ca1cb : SUCCESS in 16m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74b0c1bb075846f4a694227468ea2605 : FAILURE in 3m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/058e0bdb78034f08981e505fba8ca429 : FAILURE in 18m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a481c37b8cfc4257b5be53acc285c078 : SUCCESS in 52m 53s","accounts_in_message":[],"_revision_number":1},{"id":"e4d20baaf6da76f34d65df314048a74130beaf6d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-04 20:29:02.000000000","message":"Abandoned\n\nNeeded this to pass probe tests, but needed to pin tox to get everything else to pass. Did both as https://review.opendev.org/c/openstack/swift/+/869135","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"192e6362b46db3bf2f44ca0dee6a7f409c1b3f47","revisions":{"192e6362b46db3bf2f44ca0dee6a7f409c1b3f47":{"kind":"REWORK","_number":1,"created":"2023-01-03 23:22:28.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/32/869132/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/869132/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/869132/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/869132/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/869132/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/869132/1"}}},"commit":{"parents":[{"commit":"2d7c1dc6dd8a672c1f89a46b2423898d33ff63d2","subject":"CI: Fix our usage of tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2d7c1dc6dd8a672c1f89a46b2423898d33ff63d2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-03 23:21:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-03 23:21:13.000000000","tz":-480},"subject":"Simplify our warning suppressions","message":"Simplify our warning suppressions\n\nThe message filter is a regex, so any prefix of the message will\nsuffice. This allows us to also drop a new message seen on CentOS 8:\n\n   CryptographyDeprecationWarning: Python 3.6 is no longer\n   supported by the Python core team. Therefore, support for\n   it is deprecated in cryptography. The next release of\n   cryptography (40.0) will be the last to support Python 3.6.\n\nChange-Id: I84dc4dd72d88b70c0dc641767527eb4fadb2403e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/192e6362b46db3bf2f44ca0dee6a7f409c1b3f47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/192e6362b46db3bf2f44ca0dee6a7f409c1b3f47"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
