)]}'
{"id":"openstack%2Fswift~869135","triplet_id":"openstack%2Fswift~master~I316170442c67c1b4a5b87f9a1168cc04ca2417b8","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-01-04 20:28:13.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I316170442c67c1b4a5b87f9a1168cc04ca2417b8","subject":"CI: pin tox at the project level","status":"MERGED","created":"2023-01-04 01:33:48.000000000","updated":"2023-01-04 21:50:54.000000000","submitted":"2023-01-04 21:49:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869135","meta_rev_id":"c1988ace76073794028aa36218e5bd20efde0524","_number":869135,"virtual_id_number":869135,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-04 21:49:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-04 20:28:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-04 20:28:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-04 02:50:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-04 03:39:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-04 20:28:13.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"}],"messages":[{"id":"a889855a062fa9f0fa513ffe7d96d0cb892b9b64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-04 01:33:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"448ff90a6dab344a43bb4b8fbf7d4b6a27c1a1b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 02:50:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/447abb9b539742589eaaab5e66bd0f4a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e621869d30524fbb9e9ce00626c84614 : SUCCESS in 50m 29s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/970e071ad1ef4f4ab61dd1236eded3a4 : SUCCESS in 1h 01m 22s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f324b3bb35bd486081a67dd03f33309e : SUCCESS in 1h 14m 47s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/902dbd7c77354b41b290ac97ce4f206a : SUCCESS in 1h 01m 42s","accounts_in_message":[],"_revision_number":1},{"id":"9af5d7721132391a4567db6f27ccbdba7a73a597","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 03:39:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9fb6cf48b5844bd4ae8b498a68555561\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8a6071e3731a44d48ee5d34a1a124bbd : SUCCESS in 1m 54s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7299535b8a164c4b831cbff0bfcae0fe : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/444096bfa1fa44cea1fb2ab7f2bf8045 : SUCCESS in 6m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ef5254c958354cf79c0d9edd9f8505a1 : SUCCESS in 2m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f01e228aca574b31bab402c1dc24f6dc : SUCCESS in 4m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/a91f37118f1c417cb0879d9f140c5aa5 : SUCCESS in 47m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/57f00691904b44188d2aa49867d13b57 : SUCCESS in 38m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/492a5b1f61254bb8a1a7e320356cf9fe : SUCCESS in 43m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/09ab93af8ba4427b9456c1a350525c9e : SUCCESS in 15m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3ac5b42c683c4eb0bf8492471aca427d : SUCCESS in 19m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/75da61a65127481ebcee88d25f96c672 : SUCCESS in 16m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/978a301873ef40099937bee2c4938d57 : SUCCESS in 7m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5035813e48864152b7f9b85e2515d636 : SUCCESS in 4m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/22a9876148dc435c8e774838365bf125 : SUCCESS in 13m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d67962cd95744d03a7d5ec203dabce20 : SUCCESS in 11m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/92a442c5d273404ba03fba213f975a7f : SUCCESS in 13m 31s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5485b19b318945b89e2855636acfcbc0 : SUCCESS in 16m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e5686269c29c42b8b503efdafcd3e4c2 : SUCCESS in 16m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/aa9e7c907c704540878674d1fe779297 : SUCCESS in 15m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/10662912378846f39843559b18ef2259 : SUCCESS in 15m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0816cd718fde48fda6693accf8dad868 : SUCCESS in 14m 53s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4fdd45777d594a82824b9ad2d7611aa2 : SUCCESS in 16m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ce087df11d214cc0a0ac67c938e02b47 : SUCCESS in 21m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2e8c7be1640243a6a99ff4896b4c97de : SUCCESS in 6m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b4ce9c723220493a8b270a109ade938b : SUCCESS in 17m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3eb0946c87de44dda648e217cb41f1d1 : SUCCESS in 7m 03s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8a0738a0daf143d989a3ad229fd33942 : SUCCESS in 1h 21m 53s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/8ec3bf5dbb1d4e2bb448cc6bf6897264 : TIMED_OUT in 2h 00m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/263d0f90aa4d440cb40016444d66c26f : SUCCESS in 49m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/17c0fc4a54af445b9ed45974b6ef1061 : SUCCESS in 36m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/220a71f9a14a43e483dceb802d697b85 : SUCCESS in 23m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a15db383c764033ab9ed224312037b1 : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/952c15e551db4823a325e461531fbc9e : FAILURE in 29m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cdf28df3a4834cdb868b4b935852864f : FAILURE in 56m 37s","accounts_in_message":[],"_revision_number":1},{"id":"6c6b40d020694f4fda3a99ed003c7c5c3b1d2313","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-04 18:53:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fe87fd877e67148325159ce02ab32f39d54f58d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 19:50:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f34b7ef9f9a8487182ad51f534a45bde\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/47d16adb333f49bcb7e6011ea4bc2c60 : SUCCESS in 55m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/af2f3c3a25c64d03839a7db040c5ccf1 : SUCCESS in 31m 24s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8941a921d6d5451c96a0c009d5275703 : SUCCESS in 30m 05s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/296418321ccd434bb4b3b447c7194440 : SUCCESS in 29m 14s","accounts_in_message":[],"_revision_number":2},{"id":"873f6b34ca000cb96d0fb576b5417b7df5e75c49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 20:25:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bf97a981aa240ca87d9d73f0558ac09\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d4a1fc4757d04f7d8ba9a0ba7e4ab85e : SUCCESS in 2m 42s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9f6fcefc0e7643c983a3e59fcb826dd3 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b83eb8ddf2d440c9322563293a38b2a : SUCCESS in 9m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b42361c769b84817b8fc22614f06c074 : SUCCESS in 2m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f965cff85e034c38ba05098823899cec : SUCCESS in 5m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/e697ade37b4b42ee81618f3343b662fc : SUCCESS in 48m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c805367542574edebee10f9bfc2a9626 : SUCCESS in 41m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e987801bc78c41e99ad6b719c702317d : SUCCESS in 49m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0cd38de86f1944e2beb3604afe29999c : SUCCESS in 29m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/67db10a9983e41bda94936d384266f42 : SUCCESS in 29m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df7bdc82800e45b09ea98ccfe3b4fd33 : SUCCESS in 18m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a90a0bc85f804b69890220de68627592 : SUCCESS in 9m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/64a32f7d025c45c3b77d1c6c940cb216 : SUCCESS in 5m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f9200f84387444ffb645451c0d9bd7f8 : SUCCESS in 27m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/227fff08657e4c53828f459625065fca : SUCCESS in 15m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/592c2da36e2946cebc4fdf5eb316c7f0 : SUCCESS in 17m 37s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/37198da7383047b68e9e32d3e46b8767 : SUCCESS in 15m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d060efdd6462424cbe023541c59e8f39 : SUCCESS in 15m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/088f838fba63436da9bf3a2d5162317a : SUCCESS in 18m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8d886be3249f429fabe24ae59092c0e2 : SUCCESS in 18m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8db84bb1110e41519af57c714f868d41 : SUCCESS in 16m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/bcc43c041c66490187dd01ace6f569bc : SUCCESS in 15m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4e4aa232533f44c1a450638b53ad4f47 : SUCCESS in 14m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/86ae0219a61b47b3a71d30ae74dd2eb4 : SUCCESS in 7m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/18da414dac234b66b268d8a6eaf92a0d : SUCCESS in 19m 28s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/24ad64dbe6914ab9841b276aa402882e : SUCCESS in 9m 00s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/53c7b9c013e648a38c9d98b047269542 : SUCCESS in 1h 23m 34s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d27d5df4576a47f19ac712646f043056 : SUCCESS in 1h 25m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/67cb5e144ac84ab3a04762e770e21dd4 : SUCCESS in 49m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/817e14ce19294651933c83a9ab0bca4f : SUCCESS in 43m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44c77cc7a2744d18a20a6cf9b58c51ac : SUCCESS in 15m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/788f3c3c468a429ca511ac15f2ab56d5 : SUCCESS in 4m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/85de29db368242df8dcba053168eaa67 : FAILURE in 19m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a662726e5c95405c8402a05d3f755653 : SUCCESS in 57m 03s","accounts_in_message":[],"_revision_number":2},{"id":"007bff55077531b273ab5b0ffb7222715af60770","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-04 20:28:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"93dd836e6aa9948b04c94b98a4717f630bbfc785","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 20:28:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a99d248a205d5b203cc5d594a153ab7d40c83c0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 21:49:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c4cb59fa88c4101be06e4c92cb808ed\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/99baea48840b4fe2968aaf6c8e56d3e9 : SUCCESS in 1m 45s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7378f61ccae6427fb3b15b4294fd47d8 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e03864651094738aef9358ace01d141 : SUCCESS in 8m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9f72b3f667a04e368d7f7ae691925f2c : SUCCESS in 2m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16b113d11b464e04a22eecbfca375784 : SUCCESS in 5m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/d5ea8f1836bd40ce859b062105ac076d : SUCCESS in 46m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/81c28abae0564ce49cb268586c913cae : SUCCESS in 52m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2387ae5fc91342cc831c382acacfe435 : SUCCESS in 46m 11s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/13e29e2c9cc0459484c9b81bc0d1a526 : SUCCESS in 10m 43s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/3534bc67bab1484cb4e0d801af8c6ea8 : SUCCESS in 4m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ab56afdb052541e3af09ef0a494875db : SUCCESS in 16m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6796697b9dde4ed489cb5a54bdc570b8 : SUCCESS in 12m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/56d7b6cb1fb2400086371beb7851c42c : SUCCESS in 16m 29s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d71e32dcd9df4a36ac71306c8b8c8c95 : SUCCESS in 15m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e23984a5de354bdf8efb7cb5b4d26074 : SUCCESS in 17m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6ccaf99a6bf448dd9253d7ee1a2d2e9f : SUCCESS in 15m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f64151209242496791341e879cd101b5 : SUCCESS in 16m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/509e16953a434dd99077f86f32eb5262 : SUCCESS in 15m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/92da39df2cbd4a9e886abc415774f6dd : SUCCESS in 17m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f49ec65c3cbc4579a2282c695be27af3 : SUCCESS in 13m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/30fdebc15f784775aaa982cdc575f8fa : SUCCESS in 7m 22s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f427e2d46d5e4214902d15f3eb93ec4d : SUCCESS in 6m 53s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/84606a24f89d425791dee365cf640c03 : SUCCESS in 1h 11m 51s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a486df1da6954dbca971f3e33a18a09a : SUCCESS in 1h 15m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/28e9487480944078a6dace158c2507d9 : SUCCESS in 40m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e6d0716524e54e159794dda1a28ff697 : SUCCESS in 44m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1815ba8b22174e078849e54b26a5a739 : SUCCESS in 17m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7946d89f7d2947228c4a907627d7fc6e : SUCCESS in 3m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/755581ea37c547d7b04284147c0795b6 : SUCCESS in 46m 47s","accounts_in_message":[],"_revision_number":2},{"id":"ff5fd80ad72b06e2daf58230395903487e0ab269","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 21:49:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c1988ace76073794028aa36218e5bd20efde0524","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 21:50:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e179699b3eac427b8f5fc087204879ef\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68d737873f734f26b6f11656d50e7d2c : SUCCESS in 43s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/b8b4e6a8708144efafa90bab18a84ac1 : SUCCESS in 33s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/2476f172b7974aa499138a2573fb5d3a : SUCCESS in 40s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f77172e2ad6fef4b226637b8ae099a6faf810978","revisions":{"3018b542974149b1a719d2d9c19419c87cc6018c":{"kind":"REWORK","_number":1,"created":"2023-01-04 01:33:48.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/869135/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/869135/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/869135/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/869135/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/869135/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/869135/1"}}},"commit":{"parents":[{"commit":"2d7c1dc6dd8a672c1f89a46b2423898d33ff63d2","subject":"CI: Fix our usage of tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2d7c1dc6dd8a672c1f89a46b2423898d33ff63d2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-04 01:32:14.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-04 01:32:17.000000000","tz":-480},"subject":"CI: pin tox at the project level","message":"CI: pin tox at the project level\n\nNo sooner do we fix the gate than tox has a new release that breaks\nit again. Let\u0027s give them a bit to settle down; in the mean time, stick\nwith 3.x.\n\nSee https://github.com/tox-dev/tox/issues/2811\n\nChange-Id: I316170442c67c1b4a5b87f9a1168cc04ca2417b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3018b542974149b1a719d2d9c19419c87cc6018c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3018b542974149b1a719d2d9c19419c87cc6018c"}]},"branch":"refs/heads/master"},"f77172e2ad6fef4b226637b8ae099a6faf810978":{"kind":"REWORK","_number":2,"created":"2023-01-04 18:53:01.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/869135/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/869135/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/869135/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/869135/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/869135/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/869135/2"}}},"commit":{"parents":[{"commit":"2d7c1dc6dd8a672c1f89a46b2423898d33ff63d2","subject":"CI: Fix our usage of tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2d7c1dc6dd8a672c1f89a46b2423898d33ff63d2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-04 01:32:14.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-04 18:50:01.000000000","tz":-480},"subject":"CI: pin tox at the project level","message":"CI: pin tox at the project level\n\nNo sooner do we fix the gate than tox has a new release that breaks\nit again. Let\u0027s give them a bit to settle down; in the mean time, stick\nwith 3.x.\n\nSee https://github.com/tox-dev/tox/issues/2811\n\nAlso simplify our warning suppressions. The message filter is a regex,\nso any prefix of the message will suffice. This allows us to also drop a\nnew message seen on CentOS 8:\n\n   CryptographyDeprecationWarning: Python 3.6 is no longer\n   supported by the Python core team. Therefore, support for\n   it is deprecated in cryptography. The next release of\n   cryptography (40.0) will be the last to support Python 3.6.\n\nAs we\u0027ve previously seen with cryptography warnings, this can slow down\nour probe tests to the point that they time out.\n\nChange-Id: I316170442c67c1b4a5b87f9a1168cc04ca2417b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f77172e2ad6fef4b226637b8ae099a6faf810978"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f77172e2ad6fef4b226637b8ae099a6faf810978"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
