)]}'
{"id":"openstack%2Fswift~869268","triplet_id":"openstack%2Fswift~stable%2Fyoga~I316170442c67c1b4a5b87f9a1168cc04ca2417b8","project":"openstack/swift","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-01-05 16:16:05.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I316170442c67c1b4a5b87f9a1168cc04ca2417b8","subject":"CI: pin tox at the project level","status":"MERGED","created":"2023-01-04 23:58:47.000000000","updated":"2023-01-05 20:35:45.000000000","submitted":"2023-01-05 20:31:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869268","meta_rev_id":"93c15906b174c86d065da695f1c4a0e27981fe29","_number":869268,"virtual_id_number":869268,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-05 20:31:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-05 02:04:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-05 04:58:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-05 01:05:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"54d453de9d6a7ca1caec8f7bb23001102e5374d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-04 23:58:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"feef91021205d1c4f1d1fb6e67aba72b745d6f40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 01:05:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dda8d0f46db74d13b11f617f24f58dfe\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/165459e522dc43539e17ac4850143b24 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1db4a500bc544421b75c00c28a069b94 : SUCCESS in 7m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/948f0657d8784142a40e5b98029f6967 : SUCCESS in 3m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/526ed86457134402b33ebc9febf06b2a : SUCCESS in 4m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/dcd64f7f3d1f4df68075277766d31f29 : SUCCESS in 1h 01m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/724a0a540ebd43b99014dcc156dcf4d9 : SUCCESS in 43m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6258e34f950d4ffb8d70108de05f10eb : SUCCESS in 39m 18s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/28d214e3849245d5980bb3c123f647c7 : SUCCESS in 18m 38s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/45dd5f8be612440780d5d3f70470b0e9 : SUCCESS in 18m 23s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/5b616807e4c74bbfafc60e599a2f1e54 : SUCCESS in 16m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e5bf3b6aae3f421195893c4cfbe0fe31 : FAILURE in 4m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/439d4c07934648ebbdf37ab6dd525bef : FAILURE in 3m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a476dfc9b7954da3a1519a69287fb695 : SUCCESS in 15m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2477766685574f7cb9896b34b23890b2 : SUCCESS in 12m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7bda79d7f1f2416895458905caafabc5 : SUCCESS in 13m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/01b6f1238dc24cac9ce54da3665c690b : SUCCESS in 14m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ff292101093244c5ac8ea922bd804b6b : SUCCESS in 13m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e7ea0d82a4c64bba98278c0ddc93a2f0 : SUCCESS in 18m 13s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8fdf43874a3a4164ac985fedc0ce515a : SUCCESS in 11m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fc9c151a5c69454483f90854cd7c62c0 : SUCCESS in 14m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f125248ccdce46b791cb424ff78fda2c : SUCCESS in 15m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4d6d6452991243848cd7a10dfd438381 : SUCCESS in 5m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/259dac3db6f54e0b95f744994f1291d8 : FAILURE in 7m 47s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7e0d0bf8a2144727bfd223eecb057254 : SUCCESS in 49m 26s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0e1cc1024d0a478e8abb44861a21d720 : SUCCESS in 52m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/95dd4f723702481d868686ab8009f421 : SUCCESS in 33m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/845668acfae74dc387cf742c349b01b1 : SUCCESS in 36m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e1a03756b9e84a7993b72d940cad724b : SUCCESS in 15m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8aa42b06f1c4416ea8259f9c82a379bc : SUCCESS in 4m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3d708fcfcbcb404199e19adf637d3938 : FAILURE in 10m 43s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00b75a3d32bc42139ba54d1c2507b6a6 : SUCCESS in 42m 57s","accounts_in_message":[],"_revision_number":1},{"id":"eee1a368502ad5453a11f4bf52cbebef466e5e78","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-05 02:04:12.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"65f43a09ee18c860d4132088b8351ed35e985e84","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 03:02:13.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7a0e49389de4f5fb8a5637550fefd7b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8923c4855cce41f6ba052715aaffae30 : SUCCESS in 1h 21m 20s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/16bff391a1f14d31bebf408d201d5d84 : SUCCESS in 35m 58s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/cebd061853e549c1932119d496d85e01 : TIMED_OUT in 3h 01m 06s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/fa0e1920dd9e4129b47645237d0c3973 : SUCCESS in 36m 22s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/964211090ffc4b5f83f9ba493848da9b : SUCCESS in 1h 26m 16s","accounts_in_message":[],"_revision_number":1},{"id":"1786af016a6b7a3796355be89e3fb4459aec5d7a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-05 04:58:37.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c09c20739577134e61522ec1a5158837f226a57d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 04:59:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"67d67feb592b836a8d8e12524e11101ab73541ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 06:07:47.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e11fdde458347e9a01c696bfcfedf81\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6e682ea0040f4493919901a31fea3428 : SUCCESS in 1m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c13a8ff6d35e4d238f0a7940209ca190 : SUCCESS in 7m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/02b357002a0e4fe9a67e6d1ac4cf5cc4 : SUCCESS in 2m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bcf7ad69b46f4e919aedbd3da4c92410 : SUCCESS in 4m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/dea959792a374d9d8b63bbddce065981 : FAILURE in 27m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f45d7adcd7254fd5a067d16720d33afa : SUCCESS in 38m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1fb6e5cae4fe4b32b823d8341f8457cd : SUCCESS in 44m 10s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/9975845cb43744d4bf999054e2bb4c72 : POST_FAILURE in 5m 57s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/7534bda9cb3b4c41a09f5db840b275f3 : POST_FAILURE in 4m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3453434c053241108de091fd5e0cf4d0 : SUCCESS in 13m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b6db48525e9d44b29111d2a891f6f0b6 : SUCCESS in 14m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2d1b71b2af9a4a8095d5c40036a3e1c3 : SUCCESS in 11m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/170a72fd31aa4b6cb9b9f9416f5b659a : SUCCESS in 14m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1335a78d61f54f98a0f4479812dbf429 : SUCCESS in 15m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/60e08c7896b648de9ee2ed6e6c0cfd2e : SUCCESS in 14m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6168fc40cc9b4e0a8f6061fa6f8e7601 : SUCCESS in 12m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d91efec85717468d8974bf225394e430 : SUCCESS in 14m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/602725cfdfda48cfac37f70024554d9f : SUCCESS in 14m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1d42f448a259467cae0739b34afdf662 : SUCCESS in 5m 57s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9ce63459d94d44a1946d0fd9d7964c53 : SUCCESS in 44m 38s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0a6ff50ca3374927a96dad1c9442cd44 : SUCCESS in 59m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/13f2321bd3514e95bbffdbe7fb5861cb : SUCCESS in 29m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/595069f954414a23a613dd89a4650dbf : SUCCESS in 39m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dfe846724ddf4a909837c547932f1375 : SUCCESS in 13m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65b35bb527c04d9087fb5cd5f83ddc01 : SUCCESS in 5m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0651cf4c85764019a49cc3919d8d1ee5 : SUCCESS in 43m 59s","accounts_in_message":[],"_revision_number":1},{"id":"5cdf534ac506a73d4ee941c648bef954e7d2b247","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-05 16:16:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"60a9354d6f1e36a2045702f4676f2390e038b775","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 17:29:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0db67074d4f84444aa2d9ccbe26a5958\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b22d4003c1e84ee29a528ba541cb2124 : SUCCESS in 1m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f40e26f17552405e98c3d5510dd250fc : SUCCESS in 7m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c0ab4c1d9ae846d29fadf66227783064 : SUCCESS in 3m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f27effe9eeb44df5ba871a836151fc1d : SUCCESS in 4m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/ac91dd2228a045a1953f37bc38c5ddb5 : SUCCESS in 56m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7d3ee29da4db4e27bc2f153ae7b8d65f : SUCCESS in 53m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e8e526849204124bf6795904d1da40e : SUCCESS in 57m 35s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ec81f916566049e08db16c31c6a413a5 : SUCCESS in 18m 21s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/15f83c60fad64b8e8a58d9744fd31a04 : SUCCESS in 32m 10s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/b0a2d5e5095f482489e2f5f727d35a2e : SUCCESS in 23m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e9c0e2082246496abb6754b23ac5f19d : FAILURE in 3m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2f07c5931d5247219b4389219e371583 : FAILURE in 3m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c59bfd1f7e604112b4a0a77ccfef54e9 : SUCCESS in 18m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4dfa8d9a23e748be88a95e21365955b8 : SUCCESS in 15m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/af78791d96e146bfb1222aef06da7d7e : SUCCESS in 14m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/34781ad5968b469a887664e212ba78b5 : SUCCESS in 15m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d73c494fd9134851879e56e2a7a9d082 : SUCCESS in 29m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/97e9e35bffc347c7a36d90004382cde3 : SUCCESS in 17m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1ee7bbff4be94a319ad7cde8fd4bb7b7 : SUCCESS in 13m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d99cc19535664fc3a38d96331aac7911 : SUCCESS in 29m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/58bd726213c34324a0f9e91fc4fd0e58 : SUCCESS in 14m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9872dca8dcd940c2a1e5cc7d132aa09d : SUCCESS in 6m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c2821e4bad514eeea2b133c0f6dd2696 : FAILURE in 7m 39s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ef0eefc8abd641e5a4eb8b229fac28b6 : SUCCESS in 1h 11m 47s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/316c45d2fc2e44f58b08fc11348ec918 : SUCCESS in 51m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/17b6e35b154e4da49c2306f79b34fc98 : SUCCESS in 55m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/320da97183484ed882c536078a999b48 : SUCCESS in 54m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a0176a3266234cd597672ec161d508a5 : SUCCESS in 19m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25137e61b51347e8a8b50a097465bd68 : SUCCESS in 4m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cb4bf710058949809d22408b6e2605dc : FAILURE in 7m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d349bf92edaa463a9fafb41ea57684a2 : SUCCESS in 53m 24s","accounts_in_message":[],"_revision_number":1},{"id":"2004b4055310aad82d9c9280b3a809093863e713","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 17:31:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1e71f224d001a3ea1e8d80136cec5368fcd5c3e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 19:44:34.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0771ac70251148a8b7787441fbb9c8d7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1675e6134a8a484caba8fbb8d4959f4b : SUCCESS in 51m 33s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cbbbc2781427446581bea576a17ee7fc : SUCCESS in 59m 44s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/a644e2fc201f45089322e3adb4572775 : TIMED_OUT in 3h 01m 12s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4199a75bb8af49829c240737043a40a5 : SUCCESS in 1h 21m 36s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/cee379e161754444be64f12b4cd15c3c : SUCCESS in 1h 29m 13s","accounts_in_message":[],"_revision_number":1},{"id":"5f9db2b8f4781b0366f772327c3ce1c1d66ad12f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 20:31:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b47356f37bba4637a51f20d6cc0b65a3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c85da79ece514771bc3c709b9252fd58 : SUCCESS in 1m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b560ef8399b34509905822d741a94312 : SUCCESS in 8m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c829fac87a4342fc923eab01cad7190b : SUCCESS in 2m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e375feb0749486c8c4b4c94971a3e24 : SUCCESS in 6m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/8f4c4650e19e4dd0b1924055f214fbd7 : SUCCESS in 1h 02m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/14bec47a731b41e382de0d93c2582b42 : SUCCESS in 57m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a417375108344899428adc6c5e7a033 : SUCCESS in 1h 07m 38s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/671e34f507d44a1480da30516eae92f3 : POST_FAILURE in 5m 12s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/1f791e7996b045e5b20f9d1f4b9b954c : POST_FAILURE in 5m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/14bd6cc722af4cb68449acb46ec26971 : SUCCESS in 16m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/abe042f6dd1d471590cb58ee6282f5f3 : SUCCESS in 15m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6fb3dda7cada45d3ad1b8dba40841dc0 : SUCCESS in 15m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a7c5045e97a3490a8cf51b3fb9fed611 : SUCCESS in 17m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/02de27233a094d8e975bfcf3fc78001c : SUCCESS in 19m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/71a32683b29a4fc694019dc08fa779b0 : SUCCESS in 18m 40s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bae99353ab9041a292d186f764ca71af : SUCCESS in 15m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d0e03584715b47c0b6e40ffc1d245470 : SUCCESS in 16m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8656e5fee1674c93b73804230dffec7e : SUCCESS in 16m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/093a42b489324621a59faf0e465570af : SUCCESS in 6m 20s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7fcc5f6c9860447ea38099470a77ce5b : SUCCESS in 59m 13s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/02106a1e75f14d8e9e15d24b3f2728cd : SUCCESS in 1h 06m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0aa1437610c143d985bceca8326796d8 : SUCCESS in 41m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6dd35126141740158c7a3a459c6970e7 : SUCCESS in 38m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61353afff6604d00ada9bc91a56e775e : SUCCESS in 15m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c377fb3511c241dfa163b99ed50a7782 : SUCCESS in 5m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff13882548694cb589e64a4a6fe5559a : SUCCESS in 54m 59s","accounts_in_message":[],"_revision_number":1},{"id":"f000bb83b7e31a6526ca2bf8e4338eb70912bd27","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 20:31:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"93c15906b174c86d065da695f1c4a0e27981fe29","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 20:35:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e930c221d2a94772b883a3752a19ec21\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5654c543428647bb863fb3dbf500c31c : SUCCESS in 47s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/5d0736ae6b5d4b109146d34b26a7f2ad : FAILURE in 34s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/e40d51eab766486a9519b1331cbf2604 : FAILURE in 28s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb994ea501d96fdf73d4df9479a3b7d51e2d5744","revisions":{"eb994ea501d96fdf73d4df9479a3b7d51e2d5744":{"kind":"REWORK","_number":1,"created":"2023-01-04 23:58:47.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/68/869268/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/869268/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/869268/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/869268/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/869268/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/869268/1"}}},"commit":{"parents":[{"commit":"a6a85919a579bb194fc290fe719bf03269684de9","subject":"Inline parse_request from cpython","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a6a85919a579bb194fc290fe719bf03269684de9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-12-30 18:44:15.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-04 23:55:37.000000000","tz":-480},"subject":"CI: pin tox at the project level","message":"CI: pin tox at the project level\n\nNew tox has a bunch of incompatibilities -- on master it\u0027s worth fixing\nthings so we can work in the new world, but on stable branches it seems\nsimplest to keep ourselves in the old world.\n\nDo it at the project level to ensure the pin applies to the\nperiodic-stable jobs.\n\nAlso simplify our warning suppressions. The message filter is a regex,\nso any prefix of the message will suffice. This allows us to also drop a\nnew message seen on CentOS 8:\n\n   CryptographyDeprecationWarning: Python 3.6 is no longer\n   supported by the Python core team. Therefore, support for\n   it is deprecated in cryptography. The next release of\n   cryptography (40.0) will be the last to support Python 3.6.\n\nAs we\u0027ve previously seen with cryptography warnings, this can slow down\nour probe tests to the point that they time out.\n\nChange-Id: I316170442c67c1b4a5b87f9a1168cc04ca2417b8\n(cherry picked from commit 5637dddcdd334ffe75f63546b87d09964961e273)\n(cherry picked from commit 191bb266dc9866fb71ab66daaaeaa477b4a038e5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb994ea501d96fdf73d4df9479a3b7d51e2d5744"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb994ea501d96fdf73d4df9479a3b7d51e2d5744"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
