)]}'
{"id":"openstack%2Fswift~869276","triplet_id":"openstack%2Fswift~stable%2Fwallaby~I316170442c67c1b4a5b87f9a1168cc04ca2417b8","project":"openstack/swift","branch":"stable/wallaby","hashtags":[],"change_id":"I316170442c67c1b4a5b87f9a1168cc04ca2417b8","subject":"CI: pin tox at the project level","status":"MERGED","created":"2023-01-05 05:04:30.000000000","updated":"2023-01-05 17:32:53.000000000","submitted":"2023-01-05 17:30:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869276","meta_rev_id":"77a038f23c2923e151fdf5792a0835196428e914","_number":869276,"virtual_id_number":869276,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-05 17:30:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-05 16:16:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-05 16:16:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-05 06:07:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"93c7e9c96e0512de40e4d8ef91076c9c13d7b7ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-05 05:04:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a28b6fe39a718195d75aaf82cb5b50517cdc2d36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 06:07:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/898eb0ebfcc04d71b449479a08470adb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/774ca4ea715f48b3821d0319da4cb3be : SUCCESS in 2m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f39cb423da444d91b4098b7f9c17ecdf : SUCCESS in 8m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c7b4d348304b40c99f66db2acd86c505 : SUCCESS in 3m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b68f6a1cf5a248318a683747fd5578b8 : SUCCESS in 5m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/9832ba9a4b2846b68940c03c408955ff : SUCCESS in 1h 00m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e75f435687074f38b043af5fe5f96ba7 : SUCCESS in 37m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba17a1c75ae54be89358c02688b404dd : SUCCESS in 58m 41s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/5e5f340cb4134fe88fac1172c9fe6f90 : SUCCESS in 21m 35s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/90ea83994d4d489ca2004f359d97875f : SUCCESS in 21m 12s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/37f1510224df4299a1791aa84c65d36b : SUCCESS in 22m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/89ed0a10a4aa42808b042dbd85acc9f4 : FAILURE in 3m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d14a04cc41274d26bc519bbd763cccda : FAILURE in 4m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6692e691ad4f4348947c7d273d7ad493 : SUCCESS in 17m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/94eb2443dcdc405a826ee17637df91a7 : SUCCESS in 13m 16s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ac36180bcad944919757e6222ae0a010 : SUCCESS in 17m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d7702b2722544515a40a44b6f75601dd : SUCCESS in 16m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dbe1064fffb146bd8ffbbdd4b82218a5 : SUCCESS in 13m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5bf46388a39b495bb8d789da9d04cfc4 : SUCCESS in 14m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8679ddc9f43244bab93dbccb55021408 : SUCCESS in 14m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8fd06fa1e40549fda31ca78c7da09b98 : SUCCESS in 17m 25s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/28181f9506754f049a7706a3422ca9fc : SUCCESS in 20m 07s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8e539e2b94cf41948dbc86eba3ba9d74 : SUCCESS in 19m 56s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3d5b7655fd3a4e4d87d2c020448d0304 : SUCCESS in 16m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/950d7c6f68d54ad594b05530db03c863 : SUCCESS in 5m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1675c17d71b7458ba7586a0950046864 : FAILURE in 5m 51s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cbabe71eab5c45c6a8f75d8820cd7b7f : SUCCESS in 50m 18s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/aa830a67a9624423906fdbf8ccf45ad2 : SUCCESS in 52m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ed1347d4be164a0b80fdad9e2a342990 : SUCCESS in 40m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/49b4baffeb36455799c1d5c5e5d0c9cd : SUCCESS in 32m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43275782e3554480bd918c594fd30c46 : SUCCESS in 12m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1d2f83cef24489499f4c276d92419fc : SUCCESS in 6m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/42df3a0d6a974dca8dc3b7e523c3e984 : RETRY_LIMIT in 5m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ccf651f83cae4589a1ff41a22155d9b0 : SUCCESS in 55m 41s","accounts_in_message":[],"_revision_number":1},{"id":"f55dee2d06042be7143275eb90a132a720ca85cf","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-05 16:16:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ebdcd8d1d1d4b47e7510bba2ddded5c8e86a9d78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 16:17:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"10fc793bf87204760627d51836b8c9f797f31d71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 17:30:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d87d9257e89e44e0ac1d408812b8238d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/02f074f0376e4773b7941d1ae0070448 : SUCCESS in 2m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80417fc49654435a993bb18a21f317f3 : SUCCESS in 7m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cb71bc78971c46759d8dd9b282daa912 : SUCCESS in 2m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7b00a08b7a04bb3b70b4f7f04c50ffb : SUCCESS in 5m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/36410ec4bc58482282a74f99c398cf00 : SUCCESS in 1h 05m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f2f78bf946ee480f84b7e1f580a010a2 : SUCCESS in 56m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/72e65b78ffc44ec3ad5ffba836cd5f38 : SUCCESS in 52m 14s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/8565591527b94db5a3980310a26567a3 : POST_FAILURE in 5m 15s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/4763fd1df10f4d6d9e50acb74801e730 : POST_FAILURE in 5m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ae897797e2754f42b495c6b407340260 : SUCCESS in 11m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/afabde7bfb5641f8992f0391ae45f397 : SUCCESS in 24m 42s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b07e17400b1d46bb95c08b716b42de83 : SUCCESS in 15m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a7b119515382447ab3d3a2a9b97f45c5 : SUCCESS in 14m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e9c9ead4944f498dadf7095db0c4c00a : SUCCESS in 14m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/32eb9f485e1043118135185b7fff4ce8 : SUCCESS in 17m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f4dc541e0ff2432c8efa04f03e9a7658 : SUCCESS in 31m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1a111c5ae37f4a29b83088531cf91967 : SUCCESS in 19m 19s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9c316f3464fa43d1ac9f14760cc99436 : SUCCESS in 18m 55s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/0d8fa954e13842d8a12c7314fa7ec010 : SUCCESS in 17m 20s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a5cb727ec9de49488d090ba1a41b9fa0 : SUCCESS in 18m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/84093f5fc69c48aca2fdbb41155facb1 : SUCCESS in 7m 04s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e8bc1a329f954eae98bf60c77b16eb7a : SUCCESS in 58m 26s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f5bb93b4054546c382852fdc7de922be : SUCCESS in 57m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6ef553e21585483cb38b59b360c8ea2c : SUCCESS in 41m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6f83c99cf9764e96963b428349cefcb3 : SUCCESS in 50m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63c8e19379d74a659b1974431c84621e : SUCCESS in 14m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bd9b2a2fd93478fa1a804c06a897102 : SUCCESS in 5m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ead73024904c4b0f8f4578d0e12b5185 : SUCCESS in 1h 02m 16s","accounts_in_message":[],"_revision_number":1},{"id":"05a918ffa9a5823ada0b8fbec1b4eebcaf1b640d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 17:30:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"77a038f23c2923e151fdf5792a0835196428e914","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 17:32:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a84d90a7e9a48a5ba599b31d2981d5d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1890ab41fb8543308c501671ee8c8f20 : SUCCESS in 43s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/7cde829bdcbc4345a447923ff4705a50 : FAILURE in 32s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/c6794a1f9f524dc3a883b227c1996058 : FAILURE in 30s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"be6353840b98175b0103cfb4eae833354451e0c5","revisions":{"be6353840b98175b0103cfb4eae833354451e0c5":{"kind":"REWORK","_number":1,"created":"2023-01-05 05:04:30.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/76/869276/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/869276/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/869276/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/869276/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/869276/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/869276/1"}}},"commit":{"parents":[{"commit":"021ec6f685d06232bec970174b4a3c112808c753","subject":"Merge \"Inline parse_request from cpython\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/021ec6f685d06232bec970174b4a3c112808c753"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-12-30 18:44:15.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-05 05:03:12.000000000","tz":-480},"subject":"CI: pin tox at the project level","message":"CI: pin tox at the project level\n\nNew tox has a bunch of incompatibilities -- on master it\u0027s worth fixing\nthings so we can work in the new world, but on stable branches it seems\nsimplest to keep ourselves in the old world.\n\nDo it at the project level to ensure the pin applies to the\nperiodic-stable jobs.\n\nAlso simplify our warning suppressions. The message filter is a regex,\nso any prefix of the message will suffice. This allows us to also drop a\nnew message seen on CentOS 8:\n\n   CryptographyDeprecationWarning: Python 3.6 is no longer\n   supported by the Python core team. Therefore, support for\n   it is deprecated in cryptography. The next release of\n   cryptography (40.0) will be the last to support Python 3.6.\n\nAs we\u0027ve previously seen with cryptography warnings, this can slow down\nour probe tests to the point that they time out.\n\nChange-Id: I316170442c67c1b4a5b87f9a1168cc04ca2417b8\n(cherry picked from commit eb994ea501d96fdf73d4df9479a3b7d51e2d5744)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be6353840b98175b0103cfb4eae833354451e0c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be6353840b98175b0103cfb4eae833354451e0c5"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
