)]}'
{"id":"openstack%2Fswift~869278","triplet_id":"openstack%2Fswift~stable%2Fussuri~I316170442c67c1b4a5b87f9a1168cc04ca2417b8","project":"openstack/swift","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-01-05 22:13:06.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I316170442c67c1b4a5b87f9a1168cc04ca2417b8","subject":"CI: pin tox at the project level","status":"MERGED","created":"2023-01-05 05:15:46.000000000","updated":"2023-01-05 23:30:00.000000000","submitted":"2023-01-05 23:28:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869278","meta_rev_id":"3493e32af88db8587ee837fbaf4d2cfc590e0899","_number":869278,"virtual_id_number":869278,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-05 23:28:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-05 22:13:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-05 22:13:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-05 06:14:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e6160a2d21e4bf8ee749ad83c2c80874d7bd83e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-05 05:15:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62dab2928f2a8872ed922cb95f8fd5e38f605ac3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 06:14:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/09d76185e33d438ba6f70884be7fe7cc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/87070d2a33a54675a20ca7be4777cc59 : SUCCESS in 1m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8238eaea256f4830b030557b84c126fc : SUCCESS in 7m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1a73f87a0589460098c590edba00b34a : RETRY_LIMIT in 1m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a52b0997d414eb18c3d82424b9f90de : SUCCESS in 4m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/a1aeb979b1a74b59b900df28ee61a367 : SUCCESS in 57m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/557148000154465aafe2c0e6b00e059f : SUCCESS in 31m 20s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/15abd2c432fe41648f86f6da5ddf3fcc : FAILURE in 3m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/43b8c26c9f5441b2b719543d2cb5f1db : FAILURE in 3m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fdee3502988c4943bd51478ba94ed596 : SUCCESS in 13m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0465a6c10f1e41b696bcc3c2a67884a9 : SUCCESS in 15m 29s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5f2f4282f49048bda59dee011e5634cb : SUCCESS in 14m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d27d692a654d4c1bbc0ee63ee5ea8098 : SUCCESS in 13m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c80424acde784418886dcf35c447843c : SUCCESS in 8m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5ee95a9ea87f4185add40064da9450f9 : SUCCESS in 11m 20s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/f79a40cf935541dbb41421f71fc0d7d3 : SUCCESS in 8m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/188b16325d614422a78178ed73c44789 : SUCCESS in 9m 56s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/ecce5d9fa9174452bca2540b78805bdf : SUCCESS in 13m 57s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7f835cf16f014911929f93b6c6be75f6 : SUCCESS in 10m 14s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/9ac3078165b84f0fb3eb78de6fa07a3f : SUCCESS in 10m 55s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/6ca22a6de17044009b093838c26f4f01 : SUCCESS in 10m 11s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/cf320c48094e47a7b2da66752fbf8b40 : SUCCESS in 10m 02s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/1d13c1f240124673b6afbb5216fd9857 : SUCCESS in 16m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c1e81c618b4147d9935bd859cc0ffd12 : RETRY_LIMIT in 3m 24s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2abfb0fffdf44ed1a9d9a555c9ad2d17 : RETRY_LIMIT in 3m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a1f7bbc1308c404793604666bb0fa65d : SUCCESS in 37m 44s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/84b2974c8cb74c62911cefba3a7c0997 : SUCCESS in 21m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/503d0a096f26425fac8e3665dcaa844e : SUCCESS in 37m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f16f21e9abc14eaa8f0a86e7f4f016e0 : SUCCESS in 12m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cc5b40488dc4f4d95255ba597ecb15d : SUCCESS in 5m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2bb1409bd2f24cfeb7b6595879f6c5e4 : RETRY_LIMIT in 5m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa7294c545304bd2aaf2386bf41bb160 : SUCCESS in 57m 47s","accounts_in_message":[],"_revision_number":1},{"id":"54f52343ee51345d50bbf85375b19524bfcc1958","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-05 20:39:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7fa89a0fc56447230dca4e2db5ebbb4c8d53c4e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 21:59:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b400957dab345f684b6eeb5dae2acf2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b0da05b79b91450f851bd709f011bc88 : SUCCESS in 2m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91fc036e9e09492e9eed64615e075028 : SUCCESS in 7m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e07e8db6588442eebe59c06d1029f81d : SUCCESS in 2m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0699bdb59ac8492f966a3ae46101031a : SUCCESS in 4m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/5d36b3369a794344b20fc85c220b2853 : SUCCESS in 1h 14m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/29671bbb0b5840809e8905e42a9f3836 : SUCCESS in 44m 42s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2a08838339144c7c99d9f3a363ddd2cd : FAILURE in 3m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2c183cd77658427daa67b6d2e14c32a1 : FAILURE in 3m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ae739a1e0ac84df4bf9a5580ff2d2345 : SUCCESS in 12m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0cd3e5a08af54be6804b8004afff176b : SUCCESS in 12m 30s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7da1d3a3270b4b52ae5a09a25aaac28a : SUCCESS in 12m 33s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/91916111579346db9ed1af9d8d96dcc8 : SUCCESS in 13m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/70379c0085b447ecaccc1a8b256585fb : SUCCESS in 8m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ef7cbed645544e438366f5b4d31e5dd0 : SUCCESS in 13m 31s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/ab89eb64defa48e9b136d00b9b6fd563 : SUCCESS in 10m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6d147eb34441464791aa6911b394ded4 : SUCCESS in 11m 58s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/3bf0eb2fe8d24bd7998d9020bd17fa38 : SUCCESS in 14m 45s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9aa603a3adcf4713b40989433d608a04 : SUCCESS in 10m 01s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/98c5d67dee6c471f9955a64e50984c55 : SUCCESS in 13m 22s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/21de013d6d6a465d83f6a90651e4e90e : SUCCESS in 10m 04s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c91bd9afeed041c799022bb45131552c : SUCCESS in 11m 18s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/c55dfd390186476eb4637bb71ea1b56c : SUCCESS in 16m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/054808f8683841b08fc6f0322dc1ab80 : FAILURE in 7m 52s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/79df244b36704d03bf02947fdeb47db3 : SUCCESS in 46m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c3e1d82eb3b94ca6bb0b2716dc3e82f9 : SUCCESS in 32m 27s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/08d881a4685b4f28924e7ada0893aa55 : SUCCESS in 32m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6d2b1e8befc4414ea973104b47446bf2 : SUCCESS in 35m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2c9a8979698411589084afebb698000 : SUCCESS in 13m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ee7bde5954945b8879116dbe0468bda : SUCCESS in 5m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a9738221afea40209f8dc87c03851f54 : RETRY_LIMIT in 6m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e8e8ead7e7b4ff997f8065985646992 : SUCCESS in 1h 00m 43s","accounts_in_message":[],"_revision_number":2},{"id":"041eb2aac8ae7e344be1a62ef7562e245289192e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-05 22:13:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6a09e90c595c73a8b61d3788d12661959cbd7708","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 22:13:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"95c5763aa3bbd83f9c2bc0c02229af9ab81d915e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 23:28:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90891ad9c528445099822b4817b07eed\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/481ae74c46a74547b1f9f2942be3f441 : SUCCESS in 2m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25bffc4545f44eb5b8bf173af746ad2e : SUCCESS in 8m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/38e3c9a446bd455d8917aac03cae306b : SUCCESS in 2m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dedcf7149f1047c598575ea5c0761391 : SUCCESS in 5m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/a228bcbcdca44957a776391695096c57 : SUCCESS in 1h 14m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c89ebebfde7b49acbd41639b95769bac : SUCCESS in 49m 51s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/97008ccd287742b7993f8bf8ceeade72 : POST_FAILURE in 5m 13s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/9aa8c7916e1f4ebcaf635667485d05c9 : POST_FAILURE in 5m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/58a5dbcb9d6c4a3abbacbbcc642ac53a : SUCCESS in 14m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b5e4341107b84748aecc1fe23a8ba1d9 : SUCCESS in 13m 58s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/684eb99892ed4797bb00185e0309f16d : SUCCESS in 14m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bf4989118add43adb3e6fbe886dd2e83 : SUCCESS in 13m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/edced8e2ccc44b71967f57551892328c : SUCCESS in 9m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c8f2e223f6b24cec8c46806594a9a8c0 : SUCCESS in 12m 05s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/d430de29091940a489d3189220bb2849 : SUCCESS in 9m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/86437ff69b42425d87238b2c856d8574 : SUCCESS in 10m 24s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/e58f9566e8c746d9ab10a2c94923e8c5 : SUCCESS in 15m 03s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9eacbb85011a4c02a7a98e9c24d8f123 : SUCCESS in 10m 17s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/b56d1d7a9aa54dc5a1c7f010b0622bca : SUCCESS in 10m 03s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/1d3d8069611b4d238aae1e3ce78127ec : SUCCESS in 10m 50s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/764de43994ae49b4b024db55a713de9d : SUCCESS in 17m 16s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e82867b71784447e9781a35835592a5c : SUCCESS in 45m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e6c02512b2ef4eca8d6e4ec73ac5302b : SUCCESS in 34m 51s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/e116721410d849ecac8e10b0c14978d7 : SUCCESS in 33m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/980e09617b7b479e8a9845d376dddd82 : SUCCESS in 34m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf7277964e7844968ab2988f408eeea7 : SUCCESS in 14m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9717a57d6a6c4e0299e36e66156dce99 : SUCCESS in 5m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ccba4f9aca104a8a804c91c292c09eac : SUCCESS in 1h 00m 03s","accounts_in_message":[],"_revision_number":2},{"id":"889b382e1099faaf2a75d95eaff4d4422c2d3032","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 23:28:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3493e32af88db8587ee837fbaf4d2cfc590e0899","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 23:30:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1280c209ee9b4198905c5c1bf670b0f9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/479af68deca14408b97502178556729e : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7527f59c05a4af98dc08bd964718c0b : SUCCESS in 42s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d83a3be510f8452aa4b4d50582485601 : FAILURE in 28s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/87819058b41f417ebe07c10465fd83b9 : FAILURE in 27s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"842081d6eae1a624b0056501b4bbbb7875a8a3c7","revisions":{"9f86527933172351d8daa0f9893903ec52426f2c":{"kind":"REWORK","_number":1,"created":"2023-01-05 05:15:46.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/78/869278/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/869278/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/869278/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/869278/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/869278/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/869278/1"}}},"commit":{"parents":[{"commit":"430bb7f5681f80ad96c7d7aca2b40023508d6843","subject":"s3api: Use constant-time string comparisons in check_signature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/430bb7f5681f80ad96c7d7aca2b40023508d6843"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-12-30 18:44:15.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-05 05:15:19.000000000","tz":-480},"subject":"CI: pin tox at the project level","message":"CI: pin tox at the project level\n\nNew tox has a bunch of incompatibilities -- on master it\u0027s worth fixing\nthings so we can work in the new world, but on stable branches it seems\nsimplest to keep ourselves in the old world.\n\nDo it at the project level to ensure the pin applies to the\nperiodic-stable jobs.\n\nChange-Id: I316170442c67c1b4a5b87f9a1168cc04ca2417b8\n(cherry picked from commit eb994ea501d96fdf73d4df9479a3b7d51e2d5744)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f86527933172351d8daa0f9893903ec52426f2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f86527933172351d8daa0f9893903ec52426f2c"}]},"branch":"refs/heads/stable/ussuri"},"842081d6eae1a624b0056501b4bbbb7875a8a3c7":{"kind":"REWORK","_number":2,"created":"2023-01-05 20:39:44.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/78/869278/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/869278/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/869278/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/869278/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/869278/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/869278/2"}}},"commit":{"parents":[{"commit":"430bb7f5681f80ad96c7d7aca2b40023508d6843","subject":"s3api: Use constant-time string comparisons in check_signature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/430bb7f5681f80ad96c7d7aca2b40023508d6843"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-12-30 18:44:15.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-05 20:38:40.000000000","tz":-480},"subject":"CI: pin tox at the project level","message":"CI: pin tox at the project level\n\nNew tox has a bunch of incompatibilities -- on master it\u0027s worth fixing\nthings so we can work in the new world, but on stable branches it seems\nsimplest to keep ourselves in the old world.\n\nDo it at the project level to ensure the pin applies to the\nperiodic-stable jobs.\n\nAlso, skip installing python-dev on jammy.\n\nAlso, pin PasteDeploy to 2.1.1 on CentOS 7. This was the last version to\nsupport py27.\n\nChange-Id: I316170442c67c1b4a5b87f9a1168cc04ca2417b8\nRelated-Change: If69ae0f8eac8fe8ff7d5e4f4f1bff6d0ea9e7a8b\nCo-Authored-By: Matthew Vernon \u003cmvernon@wikimedia.org\u003e\n(cherry picked from commit cc033154ad4a4f345258457f3ceed9143fb3d46d)\n(cherry picked from commit eb994ea501d96fdf73d4df9479a3b7d51e2d5744)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/842081d6eae1a624b0056501b4bbbb7875a8a3c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/842081d6eae1a624b0056501b4bbbb7875a8a3c7"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
