)]}'
{"id":"openstack%2Fswift~869390","triplet_id":"openstack%2Fswift~master~If6af519440fb444539e2526ea4dcca0ec0636388","project":"openstack/swift","branch":"master","topic":"memcache_metrics","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2023-01-20 00:20:08.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-01-06 18:20:52.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"If6af519440fb444539e2526ea4dcca0ec0636388","subject":"Memcached: add timing stats to set/get and other public functions","status":"MERGED","created":"2023-01-05 16:58:35.000000000","updated":"2023-01-20 00:20:08.000000000","submitted":"2023-01-06 21:15:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":22,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"869390-memcache_metrics","meta_rev_id":"12cb572642d0d9094b3c14d9d8e5703e9848d7bc","_number":869390,"virtual_id_number":869390,"owner":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-06 21:15:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-06 18:20:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-06 18:20:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-05 18:19:03.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2023-01-05 18:26:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-05 23:29:20.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2023-01-11 16:12:04.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2023-01-19 14:14:36.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"}],"messages":[{"id":"87b44db7539ba7f1c51da05682406f524d72edd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-01-05 16:58:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a0978ca126f2f652f1d2c32742c7a9fe0a78ae9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-05 18:19:03.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f3ec21c49d2a348436f9f892b3593aca31b61201","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 18:26:46.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/42e392b0f99c4e8995381340be152804\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2aeb43a0944e4704badc7ce39481dd25 : SUCCESS in 9m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/705a64547e9d4e919c3475ff9669ab39 : SUCCESS in 1h 03m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5524dd411cfd4c5dbdd06c1551a728b8 : SUCCESS in 46m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e7775e415bf641b6928e72a4834dfb6f : SUCCESS in 1h 01m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cd166f66fd3c4bbe9cc32b548c9d5ee4 : SUCCESS in 16m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bf25de0b62894d79a58b4f2ceb504f07 : SUCCESS in 17m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0aae730a76cb4f30b0325d903f6eaf5f : SUCCESS in 17m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/44addb48732f4285a21bbb769d771988 : SUCCESS in 10m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f1537b11b767468b9cb35a4836c1db70 : SUCCESS in 4m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/18a10cbb7f5a4dbba6285fcba65e14a8 : SUCCESS in 18m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c9f57ed3cbcd49c28060414ede9d4e7a : SUCCESS in 15m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/89ffa8fb2b6e4f2380e798f370daa1d2 : SUCCESS in 14m 19s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/90c475e64f38403a90104165710ce379 : SUCCESS in 17m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d602a90cb1614a9f943f828d58e41201 : SUCCESS in 15m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/004725c2475b449ca94174ca300d6cd5 : SUCCESS in 17m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a632008796384a779f85939934ecf511 : SUCCESS in 16m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/109ead45121c47939e8e9fd9f217caee : SUCCESS in 24m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/40428b513d9c47ce8ce5ee3953866078 : SUCCESS in 22m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8561cab53a6a4b8fb90df5e09fa16489 : SUCCESS in 16m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b3e1cd83f3f4421190f72bd7686e7365 : SUCCESS in 7m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d6d0d2b6e84e479990eecde42dca985e : SUCCESS in 14m 52s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/959fc984b4b9465b84f69c07311d96df : SUCCESS in 7m 57s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c64667f98198470e8dfa8152558747a9 : FAILURE in 1h 23m 54s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/fb4eb28759884d42b957c1da0848fa22 : FAILURE in 1h 26m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b8b94d5ced2f4d2ca95dee06c7f282ec : SUCCESS in 41m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4d757f5df85d4bb0afc048b1dbdf470d : SUCCESS in 41m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2c78ad04fe84d739b83e5c318134bb2 : SUCCESS in 15m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbd410a56acc4114b3eea20a0e19c713 : FAILURE in 5m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b9caf14f44954c88ad2ff67524a42a17 : FAILURE in 20m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e5cc50b6b9746a88b8f4438b933c745 : SUCCESS in 51m 03s","accounts_in_message":[],"_revision_number":1},{"id":"5a2e4306e3320ccf0039c9b4f49b92f599f6b315","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 19:47:22.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81639bb895624cd5ae328dd9a8e04826\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b5f3bb31467f45ff8eaf032f3147dec6 : SUCCESS in 29m 15s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/92672c089f5f43108d8997125a8db057 : TIMED_OUT in 1h 30m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b78b75bf5fbc464898fe2d1db5f2184d : TIMED_OUT in 1h 31m 03s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/355a01c0a796420980ac22d7b3cfcb82 : SUCCESS in 1h 13m 12s","accounts_in_message":[],"_revision_number":1},{"id":"3f974a1ed1f4c3335c6455119f53b7a5b7b881e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-01-05 23:14:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e797a1874b11081cea4af579bf1fd80e78bbab38","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-01-05 23:16:12.000000000","message":"Patch Set 1:\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8b99a3d7704f3a8e03bc1e38429877c78fe145da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-01-05 23:23:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5c7575b09437b418c574ea1518397b6948a57b3e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-05 23:29:20.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5881439e70cdf0d20ae0d5ff67da74e6d9fa8042","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-06 00:05:09.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2fa04e791e44aab83bd96d391e9f7ae34028582f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-06 00:05:29.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2a7b62bb3a3bb0f033f0c921a9d948074b313dcc","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-01-06 01:09:45.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0b2a112ac2c44a460b41218f40a3249f01507633","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 01:39:52.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f1445ec8530452ab45ccc553ede14d6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c15d066d97c48b9887673859bb9fe7b : SUCCESS in 8m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/167c151b71dc445291bbc734538006d4 : SUCCESS in 1h 05m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ac2e89c6d96a48bc95147c1c951c5d07 : SUCCESS in 54m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a5be1d1f87354b7b8e461b35985d1fd8 : SUCCESS in 1h 01m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/60c7a70794804279aa5a820b06a40acd : SUCCESS in 19m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e11c70bcdc864a86aed613a80deb3c64 : SUCCESS in 19m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dcbd3586c62843e39d6911f02568a47a : SUCCESS in 20m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/63ea78f04f214bcaaa3df1241c732686 : SUCCESS in 8m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b78d082803374448b7bc16a80a3441ae : SUCCESS in 4m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/824429260d364d9b9258f52fb968d9c5 : SUCCESS in 17m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/01f7105890dd4b26b66bfa6f1d0108e7 : SUCCESS in 17m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6e82fa6e0b0d43e39ab11ff1e9d03c9a : SUCCESS in 16m 34s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9bcb048008b546b58729bfb0e9a72944 : SUCCESS in 16m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1343894288ec452fbbea2f00d4ac3f1a : SUCCESS in 17m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/64ae4f5646af4a058fc849db53d1a70c : SUCCESS in 16m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/880f024638fc4e5d89d2667fef938432 : SUCCESS in 15m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/21137a7a69984c94a5b0d54cfbdc180a : SUCCESS in 14m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5c18b68bd2994c23a8d7fe4a7c7186b3 : SUCCESS in 13m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/08c6f3bc2f2a4ec8bbdd088f48d0e1da : SUCCESS in 14m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b5ee6a3371834adbb5680c4af304b1a7 : SUCCESS in 6m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e0f2d4e990d1460a8e96050e277383d7 : SUCCESS in 16m 52s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fc183409151449d58aab41874c011325 : SUCCESS in 7m 59s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d80344a617d14d3b8192338a29cb4a27 : FAILURE in 1h 23m 11s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0bd912bd880147a494e7bc8d12a5abe8 : FAILURE in 1h 28m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9fa62332a896439a8b266ac7cdaa00ba : SUCCESS in 50m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4a25d02419ea477ca776b97a973215f5 : SUCCESS in 47m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e99e37aa5ca4b7ab78c419a5ebe2517 : SUCCESS in 18m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f73122bb39c4483b4d455e0beeb5401 : SUCCESS in 4m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/56accae38b8846ffb9b778e6092ff4bf : FAILURE in 22m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61d3f45df76f43bf858f2d45bb8336e7 : SUCCESS in 56m 01s","accounts_in_message":[],"_revision_number":4},{"id":"c171ea44e3783398f7ae35160eef93b8240e6fc6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 02:17:34.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d550b5598af45138d393536b061e7ad\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c14f51adf4ca432893a233f0bb9d7421 : SUCCESS in 1h 01m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f668947a58144d14a76fa802bdf73711 : SUCCESS in 1h 04m 19s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e93a554c700e47dbb09a3d68b8a82ce3 : SUCCESS in 1h 00m 38s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b95b1daa277e40e990ac39b969b014f6 : SUCCESS in 1h 16m 26s","accounts_in_message":[],"_revision_number":4},{"id":"1317d6ceaed494307ceaf928298b0c2943404407","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-06 02:54:23.000000000","message":"Patch Set 4: Code-Review-1 -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8114903d24265cb61d92446a91902d457189b12d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-01-06 04:34:53.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"11aca6dbdca39336579a4b724a33efba0d9becd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-06 05:04:18.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ae18d07cf518b5b5b638ff547e12ccc09f3bfe21","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-06 05:05:57.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"02de136fd2a7c5f343f5dd308f38a808d596660f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-06 05:08:09.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"51ae6afe180a3977728a3adefd481247e68383f6","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-01-06 05:08:32.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"5977020aa1b47a94f6eed03a0c37ed88871b7b0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 06:29:00.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d34a6a6fe2204dcfb07d14063c34fd51\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5a2f7ccd4cb48b18d9e5e30b542a44f : SUCCESS in 8m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/6c0f520f4b824236b5b1dfcef83860ed : SUCCESS in 1h 03m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/24138305b5204260a76987161bffb05c : SUCCESS in 38m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7529fe2242474362b8b36a56e6dc4aa4 : SUCCESS in 57m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/936d8d6fd96d4b68946f418d87fb23f2 : SUCCESS in 25m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2d406ec46f0e45cd9de96660d80487ba : SUCCESS in 19m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ef6190130cfe4378ada3b895aadc6441 : SUCCESS in 28m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/813e56a9051042c287f53d018a9ad46a : SUCCESS in 7m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/15dfa4244e4d44da92ae6ca436a38e4a : SUCCESS in 5m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d9526fd582f04ceb8a56f8529fd2e540 : SUCCESS in 19m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/755a6bc2333f4def847cfcf6f0b5ab12 : SUCCESS in 22m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fc0a2507af4e4ef2948283ba6128e3f2 : SUCCESS in 20m 55s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/cd84824f548f4fc0801f06b31504e5ea : SUCCESS in 15m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e05cf9b64ccf4099a7f1a0ec7eb6de9f : SUCCESS in 21m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d25d82b4bf834e75b0838eb5b13481f7 : SUCCESS in 26m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f6a701fa285d4567aef34142ebf1075e : SUCCESS in 19m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/182d5b0c6ae34f7a9fcf2f212b830ab2 : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4f01c7cab4744076be0fa2d823365fc9 : SUCCESS in 23m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c0e9ae340b4045f4b7597c5ca73a804e : SUCCESS in 24m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/58c2924c8ba946098128d0cf49ad892d : SUCCESS in 7m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/66e25d37b5464aca9631b39af6ef5953 : SUCCESS in 20m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3eb9846b1d2e48eaae275f0601c45b72 : SUCCESS in 7m 57s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fde30ca9105c4ee9aa7846bbf24e71e9 : SUCCESS in 1h 12m 16s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/880104834f2741eb8d1e422f69e5ee12 : SUCCESS in 1h 18m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/80adef77b18c49dab35f304d7fd55eb1 : SUCCESS in 38m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c7dfe296b4484fd3a98f1ef35daf2108 : SUCCESS in 45m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b9fd5d14103c483ba974e08b72d36d4b : SUCCESS in 21m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1754f9f29cd94762bf02af0a14befcc7 : FAILURE in 4m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e751174de9a64016a9ffc725403be5a9 : FAILURE in 29m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3867fb87cb7446bf98469a01152195d6 : SUCCESS in 1h 01m 14s","accounts_in_message":[],"_revision_number":7},{"id":"94e65534243bcd22276a5431a88c78dc4f0e2721","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 06:41:22.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b8f5f311132426699491e8ecd3a4f91\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/429a8a79102f49599c89d140004f0279 : TIMED_OUT in 1h 31m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/011deb8ad5ef4d51a6f8b36aa3dabd90 : TIMED_OUT in 1h 31m 01s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c0d5db15ca184499a59245c5424439cb : SUCCESS in 1h 15m 08s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a07e55e79c0f45a6ab84bffa0bb304ce : SUCCESS in 1h 16m 04s","accounts_in_message":[],"_revision_number":7},{"id":"a1f115ee85f7d80317d7557c971f4626bc4c5c00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-01-06 18:02:24.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"675956604abae6bc92d4a81d5b29e38aad610deb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-06 18:20:52.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"183af4752307ea34355e5cfd201f1d25a196fb8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 19:53:32.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e742df30d0df4d80b6c979ed187fcace\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d77b45d4c5a44ce823e8e577df802b5 : SUCCESS in 8m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/ff9f0764bcac48a085e5780a11f7d8a0 : SUCCESS in 1h 01m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/221fb7a5eec84b09b556b4a03ee4e3b1 : SUCCESS in 41m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/44d802b56164452a913c11c3312f4cb5 : SUCCESS in 1h 13m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/54896317c6f64b98adcf84aad9775c10 : SUCCESS in 18m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9cdacda4bb2b4303935d8dbb2e6b620b : SUCCESS in 21m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/99617d09362948a194bda1d78a0a7442 : SUCCESS in 17m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6386771429c34b238d6b63dc4052cbf9 : SUCCESS in 9m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4754d09329f34d039ffcc3a0e4d54365 : SUCCESS in 5m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/77d930c3ff9348d697d08f529a14327f : SUCCESS in 15m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b1ee1812ad82476d826c86cfe467a363 : SUCCESS in 17m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7c66313955134819bf502397e0d2fc97 : SUCCESS in 15m 11s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/61accd74c5f4465986381a178982045e : SUCCESS in 18m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b1e93d3b260d4d30b87eb3e238d7a514 : SUCCESS in 17m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9c209d327d1a4696973b02dc603ddeb3 : SUCCESS in 17m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f9dea08a711844a8a95ca2df8b479c89 : SUCCESS in 17m 24s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cb61c5f5ce06405fa5686cb889e2e6f9 : SUCCESS in 15m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/98133ffb80b144a5ba28d8c97e36ade6 : SUCCESS in 14m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/46c324f6138946a187fdbf0b076877e8 : SUCCESS in 16m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4e68606da4e84148880a9016d15304ca : SUCCESS in 8m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fe4b93b0d45b43d49b5be3c6f84f9df8 : SUCCESS in 17m 56s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e19e85b551a14d018ab94f657c5304d5 : SUCCESS in 7m 35s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6b2a1e457fe545f5a6a51db1775f1ae1 : SUCCESS in 1h 15m 06s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9e0b2a72b6c7461b99bd305a9e52f955 : SUCCESS in 1h 45m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2b198824abd54eb0b622596e1e3d142b : SUCCESS in 45m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/61c2625c512f45fb9561f95dc067cd86 : SUCCESS in 48m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/940b0b92e4994554af8bdec6b56fd979 : SUCCESS in 25m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb0a3e91a6274b3dbc50be7ebf413815 : SUCCESS in 3m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2ea7574ed13641a29286842cdaeecc85 : FAILURE in 37m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d2acef16ac144263b03903e36ffd26c9 : SUCCESS in 40m 02s","accounts_in_message":[],"_revision_number":8},{"id":"46e08c2645f56583114bbcbd489c22b3360561fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 19:54:08.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"3434117ef0d97d2ee63d65c47f2a4005561e87da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 20:27:08.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa53c0cc371547f9bdb14122e797d4ef\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/079a26f9165e4a4091733c16963e30e2 : TIMED_OUT in 1h 30m 54s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6d5d04d5bd9348ee9745a589eb4c0506 : SUCCESS in 1h 04m 43s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/26a07d0df508464e965c71563527bc9f : SUCCESS in 1h 11m 48s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/40ef09ae2b98441492bfc32b3c394bdc : SUCCESS in 1h 02m 17s","accounts_in_message":[],"_revision_number":8},{"id":"4b8fff829a55ca09781a35cede1e0ed27fd95a18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 21:15:37.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6e5f70f326340a7973ffa8110d3b31e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ced4269b7464fd884facaa838542f1a : SUCCESS in 8m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/22e4e78dccfd4c67b2d3ef9706dba67c : SUCCESS in 48m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/683f0ae24f004419974393cd58458293 : SUCCESS in 41m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/82364a24d51442dcadce7a7b00412274 : SUCCESS in 52m 30s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/6bd394d79f374f04b640119d4bab32e5 : SUCCESS in 9m 43s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/ad89bcfe038f44ef8fd79724fe9d6697 : SUCCESS in 5m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0102a88a5d00473b8081508399f138c1 : SUCCESS in 21m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d2520becbf52493f814d3068678b0bd7 : SUCCESS in 15m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e69a07f59b13487794a50d25a0badfab : SUCCESS in 21m 36s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8180b9afdc5a4cca8b6e872c9bfc611c : SUCCESS in 13m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d2fe66f0bdf3474192a377d600b04bd6 : SUCCESS in 21m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/32ee073a2b824827adaf2187dac67d9d : SUCCESS in 16m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b4bb772d21144c3b816b97efb7f941d4 : SUCCESS in 22m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f3f3be5c821941ec87eedfda8ecc8b27 : SUCCESS in 15m 20s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7b7ae67fae2e4cc7bb0001ed0e785cff : SUCCESS in 12m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/efeb0421973f4f15a798de89794bf29e : SUCCESS in 21m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2cac3258946d47e4af98aead54abdde8 : SUCCESS in 6m 20s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/13ea718161a64835856aec9948e119d7 : SUCCESS in 7m 23s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/42f9c1cdd97445a9915a5eee6742c540 : SUCCESS in 1h 14m 02s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/202583acfd374c47a90d5975488a6988 : SUCCESS in 1h 21m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5d06de03f4654165bca666b9cc8d52e4 : SUCCESS in 38m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8e34bb71cc70460c867e1c1d7d5fc3f5 : SUCCESS in 46m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/472a06c0ed7543e7a8813ba150a41b4f : SUCCESS in 12m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c4a43dfbb164311953f5c259b0afa75 : SUCCESS in 4m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d22d35eaee94e30a3fedc4b6e4dcf5f : SUCCESS in 41m 40s","accounts_in_message":[],"_revision_number":8},{"id":"3a4fda7d6f48a552a7551b290b04aaa0a4e40945","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 21:15:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"25008513de9014d956bdea1db5e701dee725f628","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 21:16:40.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a853439aee934a72af5b7b0e6e81699f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27fc3f9fe6b042ac8f96346d0821fdcd : SUCCESS in 43s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/a26fa77573df461385c43260198136ab : SUCCESS in 31s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/f1882f18a1dc4b418d537c8a5a6d8e7c : SUCCESS in 34s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"0fa397d80e5a65c64588c2b4209cb36a401e9e98","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-01-11 16:12:04.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"ddba855af5cc0a9bfb4a732e464f017f93a13dba","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-01-19 14:14:36.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"12cb572642d0d9094b3c14d9d8e5703e9848d7bc","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-01-20 00:20:08.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"b4124e0cd25b7f50b7d6be284fe2b936d7191acc","revisions":{"85c0ce297257ce3b25af35364abd79a2ea414dce":{"kind":"REWORK","_number":1,"created":"2023-01-05 16:58:35.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/90/869390/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/869390/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/869390/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/869390/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/869390/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/869390/1"}}},"commit":{"parents":[{"commit":"87e9ca7a66236e35de84f76e3f470f0b34d204c7","subject":"Merge \"Sharder: add a new probe test for the case of slow parent sharding.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87e9ca7a66236e35de84f76e3f470f0b34d204c7"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-01-05 16:57:34.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-01-05 16:57:34.000000000","tz":-480},"subject":"Memcached: add timing stats to set/get and other public functions","message":"Memcached: add timing stats to set/get and other public functions\n\nChange-Id: If6af519440fb444539e2526ea4dcca0ec0636388\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/85c0ce297257ce3b25af35364abd79a2ea414dce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/85c0ce297257ce3b25af35364abd79a2ea414dce"}]},"branch":"refs/heads/master"},"8becc1251c7aeb05cb70ddc5e282065dd97cf5c5":{"kind":"REWORK","_number":2,"created":"2023-01-05 23:14:09.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/90/869390/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/869390/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/869390/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/869390/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/869390/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/869390/2"}}},"commit":{"parents":[{"commit":"87e9ca7a66236e35de84f76e3f470f0b34d204c7","subject":"Merge \"Sharder: add a new probe test for the case of slow parent sharding.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87e9ca7a66236e35de84f76e3f470f0b34d204c7"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-01-05 16:57:34.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-01-05 23:13:57.000000000","tz":-480},"subject":"Memcached: add timing stats to set/get and other public functions","message":"Memcached: add timing stats to set/get and other public functions\n\nChange-Id: If6af519440fb444539e2526ea4dcca0ec0636388\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8becc1251c7aeb05cb70ddc5e282065dd97cf5c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8becc1251c7aeb05cb70ddc5e282065dd97cf5c5"}]},"branch":"refs/heads/master"},"242ffe8fb59d826708b378ed93e3bf5410cc6636":{"kind":"REWORK","_number":3,"created":"2023-01-05 23:23:27.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/90/869390/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/869390/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/869390/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/869390/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/869390/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/869390/3"}}},"commit":{"parents":[{"commit":"87e9ca7a66236e35de84f76e3f470f0b34d204c7","subject":"Merge \"Sharder: add a new probe test for the case of slow parent sharding.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87e9ca7a66236e35de84f76e3f470f0b34d204c7"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-01-05 16:57:34.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-01-05 23:23:20.000000000","tz":-480},"subject":"Memcached: add timing stats to set/get and other public functions","message":"Memcached: add timing stats to set/get and other public functions\n\nChange-Id: If6af519440fb444539e2526ea4dcca0ec0636388\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/242ffe8fb59d826708b378ed93e3bf5410cc6636"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/242ffe8fb59d826708b378ed93e3bf5410cc6636"}]},"branch":"refs/heads/master"},"359870c09ec770b5d34e54cb26907b35354c0256":{"kind":"REWORK","_number":4,"created":"2023-01-06 00:05:09.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/869390/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/869390/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/869390/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/869390/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/869390/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/869390/4"}}},"commit":{"parents":[{"commit":"87e9ca7a66236e35de84f76e3f470f0b34d204c7","subject":"Merge \"Sharder: add a new probe test for the case of slow parent sharding.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87e9ca7a66236e35de84f76e3f470f0b34d204c7"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-01-05 16:57:34.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-06 00:05:04.000000000","tz":-480},"subject":"Memcached: add timing stats to set/get and other public functions","message":"Memcached: add timing stats to set/get and other public functions\n\nChange-Id: If6af519440fb444539e2526ea4dcca0ec0636388\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/359870c09ec770b5d34e54cb26907b35354c0256"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/359870c09ec770b5d34e54cb26907b35354c0256"}]},"branch":"refs/heads/master"},"bb68485eb8029d2503b7efa830bbab95633891b8":{"kind":"REWORK","_number":5,"created":"2023-01-06 04:34:53.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/90/869390/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/869390/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/869390/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/869390/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/869390/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/869390/5"}}},"commit":{"parents":[{"commit":"87e9ca7a66236e35de84f76e3f470f0b34d204c7","subject":"Merge \"Sharder: add a new probe test for the case of slow parent sharding.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87e9ca7a66236e35de84f76e3f470f0b34d204c7"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-01-05 16:57:34.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-01-06 04:34:42.000000000","tz":-480},"subject":"Memcached: add timing stats to set/get and other public functions","message":"Memcached: add timing stats to set/get and other public functions\n\nChange-Id: If6af519440fb444539e2526ea4dcca0ec0636388\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bb68485eb8029d2503b7efa830bbab95633891b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bb68485eb8029d2503b7efa830bbab95633891b8"}]},"branch":"refs/heads/master"},"2cac39162cba12d0fd394725fca1c545dc628b81":{"kind":"REWORK","_number":6,"created":"2023-01-06 05:04:18.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/869390/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/869390/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/869390/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/869390/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/869390/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/869390/6"}}},"commit":{"parents":[{"commit":"87e9ca7a66236e35de84f76e3f470f0b34d204c7","subject":"Merge \"Sharder: add a new probe test for the case of slow parent sharding.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87e9ca7a66236e35de84f76e3f470f0b34d204c7"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-01-05 16:57:34.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-06 05:04:10.000000000","tz":-480},"subject":"Memcached: add timing stats to set/get and other public functions","message":"Memcached: add timing stats to set/get and other public functions\n\nChange-Id: If6af519440fb444539e2526ea4dcca0ec0636388\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2cac39162cba12d0fd394725fca1c545dc628b81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2cac39162cba12d0fd394725fca1c545dc628b81"}]},"branch":"refs/heads/master"},"235155e13b049f39417af8085fa01ab887e9094e":{"kind":"REWORK","_number":7,"created":"2023-01-06 05:08:09.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/869390/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/869390/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/869390/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/869390/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/869390/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/869390/7"}}},"commit":{"parents":[{"commit":"87e9ca7a66236e35de84f76e3f470f0b34d204c7","subject":"Merge \"Sharder: add a new probe test for the case of slow parent sharding.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87e9ca7a66236e35de84f76e3f470f0b34d204c7"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-01-05 16:57:34.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-06 05:07:42.000000000","tz":-480},"subject":"Memcached: add timing stats to set/get and other public functions","message":"Memcached: add timing stats to set/get and other public functions\n\nChange-Id: If6af519440fb444539e2526ea4dcca0ec0636388\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/235155e13b049f39417af8085fa01ab887e9094e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/235155e13b049f39417af8085fa01ab887e9094e"}]},"branch":"refs/heads/master"},"b4124e0cd25b7f50b7d6be284fe2b936d7191acc":{"kind":"REWORK","_number":8,"created":"2023-01-06 18:02:24.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/90/869390/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/869390/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/869390/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/869390/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/869390/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/869390/8"}}},"commit":{"parents":[{"commit":"87e9ca7a66236e35de84f76e3f470f0b34d204c7","subject":"Merge \"Sharder: add a new probe test for the case of slow parent sharding.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87e9ca7a66236e35de84f76e3f470f0b34d204c7"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-01-05 16:57:34.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-01-06 18:02:15.000000000","tz":-480},"subject":"Memcached: add timing stats to set/get and other public functions","message":"Memcached: add timing stats to set/get and other public functions\n\nChange-Id: If6af519440fb444539e2526ea4dcca0ec0636388\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4124e0cd25b7f50b7d6be284fe2b936d7191acc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4124e0cd25b7f50b7d6be284fe2b936d7191acc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
