)]}'
{"id":"openstack%2Fswift~869406","triplet_id":"openstack%2Fswift~stable%2Ftrain~I316170442c67c1b4a5b87f9a1168cc04ca2417b8","project":"openstack/swift","branch":"stable/train","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-01-05 22:14:32.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I316170442c67c1b4a5b87f9a1168cc04ca2417b8","subject":"CI: pin tox at the project level","status":"MERGED","created":"2023-01-05 21:05:17.000000000","updated":"2023-01-06 01:13:54.000000000","submitted":"2023-01-06 01:12:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869406","meta_rev_id":"03de2ff40b023d8847044cf396ac85d635de70d6","_number":869406,"virtual_id_number":869406,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-06 01:12:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-05 23:17:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-05 23:17:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-05 22:02:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"080095b5e1ee149a4be62a555080dff2d094c117","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-05 21:05:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d820c4500d4ec3615d7d8492f137a3cd9b4f1e2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 22:02:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3f01188d48c47969b215f3e4cc40090\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2fee639a871f49eaa687626b7d47f7af : SUCCESS in 2m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/446556357c8f4ae9ae73d70c3a85d510 : SUCCESS in 5m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/73ae54926fe545319386310b9917870b : SUCCESS in 3m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b55f7f1c286146c5b5e9cc418a52787c : SUCCESS in 5m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/6eb137d72a044d45b3c33e089efb9933 : SUCCESS in 53m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6b29991485ee4026a5b56f6c39f726a8 : SUCCESS in 37m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3a838bf7c9f46be868f7ce5273306d0 : SUCCESS in 39m 46s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3fe8743925a74622a7497b4a0b2f085d : FAILURE in 3m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9c50efe18423468198396a11082360d1 : SUCCESS in 10m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a230958c021e4507a7185e369297a2d1 : SUCCESS in 10m 28s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6d338331679e4b898803ab9a25efd5f3 : SUCCESS in 12m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e010cdbfda5c41d68686f062f3e99983 : SUCCESS in 9m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/11acc31de02943a5aaa8d9e2465b6749 : SUCCESS in 8m 39s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/4f53eaf5109c4257afcf1b4e20474cca : SUCCESS in 8m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bf67dabebce64e2091982000423037cc : SUCCESS in 11m 12s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/1ab19ea08de449d89c01ddcce8116cd2 : FAILURE in 12m 16s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a1e82c9cc25d4f6a857a8845c292ab48 : SUCCESS in 9m 31s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8ddd3828b02e4f2bb6ecd9da8a6766ef : SUCCESS in 9m 37s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/a0ef3283fa42461b9aeed2c4b010e150 : SUCCESS in 9m 20s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3c9a169b7feb44afa4800dee8f376523 : SUCCESS in 9m 06s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/c8ebaf79ce334374abf399d8f211594e : SUCCESS in 12m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/51ef772f028143828590859afedb29cd : FAILURE in 7m 13s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1654a25e10c4410c86976e817102c071 : SUCCESS in 31m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/834ce4910f544dc2b9941f2c9aeb0ea6 : SUCCESS in 23m 35s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/9f825c95b5e548ce873d09ccea186d4a : SUCCESS in 23m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dfd1db426c0e48c8a38a9e9aa29b766c : SUCCESS in 21m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47db3fd5c74f4942b02887f40be73f85 : SUCCESS in 11m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea036a8b510748bea952199cae5a54ec : SUCCESS in 4m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4b443dc48a0f410aa5bbec85c1438363 : RETRY_LIMIT in 5m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/68c950fc763c4fbca9ae5bac9c3e38a3 : SUCCESS in 45m 03s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2014eded1b294316aee34feb096874ea : SUCCESS in 48m 08s","accounts_in_message":[],"_revision_number":1},{"id":"abe3152aceaadd50cad4554b435ef15cb63c02b5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-05 22:14:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9554750f51f7f84e907a01f3a71cdfc6d21dc7f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 23:16:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/531ce5ea3ef84b4ca641a2df0af807ed\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a970078a76b24e9aac7eb852687d1834 : SUCCESS in 2m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eae9b11314f2427293dab647100e036f : SUCCESS in 5m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/83646e606a97449182ecd2c892d879c2 : SUCCESS in 2m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb633c12a59e4003baec79a1e4f45507 : SUCCESS in 5m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/85e71f4a576044819edfec6c0f4c9415 : SUCCESS in 43m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/75f8035a0c664e198ef6672c1e51dc1e : SUCCESS in 28m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2a1bd6704ec74b2788dcedee560266bf : SUCCESS in 37m 26s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3198fbd37f744953ba841b6cc9287e93 : FAILURE in 5m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ac3b70f17fdc4e41911c25eef777e0bb : SUCCESS in 9m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5b348e9601ce413b823810435fb4b1be : SUCCESS in 12m 09s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b4ff9766f7684134bcebc4c484a327a6 : SUCCESS in 11m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/24d58b14bf0349b6906405919f4a1194 : SUCCESS in 8m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/db958c1837064a48acb74204bc1c08df : SUCCESS in 9m 11s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/64a57d82ad3441dba879ecfd88ce5283 : SUCCESS in 9m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/949c45b9111643fd93b5517ec5aca8b2 : SUCCESS in 9m 34s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/cc1072f1ba3845ff8d08f974bc62e291 : SUCCESS in 10m 46s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6eeb09fd4d574db9a54557a021116571 : SUCCESS in 10m 33s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1b8a9fc0a29542c08cdd074d7db86791 : SUCCESS in 8m 37s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/82629a104db44249a47a1dffe259b7a0 : SUCCESS in 9m 25s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ce9f212aaddc4957ab578f11182831ef : SUCCESS in 11m 18s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/19e284a3f84f4a0093a041b4e0bce73e : SUCCESS in 13m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7ca7d022df9c45e9a497d1a2a60198ca : FAILURE in 5m 56s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1aab71a3915c485f9409a2ca5c12befe : SUCCESS in 37m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a01ed93984c44fc9824be6352ebc47f1 : SUCCESS in 31m 05s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/7bb5761779ea400c96551787613ebe94 : SUCCESS in 46m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3e0836e054b14ca68e98bb866f17ca5d : SUCCESS in 18m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8cf19a85d4c6479dbe548e4bd8ff7e6a : SUCCESS in 11m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50499024917c43a08fecbe60f514350e : SUCCESS in 5m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b6d4d00002c542c091419c98d062423d : RETRY_LIMIT in 7m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4125abeaf1564d8ab6502aedd2c838fc : SUCCESS in 47m 36s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/15fc7766704346ee93e8a0f8ec52bd24 : SUCCESS in 47m 55s","accounts_in_message":[],"_revision_number":1},{"id":"d0d313467e62a9bd4891bb5c7e48481649f68909","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-05 23:17:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"45dc991e2e15b1e43ad332c1bd8450e986cdf470","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 23:18:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"543f9a4b0ff6256f5c6cd1dc414c679ebca1bb44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 01:12:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6542d142b47e4a9fb5cf17f409f1493c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fb75a48aab0149f38c8a733fc44fa912 : SUCCESS in 1m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2fdfd7e623e4b7abb16a0cadb3e2fc7 : SUCCESS in 4m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ff136b9810544d33a7c2448188583e8f : SUCCESS in 2m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7915d3350cd44a6b397e766570534a3 : SUCCESS in 4m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/0688a3b0986b4d278ac00c34c54bf23d : SUCCESS in 44m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c2440940b0444677ba99cba3c48c1de8 : SUCCESS in 30m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3029e3b2abdc46daa736c8a61e30d752 : SUCCESS in 44m 05s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/834db40557aa4c2281049d513fad050e : POST_FAILURE in 4m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/858102b7a056426daee3c28ad7b149c2 : SUCCESS in 11m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/975a5495c22649d3ab314ef535ceb4dc : SUCCESS in 9m 28s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ea1074cdd86c45db8d214cf2302501bf : SUCCESS in 9m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5e69dd27b85146f7a844f9246288bd6f : SUCCESS in 7m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/de4994508c4141c9b3b47e644018d5a2 : SUCCESS in 9m 07s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/977f8ae1124042169096ebc48b7cae70 : SUCCESS in 8m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fbbc412581f34283af6025506ea1634f : SUCCESS in 9m 28s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/2a66e51b179a491c91f2efc4e1a8fd97 : SUCCESS in 11m 47s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ff220027a0084b6bb63d89150a760036 : SUCCESS in 8m 18s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/cf4ec3b071834e509ea8ffb64a856cd4 : SUCCESS in 9m 25s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ef0eba6ef5f84cbebb670a06583b1fcc : SUCCESS in 10m 38s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/cabe9eaec988495eaa79bff93bfe6359 : SUCCESS in 12m 08s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1d7a79fe4f6b4dc6911364a7b36f8777 : SUCCESS in 37m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3c5f30acf32b422d84c41f49401fca1d : SUCCESS in 18m 28s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/80fcc3ff20ba44ffac0ca12e9751dd00 : SUCCESS in 34m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/aa69625963d143f487c12422c8b67175 : SUCCESS in 31m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d0ef8e5c1a64cfe9b6eaf5588270c85 : SUCCESS in 11m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a4f16aa62de43bf9249a92c7e5d6ef6 : SUCCESS in 5m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6bcf811e564d4f33b7f7ecbf30bcbfbb : SUCCESS in 56m 27s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0ac5b18f1f3e4319b509e655476f9884 : SUCCESS in 39m 54s","accounts_in_message":[],"_revision_number":1},{"id":"344be70605195ecdac85f6c0a3cee662a545757a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 01:12:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"03de2ff40b023d8847044cf396ac85d635de70d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 01:13:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9182f40c48584ea5a058bf88a3922a95\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d6bdc76ff934af0b85e50b43dab105d : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db0771b5b4e6496787e3ea096a934896 : SUCCESS in 40s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/5320ef3a69d74cc7a2dfcbbf3fccadc8 : FAILURE in 27s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"78bc4d090ed3cb83462b7daa864e726eccbcac2c","revisions":{"78bc4d090ed3cb83462b7daa864e726eccbcac2c":{"kind":"REWORK","_number":1,"created":"2023-01-05 21:05:17.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/06/869406/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/869406/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/869406/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/869406/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/869406/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/869406/1"}}},"commit":{"parents":[{"commit":"2829194bbf78058f1543819cad63fe596d0e358f","subject":"s3api: Use constant-time string comparisons in check_signature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2829194bbf78058f1543819cad63fe596d0e358f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-12-30 18:44:15.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-05 21:05:09.000000000","tz":-480},"subject":"CI: pin tox at the project level","message":"CI: pin tox at the project level\n\nNew tox has a bunch of incompatibilities -- on master it\u0027s worth fixing\nthings so we can work in the new world, but on stable branches it seems\nsimplest to keep ourselves in the old world.\n\nDo it at the project level to ensure the pin applies to the\nperiodic-stable jobs.\n\nAlso, skip installing python-dev on jammy.\n\nAlso, pin PasteDeploy to 2.1.1 on CentOS 7. This was the last version to\nsupport py27.\n\nChange-Id: I316170442c67c1b4a5b87f9a1168cc04ca2417b8\nRelated-Change: If69ae0f8eac8fe8ff7d5e4f4f1bff6d0ea9e7a8b\nCo-Authored-By: Matthew Vernon \u003cmvernon@wikimedia.org\u003e\n(cherry picked from commit cc033154ad4a4f345258457f3ceed9143fb3d46d)\n(cherry picked from commit eb994ea501d96fdf73d4df9479a3b7d51e2d5744)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78bc4d090ed3cb83462b7daa864e726eccbcac2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78bc4d090ed3cb83462b7daa864e726eccbcac2c"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
