)]}'
{"id":"openstack%2Fswift~869502","triplet_id":"openstack%2Fswift~stable%2Fstein~Iba833ba18d6b80ce9f061bce59644711525b6603","project":"openstack/swift","branch":"stable/stein","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-01-06 23:32:40.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Iba833ba18d6b80ce9f061bce59644711525b6603","subject":"Fix probe tests on stein","status":"MERGED","created":"2023-01-06 18:55:52.000000000","updated":"2023-01-07 00:49:34.000000000","submitted":"2023-01-07 00:48:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869502","meta_rev_id":"2ff9716e5a1a6e4f9e8786e3cb70df5c63cf5c1d","_number":869502,"virtual_id_number":869502,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-07 00:48:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-07 00:12:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-07 00:12:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-06 19:36:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c5f67a11c452519a05b72124c7e9512bf7f6d306","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-06 18:55:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d87055c53eaafe1158ff690697e26b4612bdb7cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 19:36:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b7772d810e64442392a08a0bdcb9a95c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1262ff3243154a818fb51cd24e7e9f86 : SUCCESS in 5m 21s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/34bc721c820342c9971afd507b3d8e34 : SUCCESS in 11m 00s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/c9385ca42d61409fa9c0f72f464399cb : SUCCESS in 7m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/52f4bb0a3c414013aecfe9d209e4e684 : SUCCESS in 8m 03s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/909ec7e7610142eb826dc2f63b6d9d41 : SUCCESS in 9m 40s (non-voting)\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/03e708cb3beb4058b3019cead984a92b : SUCCESS in 8m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a4df273a9f7843e8a08bbc7dc4024a14 : SUCCESS in 8m 26s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/45425c819a1b4515bca9ad06295d3943 : SUCCESS in 8m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e2bf6fe177054c1588f46e6d07c30151 : SUCCESS in 8m 42s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/e78c4e4947f74bff915526412a0ce46e : SUCCESS in 9m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e9ba2d4eed7f46d4b5145d08244f0d8f : RETRY_LIMIT in 3m 14s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ef3df82958b241bb803d38e30f57fc0f : RETRY_LIMIT in 3m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fc89c57f70c941549da0c363d9e538c6 : SUCCESS in 35m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e2825ccffb8f4cd5834760880596b8c4 : SUCCESS in 30m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34bfd3a053d5457685a43e489a9321fc : SUCCESS in 10m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a244c225f0f24aa682fc946a8dd6132d : SUCCESS in 5m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/33f20e2c33284371aae6f0d98c874a9c : RETRY_LIMIT in 5m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3bea5b385b73cea730ea1aa20881b118d93eb1b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-06 22:29:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9cd5dc0d29bbddfed2c58d6c6b41be71dd3b46b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 23:03:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7abb15b010ed4491b07cf8f8111689c3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fd6576ecf3e401c8b361cf9d0659111 : SUCCESS in 4m 12s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/36b616931af94aa6a2c673bbc099e9e6 : SUCCESS in 10m 10s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/b57b1158d47a494099d12664fd6a4e4e : SUCCESS in 8m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/28a3e53c543545eb872a39992661ff31 : SUCCESS in 8m 32s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2bc396d0b0dc46efa21c570c6211da86 : FAILURE in 7m 01s (non-voting)\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3738489c2a244979ad341c11895167fd : SUCCESS in 8m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e683f8cbd7984677bbdfb02ce25756dd : SUCCESS in 8m 00s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/db4988f7a90940f493bfb4cbf6b6813e : SUCCESS in 7m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0065d22fe978421c8e191251d7556ec9 : SUCCESS in 7m 10s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/59169c1667d54cac9843b8b8ee0811ee : SUCCESS in 9m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ad7673fa4e044571af058b99a7381262 : POST_FAILURE in 3m 38s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/42e1809eb03e4dd59417948e1795a7b5 : FAILURE in 20m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/76c37958ae644d87a887f33bebc96af6 : SUCCESS in 29m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/316adaff8d4445ddb8c8fdccff166bea : SUCCESS in 18m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d8d203d64e604a47b560c7c51fc88a43 : SUCCESS in 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23ef7be7f0d842a9ac162a89e3882f68 : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/def758f966db42659c1d09b04aa4f853 : RETRY_LIMIT in 8m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7d3bddaa3b506909c7c0613db9ebab6d4aceb2a7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-06 23:32:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"20ad45e76bceb0d07ed7daededbb1a57cd847135","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 00:07:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8b6e5bc4c794bf48b7471963cfd3cae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a8e0a643f264331941a10bacd92f616 : SUCCESS in 5m 05s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dfca400d27184f9a99fa34063a1c835a : SUCCESS in 9m 24s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/0384baab4ad6468c80d9d6e7dc054d7e : SUCCESS in 9m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1fbfac534a8e4d94b9cb6259ff32589f : SUCCESS in 7m 11s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c154f9fb66d8410287e131b5ca7f3cde : SUCCESS in 8m 38s (non-voting)\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/95a1af9e0c11428bad1ad9b7686dbd03 : SUCCESS in 7m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ffae507c6e894c5399c8dbeb2084f031 : SUCCESS in 7m 47s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/4e469c026c1c4a85a23c279aca124004 : SUCCESS in 7m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3330083ea61640c5b79f63f860be187c : SUCCESS in 8m 53s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/0f1964b282fe47a09748f69f45362801 : SUCCESS in 9m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8ceb33a858ee48ea8e44cc51c5ef375c : POST_FAILURE in 4m 09s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/56ad1995ffdc4d48ab724f7f1919733d : SUCCESS in 30m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/efeae1a868404ad09b10409d02576b30 : SUCCESS in 21m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/713badb080d64bdb8d80635a4fb37c63 : SUCCESS in 20m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da25625a90a54226a91c2f48698e54fe : SUCCESS in 10m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e4acbf4ad554460a47c15bd9f9a6a6e : SUCCESS in 4m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ee814bc2106540ed9141923876075dee : RETRY_LIMIT in 5m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d82669f7ff5d81c63a380cdf74bb505672367b5c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-07 00:12:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"689176e18deaed90ad92f3bda6d92c6e5e49f16a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 00:13:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"84aec492dd25191c5b7831296ac5478bf7854736","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 00:48:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac34eefce04f414abc1333ab9628ecf2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89c40374c94a47c3a18b65e5613c690e : SUCCESS in 4m 37s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/544016e35f4d47ceb44a2bb267091fb1 : SUCCESS in 9m 31s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/d139e105a68d422c908818bd45080edc : SUCCESS in 9m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5cafc14711e94b53ad42593b4da4a199 : SUCCESS in 7m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7ab0f966fb4549448063a20b0720ecc6 : SUCCESS in 9m 18s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/907698f757254471b40e3b83d8b55d23 : SUCCESS in 7m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/78a69b28e86543199b1233a685723cbb : SUCCESS in 8m 38s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/28db6658252f457dbdfdb72e53fe6167 : SUCCESS in 9m 12s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7510cb43e4384cb8b60eefa5f1e596de : SUCCESS in 24m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a6a0b7356b6f4ffca3299f81a4e55093 : SUCCESS in 30m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d3dd6e31e6eb42c38f155dad67449671 : SUCCESS in 30m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6ee61487f92471d97aea916d42f1f98 : SUCCESS in 9m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aa9dd4dd1b6448b9ad6ca97fd247cf7 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":2},{"id":"5a29f3a0b2bd9376da3c76fdb3ab18457494e05a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 00:48:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2ff9716e5a1a6e4f9e8786e3cb70df5c63cf5c1d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 00:49:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a35050ed115142288280c3b20d4d4015\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f33b4b3939c94ef78387ba4c561e1686 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"eb2e840edd65a8a79d85bc7ca8527b4bb7ba24e6","revisions":{"8f7a16f616cdd6357c9c701d73fa6e6f55eb42fa":{"kind":"REWORK","_number":1,"created":"2023-01-06 18:55:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/02/869502/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/02/869502/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/869502/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/869502/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/869502/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/02/869502/1"}}},"commit":{"parents":[{"commit":"cd79c4ed594b651ff083c4e2b639bff1daf1626e","subject":"Quote paths before sending them to swob.Request.blank","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cd79c4ed594b651ff083c4e2b639bff1daf1626e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-16 18:34:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-06 18:55:32.000000000","tz":-480},"subject":"Install pip from PyPI","message":"Install pip from PyPI\n\nThis is a combination of 3 commits:\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nUse ensure-pip role\n\nHopefully this will fix the currently-broken probe test gate?\n\nDepends-On: https://review.opendev.org/#/c/736070/\n(cherry picked from commit 51a587ed8dd5700b558ad26d70dcb7facc0f91e4)\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nUse latest pip in gate jobs\n\nWe\u0027ve seen failures with probe tests lately where dnspython 2.0.0 is\ngetting installed even though it doesn\u0027t support py2 anymore. I think\nusing latest pip should be better about noticing that and installing the\nlast 1.x release intead?\n\n(cherry picked from commit 1c91547643f9c85f152d1d8c15431fec8334e214)\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nPrevent upgrading to pip 21+\n\n(cherry picked from commit 0b870eb94dfe9769a07695b296eaa6758820a5f3)\n\nChange-Id: Iba833ba18d6b80ce9f061bce59644711525b6603\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f7a16f616cdd6357c9c701d73fa6e6f55eb42fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f7a16f616cdd6357c9c701d73fa6e6f55eb42fa"}]},"branch":"refs/heads/stable/stein"},"eb2e840edd65a8a79d85bc7ca8527b4bb7ba24e6":{"kind":"REWORK","_number":2,"created":"2023-01-06 22:29:03.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/02/869502/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/02/869502/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/869502/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/869502/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/869502/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/02/869502/2"}}},"commit":{"parents":[{"commit":"cd79c4ed594b651ff083c4e2b639bff1daf1626e","subject":"Quote paths before sending them to swob.Request.blank","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cd79c4ed594b651ff083c4e2b639bff1daf1626e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-16 18:34:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-06 21:11:41.000000000","tz":-480},"subject":"Fix probe tests on stein","message":"Fix probe tests on stein\n\nMostly, this involves installing pip from PyPI, and making sure we use\npip (not setup.py) to install swift. Plus solve a small issue with\ncrudini, iniparse, and system packages.\n\nThis is a combination of 3 commits, plus bits and pieces of two others:\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nUse ensure-pip role\n\nHopefully this will fix the currently-broken probe test gate?\n\nDepends-On: https://review.opendev.org/#/c/736070/\n(cherry picked from commit 51a587ed8dd5700b558ad26d70dcb7facc0f91e4)\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nUse latest pip in gate jobs\n\nWe\u0027ve seen failures with probe tests lately where dnspython 2.0.0 is\ngetting installed even though it doesn\u0027t support py2 anymore. I think\nusing latest pip should be better about noticing that and installing the\nlast 1.x release intead?\n\n(cherry picked from commit 1c91547643f9c85f152d1d8c15431fec8334e214)\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nPrevent upgrading to pip 21+\n\n(cherry picked from commit 0b870eb94dfe9769a07695b296eaa6758820a5f3)\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nRelated-Change: Iae86d8838854023010686d3d4bed3befe6160ca5\nRelated-Change: I4be1e86e3291ad1619c695fb93d7cadf053b556d\nChange-Id: Iba833ba18d6b80ce9f061bce59644711525b6603\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb2e840edd65a8a79d85bc7ca8527b4bb7ba24e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb2e840edd65a8a79d85bc7ca8527b4bb7ba24e6"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
