)]}'
{"id":"openstack%2Fswift~869519","triplet_id":"openstack%2Fswift~stable%2Fzed~I8b4527ffff557ab227910816a729579c06557911","project":"openstack/swift","branch":"stable/zed","topic":"tox4","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-01-07 07:11:46.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-01-07 05:43:53.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I8b4527ffff557ab227910816a729579c06557911","subject":"[stable-only] Pin tox\u003c4 for stable branches (\u003c\u003dstable/zed) testing","status":"MERGED","created":"2023-01-07 02:40:53.000000000","updated":"2023-01-07 07:12:54.000000000","submitted":"2023-01-07 07:11:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869519-tox4","meta_rev_id":"6a4f01d3421ddecaa442d95bd51b588a67e9eb88","_number":869519,"virtual_id_number":869519,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-07 07:11:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-07 05:43:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-07 05:43:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-07 02:50:45.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2023-01-07 03:50:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-07 05:43:53.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"3f4cf418edea6fd60a26cca27fe03c28c4216946","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-07 02:40:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0861a1fbf0390003cfb9026da86c1f2116bc5eb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-07 02:50:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3ef4fa9daa54b24a22d9ff46a98dfde8982885e1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-07 02:58:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"000c66b0c9b627f189de43413ce81015033718f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 03:50:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc5f01c3211044fea29eb30e934c46fa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc20dd87dbb24b549f82b261a5ed55e9 : SUCCESS in 9m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/50dd52c63bce4276bb5e8bea7f4f1e99 : SUCCESS in 53m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a144ca0e942b4104bbc90f0e1360a251 : SUCCESS in 43m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a5e5ee79380d4882acb9c3b0781368ae : SUCCESS in 46m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/339b06de2c7a4e74bd4bfd85d5b0e82f : SUCCESS in 19m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/77d236b549754e22926b678be23f3888 : SUCCESS in 19m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/adcc10ab83c44a4da994837b51674ec5 : SUCCESS in 21m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e42c767eec0743d9b0e70b91e289b388 : SUCCESS in 7m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/efcee9fda2cb4d4f9ff0f4d561d20c04 : SUCCESS in 7m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cb693735e3a848ac91daf78a06ff7325 : SUCCESS in 16m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a3a9d782539f45d48baf45609f412906 : SUCCESS in 12m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/26951bbfd3da425ebfa9921e007f889c : SUCCESS in 17m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9c715cf32d1c4f26b5de985ea59d6681 : SUCCESS in 17m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/196326244f454712aad92dc73a400ca4 : SUCCESS in 15m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/91c9348915f04983b4d0b140c0d4e845 : SUCCESS in 14m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/74b42d7771894026ab99805cb03e3931 : SUCCESS in 14m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/639a1748bdbd468c8eb58470694b0f26 : SUCCESS in 43m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4284a722a7234749a3e5a7c151ee914f : SUCCESS in 14m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/642cc232a3ee4639bb61df2f120e5e94 : SUCCESS in 7m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1160d76e24d3422a9debff23c7312cec : SUCCESS in 15m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c1cd288901b14e06b7fe1545f52364a2 : SUCCESS in 7m 32s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/37b5ee30a39e4386b974118827c73345 : SUCCESS in 56m 49s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0914b23c921d4b9ebadf77a17f67aee6 : SUCCESS in 59m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/082200d305944fedbcd6ae14ca032634 : SUCCESS in 46m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f18d3e25888248b6b3843b04ebcdfe91 : SUCCESS in 45m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f91e77af39247db821348bbeb9ce8fa : SUCCESS in 19m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fcb808da1fa4ddc8a55b305a61774d8 : SUCCESS in 6m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1ddebbe9f0404d748d4182339aa2575f : FAILURE in 19m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/027ccb03218e4806b824c3832a11ef1e : SUCCESS in 1h 01m 25s","accounts_in_message":[],"_revision_number":1},{"id":"57c0dd8399cd3a9803ab413bf7833579a6041b94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 04:01:42.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e5a55c267b9483898820e5be40d89e6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/efead5c1380842f28a212dbd2cc8076d : SUCCESS in 1h 05m 37s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e6b916b33d2d42edb6fba2bd1be7743e : SUCCESS in 52m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/445bcc38a63944b59b93df6ffd12da84 : FAILURE in 1h 16m 42s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2170ee1f160a45d48b18f2e901bda5ea : SUCCESS in 1h 02m 15s","accounts_in_message":[],"_revision_number":1},{"id":"caebe0da2340f7012a488d9807b57474a28137ce","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-07 05:43:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aacc3833bbd0ca33204d2ace3682b14b7a4a2186","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 05:44:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"522872bd4157cf8d9f48a5b50c37ebddee8812ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 07:11:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05ae11820bb64016862032d263b7ad44\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d87fcde60603499aaf07af44c20faf59 : SUCCESS in 7m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/86a206e981754154833874c22698f24c : SUCCESS in 42m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2ca9cca0245d4faebe0b03372d6614fe : SUCCESS in 52m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1dbc244ec39342a2abd5e2a15f8a4327 : SUCCESS in 42m 55s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/2cabb901ef4d4e7d9784ea015aeb2870 : SUCCESS in 8m 48s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/f56c14798e074fa885d3cbbcf7ec5435 : SUCCESS in 5m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0aea1ec220044744ad39cb4e5292ce1d : SUCCESS in 12m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7eeebfbb076a490b970e5bcc618a20c0 : SUCCESS in 16m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5eb246b148324e8b96f9a4499f64041f : SUCCESS in 12m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b31816da6c7b47f1b5579a277b36bb7f : SUCCESS in 16m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4a516b61ea9a41bf8cf6b7cf11367780 : SUCCESS in 17m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e344dd8ed8aa4638af2ba6dbdd9155d4 : SUCCESS in 14m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/25cbf23545d647f48bf22dd6979233ed : SUCCESS in 15m 20s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3d01cef2981f43e58a00b39ca1ae742a : SUCCESS in 16m 09s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7658be5b402241d7987f9a09ef2f373c : SUCCESS in 14m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b25957b069fe41d58ff00cf4a669b139 : SUCCESS in 6m 29s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ead96ee4763845d5bbec5d93911cbdce : SUCCESS in 7m 11s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3a24b79d13944399b1ffea54debe3d26 : SUCCESS in 46m 55s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/8422eca1e8fc4a2ba5768e1b5db339ce : SUCCESS in 51m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/355d2c5dadbf444a882c70c9600b23b7 : SUCCESS in 45m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bd050eae8c2a406090de61384da86002 : SUCCESS in 33m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d03a609ff550457a97d8dc19a2ba158c : SUCCESS in 15m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9eacd95164e0425fa28ac526fd45a47c : SUCCESS in 4m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/91401960ce3c4e859b49c69f11756f56 : SUCCESS in 33m 50s","accounts_in_message":[],"_revision_number":1},{"id":"ffe90d7600dad545415b8a67967abca28b6b5260","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 07:11:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6a4f01d3421ddecaa442d95bd51b588a67e9eb88","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 07:12:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fefd55a793f4d9d99f0cbe9281181a2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76884e0a7ca149f98d08b5920067609b : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/53459abfb768473983a967607cdc83d9 : SUCCESS in 36s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/52f0352db2f64096bbac5878095cc362 : SUCCESS in 35s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a2d363de61718d4b5a15e2c553946b71cd2464ed","revisions":{"a2d363de61718d4b5a15e2c553946b71cd2464ed":{"kind":"REWORK","_number":1,"created":"2023-01-07 02:40:53.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/19/869519/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/869519/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/869519/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/869519/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/869519/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/869519/1"}}},"commit":{"parents":[{"commit":"88e3f7be98f4710b67cbe8582afe47a18f521cad","subject":"Fix Zed CI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/88e3f7be98f4710b67cbe8582afe47a18f521cad"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-07 02:35:58.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-07 02:35:58.000000000","tz":-360},"subject":"[stable-only] Pin tox\u003c4 for stable branches (\u003c\u003dstable/zed) testing","message":"[stable-only] Pin tox\u003c4 for stable branches (\u003c\u003dstable/zed) testing\n\ntox4 need changes in tox.ini file to be compatible with it.\nThose are ok to fix in master brnanch but on stable branches\ntesting we do not need to fix them and should keep running the\ntox which was used in when those stable was released.\n\n- https://lists.openstack.org/pipermail/openstack-discuss/2023-January/031668.html\n\nIt is pinned same way in devstack side too\n- https://review.opendev.org/q/I9a138af94dedc0d8ce5a0d519d75779415d3c30b\n\nThis only pin for jobs running for stable branches \u003c\u003dstable/zed.\n\nChange-Id: I8b4527ffff557ab227910816a729579c06557911\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a2d363de61718d4b5a15e2c553946b71cd2464ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a2d363de61718d4b5a15e2c553946b71cd2464ed"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
