)]}'
{"id":"openstack%2Fswift~869522","triplet_id":"openstack%2Fswift~stable%2Fwallaby~I8b4527ffff557ab227910816a729579c06557911","project":"openstack/swift","branch":"stable/wallaby","topic":"tox4","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-01-09 00:29:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8b4527ffff557ab227910816a729579c06557911","subject":"[stable-only] Pin tox\u003c4 for stable branches (\u003c\u003dstable/zed) testing","status":"MERGED","created":"2023-01-07 02:54:31.000000000","updated":"2023-01-09 00:31:01.000000000","submitted":"2023-01-09 00:29:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869522-tox4","meta_rev_id":"1acf6fac041dc4baa63ec669be8ed71985e29079","_number":869522,"virtual_id_number":869522,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-09 00:29:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-08 23:20:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-08 23:20:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-07 03:55:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-08 23:20:52.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"26830c51946b1fe6e3839e713d742383aa5cce7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-07 02:54:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"473991a2f0a31e4e23da2bd1de62f8f84ed98c39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 03:55:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a5fcdfc50f144efb221d3dd352108c2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7cbe12dcfb14d65acbb13f6bf58eff8 : SUCCESS in 7m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/2fda08cf729b43a295ca5d39c31cd5e3 : SUCCESS in 1h 00m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0459c47fe7cf49a5b6d4d35f6908ddb0 : SUCCESS in 41m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b57ae2a71381402a99652807b5f15393 : SUCCESS in 48m 16s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/54fcbd4f99e24ddab4be74709b7ef430 : SUCCESS in 20m 31s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/4eba3126c4154c7896fa17cdb15aafed : SUCCESS in 18m 17s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/59752a34efdd42c08e71fb4d69f2a0f1 : SUCCESS in 19m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5f7c2ab9e40a4e2db6db69ef34cacef2 : FAILURE in 3m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/57c240d237c64439875bc6638a53d198 : FAILURE in 3m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a2fbb939088e4ca581007f06d98c9958 : SUCCESS in 12m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d24aab1a9154433c9dbc0ef0d7761504 : SUCCESS in 13m 15s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8f8dd59e35874213a174b4f4bb1e31ce : SUCCESS in 13m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d75c7aab377146ccbc3e86aca27d466b : SUCCESS in 13m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/781dcb276cdd48358999747dd1fec512 : SUCCESS in 14m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9fd08f2d51eb4457bae5e2ad3df95045 : SUCCESS in 16m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2f94439c0a0a42438bbecd765551cbb0 : SUCCESS in 15m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0091f2da02c34d46842d5163b43316fa : SUCCESS in 18m 39s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/477dba94e9ec406b847a7bccac67426e : SUCCESS in 15m 50s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1563e1d428e245fbaa2cfa37193af162 : SUCCESS in 17m 15s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/88bae70d758241699128e49e2f23f1a8 : SUCCESS in 23m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b91b133c1464488ea11e36dcd3776435 : SUCCESS in 5m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1eb789dc671c4775a5b606d2d4de3f7d : FAILURE in 7m 49s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fb55ca906c6c4971893d52f1a559d1b3 : SUCCESS in 45m 06s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/27a3875bf5f347b991d00811fb506ab6 : SUCCESS in 49m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9f1f53305aa045749d97f5de9e076b6e : SUCCESS in 40m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f3d6c347c1824447a0040799a380ecf7 : SUCCESS in 33m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6475dd75bef9439b8fa330fdbeca3b38 : SUCCESS in 15m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83bc1c3ef0b34d59b3d53a621db63a97 : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/67a2dba7d759423e9465bdcdf196869f : RETRY_LIMIT in 5m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b5d4517421194b86a2426af56d441414 : SUCCESS in 49m 37s","accounts_in_message":[],"_revision_number":1},{"id":"fdb2cec7c96e2cfb6b3add4bce568d1530fda5e8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-08 23:20:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"646cc7d788c3b6cc3c874491effcc6ba3ff83542","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-08 23:21:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a83f4c68776b7e09f61d2309c2a2150911bec857","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-09 00:29:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96eaf132c972478c8ef4bea4b3921867\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca114bd81e494d39949fecdcbabc1650 : SUCCESS in 7m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/410d101042d8459da9a17cd60f630ced : SUCCESS in 1h 03m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/033a6788302c4ce4b5b36efeae9e0b63 : SUCCESS in 50m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49ac27c2f3af47e5aa7317d5d50e0674 : SUCCESS in 1h 01m 24s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b69436ef24d14a4d9fc3bdbce46444c9 : POST_FAILURE in 5m 24s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/e7f74f62217245008c94605e8e890c8e : POST_FAILURE in 5m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0784b0586630429aad0979b5f461be5f : SUCCESS in 15m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b09ee6bec5684a0b9de85c521fafc399 : SUCCESS in 14m 21s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d0bbac0d47614aeab1c8aaced3b9f9e9 : SUCCESS in 14m 06s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/cf26a272e2ab4a3c9714e32a1b7b6753 : SUCCESS in 15m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c4c8512818734ee2bcbbc83e974c22a9 : SUCCESS in 15m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a37b76e8e67345a3b034be77b900fa2f : SUCCESS in 16m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/80e8863ae86544ab8ebf43b3327a0af9 : SUCCESS in 14m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7be3db336fe64cc082bc12e4913a9892 : SUCCESS in 14m 10s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/cfdaf12e26dc4f32bdc7078a4cc7a6d4 : SUCCESS in 14m 29s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/f5b04c4343ae4ef0bf57177d9f9c5c7f : SUCCESS in 14m 58s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f1ea757916e8482295811bc4b39cb11f : SUCCESS in 13m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b879ac74da4c42dab0edead302578934 : SUCCESS in 5m 08s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/52d21275136e4a2dab7b04b5c0f8184d : SUCCESS in 35m 33s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/92dffa7bf2a24ecbad9fecba8fb03bfe : SUCCESS in 41m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bc943f21b50a416c8813f9212bc9c274 : SUCCESS in 27m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/847e70c93d7f4b028dd6c3c652270372 : SUCCESS in 25m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae70f96874f64096a3d9469be8c7b696 : SUCCESS in 11m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b0eb5ef9cd741f09759a53c6ea31246 : SUCCESS in 7m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc1a2ebe09fe48659d0b659f80bb0271 : SUCCESS in 41m 50s","accounts_in_message":[],"_revision_number":1},{"id":"30a37d091e8d02fe96fbee627492e3311becba28","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-09 00:29:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1acf6fac041dc4baa63ec669be8ed71985e29079","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-09 00:31:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ee767efb0be4c89afff4260dfd366c9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e78b7047efe04ce185341f0db3a133fe : SUCCESS in 55s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/12c639b1a5bd4a9d96b6e41bf9a5ecb7 : FAILURE in 29s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/0be20f86820f417e9a4cf7c6cbf5a24d : FAILURE in 31s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"598f44320d082ce57aafff439ed460f9bae81051","revisions":{"598f44320d082ce57aafff439ed460f9bae81051":{"kind":"REWORK","_number":1,"created":"2023-01-07 02:54:31.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/22/869522/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/22/869522/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/869522/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/869522/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/869522/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/22/869522/1"}}},"commit":{"parents":[{"commit":"6a1a8ce14be8b4625eaf4f2b4d18ddfbdbbb4289","subject":"Stop partial()ing hashlib.new","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a1a8ce14be8b4625eaf4f2b4d18ddfbdbbb4289"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-07 02:35:58.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-07 02:54:13.000000000","tz":-360},"subject":"[stable-only] Pin tox\u003c4 for stable branches (\u003c\u003dstable/zed) testing","message":"[stable-only] Pin tox\u003c4 for stable branches (\u003c\u003dstable/zed) testing\n\ntox4 need changes in tox.ini file to be compatible with it.\nThose are ok to fix in master brnanch but on stable branches\ntesting we do not need to fix them and should keep running the\ntox which was used in when those stable was released.\n\n- https://lists.openstack.org/pipermail/openstack-discuss/2023-January/031668.html\n\nIt is pinned same way in devstack side too\n- https://review.opendev.org/q/I9a138af94dedc0d8ce5a0d519d75779415d3c30b\n\nThis only pin for jobs running for stable branches \u003c\u003dstable/zed.\n\nConflicts:\n this file is not present in stable/yoga and less\n -    tools/playbooks/s3api-tests/run.yaml\n\nChange-Id: I8b4527ffff557ab227910816a729579c06557911\n(cherry picked from commit a2d363de61718d4b5a15e2c553946b71cd2464ed)\n(cherry picked from commit 8cf39653dc812a13bfcaf788512bb0e2cf5c8c49)\n(cherry picked from commit 8aface90d936e58c7cdc4f896cef62e887706e63)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/598f44320d082ce57aafff439ed460f9bae81051"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/598f44320d082ce57aafff439ed460f9bae81051"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
