)]}'
{"id":"openstack%2Fswift~869523","triplet_id":"openstack%2Fswift~stable%2Fvictoria~I8b4527ffff557ab227910816a729579c06557911","project":"openstack/swift","branch":"stable/victoria","topic":"tox4","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-01-21 01:23:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8b4527ffff557ab227910816a729579c06557911","subject":"[stable-only] Pin tox\u003c4 for stable branches (\u003c\u003dstable/zed) testing","status":"MERGED","created":"2023-01-07 02:55:10.000000000","updated":"2023-01-21 01:24:31.000000000","submitted":"2023-01-21 01:23:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869523-tox4","meta_rev_id":"38533b0cd601ac2de41c3e6b539c56c6434c5982","_number":869523,"virtual_id_number":869523,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-21 01:23:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-09 05:36:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-09 05:36:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-07 03:56:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-09 05:36:57.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"c207fd201e3e5d18c670e9663787a3a4d0b4d14a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-07 02:55:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d14e3921882b5f80f5ff592fad67ce3daf32b06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 03:56:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65be738e60614f72a48ae7cc2edd42e7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38a4f83fdbfb4d43944057b4dda228b7 : SUCCESS in 9m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/5029e162dbd9403298218bd79bac346b : SUCCESS in 53m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b0c1395d2182485a955af7e8377f3378 : SUCCESS in 41m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/64b5eea58c7c41daa43d6a5487da7f25 : SUCCESS in 1h 00m 32s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/31f121b012fe4c25904523fd0c0f1812 : FAILURE in 3m 06s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6258368a4c31425e96729b2bf24dbecc : FAILURE in 4m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ca23d49790ac421e9adefa59881a82e1 : SUCCESS in 12m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f7443aa0810e43e9a22e5169b5b5b275 : SUCCESS in 12m 32s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a0c76579147c4b43882470bd3fd8ea0c : SUCCESS in 12m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2d8ac70edac543aea707f231806db24c : SUCCESS in 13m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/59bacc9f073f419daabc7359aa6f5d3a : SUCCESS in 15m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2edbab6db5404e059a7f9665ec932277 : SUCCESS in 17m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c3a61de7ddb648c48193c8b9a817a147 : SUCCESS in 15m 43s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d84d08be9dc94fc7bbd22ec812bc7d52 : SUCCESS in 16m 36s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d4263e04fa2a49c49ffc32ea477308ca : SUCCESS in 15m 55s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/efbe93d999484d43be445070a4f66676 : SUCCESS in 17m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5c5f7b06f39d421993d277eaf0bb62ee : FAILURE in 6m 37s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b7549116fc1f46939b7c0df5bf03ff31 : SUCCESS in 43m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8055a42051fc425db59a5153af2df1e1 : SUCCESS in 40m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9865f478db8a4e08bd65e2c66f78730e : SUCCESS in 39m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/299b0fc0ceb9490e96f6b8d333596155 : SUCCESS in 12m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80cc72e6b3d142d19dd93cd3322dff0a : SUCCESS in 5m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b207427e29a94cae8e37a663cfe9f9af : RETRY_LIMIT in 7m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5bc58b55fc464a3895b8855a64840997 : SUCCESS in 49m 36s","accounts_in_message":[],"_revision_number":1},{"id":"887a97e542186d10274f660456cb65591b83bd76","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-09 05:36:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"89378a03a3b134c290dd2773722d82ef4c9f9983","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-09 05:37:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"74db112f260be7c45d3e28abea1fc798b72888c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-09 06:55:42.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7c312726b7a4c6bb11e608e605463be\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05a24466f549461ea3f0e1024330edb2 : SUCCESS in 8m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/2d5295c6fae448a39b29c9df71a3d9fd : SUCCESS in 1h 13m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d42620bd4c70424c886f3737ff3f6857 : SUCCESS in 47m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/07b083c93b7645458b6146459e1cb3fc : SUCCESS in 1h 00m 37s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/6b58d65f279744438c39f82501ab9a2e : POST_FAILURE in 4m 52s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/29133cc73bcd4215ad84258b6555d6a5 : POST_FAILURE in 4m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4657aad64fb94480801720f0974807e0 : SUCCESS in 12m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/68cbaf7b4bbb44b5aaa73b32a5cc1e60 : SUCCESS in 12m 27s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0e52d2b9e96843e59763c8f2752033bc : SUCCESS in 13m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/990880812999460199bf858e63141d33 : SUCCESS in 13m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a7f6639c22354823bdc49f556ab3c8f7 : SUCCESS in 14m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/494173f169374d1cb6dc9d1782656597 : SUCCESS in 17m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/eb3856d2a2924e7a8eeb39aea6311174 : SUCCESS in 17m 09s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/bd0dc41197544dcdbee4079588eb7d0d : SUCCESS in 17m 16s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9064040285834a40959d651b83baf254 : SUCCESS in 16m 10s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c8bdc2f2103d488ab34f638d365b4423 : SUCCESS in 49m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9e445672b1314702b48b2e7115272fbe : FAILURE in 14m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bc854d75c06f43f78e4debca5935b978 : FAILURE in 14m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b77aed48de774815a155c53c11841f64 : FAILURE in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7943d0d65ec42639994c0094f649cc7 : FAILURE in 4m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fd1ce975fd0a4748b264f8d4cc6d3fee : SUCCESS in 57m 22s","accounts_in_message":[],"_revision_number":1},{"id":"0459b5876042fca8b17403249fb14011d7cb5b8b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-20 17:04:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"025ed60952cfa9600bcbb9a09dc7d4c80a6bebe7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 18:56:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfa23cf302ae44c89129482e1f55bf3a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa63b2f5f878462faaadfc6f22065eac : SUCCESS in 7m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/e5d951d66b4842b0af426b4538cb7b2d : SUCCESS in 1h 01m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b018425f6ded4f12a270be2010c184e1 : SUCCESS in 35m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2636f2df81404d26b392203450bc4cad : SUCCESS in 1h 00m 11s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d9d1adaec8474488a2528ed1cb54d94f : FAILURE in 3m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e1e3e668f0514bdabcf566536cf2ccc3 : FAILURE in 4m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7b77cbac9e2944739a8e1be0b3c84795 : SUCCESS in 12m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5cb210586583424188ed00237d7876c1 : SUCCESS in 11m 20s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d36dbb77fa014619b3ae5064e8c2d204 : SUCCESS in 12m 16s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0a09937eaf2b42cf8c8bcf2ee7737c66 : SUCCESS in 13m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9bee54e0f4c74d43af94530249f9c5f4 : SUCCESS in 19m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4b50a72fdd8748eaa8ebb2bcd518697f : SUCCESS in 15m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/66318a701177458c9adf4c104b55cb8d : SUCCESS in 18m 10s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5a7279107bd7469ea92389b4c6a35fc6 : SUCCESS in 15m 39s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a02bec2af875424da1a8ce41428aca3f : SUCCESS in 17m 14s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0ae9b45379164020b7794e3287f4b423 : SUCCESS in 25m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b0d2dcf093804ad5886518163fb2e58e : FAILURE in 6m 49s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2a3d64566c314cc4911290eccf68f5a6 : SUCCESS in 42m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/82e22a9ddb194cd291a88fac7b8958a8 : SUCCESS in 43m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c99312c1baec42ae8f16c4bee8ea5d14 : SUCCESS in 48m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e7ef9187faec4e5c84a00e4dd6d75bd4 : SUCCESS in 17m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ba4ad862bd44629a224a41dbf1d550d : SUCCESS in 5m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bfe3fe2dc12545afbbe2b07625185a5f : RETRY_LIMIT in 6m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d0e3a77496ed43cdada5c010524fc062 : SUCCESS in 44m 18s","accounts_in_message":[],"_revision_number":1},{"id":"54474c8332d6d289c1a18e8db29067bf6cc846d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 18:57:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"40e9f8e66178c3f46f9d0b1fcb13bb3e2d156f9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-21 01:23:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bfc38f2ed2b4084b4a823c5d3478da2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1786f00d5214067939e4dd3906c75eb : SUCCESS in 8m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/a137729689d74f9a8c65af3b1199b2af : SUCCESS in 1h 04m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d318d6e1eb3740ee9c625a639d37f55b : SUCCESS in 35m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/080af64f76aa487bb1130061cbaa26e9 : SUCCESS in 59m 01s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f4941d1725014e79873e6ecaebb889f4 : POST_FAILURE in 6m 21s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/c740c9cde4454b85bf26244d22f2bebf : POST_FAILURE in 6m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ca8f7e92509048d493bc8089337939d5 : SUCCESS in 13m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9d50290682dc4120a6f8b252ef6d3420 : SUCCESS in 13m 18s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cdc227389a99432f9df2adb065837d88 : SUCCESS in 13m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f518d966b3234db1827bfdae54d9194f : SUCCESS in 13m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f0ff5ce64cac4f0897db58f74c75ec49 : SUCCESS in 18m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/050ec7e25d1745aaa31051c2b48d9fc0 : SUCCESS in 17m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1db7c0b8996e4aafb8cd36c2abb7a4ba : SUCCESS in 18m 42s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7ec90e1695f745eabf37b1ccc51c5d1b : SUCCESS in 17m 56s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ca8de8c17ea546409d553ee2e7d290a2 : SUCCESS in 18m 06s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0f7dab1d5c8b459ea3593970d3ab1020 : SUCCESS in 51m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/86ad4c98e4244ffe96cfe862227e49ed : SUCCESS in 39m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/965b811bb51b452cad90383325f24f05 : SUCCESS in 45m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e7283a25a5b4cf8b85cc2e20eeafda8 : SUCCESS in 13m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aeda879eb52458187b257730f0b3a92 : SUCCESS in 6m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9469ed91ba1642a39c7906bfd1c816f8 : SUCCESS in 52m 29s","accounts_in_message":[],"_revision_number":1},{"id":"f6f9322a594540e7e74bc0995749d5ed67e21f1f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-21 01:23:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"38533b0cd601ac2de41c3e6b539c56c6434c5982","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-21 01:24:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a29b123adc5f483a83f1a311180b7da3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5712afe517d1471f89d85f4fdec0cb6c : SUCCESS in 41s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/7a8498a1290b4a8298770135482bee76 : FAILURE in 32s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/82765c15e8d4417cb1efe3108d9e14a2 : FAILURE in 30s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1ab1f9c9a2dafcec6901c458bea33d1f40da3357","revisions":{"1ab1f9c9a2dafcec6901c458bea33d1f40da3357":{"kind":"REWORK","_number":1,"created":"2023-01-07 02:55:10.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/23/869523/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/869523/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/869523/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/869523/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/869523/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/869523/1"}}},"commit":{"parents":[{"commit":"4ca7bdf93d471435278e0c1a3b1486f18fd67c0c","subject":"s3api: Use constant-time string comparisons in check_signature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ca7bdf93d471435278e0c1a3b1486f18fd67c0c"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-07 02:35:58.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-07 02:55:02.000000000","tz":-360},"subject":"[stable-only] Pin tox\u003c4 for stable branches (\u003c\u003dstable/zed) testing","message":"[stable-only] Pin tox\u003c4 for stable branches (\u003c\u003dstable/zed) testing\n\ntox4 need changes in tox.ini file to be compatible with it.\nThose are ok to fix in master brnanch but on stable branches\ntesting we do not need to fix them and should keep running the\ntox which was used in when those stable was released.\n\n- https://lists.openstack.org/pipermail/openstack-discuss/2023-January/031668.html\n\nIt is pinned same way in devstack side too\n- https://review.opendev.org/q/I9a138af94dedc0d8ce5a0d519d75779415d3c30b\n\nThis only pin for jobs running for stable branches \u003c\u003dstable/zed.\n\nConflicts:\n this file is not present in stable/yoga and less\n -    tools/playbooks/s3api-tests/run.yaml\n\nChange-Id: I8b4527ffff557ab227910816a729579c06557911\n(cherry picked from commit a2d363de61718d4b5a15e2c553946b71cd2464ed)\n(cherry picked from commit 8cf39653dc812a13bfcaf788512bb0e2cf5c8c49)\n(cherry picked from commit 8aface90d936e58c7cdc4f896cef62e887706e63)\n(cherry picked from commit 598f44320d082ce57aafff439ed460f9bae81051)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ab1f9c9a2dafcec6901c458bea33d1f40da3357"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ab1f9c9a2dafcec6901c458bea33d1f40da3357"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
