)]}'
{"id":"openstack%2Fswift~870853","triplet_id":"openstack%2Fswift~master~If33532232572e802b2099f39f4231ee6add6d5a9","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-01-17 22:06:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If33532232572e802b2099f39f4231ee6add6d5a9","subject":"Fix docs build","status":"MERGED","created":"2023-01-17 18:33:01.000000000","updated":"2023-01-17 22:07:37.000000000","submitted":"2023-01-17 22:06:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"870853","meta_rev_id":"67aba7e09daa9030a7d42c1741c996eb6c524b84","_number":870853,"virtual_id_number":870853,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-17 22:06:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-17 19:01:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"value":1,"date":"2023-01-17 19:06:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-17 19:01:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-17 19:06:18.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2023-01-17 20:16:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"08eec85c986014962229b3a8299fa32ea1e45013","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-17 18:33:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c2ef6f1828dabeb5b1ecad75e6f3c0197b9f1b5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-17 19:01:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b4f15d9714794f6d413c1aa9d0394ffcf4339ad5","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-01-17 19:06:18.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0b2f78d3532f42ae8d0df51660c3ea6ab70618bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 20:16:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab324b29c92c405ea2d00ee6ca66b04b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/66e7163490b64c6db261a376dc335d1a : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f414d7e7c9644acdaae113e4a31e874b : SUCCESS in 9m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09a8e34d797946309cadbdd74c469fd0 : SUCCESS in 5m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/22c6dcf797824361acee5ed1a70eb111 : SUCCESS in 1h 09m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/863f71b9196042b9915fb0b90bb36259 : SUCCESS in 42m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b3f20a91a1840b08e19d569f9fdd3ae : SUCCESS in 1h 07m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/258afe65fa7947b2ba99e92adbfa2d81 : SUCCESS in 21m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4dffe0d2017a4934a841f5a3ce4e187f : SUCCESS in 20m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dbe4f1eeff6b4b64a6b5b30c65063def : SUCCESS in 21m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/85a480601de5466fa65d5eb681aa3ecd : SUCCESS in 8m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e1363446fee941fc878ad4e1d0a2356d : SUCCESS in 5m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5449db6812f844f59e237e03dd7ebf27 : SUCCESS in 16m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5226a0884f8945b0b9256b98965c428b : SUCCESS in 15m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/77c4062e712f4b84b67fcd03f1b316b3 : SUCCESS in 16m 26s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/49506b79bdf64f269bb51f58debe1a52 : SUCCESS in 16m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/aa58e14387844d8ebade70b549b874fd : SUCCESS in 16m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/be6d93df50714f238d55e9400636b18b : SUCCESS in 17m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1f449cd363f844f0bf19a4df7284bb02 : SUCCESS in 21m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e391ada907bd4483af829ec9c751a99b : SUCCESS in 17m 06s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8c4f596d790e4a06bafdd6254f275a9c : SUCCESS in 17m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c7015f465f5e44d0ad16948b2e6ca369 : SUCCESS in 18m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7c83a8fa000e4c37aebff4f5e0ec508d : SUCCESS in 6m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7cff558531d04d648ac7e95c349a35fa : SUCCESS in 19m 06s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3efdd1b4436d49e3ad2342b4530061da : SUCCESS in 7m 15s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/683f96b72c2d4ff39c3ed156d6a3fc15 : SUCCESS in 1h 33m 05s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9f9659ec4fe346a19914f01e97a68c6a : SUCCESS in 1h 27m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6f67061a9d6b447fa7633790e59ca62c : SUCCESS in 1h 00m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f2327ba76a2a438c90a6e2ab79a4f8cd : SUCCESS in 58m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9fba34312c22428c9be6aba8a963abd6 : SUCCESS in 18m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cdafeee44194111b291e0a0106d2892 : SUCCESS in 4m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dac63cf8eb2641269861733ff45468d2 : FAILURE in 25m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25742e0a921442ff945f4a767b0c3948 : SUCCESS in 1h 01m 12s","accounts_in_message":[],"_revision_number":1},{"id":"ae01e8849080a7963bf4affc0d66b662719ba2ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 20:17:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a393bb286a59154e5431b95532f70dddce301a92","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 21:52:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d60893dc82344869b7d9e57b046b5054\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9d69207ac7194005bb1161f3cd9f4e69 : SUCCESS in 43m 08s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/79904bd2932043d09720bdbbd3ce00e0 : SUCCESS in 44m 42s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e0529b5ddd9e4949a1a9967aa37a02cc : SUCCESS in 48m 14s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4a4bdcf34b59496c968a0fa726471838 : SUCCESS in 1h 07m 18s","accounts_in_message":[],"_revision_number":1},{"id":"ee9f2d304bc31101e0c20879d253071a1935cbf8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 22:06:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b413b288511b4bbfa03af40130165fdd\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f1b87a2e6dd24a9da04287c5b67b1c53 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/974ed154fc3e4acca0f00741574d8eb7 : SUCCESS in 8m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b9c447899b54be2a7e964a84f8d4953 : SUCCESS in 5m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/043388011bc04af7970ea7ed83c00729 : SUCCESS in 50m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/97335994de04471b89ee156039ba23c1 : SUCCESS in 51m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9597f1d976de46ae9171706ac2c52075 : SUCCESS in 1h 04m 08s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0efe56edcc5f407ab7d078c0ab804773 : SUCCESS in 10m 57s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/84b06c1e50e94116893c249347146c8a : SUCCESS in 5m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/306364171cb646d08499c014b9fe0e42 : SUCCESS in 16m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/52c59e68bd85473f905747a0f0fb578b : SUCCESS in 14m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3a86f410b3d649ae8f237bc612e680d8 : SUCCESS in 14m 41s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/50666c1c7a1548379a2bf6202552e36d : SUCCESS in 18m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b274d8c49caa4eada34c458b8e221364 : SUCCESS in 20m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/71040c25038d406eb66cd4108f5a903e : SUCCESS in 17m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/991b7c3ca36f49dcac8c37e8a809904e : SUCCESS in 17m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/844fd60201934c4d8dad8df23eee4b50 : SUCCESS in 13m 01s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/348c5e2b2fe046389d6b18cd7f8530e1 : SUCCESS in 15m 18s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/bd626241fab846b2a1420fab3ea92d3b : SUCCESS in 14m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7c4786356f5d45fc8cca6a7bfce8c380 : SUCCESS in 6m 37s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cab2f0ba6272457dba14f00cdbed0c83 : SUCCESS in 7m 14s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9eef362d0428482c8cc528d59595d64b : SUCCESS in 1h 12m 23s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/75decc983c8d431a90a862fcbf060c17 : SUCCESS in 1h 21m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cea01e80768a476f836f55707082dc2a : SUCCESS in 53m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b03e7fd77a324cf389132eda1f7d6268 : SUCCESS in 50m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70dc86c775bd48059f538a554eba4b38 : SUCCESS in 18m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75a87faa32fa473180568f4fbabbfaad : SUCCESS in 4m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec67fadb54914c6fb4903557a1389cb1 : SUCCESS in 38m 20s","accounts_in_message":[],"_revision_number":1},{"id":"ab5d91f3c03959e1c9c3362f7511f8fe4704d485","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 22:06:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"67aba7e09daa9030a7d42c1741c996eb6c524b84","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 22:07:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7aec471b82c546f98f0ecd3b141d60c6\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/946c8cb980c440da9f6cf0f45266e455 : SUCCESS in 52s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/556c03abf3b24422a17827caaaa709f7 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3ef657a43874b19953d907ee7bea3f4 : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/46a5cc2c785d4c2a94ffea39bbd03e34 : SUCCESS in 33s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/8d99f52e6f784281ab3b1381f20c33a3 : SUCCESS in 31s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dd9b687f715f0ef6d87b420a0bad3fe8f54f0954","revisions":{"dd9b687f715f0ef6d87b420a0bad3fe8f54f0954":{"kind":"REWORK","_number":1,"created":"2023-01-17 18:33:01.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/53/870853/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/870853/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/870853/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/870853/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/870853/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/870853/1"}}},"commit":{"parents":[{"commit":"d22a75405297a148d7dc1e2c8d703678df065f63","subject":"Merge \"tests: Remove references to soft_lock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d22a75405297a148d7dc1e2c8d703678df065f63"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-17 18:23:25.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-17 18:29:28.000000000","tz":-480},"subject":"Fix docs build","message":"Fix docs build\n\nWe need to install swift to build docs anyway, so make sure we include\nswift\u0027s requirements as well as the docs-building requirements.\nOtherwise, swift\u0027s requirements will not be subject to the specified\nconstraints file.\n\nChange-Id: If33532232572e802b2099f39f4231ee6add6d5a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd9b687f715f0ef6d87b420a0bad3fe8f54f0954"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd9b687f715f0ef6d87b420a0bad3fe8f54f0954"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
