)]}'
{"id":"openstack%2Fswift~870862","triplet_id":"openstack%2Fswift~master~I5590b9d7402c9d128278f92fd522bf7364d781a4","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-01-17 22:15:51.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I5590b9d7402c9d128278f92fd522bf7364d781a4","subject":"Clean up project URLs for PyPI","status":"MERGED","created":"2023-01-17 19:57:32.000000000","updated":"2023-01-18 03:44:13.000000000","submitted":"2023-01-18 03:42:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"870862","meta_rev_id":"1f089824330dbbee01f7380ffb428ff62d304a7e","_number":870862,"virtual_id_number":870862,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-18 03:42:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-18 01:52:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-18 01:52:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-17 21:39:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f0826ff5908874cb6d75823cfa0f8086792b4696","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-17 19:57:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6465c3568c230d0f718964ee5fc4c84c555f47f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 21:39:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d9fb7d17766346a992ab5a7a2ebbf0e0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d28eeaa4155e469d92800b91e4eab02a : SUCCESS in 1m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fb7f6bb6192471d86a101f3b320ab4e : FAILURE in 4m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/0daa9c1f5f9b4351910024435c7303a6 : SUCCESS in 55m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7f05965eeb7b4755af1a88e53a042203 : SUCCESS in 46m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a6457e8feb6464498d16c90b3261659 : SUCCESS in 55m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3d78d160b3814b31bb4655052cbc95ee : SUCCESS in 17m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b3cec589636c4b43adc0e13c330c4dde : SUCCESS in 17m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3c44d870ac60491489940ea642847217 : SUCCESS in 20m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b58b324a19e64eadb3314f128ab6e94e : SUCCESS in 9m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5887f0d0bb734a18adf9328643f99591 : SUCCESS in 4m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/211b197389d24db4b8eaf49398f2b74c : SUCCESS in 18m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5a2616ad88824a8f84b09a263d7f2f3b : SUCCESS in 14m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bc7301c53448487b840275e98cd40a74 : SUCCESS in 15m 17s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0155977f62a44436b66764b41c86d61a : SUCCESS in 15m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f304730f52534270b165ae1cbd071bc4 : SUCCESS in 15m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/be966808e46e4f7aa6926bb5d0b0ac67 : SUCCESS in 17m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cad8673f0a3d431290941d90185d1c7d : SUCCESS in 18m 40s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d5f9cdbf59ab4fb591650513b0417fea : SUCCESS in 13m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c3d761abad034724ac19bbc46b4e2dad : SUCCESS in 14m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9877e650dae44fb79144399a48e9d8e1 : SUCCESS in 15m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c1529d44ac2a48a7ad1d177511846160 : SUCCESS in 5m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dbe154264d114e5fb048e1aaf9a3d082 : SUCCESS in 16m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ff7648d46ef44134b9b7111d775a1a3b : SUCCESS in 7m 58s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4e36580faadb4f019a09ced52c68c6e6 : SUCCESS in 1h 19m 55s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/53b4b9f78b1c4fd685150384ea3b5dd3 : SUCCESS in 1h 11m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cc3e83746dbb449281f860c8fda7e1c3 : SUCCESS in 53m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/24bf068389cc4f2ea3e687912e1e33e0 : SUCCESS in 48m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/abb93284455d4691ab364f9a6be3e684 : SUCCESS in 19m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d61b084e1d945f9a06595c83309e8a4 : SUCCESS in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7c8a2bbcba4c4844ae322425e19f48e8 : FAILURE in 19m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e666f03554e48a1b60c76b81556e413 : SUCCESS in 48m 39s","accounts_in_message":[],"_revision_number":1},{"id":"2581ffdf95f8de3ed837563199bbbc7ab0e94a28","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-17 22:15:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1615ff4904567deed4bb6d7277ba26ed4b7b8761","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 23:51:02.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33ba520a4848416db61b54e269d1fceb\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9a99164a718042e9b0c483b9ee57e912 : TIMED_OUT in 1h 31m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/32560b3ffd8241b2a3b09e1c876ba497 : SUCCESS in 1h 05m 08s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/70a2b21b88d147a89e17492b3a6eeafd : TIMED_OUT in 1h 30m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6eb1d1f634474e4b86e68ffad93d16e1 : SUCCESS in 1h 30m 32s","accounts_in_message":[],"_revision_number":1},{"id":"1e6c1153ffb23a843090a1e5bbe79acf6905d26c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 00:19:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/151009ec3f1f45018680384a9499c27a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3026027c091447009578f3bfa01f5a91 : SUCCESS in 1m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe61b0eab8914953bcdd6c50d7bfd824 : SUCCESS in 9m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/a0f97667d4d743ceb3d3fd17f4805887 : SUCCESS in 55m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3bb4a44f655947dfbe3b01e418d14cec : SUCCESS in 49m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f6d70eafe1f44809aea1e5c16b90b0a : SUCCESS in 48m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3a1a2c51b25647c3b3a7566c4e9df657 : SUCCESS in 17m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6c3b777aee084484b65857b64cde9dce : SUCCESS in 18m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/18eda12742b44b539b91f86b1e139d10 : SUCCESS in 17m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/28ad5079cabd4bc99eb9856e79244b28 : SUCCESS in 8m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e0c387d5d24c421fbd7f73cf7d76aedc : SUCCESS in 4m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7b074d97e3764965af65d2c0875235f7 : SUCCESS in 17m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fd1b15c64f1d48eca44c2e7844bbdd1b : SUCCESS in 14m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/efccf6a6d9f14a26861b5625fb1c1c39 : SUCCESS in 13m 57s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0e9b168b08fd4377b796f5200048cead : SUCCESS in 15m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d219ba4cfb634db8be8e517117bb3632 : SUCCESS in 16m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/991266f3780a4134a1bda8ecf1572dd0 : SUCCESS in 16m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/90631ddcd4164da889aac1108738893a : SUCCESS in 21m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5d3f355dbc364a1cb399eb48f8a98ac8 : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9ca24795f8bd4da68037fa5e06f712f5 : SUCCESS in 13m 59s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/97025ac937d44e799877c97455cd688e : SUCCESS in 15m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/86c09621c4d74f9081e4e2556551a414 : SUCCESS in 6m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1979398b65ce4d01824903829fd0a017 : SUCCESS in 17m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9ffbd2b5eeec40129fa6be23c279c40b : SUCCESS in 8m 16s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ec3430bdeb8148c3b86a2163b00de68f : SUCCESS in 1h 13m 54s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/34f1310035f9424193c95c5bc9ad7ddb : SUCCESS in 1h 28m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/94edd76f095b49c1b8c60e75f8b20d64 : SUCCESS in 58m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fc73923c17e7459c820cadb24a081cb1 : SUCCESS in 46m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6abd5449a49745e7a242ddcc2d67570a : SUCCESS in 15m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/850525b0f0d8491781e8738972511e9c : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bce30dbc5c5d4b34b8a05d7476517a30 : FAILURE in 20m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5994a415fb4c4796b00426ac548ba04f : SUCCESS in 49m 03s","accounts_in_message":[],"_revision_number":1},{"id":"fe62bb862b43f63d569a8187f2df035ae24d23be","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-18 01:52:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e575ae4a528565a63fac7b2a6a0467b3575b8660","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 01:53:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"49ae0e1770626df8dfbaf4f4291c9b7744688f37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 03:42:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b431ec4a5e9249548d4378bee60579c9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d830bae8bd6e43d7a11c723e27aa1758 : SUCCESS in 2m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1edcdb692cd54330bcd679d1f5ffe3a3 : SUCCESS in 8m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/4767ec737fad4a4aa0e6e77b967c88dc : SUCCESS in 1h 06m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/15209a4bc847403b98c84cad3b4833f1 : SUCCESS in 54m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df6d7d52d45c4d519c560056ef4819fb : SUCCESS in 59m 40s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/abe7219256d8469f9007486620e04ae8 : SUCCESS in 9m 20s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/264707980bbb4fe5a140793d69b8fa16 : SUCCESS in 5m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8cbb4baaab9e46b187c2612d57a4272f : SUCCESS in 16m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4bd393c4f7594a36910118a68fb68059 : SUCCESS in 15m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/db16efa00a9d44c484ab7bf28bc73348 : SUCCESS in 16m 35s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/60ff6a415dc7401cb85477171d07cbde : SUCCESS in 15m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/430ebf2c029445fb8101f9090a557723 : SUCCESS in 17m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9ecb8baa539d42b79dac23fe2eae61cc : SUCCESS in 16m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f7b662c1e1d74778b20d0c44696aadc3 : SUCCESS in 17m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/52456330488d43e78dd11a09d0d4b594 : SUCCESS in 13m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/32e4546b65cb42e39c6a4de700c9ef63 : SUCCESS in 15m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0d96f7cfb17c45bd99dc3d53900af8e6 : SUCCESS in 14m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ff12de5d000c4426a6000f6ee2098702 : SUCCESS in 6m 25s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/443ddb8ad0ba4eb28b0f14ef73f6083a : SUCCESS in 7m 16s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/86d23b9877b9462b800d4deb5668667e : SUCCESS in 52m 42s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9f236219048a47adaa876d7436403720 : SUCCESS in 1h 02m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5daeb4b327a942cea42197e1889ae89a : SUCCESS in 58m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/794b5c94640d4ded9e44cc2f5f4abd93 : SUCCESS in 52m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a07de8e37c149719fc671435aa76f92 : SUCCESS in 17m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccda4f08b9e3475aa691059b430aeb8f : SUCCESS in 5m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b176e68619d84e21bfc78bc7f1194c70 : SUCCESS in 56m 14s","accounts_in_message":[],"_revision_number":1},{"id":"f756209d4802e44102f33ab65ec36d937b6099c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 03:42:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1f089824330dbbee01f7380ffb428ff62d304a7e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 03:44:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94745cf77970492c87843e209cbd95ec\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/402838a6664846c490fc61a4de08d6bf : SUCCESS in 1m 17s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/6fc702b4c87848359042913ffd20967c : SUCCESS in 32s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/33b40980dceb4189bffe42a11dd5179f : SUCCESS in 37s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cc1c649de137723f84b6b9bdd49f1534935abbef","revisions":{"cc1c649de137723f84b6b9bdd49f1534935abbef":{"kind":"REWORK","_number":1,"created":"2023-01-17 19:57:32.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/62/870862/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/870862/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/870862/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/870862/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/870862/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/870862/1"}}},"commit":{"parents":[{"commit":"d22a75405297a148d7dc1e2c8d703678df065f63","subject":"Merge \"tests: Remove references to soft_lock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d22a75405297a148d7dc1e2c8d703678df065f63"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-17 19:48:38.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-17 19:48:38.000000000","tz":-480},"subject":"Clean up project URLs for PyPI","message":"Clean up project URLs for PyPI\n\n- Switch to using `url` instead of the `home_page` alias\n- Add links to docs, LP, source, and changelog\n\nChange-Id: I5590b9d7402c9d128278f92fd522bf7364d781a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc1c649de137723f84b6b9bdd49f1534935abbef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc1c649de137723f84b6b9bdd49f1534935abbef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
