)]}'
{"id":"openstack%2Fswift~870863","triplet_id":"openstack%2Fswift~master~I570e1b00ba0f68af5e6c66933412d61f57f663e6","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-01-17 22:15:42.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I570e1b00ba0f68af5e6c66933412d61f57f663e6","subject":"Don\u0027t run reno as part of building an sdist","status":"MERGED","created":"2023-01-17 19:58:44.000000000","updated":"2023-01-18 04:18:13.000000000","submitted":"2023-01-18 04:17:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"870863","meta_rev_id":"97918ccf98c3b06491dfea95d4126b9dd6d4e746","_number":870863,"virtual_id_number":870863,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-18 04:17:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-18 01:52:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-18 01:52:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-17 21:35:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"efbc18eac04c63f02a647b1de16832712fddc308","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-17 19:58:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae5c037adb4506331734f392f2bfb591aca32c56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 21:35:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a93d49671d2b4598a22c95d930c45815\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/57a24c5825e94ebc9c5ff0bd327c562e : SUCCESS in 2m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec8aaef407214c4ebb3b900ffaa13d35 : FAILURE in 4m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/eeca0bfd2d8240379aa1bfdd4d64abc7 : SUCCESS in 58m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/eb8f1b3765234aa1a12b33a755e77655 : SUCCESS in 41m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ae8132ef7f9346aaad81b72f96e6e5f4 : SUCCESS in 1h 08m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/135bedef0ec34b758a0de8f32efc4c92 : SUCCESS in 16m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f120b7be54414d549c29c95a3bfd7301 : SUCCESS in 20m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e7aa73a85d914976b412deb90497c251 : SUCCESS in 18m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/99ef6937ee9b453b818c238752426b40 : SUCCESS in 9m 21s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2eee39fd482e4735ad09292b21abb091 : SUCCESS in 5m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/884a4131a32e4a669d48c691660ea1f3 : SUCCESS in 16m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/08cbe4bc70c84751819a4a1de7987778 : SUCCESS in 14m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/626eb74f0e52469fa379f4a5ba640d9c : SUCCESS in 14m 03s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d7275fc575734a498380fb773cbd8e2c : SUCCESS in 17m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bc6ec32005dd48e681fea2f9665da997 : SUCCESS in 15m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/11feb642a98e4799b39f2e5a7d8163bc : SUCCESS in 16m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6d730bf3ea2a43428bfa6008c612f458 : SUCCESS in 16m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d580c1024ddd46cca25a1f953648c2db : SUCCESS in 13m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1d7efd914f8a4668bc5e64d74060792c : SUCCESS in 17m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5277e6b0bb9243d39c70d9166fd6d2f6 : SUCCESS in 15m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8a299958760c4db6abc0661c569d6c74 : SUCCESS in 6m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4b81e021493045069ba97c835e0940f1 : SUCCESS in 15m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/db3f984aab144570ae1555088137e4b1 : SUCCESS in 9m 41s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/55d09967039c48439f075981b351902d : SUCCESS in 1h 26m 30s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a5f2f3455e5b463c8d3b98881ee03569 : SUCCESS in 1h 17m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/badc47d405744c479c52dc3b8f819a8e : SUCCESS in 1h 06m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8ad35b03e0a74b1b94d67bb403a295f9 : SUCCESS in 47m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e72950a5fa39447ba4276f9ee8b92c54 : SUCCESS in 16m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edcde264b74845c6ae910a8d9f4ad3ae : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/786da6b0f1274862a6a15dda9a7260af : FAILURE in 23m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41c5e4f257054c949674ec2310c5323b : SUCCESS in 54m 46s","accounts_in_message":[],"_revision_number":1},{"id":"dd3e405dcc3dd150077237c4015c1e9058737256","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-17 22:15:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1fdfe59187b27bced26a028d286711d40fb06433","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 23:54:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1dbf7f6aab624617ace54c210a940881\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3b4c0b34779841e5a32d5b8dbe3af610 : SUCCESS in 1m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/979bcb3e89764cb0ada389093aa804af : SUCCESS in 8m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/c5a59850b77b48f19df1bf6a59e7ccd8 : SUCCESS in 1h 02m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/73a6aed5eeaa43efb640f756c52a4e08 : SUCCESS in 52m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8134f8b254854794bb0daf3d7193bf1c : SUCCESS in 54m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/54ae6eafa1bc45b0933b1225d4f340ff : SUCCESS in 18m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b9f47ad4654643ae87afcdbfabd05c55 : SUCCESS in 20m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2f908b97dd554076b58e556409ca168c : SUCCESS in 18m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/734da2193786445fbe3266b65ba8955f : SUCCESS in 8m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b18ea370675b436aaa3ef7c4ca211661 : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/86252b6e0df74d31a7fe0fecea07ac21 : SUCCESS in 18m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c863d4b0a0434683887fb9d4e852e790 : SUCCESS in 16m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/107ea997e8c040078fbacf62e6ba2e59 : SUCCESS in 16m 13s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0ae11f902bc440a09da4cd78ce962ce7 : SUCCESS in 16m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d909843070ab489692f0f2683c81d227 : SUCCESS in 15m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/dd49d664fad847279a0ccb75594e9091 : SUCCESS in 18m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ab45c92f7a33476bb911bf7b99f17b2a : SUCCESS in 17m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/514bc1859370405eabbecfc6dce772d2 : SUCCESS in 14m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fae35f384f43436db188915e15d0c107 : SUCCESS in 14m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/dabf7a0967114bbf8ee57841cb82a82f : SUCCESS in 15m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/29c7028deef645c588d488b99ead0a2f : SUCCESS in 6m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d8affd6cf789482183971ed82529bb6d : SUCCESS in 16m 47s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7115d3d3996e496b8dbaba702b6ba794 : SUCCESS in 6m 49s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bb65924ccd684c95963a75064bd814ad : SUCCESS in 1h 22m 50s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3d3269595fa44a4a9524ccd18b989093 : SUCCESS in 1h 24m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a248b3a88930487cb6bcc8a8f06b6db2 : SUCCESS in 55m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8abd5ffbb0b3463ab7639e49895b021e : SUCCESS in 47m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23e058cfbcea42b49f2f311ad46875a2 : SUCCESS in 16m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08e94f7dd46c4c55b06b4fd03de4ab77 : SUCCESS in 5m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/54ba8d404b204ee58dafcf51e8113eae : FAILURE in 21m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b5482d72a3a400492f9badb9caa22c0 : SUCCESS in 49m 23s","accounts_in_message":[],"_revision_number":1},{"id":"ca98a5cccb15f4ca5718054f4a890c6774da1826","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 00:41:01.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfe5e7533b1c4a63afa8b0dc341c2ddb\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/537abab7251b45b18c668c3dd1388ba6 : TIMED_OUT in 1h 30m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bcb0cf471aa74529909d2f43d7cfdca6 : SUCCESS in 54m 53s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8a77d2564f1c4df9a2f6fb96c3ed2f70 : SUCCESS in 1h 27m 22s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/59aae4e08eee4de0b35e117fe29ae175 : SUCCESS in 1h 15m 39s","accounts_in_message":[],"_revision_number":1},{"id":"420bb4343d26a946c70956278d45d7ac221c458c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-18 01:52:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fd132d072f5d9b2306bf41fc5d5462b579f3465e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 01:53:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3a771665c4d4a3086c95b916e0c35d6bfd9d7236","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 04:17:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d33fc919b1ec442a9e016581eb38294e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e02a375ba1e34f7c9facd3e6cf78c7e1 : SUCCESS in 2m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4f96d0acb4248b2be3a68b190c54009 : SUCCESS in 9m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/031750ddd606443690417d3cf93c4a36 : SUCCESS in 54m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ac8f2cd9a31c44ac93d8d41c2285d6cd : SUCCESS in 51m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0aadf9a864cb49669b6d508994d6753c : SUCCESS in 55m 06s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/59606eda3ee64a4cb3afe81bdad9384c : SUCCESS in 9m 28s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/11cac0e0507d42c4b98fccfd0df37429 : SUCCESS in 5m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7b12587457c4448080cdbfa925e276e7 : SUCCESS in 28m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6303330d54f549f6b097339782171bd7 : SUCCESS in 27m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/34e7d3087e9a4d24bccdec46b5302e97 : SUCCESS in 24m 04s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ebe2216d32884194b3c559d94efff23c : SUCCESS in 27m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cd1edb92c0284b289847e5fac022316b : SUCCESS in 28m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c36b5136d6ff4771945d81998beb6c52 : SUCCESS in 28m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a3e4424445384a7287d2a2980aa211d9 : SUCCESS in 28m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c6ae29069bad41098820b0db6ccd6521 : SUCCESS in 25m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c8d6e501e9c640989a1bf8aeec5e1f1a : SUCCESS in 27m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/693b90445fee4c1fac76a12a8dbe670a : SUCCESS in 31m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8d3d490d281748689eedcee9f8009be3 : SUCCESS in 6m 41s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f0c7c563b4d443fba4d888fd52342fe1 : SUCCESS in 7m 27s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/73f365c523b14e2da228844d45d97776 : SUCCESS in 1h 00m 15s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/10698852ee8d4646b603759546267e53 : SUCCESS in 1h 30m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/784f6fe22ee344558c6ae5dd5e1be147 : SUCCESS in 45m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3efe4a0093f5438b959d1952f8fc15d0 : SUCCESS in 1h 04m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/88c09781fbd945ebafddbf1ef7577853 : SUCCESS in 16m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8530c106956c40d983b5fc32afad56e7 : SUCCESS in 3m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c6d9cc07bb154b0d8045a4a6001379f2 : SUCCESS in 49m 56s","accounts_in_message":[],"_revision_number":1},{"id":"5c2901aa19997e60b300d598731273495356c9fe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 04:17:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"97918ccf98c3b06491dfea95d4126b9dd6d4e746","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 04:18:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23c7ffeaaceb4f90bc6bbd28fcc92ead\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae30417332c642dfa5ad9e22b54febf6 : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/b03e442a6c0948e7aa7b145e49810929 : SUCCESS in 38s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/ddc0da1c013a4db9915e6f56bf77e7d4 : SUCCESS in 34s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"908595ee0d059ebe9bdec8323e5481139529dc3d","revisions":{"908595ee0d059ebe9bdec8323e5481139529dc3d":{"kind":"REWORK","_number":1,"created":"2023-01-17 19:58:44.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/63/870863/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/870863/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/870863/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/870863/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/870863/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/870863/1"}}},"commit":{"parents":[{"commit":"d22a75405297a148d7dc1e2c8d703678df065f63","subject":"Merge \"tests: Remove references to soft_lock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d22a75405297a148d7dc1e2c8d703678df065f63"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-17 19:58:20.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-17 19:58:20.000000000","tz":-480},"subject":"Don\u0027t run reno as part of building an sdist","message":"Don\u0027t run reno as part of building an sdist\n\nChange-Id: I570e1b00ba0f68af5e6c66933412d61f57f663e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/908595ee0d059ebe9bdec8323e5481139529dc3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/908595ee0d059ebe9bdec8323e5481139529dc3d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
