)]}'
{"id":"openstack%2Fswift~871005","triplet_id":"openstack%2Fswift~master~Ifce72ec255b1b618b9914ce5785d04ee0ebd3b8c","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-01-19 20:46:08.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ifce72ec255b1b618b9914ce5785d04ee0ebd3b8c","subject":"tests: Ensure XXE injection tests have config loaded","status":"MERGED","created":"2023-01-18 23:16:00.000000000","updated":"2023-01-19 21:50:44.000000000","submitted":"2023-01-19 21:49:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871005","meta_rev_id":"7f5e9735ce08114c7464459878ecc91e13ca045a","_number":871005,"virtual_id_number":871005,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-19 21:49:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-19 20:46:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-19 20:46:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-19 00:06:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d22830e38f2e73be6c44ca82ef190dd769bfd41d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-18 23:16:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95a27b59444125c2169879738a214ab3917962b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 00:06:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6e1ab145dc2497d8efa0aedd4c02ed7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e645d8d0807d46d8827091d6aed489cb : SUCCESS in 8m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d6d490b748d484e85762a090cac044c : FAILURE in 22m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b999f0bb56384cd4abc378d5be17104c : FAILURE in 22m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f1ada7ea50b244be8b3aa866612cae6b : FAILURE in 24m 12s (non-voting)\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/53dc825f04f74ca2a9003b67447ed7c0 : FAILURE in 21m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/67a5dcb46963496dae455dd2ae1da733 : FAILURE in 21m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f8c857db0b4d4fe58fc10e486c612de4 : FAILURE in 22m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cb2db1f1f09b4de2bca07213c8cb80c4 : FAILURE in 19m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/43c351db9fb44b80b53edfa7a8c7d741 : FAILURE in 20m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1aef3480bc0246e989a1d78bbccc29b4 : FAILURE in 16m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/81a1633f25644c85801d0121aca6cdb3 : SUCCESS in 48m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8b7c2e051ab34621a74e13bec1af35d5 : SUCCESS in 45m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38239d2e43ab4e62adda62b1a85e6b00 : SUCCESS in 5m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5f6263ae3595444b810ec95ed89c243f : FAILURE in 25m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3f203f405728da66e2cc0a54e5feb92f25182cb8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 00:50:43.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12f5739c0f304701adf217d0151f8b2a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0ec51b9ce58b4075a862485913dfde02 : TIMED_OUT in 1h 31m 02s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/102cd8e5b5f246bc812d41a43e995103 : SUCCESS in 1h 22m 46s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca95b4af4e9b4759a957ac8d7eef7287 : FAILURE in 1h 19m 32s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/790a2696f3e74d13bfe50fc30e79b664 : FAILURE in 1h 10m 52s","accounts_in_message":[],"_revision_number":1},{"id":"e6aa9e43accb0a09acbdfa7e40f1527986565407","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-19 01:03:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3765cfea98e47335fdf9b58bdd54783fed156285","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 02:10:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dcc32071ac3648a6a05a848b314c6800\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87ef41af45814d46bc9ab2ad3c7dab65 : SUCCESS in 8m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/08eaeb3bd82340668f6d14d5364178cf : FAILURE in 18m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2d399038330f4ff39e2d1dc1318e6d44 : FAILURE in 19m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/54406bf9e1af4a2fba86790c9d19e10f : FAILURE in 18m 29s (non-voting)\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dd719793eff94c37a4300df74991756e : FAILURE in 17m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a95a92c4e892484d9aea152367a96f04 : FAILURE in 17m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f73244a9eea44005ae771505d44c42c6 : FAILURE in 19m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0d1e9dc5027948b19b1c940162618fd7 : FAILURE in 13m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/49379755a19d4e64b05bd972a661f9c6 : FAILURE in 17m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8255f34f357a4bb09fc7890e26f80002 : FAILURE in 16m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5978664a2efa4b7a8b7384b83845ef0d : SUCCESS in 38m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c17fba98366242deb5c5bd52be73d527 : SUCCESS in 57m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f20930930784371a2552e5e54bcbf76 : SUCCESS in 4m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0cf812f09e634f67b7724521fca377af : FAILURE in 20m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"993a3e6264b1c349bdd8eafde4893385ceea8714","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 02:37:34.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1910004e0fd499ea3723027214f8d4c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f4678a1ab4f84d3bbda21beecd2367d9 : SUCCESS in 50m 42s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7e5774c211ee46eb8b487a35c40a76ec : TIMED_OUT in 1h 30m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/33683a2764fb4b97b121a5ee4bf734e4 : FAILURE in 58m 10s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ad9d4619ed9447ddb2774a53e6b4858c : FAILURE in 1h 06m 53s","accounts_in_message":[],"_revision_number":2},{"id":"4e66cbd97f4e7247fb5b81b53717931826239c79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-19 17:50:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2dac043c1745cce1865a33581408891d742c7c93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 18:37:47.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f2db33e24b2497fbe624d0ea5827ee9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f59960af20644275a4337969981494e2 : SUCCESS in 7m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5fa7ff7dbe3043cb8fdef6f7d92780aa : SUCCESS in 19m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/962c8d275cf044c8bbe684a4b60b46e5 : SUCCESS in 18m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d5b4f8ed52a94db9b4e01fbfab17ddab : SUCCESS in 17m 54s (non-voting)\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9997cf70383f41b893f1e88bb31f848c : SUCCESS in 15m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cbc4fc143e484eb1a95c80ccceab9718 : SUCCESS in 18m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/57b57b31c5854ef0bb41c5c9a722b603 : SUCCESS in 22m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/732d3baeb3314616ba4b78eb689c5b3f : SUCCESS in 19m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f73c71f2e38b4c1a8b21bf75657635ea : SUCCESS in 20m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8b3ee13df1f347aaa6b822466ca94580 : SUCCESS in 23m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/844ab41b371c4b85a4966aba785f4953 : SUCCESS in 44m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5b3bc6943dd5448eaad8729369c2ee06 : SUCCESS in 45m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/095df5734f1a457ea1865240e4a2bc5e : FAILURE in 3m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bca15c7c093048208519777c40523ac7 : FAILURE in 19m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c29ab2f0f915b36523a37b206b15d8ddf72dcd3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-19 19:26:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2428c989cc6314014553cda91e3f6f37306b6ac5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 20:32:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/adf0e8960f574ed0aad57d6172364e70\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bab024d56d74a708256bff5414071e1 : SUCCESS in 10m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8a1140fbff4445f1b07da1eaa83ddedc : SUCCESS in 20m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1668645cc55e4027953ea4cddbe6aa9d : SUCCESS in 20m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2e2b68dd700e4d8799e3c927084930ee : SUCCESS in 21m 11s (non-voting)\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/03977ef10d254c7d8cef6b142f6c3a20 : SUCCESS in 20m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4a74d5cf2abc4ea7a2a0aaa40dd66657 : SUCCESS in 20m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e8eb4aaf23ed47e0b073bd3867f8c8f6 : SUCCESS in 19m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c5ecd1881e76421d809d06d74412aa9a : SUCCESS in 16m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0c8951123d9e4b4b8c4a0b57af6b5f14 : SUCCESS in 16m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/357e6da61eb74294a7113a85c5ce03a1 : SUCCESS in 17m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8a2bcf33910d4c0195d7cf36072827cd : SUCCESS in 54m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c2ff9483d76e4f36920166b41079ff2c : SUCCESS in 52m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c18b8a7b55e444d863d8752ddfb26d4 : SUCCESS in 4m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f881b85c04a94d5687f47ca3a66f628d : FAILURE in 20m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"37d60410930422d178de17cb1f0965e21a6e2167","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-19 20:46:08.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"45f2bfd5744b6e4ba4756b571ce921a482f175e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 20:46:33.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"bbb798b34ec81ff99f90dfb0ea4f6ec79b4780b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 20:56:32.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51e2c907a8784e02961b4cfdd8701daf\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/80f05c65cc1d457e8cfbaba3b42f05d9 : SUCCESS in 1h 07m 06s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0bcf90488d95410ebbb565c01db5c8d9 : SUCCESS in 1h 07m 46s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0623ed65d16349a1acc6adbc58c61bf5 : SUCCESS in 1h 18m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/43eb6ea3e5174295b09389be6b0eb623 : SUCCESS in 1h 26m 49s","accounts_in_message":[],"_revision_number":4},{"id":"6b36bc855bb2a15b30af7f8307e771345b50aa10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 21:49:30.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbe51e31e6594ac6a91d71a03e3f4d35\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e88848f2ffbb4f4183e7096b0888ae15 : SUCCESS in 8m 54s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4e49436333f5472aa27877d67650b2e3 : SUCCESS in 25m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c0e0e9e3c9e6473483fea5c9f139e61e : SUCCESS in 24m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/56abab5bb0ce45dcb8e5f54f779e7a90 : SUCCESS in 24m 28s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0e59c8c508d541f3b70bd3e7ee7c0411 : SUCCESS in 24m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8472e018df2f4d5abcee140e20e8ecda : SUCCESS in 15m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ec31ebd3147d4ba18ac01429d8c3f297 : SUCCESS in 15m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/46f702ab66ef439cb419590d4ff968f2 : SUCCESS in 15m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1e7171b1ac8a4e37b81859a76eb88421 : SUCCESS in 13m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7262e99a1e784d46973b3f099e050944 : SUCCESS in 17m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/05621d024a554e6eb1fb8fbefce85394 : SUCCESS in 15m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6ccee85176b44f23911b9f228ac713b3 : SUCCESS in 6m 03s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0db59e37fc2a480d9daac7430cc9e36f : SUCCESS in 6m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e0d92bf05e46411f8cf1dd8ac2b47ff4 : SUCCESS in 57m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5b1ea35f5b9443219d64f70e08f9d4b9 : SUCCESS in 52m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50e1f9d7947f40a097c1578ceedf31a6 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":4},{"id":"d5433e6f5dc0d5741f58600e2f34b71fa50be4d5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 21:49:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"7f5e9735ce08114c7464459878ecc91e13ca045a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 21:50:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3146a8f563b48adb7ea976f977b61e7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2680a292887492686463e0a81f48012 : SUCCESS in 55s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d9ab630fb58b4006b2d51d721583aba9 : FAILURE in 28s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/0465e6bd571c498b86379ba103a69075 : FAILURE in 32s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3550e00dd9e380ba655e19047f8042bb9ae60098","revisions":{"c6b6fb35282de54091da909f8bcc658b885c0c07":{"kind":"REWORK","_number":1,"created":"2023-01-18 23:16:00.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/05/871005/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/871005/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/871005/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/871005/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/871005/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/871005/1"}}},"commit":{"parents":[{"commit":"477423f60a1579ee275c37b749cc23aa3fdc4ea7","subject":"Merge \"Clean up a bunch of deprecation warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/477423f60a1579ee275c37b749cc23aa3fdc4ea7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-18 23:14:54.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-18 23:14:58.000000000","tz":-480},"subject":"tests: Ensure XXE injection tests have config loaded","message":"tests: Ensure XXE injection tests have config loaded\n\nDepending on test order (and possibly whether there were earlier\nfailures?) the new tests may trip KeyErrors when trying to get\ns3_access_key values. Solution seems to be defining\nsetUpModule() / tearDownModule() like other functional tests.\n\nChange-Id: Ifce72ec255b1b618b9914ce5785d04ee0ebd3b8c\nRelated-Change: I84494123cfc85e234098c554ecd3e77981f8a096\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c6b6fb35282de54091da909f8bcc658b885c0c07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c6b6fb35282de54091da909f8bcc658b885c0c07"}]},"branch":"refs/heads/master"},"d7a936452981bdd5e369fb95f661ae187ca4ec24":{"kind":"REWORK","_number":2,"created":"2023-01-19 01:03:48.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/05/871005/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/871005/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/871005/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/871005/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/871005/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/871005/2"}}},"commit":{"parents":[{"commit":"477423f60a1579ee275c37b749cc23aa3fdc4ea7","subject":"Merge \"Clean up a bunch of deprecation warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/477423f60a1579ee275c37b749cc23aa3fdc4ea7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-18 23:14:54.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-19 01:03:18.000000000","tz":-480},"subject":"tests: Ensure XXE injection tests have config loaded","message":"tests: Ensure XXE injection tests have config loaded\n\nDepending on test order (and possibly whether there were earlier\nfailures?) the new tests may trip KeyErrors when trying to get\ns3_access_key values. Solution seems to be defining\nsetUpModule() / tearDownModule() like other functional tests.\n\nChange-Id: Ifce72ec255b1b618b9914ce5785d04ee0ebd3b8c\nRelated-Change: I84494123cfc85e234098c554ecd3e77981f8a096\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d7a936452981bdd5e369fb95f661ae187ca4ec24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d7a936452981bdd5e369fb95f661ae187ca4ec24"}]},"branch":"refs/heads/master"},"be97f17c26b9f6d3e01dc4ada74a62ec08ba870e":{"kind":"REWORK","_number":3,"created":"2023-01-19 17:50:55.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/05/871005/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/871005/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/871005/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/871005/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/871005/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/871005/3"}}},"commit":{"parents":[{"commit":"477423f60a1579ee275c37b749cc23aa3fdc4ea7","subject":"Merge \"Clean up a bunch of deprecation warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/477423f60a1579ee275c37b749cc23aa3fdc4ea7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-18 23:14:54.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-19 17:49:53.000000000","tz":-480},"subject":"tests: Ensure XXE injection tests have config loaded","message":"tests: Ensure XXE injection tests have config loaded\n\nDepending on test order (and possibly whether there were earlier\nfailures?) the new tests may trip KeyErrors when trying to get\ns3_access_key values. Solution seems to be defining\nsetUpModule() / tearDownModule() like other functional tests.\n\nChange-Id: Ifce72ec255b1b618b9914ce5785d04ee0ebd3b8c\nRelated-Change: I84494123cfc85e234098c554ecd3e77981f8a096\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be97f17c26b9f6d3e01dc4ada74a62ec08ba870e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be97f17c26b9f6d3e01dc4ada74a62ec08ba870e"}]},"branch":"refs/heads/master"},"3550e00dd9e380ba655e19047f8042bb9ae60098":{"kind":"REWORK","_number":4,"created":"2023-01-19 19:26:40.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/05/871005/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/871005/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/871005/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/871005/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/871005/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/871005/4"}}},"commit":{"parents":[{"commit":"477423f60a1579ee275c37b749cc23aa3fdc4ea7","subject":"Merge \"Clean up a bunch of deprecation warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/477423f60a1579ee275c37b749cc23aa3fdc4ea7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-18 23:14:54.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-19 19:24:34.000000000","tz":-480},"subject":"tests: Ensure XXE injection tests have config loaded","message":"tests: Ensure XXE injection tests have config loaded\n\nDepending on test order (and possibly whether there were earlier\nfailures?) the new tests may trip KeyErrors when trying to get\ns3_access_key values. Solution seems to be defining\nsetUpModule() / tearDownModule() like other functional tests.\n\nAlso fix up some Content-MD5 handling; if we\u0027re using pre-signed URLs,\nwe can\u0027t provide a Content-MD5.\n\nChange-Id: Ifce72ec255b1b618b9914ce5785d04ee0ebd3b8c\nRelated-Change: I84494123cfc85e234098c554ecd3e77981f8a096\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3550e00dd9e380ba655e19047f8042bb9ae60098"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3550e00dd9e380ba655e19047f8042bb9ae60098"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
