)]}'
{"id":"openstack%2Fswift~871135","triplet_id":"openstack%2Fswift~stable%2Fzed~Ifce72ec255b1b618b9914ce5785d04ee0ebd3b8c","project":"openstack/swift","branch":"stable/zed","hashtags":[],"change_id":"Ifce72ec255b1b618b9914ce5785d04ee0ebd3b8c","subject":"tests: Ensure XXE injection tests have config loaded","status":"MERGED","created":"2023-01-19 22:03:50.000000000","updated":"2023-01-20 00:19:13.000000000","submitted":"2023-01-20 00:18:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871135","cherry_pick_of_change":871005,"cherry_pick_of_patch_set":4,"meta_rev_id":"2a8352b32b38e7f4ab56a1853c52041da3782c59","_number":871135,"virtual_id_number":871135,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-20 00:18:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-19 22:44:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-19 22:44:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-19 23:05:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9696cff9bb1d5b20ca62385b8f82cb781c2cc30f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-19 22:03:50.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"43ae381378ccaf10bf2d4299dad00eb50f8c2241","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-19 22:44:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a216fc9de5ac9cc4af340f727b0b84533bff9c9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 23:05:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d75bde206ec48329dafd8b7734bf212\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76321a0b83bf49a79b522207a6da6ff8 : SUCCESS in 8m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec57d5683ad9490086af106a7827a681 : SUCCESS in 56m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0bc114a71d514ee2b61c4bfa72184897 : SUCCESS in 20m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6ae2e12ea7ef496a946aea36ed32504e : SUCCESS in 18m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ed2ee901868747658bb7118f0268d5a7 : SUCCESS in 18m 38s (non-voting)\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/89ae04e2a6604ef1b97be0408c8f66c0 : SUCCESS in 14m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/00053f10834f4fc9a5e1647b14628533 : SUCCESS in 16m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3ef1da2cdace49e2ac2d7f6857d7c72a : SUCCESS in 27m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1d3d7f2286a94c378983c9faf39297a2 : SUCCESS in 13m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/86d07240d511473787c215b0701cdf7e : SUCCESS in 16m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/66920f0d8ee641768c904a9f67a5d5ba : SUCCESS in 14m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5f6ed192eb6f44bf9023ad39f79004f5 : SUCCESS in 41m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5ad2c434056544e8a4e7f2504fd9b29e : SUCCESS in 42m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb86acb2c9d448168010e7bfb3c34244 : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/43b5d9075ab04c9dbd74c8b9372af09e : FAILURE in 19m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"95feabfda94f20a2df1e67917a4262a6d594d24e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 23:06:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b717004a23b10418ee7ac28f2cda86309df43a2e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 23:34:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bef4c100dd64abdb037ae9b27c8df81\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/34d4e66f4857460ab266a45e2c2b3675 : SUCCESS in 1h 19m 40s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2e8a5d47014e431c9adde7274bb4c069 : SUCCESS in 50m 53s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/12bfac601e8544519688ca0ca25c5b88 : SUCCESS in 1h 08m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0d196e01c95a450d9e40abbda8f09918 : SUCCESS in 1h 13m 12s","accounts_in_message":[],"_revision_number":1},{"id":"ba63b350b5f4b951dc05b99b8610d102a00b6e44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 00:18:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8eb29cd56c20475687d348cb8b84eda3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/666fe91010c1423f8b612b6d8149b7ab : SUCCESS in 8m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/de1d5f3b467a495abc3de0746e6d28e7 : SUCCESS in 57m 07s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0f7c4454aec04f69b18976a15b22a098 : SUCCESS in 11m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ab77dec9c320467e88f58ecd70ada3ed : SUCCESS in 15m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/33539397123c40b5af4e0cd39e3a1552 : SUCCESS in 14m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/29ec6e0f907548378cc82f4a462ee231 : SUCCESS in 15m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f037002d1ac94a71ad01165be622a1d4 : SUCCESS in 17m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8596071875d9489fbd01b0deec3844a9 : SUCCESS in 16m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/643570147130443da7cc69b276b38848 : SUCCESS in 15m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8c676e2ef2e54e47a82d3198b0937a9b : SUCCESS in 15m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5c6fbb7a7db945bf89930efffa2c0624 : SUCCESS in 15m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/13be335782df43adbc65e2bbc83b4efa : SUCCESS in 8m 06s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/62fa25ece8a64aba8f01a4a72ecfbedc : SUCCESS in 7m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/99c3b00eb9214b31af70ee60b312f11a : SUCCESS in 40m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/015a6282ed1b4d5e9ed2ae6de5813fb5 : SUCCESS in 50m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b15ce07ff354694add8cd97b6b8a0f7 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":1},{"id":"001f4183d6b0c5347c2038d9d631fe2899a82d4d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 00:18:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2a8352b32b38e7f4ab56a1853c52041da3782c59","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 00:19:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/622651c4615f438e9a3f5e4f9f95bcd4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea316b91d9e740a39d023e5bc1e91349 : SUCCESS in 54s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/74a8dd68206e4c68ad9c4ea6a84c604f : FAILURE in 33s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/218e3e58efe4445fb413ceb4768a8eec : FAILURE in 28s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"041cb672e8af71ba9622d22ed4c675f437b26753","revisions":{"041cb672e8af71ba9622d22ed4c675f437b26753":{"kind":"REWORK","_number":1,"created":"2023-01-19 22:03:50.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/871135/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/871135/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/871135/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/871135/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/871135/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/871135/1"}}},"commit":{"parents":[{"commit":"67785199b02d76b5b42448845e1181eb658b1723","subject":"Merge \"s3api: Prevent XXE injections\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67785199b02d76b5b42448845e1181eb658b1723"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-18 23:14:54.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2023-01-19 22:03:50.000000000","tz":0},"subject":"tests: Ensure XXE injection tests have config loaded","message":"tests: Ensure XXE injection tests have config loaded\n\nDepending on test order (and possibly whether there were earlier\nfailures?) the new tests may trip KeyErrors when trying to get\ns3_access_key values. Solution seems to be defining\nsetUpModule() / tearDownModule() like other functional tests.\n\nAlso fix up some Content-MD5 handling; if we\u0027re using pre-signed URLs,\nwe can\u0027t provide a Content-MD5.\n\nChange-Id: Ifce72ec255b1b618b9914ce5785d04ee0ebd3b8c\nRelated-Change: I84494123cfc85e234098c554ecd3e77981f8a096\n(cherry picked from commit 3550e00dd9e380ba655e19047f8042bb9ae60098)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/041cb672e8af71ba9622d22ed4c675f437b26753"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/041cb672e8af71ba9622d22ed4c675f437b26753"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
