)]}'
{"id":"openstack%2Fswift~871136","triplet_id":"openstack%2Fswift~stable%2Fussuri~I61750a1083a1c97a6222ec9040f90980ee73acc8","project":"openstack/swift","branch":"stable/ussuri","hashtags":[],"change_id":"I61750a1083a1c97a6222ec9040f90980ee73acc8","subject":"Add attrs to lower-constraints","status":"MERGED","created":"2023-01-19 23:53:55.000000000","updated":"2023-01-20 02:52:25.000000000","submitted":"2023-01-20 02:51:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871136","cherry_pick_of_change":869581,"cherry_pick_of_patch_set":1,"meta_rev_id":"c9cde520493e7419d2c716d9c583d2edd21029fc","_number":871136,"virtual_id_number":871136,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-20 02:51:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-20 01:29:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-20 01:29:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-20 01:17:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5b3911b857613b1100183f975bd59925d951af68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-19 23:53:55.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"42898b4eccc551f5896398ccdc9aee97ce10c20f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 01:17:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dc2d17d9477478486c007084807beea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc7acc1acecb4dd898f21d4fe8dc3157 : SUCCESS in 8m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2d255a2f1a4f4f039f9caf3cf58ddbb3 : SUCCESS in 2m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/d06294fbfce44412bd3f11c71179b110 : SUCCESS in 1h 15m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a5abd9ae5b6148329246a906ac6fcf8d : SUCCESS in 49m 47s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ec3859f8782c486c8b7bc86b8ed122cc : FAILURE in 3m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7d794c7dfa3140ec929d2caab73ff5e2 : FAILURE in 3m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/78555bc177c645b3acfe9c37cf3efa4d : SUCCESS in 14m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/08ae3e1a06664c1784575668a4a44da1 : SUCCESS in 13m 41s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4bb907d6f0c5477cb4835ac2db14352a : SUCCESS in 14m 47s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/244a77072051493c961bc3fd867c7f9e : SUCCESS in 13m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7fa4ae9ecbda49dda29a53182d3d4233 : SUCCESS in 9m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cc4392eab93a474488fcb2a20c3bdbb7 : SUCCESS in 12m 03s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/13124eb6585f4e248b732dd79a6e95a7 : SUCCESS in 9m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/330eb5edb0f44612bd3ae14ab899a548 : SUCCESS in 9m 55s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/acf13ac4ce8e48369e09d9da4669c1c1 : SUCCESS in 15m 05s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/091f787d07ac4db8bac43752745cb63c : SUCCESS in 9m 34s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/2fadb2ed31a0422596c48d0d72b181c5 : SUCCESS in 12m 49s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/b7c8635347a5408db97dfc32e2f2eaa4 : SUCCESS in 10m 13s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0199298c154146d6817ca29a782470bc : SUCCESS in 10m 37s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/aabff59f2d044438a9724e4ff3cda1b6 : SUCCESS in 15m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c3d6f67759ab4be88bf109309be36a0c : FAILURE in 7m 46s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3c746ddc839b4d97930563e767f78ef1 : SUCCESS in 50m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6eab02084d9644a5871aaa8ac294c9f0 : SUCCESS in 33m 35s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/fbd87065d8a54032a1612e21c8c832a5 : SUCCESS in 32m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1744fa212cad48e6a9cfd135b6fce543 : SUCCESS in 36m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e78d0bf5b8d34a10a29ba4c27cf94f96 : SUCCESS in 13m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1cb89658b4c41babeb3f2c8bfc5967f : SUCCESS in 4m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cea017cfa3e24d6fb4a4b194951a85e0 : RETRY_LIMIT in 5m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab9df8008999494eb7b65bb2b5fb7a20 : SUCCESS in 56m 44s","accounts_in_message":[],"_revision_number":1},{"id":"af953e6c1ffe0cc9b17bd2c28011d6286aeea7b0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-20 01:29:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4feafb9cc1cd66a5db1611c79101d031aa3d5180","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 01:30:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bbcf53d47f969106d1339b3cd97210693f163deb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 02:51:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63ff682871ce426994bd6015f724eabe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e3fc656eabb431c812aed1a6510db9e : SUCCESS in 7m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/328ba763b8cb4147bf9beeaa56320e2d : SUCCESS in 3m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/3cfc12a3e6d34a5d9fceacf0e4409d35 : SUCCESS in 1h 16m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a0488fcf40714770935f77e106003645 : SUCCESS in 38m 16s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/7469d6aa9ac6442bb37efa536c70e76e : POST_FAILURE in 6m 31s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/59320457d44546ec97cec2ca0276b52e : POST_FAILURE in 4m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/35335161e5954f8b80c14b4ab38dc58d : SUCCESS in 12m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/28b1b011f79b43a89fb7df1d7f49477a : SUCCESS in 13m 16s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f8281ff351c94762b3471a57eff72315 : SUCCESS in 9m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/85fa8c69eb724eb28a0286a64390eb84 : SUCCESS in 12m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8f3dc710f6814887b49c716b0feae8f2 : SUCCESS in 9m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d5efa85c9f3142a79c919c1f5701b588 : SUCCESS in 10m 25s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/3361228707144a59b8b8ca2c7ec588ea : SUCCESS in 9m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/df24028df59943778615f9aec34ddc2a : SUCCESS in 9m 42s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/5c48d2c3a205493d92b01502edb53b26 : SUCCESS in 17m 43s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/fe04fae382a84042b29280a20daee6b0 : SUCCESS in 10m 24s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/a345ad1f5a934793ac0f2d7067b78b40 : SUCCESS in 10m 42s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/92899bba323f411fa2c89b9b4326d4a1 : SUCCESS in 10m 50s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/05e0dbcbe8534efa85a88c17b06890ad : SUCCESS in 16m 11s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3d547e25d8044b7c9a85e1128e1788b6 : SUCCESS in 41m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7c72f4a396014879a929145b5107654a : SUCCESS in 34m 25s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/d758034e9a1443cf9ae1c9a4e2700768 : SUCCESS in 32m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d5d09131a72e4cf4b64955e49af1ae07 : SUCCESS in 27m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/835c0257c5b54772b84b305df275ce19 : SUCCESS in 13m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46bbbef4f4c84d21a775c3b198250d38 : SUCCESS in 5m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b56e4eb07c14c3a8cd6dba937d66b51 : SUCCESS in 46m 32s","accounts_in_message":[],"_revision_number":1},{"id":"f292b833c674f298370ce1c2cc964170056c7089","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 02:51:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c9cde520493e7419d2c716d9c583d2edd21029fc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 02:52:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2156e141c1f246c48b00ea50c3ab8ad7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a17230f52fb24a0aadc5ad1cb15f4fe8 : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/cbb36cb6cf3a46e28569c9c163081363 : FAILURE in 29s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/7decea2cce6843d1ae2e0d928e73fff6 : FAILURE in 43s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"45f6d47a7c2bc13ba58f61e09892a2a1a583b9c6","revisions":{"45f6d47a7c2bc13ba58f61e09892a2a1a583b9c6":{"kind":"REWORK","_number":1,"created":"2023-01-19 23:53:55.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/36/871136/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/871136/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/871136/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/871136/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/871136/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/871136/1"}}},"commit":{"parents":[{"commit":"28d6cf1ba8808b34e8b80728238a61f0b7000054","subject":"Inline parse_request from cpython","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28d6cf1ba8808b34e8b80728238a61f0b7000054"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-09 17:00:23.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2023-01-19 23:53:55.000000000","tz":0},"subject":"Add attrs to lower-constraints","message":"Add attrs to lower-constraints\n\nFuture work: Assuming we\u0027re going to keep the l-c job, we should make\nsure it fails if packages get installed that aren\u0027t in the constraints\nfile.\n\nChange-Id: I61750a1083a1c97a6222ec9040f90980ee73acc8\n(cherry picked from commit ee12a11e708fce6c982a85b58cd2c3899f13479e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/45f6d47a7c2bc13ba58f61e09892a2a1a583b9c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/45f6d47a7c2bc13ba58f61e09892a2a1a583b9c6"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
