)]}'
{"id":"openstack%2Fswift~871185","triplet_id":"openstack%2Fswift~stable%2Fvictoria~I7558ca9281ff32493a81c3ef85873a5f75b4d599","project":"openstack/swift","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-01-20 05:24:45.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I7558ca9281ff32493a81c3ef85873a5f75b4d599","subject":"Fix stable gate","status":"MERGED","created":"2023-01-19 21:19:32.000000000","updated":"2023-01-20 07:53:40.000000000","submitted":"2023-01-20 07:51:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871185","meta_rev_id":"dfcabebd8767cc01f183884193de4595842026fe","_number":871185,"virtual_id_number":871185,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-20 07:51:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-19 22:40:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-19 22:40:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-19 22:38:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0ed40c85b68820948f349633316e765220a64825","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-19 21:19:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55546ee957757c1c2c7af4de2c8c710f0e70fef0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 22:38:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ff162eff1334b8dac6e4d4e60fc0bf6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3cdd79e549a44e683423aadc895715b : SUCCESS in 11m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/52c20ce7d3cc4bbf997b55eca0ae0595 : SUCCESS in 2m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa2325c623bd45b7a426d4c96bff943d : SUCCESS in 6m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/879c1d84a430487a96a33ec087d6a3bb : SUCCESS in 1h 11m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/93b8e14809494265b9cc65a95069b948 : SUCCESS in 48m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2edf7e7ecc9457aa445ac7adbfe5948 : SUCCESS in 1h 14m 24s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cb87c5e6642842679951ea61318126e7 : FAILURE in 5m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/091da455a0fc415db61279602485cbab : FAILURE in 3m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6d527bcecf2846e8b9053111b58b51c9 : SUCCESS in 11m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/26bbd80a254f46d69b3ae530337e249e : SUCCESS in 14m 30s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3e7d6877aaaa4d6fb6bce6afb10a1094 : SUCCESS in 10m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9d52eff776ca48208418a0d35786fd7d : SUCCESS in 13m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a30a4d17633c4d359942e459d31a8064 : SUCCESS in 16m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/040eadb6a946406b993f6698f47fce51 : SUCCESS in 15m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/511dee39831d4a6db7eba72fe339b6b5 : SUCCESS in 18m 00s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c1f11e67c5d24d7b96eb7497bbdc3f6b : SUCCESS in 15m 04s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/501b59fd60b84126850b719599920789 : SUCCESS in 19m 02s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/137e382d47054345993cd51902836822 : SUCCESS in 16m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/13b7abe53c774729a890dcab19f810cf : FAILURE in 6m 23s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5db4048f79ff4cd78e3a4849f65d0bfc : SUCCESS in 44m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/98628733b1ab448bbf35bc12e8250960 : SUCCESS in 49m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/23f1a354b4e54ea5a1ae83f242986343 : SUCCESS in 38m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ae7fa6d92874788a94e5bc95076230b : SUCCESS in 15m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8db9423c6f4a46a0a9f3e4eba74dc794 : SUCCESS in 5m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1844cb96841a4a02ac11da9f74b679ca : RETRY_LIMIT in 5m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5950268939f54187a2cd4653b71d7a63 : SUCCESS in 1h 05m 32s","accounts_in_message":[],"_revision_number":1},{"id":"37406f4be88172e1647fc7ce7e1ea5d3a7121a93","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-19 22:40:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"304336df816937088559c687668e5b6edbf7fd62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 22:41:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"56df8bf9c276972b1fbcb884d8fdff91a28499ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 23:52:17.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e5aca6b3106a4adb9f32cc517eed2e8c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/694b2e29626746aa994975294a0f6df4 : SUCCESS in 10m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5cfa00abeac44793a885b72f879c11ec : SUCCESS in 2m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd566d37dde04881a29898e90df835c1 : SUCCESS in 4m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/84e4dec4ddf84e56b3175e142320faf8 : FAILURE in 19m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ed3bc4956f694b87b71db7c18880fdc7 : SUCCESS in 53m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb3a272e616442c58c563c311b4b4a70 : SUCCESS in 1h 06m 14s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/bb382c56df9d47119c552d11146721f1 : POST_FAILURE in 5m 02s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/e3a2de48937545228eb0c19aeb9f8bb0 : POST_FAILURE in 4m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5e7549db1c664643847811eb8b2afb45 : SUCCESS in 13m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3a6402c5dee043798036a48ff19edd96 : SUCCESS in 14m 05s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1eaf575f820c4a018b477dc459963bea : SUCCESS in 13m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/739440c3c43a440fbf5096607db2037f : SUCCESS in 19m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e6a7984f83234028b52f5aeec1429931 : SUCCESS in 25m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/34b3d3eb8fab4ae096a0da33a20a0de4 : SUCCESS in 22m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9c6bf47206464790966b0dcdf0e45777 : SUCCESS in 23m 01s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d642e3589b194e4aa4eac60f7dab80b7 : SUCCESS in 23m 59s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5af69ca919da4cc4872c1a5f6c855e04 : SUCCESS in 26m 29s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/334d56e8f3f948a4b6149cd795ddef72 : SUCCESS in 50m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/989142a3c6084e6489714e1acb3f734e : SUCCESS in 38m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/afce006588ba4fdea7fedb50940c78f9 : SUCCESS in 42m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b227a11f33b94382a623c6bf59ac9c79 : SUCCESS in 17m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08758dc1b9b445168c6a5942a1ef0d43 : SUCCESS in 5m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea2846f3c57b4493b609c97fc1c34282 : SUCCESS in 50m 38s","accounts_in_message":[],"_revision_number":1},{"id":"dd8ba8af50f75f5c7fb8955b3d0d6caa48c890be","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-20 00:05:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9ce42eff49a5ba293901a111ebac9bda33ed0c52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 02:07:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/42c26bbd77cc49cfaa3ddd59a4431fab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebcc87e0dd71455a9c6447bc55dcd284 : SUCCESS in 9m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/601922d360e743c49f7a154c9e577cee : SUCCESS in 2m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3997e5b7f9e4fd89c442bdeb1ec34f0 : SUCCESS in 4m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/4f56fa9ddc5840c98239cf909134d027 : SUCCESS in 50m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f3a79da33ccc42db96c41a01cd54a473 : SUCCESS in 38m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/190b3be688474b39aabc206fb5d21de2 : SUCCESS in 48m 22s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/10f9e683e4ea4753bc52c3906a673e9a : FAILURE in 4m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/be6ebe06db3b4646bc34035c5bea177a : FAILURE in 3m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d5a26aca25544c9aabae6839ce4fbfb6 : SUCCESS in 13m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/56b31ae193d44cd484ac86a5ea1c1138 : SUCCESS in 13m 58s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b4d02f35e62946c682d4d5bd25a99c9f : SUCCESS in 12m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d7ec47739c514635984f746306422243 : SUCCESS in 14m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/85ce0f096b304db0984773989bc0b74b : SUCCESS in 14m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7abcc082f6b441549045107b90de76bd : SUCCESS in 17m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3a6f2709d76a43f0803c77176d500658 : SUCCESS in 16m 36s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d5a5aee1b12b4e58ac4faf164f49e9a0 : SUCCESS in 18m 19s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8c12b807c2a74cfb801fbb6fd5ce2c20 : SUCCESS in 15m 59s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e3ede8ee052447d1bf50c56d8436c5a8 : SUCCESS in 18m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e68d26ee9aba4170a0188afc09c2a931 : FAILURE in 8m 15s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4c1691be15144f8c90617ed52cf6cf78 : TIMED_OUT in 2h 01m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/744ac5d9bc45461bae6f5ebc6567f318 : SUCCESS in 35m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ae0895ec18ae4e5bbb6da745777cc198 : SUCCESS in 37m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84be089f542140c4b6605467dc6d04d3 : SUCCESS in 14m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c149d99d97f6495ea085679276d888ad : SUCCESS in 6m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a03648f5df9442ca8016220c6f1087dc : RETRY_LIMIT in 5m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/48960c0d0dab497899f3668c04637de2 : SUCCESS in 43m 37s","accounts_in_message":[],"_revision_number":1},{"id":"30eeedd21880e808e72e2d56bd2d042c67684ec8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-20 05:24:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a6b367f59da588d1495ad353a0d292e794b136f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 06:38:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a407ae7623e54af998914de2982d0d9c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0860dc3c11d94f708071162c34dfec82 : SUCCESS in 8m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/172f81a5e5824c06b6cf2c2b978e4ad1 : SUCCESS in 3m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1f0b76c72464ffba5fed1a10659ebf3 : SUCCESS in 5m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/89683e040e78472a925135841eb93fcc : SUCCESS in 47m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e0b11e11af7648e9b5bca0dd4c4bee29 : SUCCESS in 42m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/be29c82e66574fe88638c4667024f20c : SUCCESS in 1h 08m 30s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bda2b6f3d5764736857cbfcc431c01d0 : FAILURE in 3m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/237b8e7e44024aa7843191765e618cae : FAILURE in 5m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a53b62e82cee4c6890815bab8dc890e1 : SUCCESS in 12m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a7e465f0f16642d284bec8ee0f024761 : SUCCESS in 11m 36s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b7075073d30a4355a4542ecddf508ccc : SUCCESS in 14m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7a4fbdcbd0784f228c2cf5342962e7e5 : SUCCESS in 13m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1e01122a330c4263b123d726d5036ca1 : SUCCESS in 14m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e6d1ad570fd345b9b0bb9c31135a50a0 : SUCCESS in 14m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4147ef81736149f7aeeb4059bd60adbc : SUCCESS in 17m 42s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5814aeb9018342b1a5b03dab80a06a79 : SUCCESS in 15m 56s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/392558ce8f55462c9bb0b95ba6a95b5b : SUCCESS in 14m 47s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/08dd1691be1e4c8ba32901efc1d28f5a : SUCCESS in 17m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5db2fef5284d4d489ebe2b77f4d0ab94 : FAILURE in 6m 59s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8a7dfbe7e3a746d99638e8c8084df44e : SUCCESS in 43m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dfcb677152754f4aafa762e0004dcf52 : SUCCESS in 44m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/522ea9f9f1a9447488d5345596ff47a2 : SUCCESS in 32m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0d9ef041e8c4499a6b891601785ad70 : SUCCESS in 13m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/110e433d90424da8a6a825bcc0e1fd71 : SUCCESS in 5m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4d454170e93044aea8716804737f75f6 : RETRY_LIMIT in 7m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d76e094e53547879c047cc0779db9a1 : SUCCESS in 1h 01m 07s","accounts_in_message":[],"_revision_number":1},{"id":"379dbd95cf95da02b83a6542b6d2a6492c802bc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 06:39:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"853a096021beeac43489c4b79b0f99fe3a67971f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 07:51:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee6c594035d141cc8cc3a60472a49f7c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9060f2d361243c893560c1022690407 : SUCCESS in 7m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0df39a1d1b794993a0c1636c62082600 : SUCCESS in 2m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b73ae9dc775b41be8af2cc6f8163c318 : SUCCESS in 4m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/ade5accddbb74b398d6d0e99dc6e1ab0 : SUCCESS in 1h 04m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5c5b7bc4b7ca432b91acc0366c5d2c66 : SUCCESS in 43m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/316233bcec884ba591df492ba0849504 : SUCCESS in 1h 07m 53s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/60ac3735c1ce42c29bd4881a7f6c5354 : POST_FAILURE in 4m 35s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/363076cfa48f4aef8182119d21543505 : POST_FAILURE in 4m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/885568358c91472b80327da2a97c7b8c : SUCCESS in 11m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/29dc5761405c4c49bf3225ddb992ed04 : SUCCESS in 12m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/10d54a3439f24129a9fb8bc0a9f77aca : SUCCESS in 12m 27s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5bf7504b7f9d444fac62cb9cd02e2588 : SUCCESS in 12m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/82e649d6492543d88e7bd9b6861b2362 : SUCCESS in 14m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3fdf4aa1518f4921964c93821b35ea9b : SUCCESS in 14m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4c672f44f6f247a0bdf5104b3497f29e : SUCCESS in 15m 42s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/836422b4cada45da8c980268548daa2e : SUCCESS in 15m 27s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5267e5f6b2cc491d85c68b06e5f0e6d2 : SUCCESS in 17m 38s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/85900d2d8aee4543af7f3f7bf09acdbb : SUCCESS in 41m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/858c32a47ed047f2b3ae62ee9dbc9dce : SUCCESS in 45m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b5cc2a8983224f06b318e4c4d4b0efbf : SUCCESS in 37m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d86edc4b618e453796b7918912bfffd9 : SUCCESS in 11m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c4cb43b483a4499955455724cc3ff4c : SUCCESS in 4m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/94a0285f9abd4b4e8e6c862a37ddb8a8 : SUCCESS in 58m 59s","accounts_in_message":[],"_revision_number":1},{"id":"e93478a5bfa1df7ecf389095309bad37d8b5317c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 07:51:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dfcabebd8767cc01f183884193de4595842026fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 07:53:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65ee6b9fc04f41ec8415fb9514436a08\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d3b6805ff8846189898d60f2fbb4f94 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6f956b1825246e69aca16164f006aff : SUCCESS in 36s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/360b2d6b2891480c9bfcf1c43cb5f80d : FAILURE in 29s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/94adee77a911401eb93099fec295f81a : FAILURE in 29s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4ee566aa2992bfded7d881c9ae71f3267ada8ce8","revisions":{"4ee566aa2992bfded7d881c9ae71f3267ada8ce8":{"kind":"REWORK","_number":1,"created":"2023-01-19 21:19:32.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/85/871185/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/871185/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/871185/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/871185/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/871185/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/871185/1"}}},"commit":{"parents":[{"commit":"e8d356925931ee5d0854957fdedda570e6210a3a","subject":"Merge \"Inline parse_request from cpython\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e8d356925931ee5d0854957fdedda570e6210a3a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-19 21:14:55.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-19 21:15:25.000000000","tz":-480},"subject":"Fix stable gate","message":"Fix stable gate\n\nNote that this includes switching the DSVM job to run tests under py3.\n\nRelated-Change: If33532232572e802b2099f39f4231ee6add6d5a9\nRelated-Change: I459d6065a3e4290768c499e7973917f7cf4ce2e7\nRelated-Change: I61750a1083a1c97a6222ec9040f90980ee73acc8\nChange-Id: I7558ca9281ff32493a81c3ef85873a5f75b4d599\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ee566aa2992bfded7d881c9ae71f3267ada8ce8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ee566aa2992bfded7d881c9ae71f3267ada8ce8"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
