)]}'
{"id":"openstack%2Fswift~871232","triplet_id":"openstack%2Fswift~stable%2Ftrain~I61750a1083a1c97a6222ec9040f90980ee73acc8","project":"openstack/swift","branch":"stable/train","attention_set":{},"removed_from_attention_set":{"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2023-01-25 00:38:41.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-01-24 23:49:47.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I61750a1083a1c97a6222ec9040f90980ee73acc8","subject":"Fix stable/train gate","status":"MERGED","created":"2023-01-20 05:35:03.000000000","updated":"2023-01-25 00:40:26.000000000","submitted":"2023-01-25 00:38:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871232","cherry_pick_of_change":871136,"cherry_pick_of_patch_set":1,"meta_rev_id":"6d974f4f82380a469dc3385bdbc2ee0582f3d8b4","_number":871232,"virtual_id_number":871232,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-25 00:38:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":1,"date":"2023-01-24 08:47:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-24 23:49:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-24 23:49:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}],"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-20 06:47:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-23 08:22:51.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"CC"},{"updated":"2023-01-24 00:30:00.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2023-01-24 08:47:55.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"}],"messages":[{"id":"78b624697619cd67a7f4ecef455e64a67315bd36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-20 05:35:03.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"c88be73799b7d1b33658e360a5407c13fa25211c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 06:47:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c889ed03100645679a2cae4d1933a6ea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d32ccda8cb514e95bd4950c2933e1bc0 : SUCCESS in 6m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2b70520fe0274566882b23173dacb47e : SUCCESS in 3m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/9279a72a4b284554a21fcc7ab73e2cc4 : SUCCESS in 58m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e374670f6cab4eeab0c6d387cab878dc : SUCCESS in 52m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ecf1dbd60563441699e3e9da95b3d104 : SUCCESS in 54m 24s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/637432d135be46b48b353820672473a1 : FAILURE in 8m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f18fa2a9f04c4b9b9ff766e30dc58e50 : SUCCESS in 12m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4a846d28142c4e759d005bffdd8258a1 : SUCCESS in 12m 55s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a83f5b58300f439da052b12e832ccb50 : SUCCESS in 13m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3f6eb1c8a08146d78136436fc78324bb : SUCCESS in 8m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/40493f3c61984e99affd741ffb4c9707 : SUCCESS in 8m 37s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/3db84bfa3a7040aea9ba1f064ef59ebc : SUCCESS in 9m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2742f0136a6a4604972693122e1d2823 : SUCCESS in 8m 58s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/6687c791157b4d75b6b6b38495f36ccf : SUCCESS in 10m 44s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b9dd09daaa5446d38e5941fc9e5f5061 : SUCCESS in 9m 37s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/102156672a744f38a49d7bea42d99eb0 : SUCCESS in 8m 59s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/13bf65065a8e4279b3d1c0f2eadce782 : SUCCESS in 8m 49s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0d2dea2f8fd64af09bd5b2011a92df57 : SUCCESS in 9m 28s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/1fc2e95f98cd406da6e6a8a6e9f23981 : SUCCESS in 12m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ba57be86b71b490ca9c617983e4f9294 : FAILURE in 7m 11s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f577248c1ad443bfa4afeabab06c2899 : SUCCESS in 30m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/523104ebf0d14f1884eebf10daaf1968 : SUCCESS in 25m 06s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/81c54dbb0d414e4b9b7cd678ae3a872b : SUCCESS in 33m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2011f50df8694fd18679b39b92be1dd2 : SUCCESS in 31m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/edca0f08cf3b44ddb4b758772a25ac6b : SUCCESS in 11m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1341dca00c44f3ebae17d629a79df21 : SUCCESS in 5m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4aadf64a14bf49d0bddff475f8888e4e : RETRY_LIMIT in 8m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b66f00f95a6b49f083e2fc7f117f7418 : SUCCESS in 1h 06m 05s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ae9d70d88064430c9782f9479f3cfa3a : SUCCESS in 49m 58s","accounts_in_message":[],"_revision_number":1},{"id":"eb4c182fa424a6e016e90192de4afba745a383ba","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-20 16:43:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b67a09e51ce1cc5ac4c77052bfd0d0ebe0bdd1f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 16:44:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"62f3ef03058c7c38e6a52b225b039ff2313ab032","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-21 01:23:25.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bfdc5604f2f54930b533be1bcf2384dd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fe7ce8c47a64e729b3b24d909ae4777 : SUCCESS in 6m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ccdc3e805c884300a8d767990780190f : SUCCESS in 3m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/727695695c2d4504ace7d820da70b7c9 : FAILURE in 54m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e08e6d8abe8f40acb154172a3b10b48f : FAILURE in 20m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/de328eb2b80a44c7870d3c91200512a5 : SUCCESS in 51m 08s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a39d6bc64eaf4b3682ee1637341db961 : POST_FAILURE in 5m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6db2c17bb0a4447993504badf8baaf6b : SUCCESS in 13m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3151ff7b3a994a9eb4e39d3c4bdf1350 : SUCCESS in 16m 59s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/206dd9d57a5b4790925e8637c8160ecb : SUCCESS in 12m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/daaae315f5aa4dfd87ecff2a51aebd4f : SUCCESS in 9m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/eabf05f04b9a4c99a8e9b2c8523e5189 : SUCCESS in 13m 54s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/638b01530edd4fd680e0375cfccda143 : SUCCESS in 9m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3b1d5fa221c3403dbbc7d11378744b0a : SUCCESS in 15m 13s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/0cbffdb2bee24383b7a28aa5a6327049 : SUCCESS in 14m 44s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c567483c3add41e19649014a59ab0b8a : SUCCESS in 13m 09s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/dc6702b9c02241d590afbcef7ca0db03 : SUCCESS in 10m 38s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a7952051ada749358405eea5d498e704 : SUCCESS in 14m 54s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/1157410b99584be39c0a9b37348b03a4 : SUCCESS in 19m 23s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e6f2d35268a146bf9ffa509a2088580e : SUCCESS in 49m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b4e6c449f9ff4100801e61f25b2842b8 : SUCCESS in 27m 18s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/3bdd268bfeee4c70879b228492a3b4eb : SUCCESS in 37m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/77ca5aa339ec435388a3e02b6a80f8b3 : FAILURE in 26m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a85a31eeed5d49c7908c6daa8ca59db2 : SUCCESS in 11m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/111f6df4522446c6bf6e9ed3afa41888 : SUCCESS in 6m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e4e82e45686465fb8c0f417ea978895 : FAILURE in 29m 50s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1222dbd684b14fb59277eb87a5a0f758 : FAILURE in 56m 19s","accounts_in_message":[],"_revision_number":1},{"id":"4cb3e32576d6618cf468f849039d21668c51c49b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-21 03:18:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0542ea131e46f887f54353ffb1843629a2607a5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-21 04:47:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8fb679336a24769ba2eb399c88f1a16\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b8d3a3ff9ef4534b42f6136acaeb25a : SUCCESS in 5m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fe2cc915bc884af693d36f2d22a1cf04 : SUCCESS in 2m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/d0caf1bf15db4d428b01d0f401e27d6f : FAILURE in 46m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a864060983db4e038dab4ed266734df7 : FAILURE in 20m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/77b8382c0422439e9595955a20ac522e : SUCCESS in 42m 20s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/714b48a273044606938f265107cebb5d : FAILURE in 3m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/158b2a588e654b5eb6e7bcb25e226c16 : SUCCESS in 18m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/81a40329717b49d99764362f0d660212 : SUCCESS in 16m 01s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/14c5640b022445e8bce42cbcb70b20af : SUCCESS in 12m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/25bdda984b9d48e8bae4b4ba5e7c3dbd : SUCCESS in 14m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b99ddb75bb1f41a888c850eb94c78110 : SUCCESS in 10m 05s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/1aa52ec0223e4acc80fbd3970bd6722f : SUCCESS in 14m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c3784aad9ff247ad9dbd3199f2bb5146 : SUCCESS in 11m 09s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/18f38c0d3d174c5f8cd258e3c4b068aa : SUCCESS in 13m 29s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a8eb205d20fb4d1699ea686f8c37cb56 : SUCCESS in 9m 10s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/175d9c3cb42a43fdb6899fb5dff14272 : SUCCESS in 10m 01s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/5e3ac5aef1a7461da3199339b5ae217e : SUCCESS in 9m 14s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9e895d2bb7b84dda9de1c8bec01d9df2 : SUCCESS in 11m 36s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/f2f1bae2ca194d22a9563cb335942d7d : SUCCESS in 11m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/063dd7a604114c4990401738fcbb02b1 : FAILURE in 6m 33s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/59bbda8d537a4a1bb74ddc226f91c8c5 : FAILURE in 38m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/620be1a8e3a24b6a97e0672f18577720 : SUCCESS in 22m 27s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/b21734c08e344d81ab5da32d3412ed32 : SUCCESS in 35m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2040c0e8711845828a22e379e05f03cf : SUCCESS in 23m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54a0213668214f4586a4039f3d4cd518 : SUCCESS in 13m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6d00135297442759d026b65ace041ab : SUCCESS in 5m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6054af37fd92469b8379f752c96d15af : RETRY_LIMIT in 5m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/df63d142723b4b0b9a037e30b2fb129a : FAILURE in 18m 53s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7df2c25856ec4372af05661400a7e041 : FAILURE in 1h 09m 49s","accounts_in_message":[],"_revision_number":1},{"id":"b54701e4386e2988cb033070a8f3dca82a217f4d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-23 08:22:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7e7816565d5ad694a99a92430b18f9efea773445","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 09:30:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/016809edde324fa6a89a450b19e37a9c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e05669f3f602424a87f163992dcf7b8b : SUCCESS in 4m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/71d4b653d4da4c62aad942309c2b94c5 : SUCCESS in 2m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/44fb5581d3004389a8b0da4a82b9ca74 : FAILURE in 59m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0b6d50f52fca41ab91357ecdd8eae5a0 : FAILURE in 29m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e45a93281f1e4a8ca65d09d278501b68 : SUCCESS in 50m 41s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4bc7e320f4bf437ba3d3c59843f2ec90 : FAILURE in 4m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0354e7e594d5472f9c570a444dcfeda3 : SUCCESS in 12m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7915f2d528594eb9a148fe44b0226ecb : SUCCESS in 12m 26s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d3bfaef4a6794aeb84736e2704652262 : SUCCESS in 12m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5425132db5f748e69521703eff76f06f : SUCCESS in 8m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/802716e8e0584535975cb30c0b7f741a : SUCCESS in 10m 15s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/36e8351ede3c41d1be1fa4a8313e7dd4 : SUCCESS in 9m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2ff5d3a809d041e48a60a909bf5d71f3 : SUCCESS in 9m 13s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/05e9248f1c924ad29305d107eea31620 : SUCCESS in 11m 46s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/44d6ea2f6d2f4af4befc91f3de5230db : SUCCESS in 10m 08s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/868f20b8c837487fb4b5b769cb776d04 : SUCCESS in 10m 22s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/8b02ccd78bee47589888eaf50067bd3a : SUCCESS in 13m 47s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/bbf32d5e234d4e56a3b11caf7795b71f : SUCCESS in 11m 36s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/6c6768bfb81f4319b5da65bd85ec26f5 : SUCCESS in 12m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6b507c6196dc43b5aedbb391a1f189db : FAILURE in 13m 52s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/007bad28aa4c4ec8a8cb248bdca04446 : SUCCESS in 45m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/773b51fd136c417f85949807c4d2e7e9 : SUCCESS in 30m 48s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/0fb3116f2a4945f08e052cb93c46c8b9 : SUCCESS in 27m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/95a90f739b8d48828d6a97c70310c467 : SUCCESS in 34m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6489f3c86ab24d0986f8d41522be52b8 : SUCCESS in 12m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92d3334b39f34d0e894380b90461a62d : SUCCESS in 11m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3d7c941f7f224b25a82d148e44fb28c3 : RETRY_LIMIT in 6m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/12e2ea704f86405f8c19249a34131d5a : FAILURE in 26m 54s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/34ff36f5917e43e7834bad834ecac691 : FAILURE in 37m 56s","accounts_in_message":[],"_revision_number":1},{"id":"8bb435a6b26e9c8b970e148fc41600518048f403","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-23 18:49:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"452bd0589a913a0e4104e576fe168ab85cb29279","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-23 19:47:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a109201618a086132a624a44cf433aa7d75d102d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 20:35:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/118aecf9b57141fcb0e7f8eb4becdbb1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e390fc6322f4713be8e12b9a5822707 : SUCCESS in 6m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/78281013a19c4e42ae10268dfa22d43d : SUCCESS in 2m 54s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/37908340f8f0474ba09671e116834f34 : FAILURE in 4m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7f2ecc285bfe408fa870cb4c25a4bb2d : SUCCESS in 22m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fe46b248ae7a411985d31dbf07390bac : SUCCESS in 18m 54s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5b2c7c69a5c84ea09bbf2a8d15a60919 : SUCCESS in 10m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bfa5c19eba0a47b29364f4ed70bd1646 : SUCCESS in 9m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fc8af0d7e035434da03453e7cd5c5ca0 : SUCCESS in 9m 18s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/6025ad4b595846f086dcf588b8f2f54c : SUCCESS in 9m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e54cb0472c9844d7afd1526a6d3999e4 : SUCCESS in 8m 51s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/cda8a9960c514d4d9ba196b85e5992d5 : SUCCESS in 13m 39s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/49cda602e1e84c3a9229591cf9ac4903 : SUCCESS in 10m 07s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/9ebdbddb84a2427e8e666a9fff5195ea : SUCCESS in 11m 16s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/893c32ebe1af4e089740c32a8953a771 : SUCCESS in 10m 16s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/add48a0330e147bc8f082b38bd89df4f : SUCCESS in 9m 41s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/7eb38e492aca4948a9b0db0ce349fd2a : SUCCESS in 12m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ec4bdeba0458470f8a1a2958eb40c70b : FAILURE in 7m 15s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3bffe6218a3146258f195bb1f67cea36 : SUCCESS in 36m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/29fd8887e65a42aa9b4c8f126e7c1508 : SUCCESS in 36m 24s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/6c8dccdcf9694a3b803a37e4d19f431c : SUCCESS in 38m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6127febcccc44e9eb624233d4f033faa : SUCCESS in 28m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75ebc63c6fbc4c0ab872d310f24646b8 : SUCCESS in 12m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40c1a70c31c6459facbb3251941215bb : SUCCESS in 5m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d00c0dc1b7e04bdaaf9de0c008b87eff : RETRY_LIMIT in 5m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"30cde6b143f6f474dc9215f2afde835cff86e19a","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-01-24 00:30:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ccf5cc9a23387b7b292c62087b4bf33f2f73caae","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-01-24 00:32:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2c8d475f6ebc456fa733b96f57f6a1bda0d34399","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-01-24 05:27:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c47e682bb7ab1362ad0bb4335209877ca71531dd","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2023-01-24 08:47:55.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"772f31264f955c8c047132a0e98379ef25a51c61","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-24 23:49:47.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b569df3f6f032755782cfda688c6cb1c8112e5a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 23:50:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a434c43dfb4837575c30df2fb83054636c690417","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 00:38:41.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8d9901f7f34448ba102de820c8dfc7a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f94da246ea54e02aa28a5d0de2e986e : SUCCESS in 4m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c2764eb748874d47a5cde396e6cf2df6 : SUCCESS in 2m 39s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/fcd9616d1ed048f3951670d3b5bf72e6 : POST_FAILURE in 4m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/71cb9d4615b7412ebd1aa06e9fc998dc : SUCCESS in 11m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/56524aa205b94cf5af827a6952e99f33 : SUCCESS in 9m 51s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e51e0556b2a648ac80d0be8fb528b877 : SUCCESS in 11m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4584d044d84e40848c9cbb37823b8950 : SUCCESS in 7m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/94c07fa20ef443fc91aa2ca2c681b0d1 : SUCCESS in 9m 20s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/efbf02d4ca2d4ef789a16deec3126011 : SUCCESS in 7m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/325c18b7a5f4495c81351a780b7a9d03 : SUCCESS in 9m 24s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/10b00bbf867a4a49be80e8491012f489 : SUCCESS in 10m 08s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/dc51f2fb3a624bb1adbd88149cfc817a : SUCCESS in 9m 37s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/3c26c058ded24bafadabe68c66338503 : SUCCESS in 7m 27s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/74b991dcb0ff4329bb63afc3df9e4aa4 : SUCCESS in 10m 13s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/f091c35a57a341d1ac848c9f94c93ca3 : SUCCESS in 11m 21s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/21a2811fbaff4132ad697a50a79446e1 : SUCCESS in 33m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8820d26dbda74aebb8bd6b64dc700b12 : SUCCESS in 18m 51s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/639103caa88f4b5389bbd9a8fc1de1bb : SUCCESS in 21m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3c30371348c74e6fa120e6e99bafc551 : SUCCESS in 29m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6459e9d598ae4797bdd727ef14e07a94 : SUCCESS in 11m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50018684f16f43d289571b65880ba74c : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":3},{"id":"986ea91e9784684b02d383985b22d1a4bea07ce9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 00:38:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"6d974f4f82380a469dc3385bdbc2ee0582f3d8b4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 00:40:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e29e4faa76dd4136be96f31010da3803\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96be540801a34ea7a44c877810dc6130 : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/3476d009c8b44512a2730677fa3c1205 : FAILURE in 27s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"91144596c670741c2f452c864c026eff5583989c","revisions":{"00c877fb0691baacc216a14be9beb7553257d9d3":{"kind":"REWORK","_number":1,"created":"2023-01-20 05:35:03.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/32/871232/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/871232/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/871232/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/871232/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/871232/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/871232/1"}}},"commit":{"parents":[{"commit":"a86270bed72cdbdf2f42eec562d9fae2d273fa64","subject":"Inline parse_request from cpython","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a86270bed72cdbdf2f42eec562d9fae2d273fa64"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-09 17:00:23.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2023-01-20 05:35:03.000000000","tz":0},"subject":"Add attrs to lower-constraints","message":"Add attrs to lower-constraints\n\nFuture work: Assuming we\u0027re going to keep the l-c job, we should make\nsure it fails if packages get installed that aren\u0027t in the constraints\nfile.\n\nChange-Id: I61750a1083a1c97a6222ec9040f90980ee73acc8\n(cherry picked from commit ee12a11e708fce6c982a85b58cd2c3899f13479e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/00c877fb0691baacc216a14be9beb7553257d9d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/00c877fb0691baacc216a14be9beb7553257d9d3"}]},"branch":"refs/heads/stable/train"},"025b81b9964ca7efc1efe72a84dc928d58c65b7d":{"kind":"REWORK","_number":2,"created":"2023-01-23 18:49:31.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/32/871232/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/871232/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/871232/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/871232/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/871232/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/871232/2"}}},"commit":{"parents":[{"commit":"a86270bed72cdbdf2f42eec562d9fae2d273fa64","subject":"Inline parse_request from cpython","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a86270bed72cdbdf2f42eec562d9fae2d273fa64"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-09 17:00:23.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-23 18:29:03.000000000","tz":-480},"subject":"Fix stable/train gate","message":"Fix stable/train gate\n\nTwo things:\n\n* Add attrs to lower-constraints\n\n  (cherry picked from commit ee12a11e708fce6c982a85b58cd2c3899f13479e)\n\n* Drop tempest and grenade testing\n\n  Recently, these jobs started failing while installing tempest. The\n  trouble seems to be:\n\n  * tempest 26.1.0 depends on paramiko\u003e\u003d2.7.0\n  * paramiko 3.0.0 is selected which in turn depends on\n    cryptography \u003e\u003d 3.3 and bcrypt \u003e\u003d 3.2\n  * cryptography 39.0.0 and bcrypt 4.0.1 are selected, neither of which\n    get installed from binary wheels and both of which require a Rust\n    toolchain to compile\n\n  For some unknown reason, this used to be OK-ish -- the stable/train\n  upper-constraints file would be respected, paramiko\u003d\u003d2.6.0 would be\n  installed with cryptography\u003d\u003d2.8 and bcrypt\u003d\u003d3.1.7, and pip would\n  simply warn\n\n    ERROR: tempest 26.1.0 has requirement paramiko\u003e\u003d2.7.0, but you\u0027ll\n    have paramiko 2.6.0 which is incompatible.\n\n  even though everything would still work. It\u0027s not at all clear how or\n  why this stopped working -- there haven\u0027t been any recent changes to\n  tempest\u0027s requirements or the stable/train constraints file, and the\n  version of pip is the same. Given the age of the branch and its\n  extended-maintenance state, dropping the testing seems best.\n\nChange-Id: I61750a1083a1c97a6222ec9040f90980ee73acc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/025b81b9964ca7efc1efe72a84dc928d58c65b7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/025b81b9964ca7efc1efe72a84dc928d58c65b7d"}]},"branch":"refs/heads/stable/train"},"91144596c670741c2f452c864c026eff5583989c":{"kind":"REWORK","_number":3,"created":"2023-01-23 19:47:14.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/32/871232/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/871232/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/871232/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/871232/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/871232/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/871232/3"}}},"commit":{"parents":[{"commit":"a86270bed72cdbdf2f42eec562d9fae2d273fa64","subject":"Inline parse_request from cpython","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a86270bed72cdbdf2f42eec562d9fae2d273fa64"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-09 17:00:23.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-23 19:47:02.000000000","tz":-480},"subject":"Fix stable/train gate","message":"Fix stable/train gate\n\nTwo things:\n\n* Add attrs to lower-constraints\n\n  (cherry picked from commit ee12a11e708fce6c982a85b58cd2c3899f13479e)\n\n* Drop tempest and grenade testing\n\n  Recently, these jobs started failing while installing tempest. The\n  trouble seems to be:\n\n  * tempest 26.1.0 depends on paramiko\u003e\u003d2.7.0\n  * paramiko 3.0.0 is selected which in turn depends on\n    cryptography \u003e\u003d 3.3 and bcrypt \u003e\u003d 3.2\n  * cryptography 39.0.0 and bcrypt 4.0.1 are selected, neither of which\n    get installed from binary wheels and both of which require a Rust\n    toolchain to compile\n\n  For some unknown reason, this used to be OK-ish -- the stable/train\n  upper-constraints file would be respected, paramiko\u003d\u003d2.6.0 would be\n  installed with cryptography\u003d\u003d2.8 and bcrypt\u003d\u003d3.1.7, and pip would\n  simply warn\n\n    ERROR: tempest 26.1.0 has requirement paramiko\u003e\u003d2.7.0, but you\u0027ll\n    have paramiko 2.6.0 which is incompatible.\n\n  even though everything would still work. It\u0027s not at all clear how or\n  why this stopped working -- there haven\u0027t been any recent changes to\n  tempest\u0027s requirements or the stable/train constraints file, and the\n  version of pip is the same. Given the age of the branch and its\n  extended-maintenance state, dropping the testing seems best.\n\nChange-Id: I61750a1083a1c97a6222ec9040f90980ee73acc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/91144596c670741c2f452c864c026eff5583989c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/91144596c670741c2f452c864c026eff5583989c"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
