)]}'
{"id":"openstack%2Fswift~871242","triplet_id":"openstack%2Fswift~stable%2Fvictoria~I84494123cfc85e234098c554ecd3e77981f8a096","project":"openstack/swift","branch":"stable/victoria","topic":"bug/1998625-stable/victoria","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-01-21 03:20:01.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I84494123cfc85e234098c554ecd3e77981f8a096","subject":"s3api: Prevent XXE injections","status":"MERGED","created":"2023-01-20 19:52:19.000000000","updated":"2023-01-22 08:37:21.000000000","submitted":"2023-01-22 08:36:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871242-bug/1998625-stable/victoria","cherry_pick_of_change":870828,"cherry_pick_of_patch_set":3,"meta_rev_id":"ba27dc66ce7b7d9f4141d6e49d8e9b8be38c456b","_number":871242,"virtual_id_number":871242,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-22 08:36:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-22 07:19:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-22 07:19:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":32333,"name":"Aymeric Ducroquetz","email":"aymeric.ducroquetz@gmail.com","username":"AymericDu"}],"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-20 19:52:19.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":32333,"name":"Aymeric Ducroquetz","email":"aymeric.ducroquetz@gmail.com","username":"AymericDu"},"state":"CC"},{"updated":"2023-01-20 21:21:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"964b78b54fa15be4d0e3cd176f872f2be074fb33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-20 19:52:19.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"5cf5f0d1bbfe826eadbccba1fb3e2183b4502345","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-20 20:24:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"138035200c36b3b8a56a8dfb69e8ac0a8e62a72a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 21:21:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9301d6812b664016be808b714e835c2d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2550e942bd7a4caabf1186313ad4cf2a : SUCCESS in 8m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/fd85f1aeac334ab8b6feb943bd4c214a : SUCCESS in 46m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9e295bd3f01646608fce998a9f4fca80 : SUCCESS in 30m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/354f5f4e31004d8c98ad53793c90edd8 : SUCCESS in 46m 38s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a8826a425dbf406c93fa4176f63b9b19 : FAILURE in 3m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c185d9e863e04da8876aca2a8c77aff1 : FAILURE in 3m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7fa5749e74124503ac95c2ea5769d90a : SUCCESS in 10m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e0911cbced6a41118a40c0bb2ce7a345 : SUCCESS in 10m 36s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/183df772f9ca45ebbec3d1dec62eb79f : SUCCESS in 12m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e3a9fb1150864c889af2b9a76cb533c3 : SUCCESS in 12m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0e69d1828f3a4fa0a67982d7abb7e30e : FAILURE in 15m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3c8e2db5974b4447a5c1529c8acbca70 : SUCCESS in 16m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a45016c4f38f4142b7cc9804dac0c4f9 : SUCCESS in 17m 30s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2416d1b4c0ac42cb928cf452160548f8 : SUCCESS in 15m 01s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/75a2f18be350430ea47c77b69fdf22e5 : SUCCESS in 17m 19s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0e7f0cfe26664cf89f0d5321984d3331 : SUCCESS in 13m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7ff8879affb14c7996a5bd9c1b7e5b7e : FAILURE in 8m 07s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4b0cd00a69a24fd58d39b90ca0ca58f1 : SUCCESS in 40m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/933814cd9cd24da096069f69ef768640 : SUCCESS in 44m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8e1e7e0ff7414be4a8e6580b32639719 : SUCCESS in 46m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/121dcc8a320e4b52b2b2bdb7a17fcdc7 : SUCCESS in 12m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96d09b8600584397a12b04a997844306 : SUCCESS in 6m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e2dd79a0344b42839b2423e66bdf2fba : RETRY_LIMIT in 5m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ebb3e1a4e69f40948e54796a1d1738b1 : SUCCESS in 54m 43s","accounts_in_message":[],"_revision_number":2},{"id":"3ff8c283a06943e4d85efdd229439b18ca13a8ab","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-20 21:54:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5b583102488ed39c4de5fbb53f2bceb251c0b9a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 23:12:08.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a1a20937c6f74baebf6a52d1989b578c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25da754ed9404e91a286afbcdd11e4cd : SUCCESS in 7m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/0aa331f9bc8e4e628006f697d0f96323 : SUCCESS in 1h 02m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/68582a1b81d34406a9b241e7b2713623 : SUCCESS in 45m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1f5fa704c01b4a7991351edecd39eed4 : SUCCESS in 1h 11m 51s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6402adab2ac04f1e9e3722e855a0b804 : FAILURE in 3m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c0104b5588d149738edb4444cd94a9fc : FAILURE in 3m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/485a6c97dc1b4732b42a7bef9a4afb03 : SUCCESS in 14m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c8f845383aaa4be4b45d0cea8ab5eed1 : SUCCESS in 12m 35s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/16f5ea2e4eb44d8bb41a3120c91d02ef : SUCCESS in 13m 47s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9ece5ecdd1d44382941b973f8aa5de51 : SUCCESS in 13m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0b2da18bf3a44428b29bd320b39267cd : SUCCESS in 16m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f04ddc279797416b83c792bbe9c105f7 : SUCCESS in 16m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e1042550259641578f4506fb3ce90fab : SUCCESS in 16m 20s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/827fa3178dde4686948eb51775970275 : SUCCESS in 15m 09s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4bf92851dec94c10a30716dc6c937443 : SUCCESS in 16m 47s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f45c1ce645834179905072cf871865f3 : SUCCESS in 15m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/384a33afde1842ba918befe64de7939d : FAILURE in 7m 41s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/39e676ce94aa40639cffe75ebf6613f2 : FAILURE in 40m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5ae241831c2c40a4878942be4c0121b8 : SUCCESS in 35m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b49f0d5c01b64572bd28e1b881aa9521 : SUCCESS in 45m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0800ee0dbf724bd29c402b7a6c27afca : SUCCESS in 12m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f7618e9e9234a7eac21b04b18c46f68 : SUCCESS in 5m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/91d79c90c4a44f0db7ab3739801fcc2d : RETRY_LIMIT in 6m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe8912d1f29645fa9243ca95a3552bf7 : SUCCESS in 1h 02m 32s","accounts_in_message":[],"_revision_number":2},{"id":"7c6e89e6a7f957208ab58d20f1a4cae08bc7d6f5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-21 03:20:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9c336a84ae69b717fc440d8b2b4dd85d10422dc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-21 04:29:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1796a22c4a644a66a92f95a5dd949883\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbbd3dbdf462466a8ebdcf1bcf7c2b12 : SUCCESS in 7m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/13e4f5d088114f67abb84a244550efae : SUCCESS in 54m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fcc4648d3dfa4e8fa00eb398976dcfcb : SUCCESS in 50m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ccd35fdc43bb4a43b91d64b2516f8135 : SUCCESS in 50m 13s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7f488ffed3a448fd8f9f2170a28f39ae : FAILURE in 3m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6363935b29eb44aa8f861561d2c7c103 : FAILURE in 3m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2686182696bb422899faa6b66cabf8de : SUCCESS in 11m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9783b5bf7f8a402ca8186a867ff0fc80 : SUCCESS in 12m 29s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0ab90c6a23094009a50ee71e97ba3d5b : SUCCESS in 12m 06s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9120bfb6b54a4c97814678b41c05b1ea : SUCCESS in 12m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a216b1cf14c9490dbc17fa66f5e2e307 : SUCCESS in 18m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/25af619ce4504a599640de6215f7da8e : SUCCESS in 16m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dbd2593adcab4925ac72646fcabb6e53 : SUCCESS in 19m 27s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a08d76b231384a91b35435edac0a9587 : SUCCESS in 19m 33s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/0015fab6b37848a58a7a51989af4328c : SUCCESS in 16m 59s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ccf77b750b754ca5aa5fa46edf18670b : SUCCESS in 18m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9e5c429cfb4b48d2909e604dc6327f2d : FAILURE in 7m 32s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7c9594f418f444c4b10b2fd0cd811a02 : SUCCESS in 46m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f18e2bfc4a9f408db71d44bd12982501 : SUCCESS in 32m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/182750f7907341e59b74a9495f660aff : SUCCESS in 38m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d802b25104814e8f850bb9c8ca2de21a : SUCCESS in 13m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6954828b38aa4e08b6cc60cfb8df8619 : SUCCESS in 5m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9958cb1cc6754548820e772b3c70120f : RETRY_LIMIT in 7m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ce20de0315c462f9e7c903296336526 : SUCCESS in 1h 01m 45s","accounts_in_message":[],"_revision_number":2},{"id":"9c393de36d8711765e4af7d55e4e713cb24cfb14","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-22 07:19:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1b329309cad4f864fc946fa14e7d737f8eb3733e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-22 07:20:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"24394acae070b37ada634d1ac71b751d920b4dfc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-22 08:36:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/661c02800bed41cab908185422c3e9eb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8f04e1e32f14282bd3620bfdd2430eb : SUCCESS in 8m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/dabdb3a3acf744e7b0180cf1ee29b6bb : SUCCESS in 56m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fd47ebf3f3cb4a999fd5d7b7908d2dc6 : SUCCESS in 46m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/85ef623c68024ec4b785d1ba9bc5ae7f : SUCCESS in 1h 08m 40s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/dbc5aeee411e4012837972ffa1dc4c27 : POST_FAILURE in 6m 40s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/7f7738026c0a484ab10376765c6bd1d9 : POST_FAILURE in 5m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/92ac27ff6d3045b2a7e243a79a8c5383 : SUCCESS in 13m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/62aa378a30b94a2a9071e935536b8cf8 : SUCCESS in 12m 35s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5d1b708dec184640b394a1b365015dde : SUCCESS in 11m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6ddf8b78aa294d3badc12f293ca02d9a : SUCCESS in 11m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/294ad51649b349f5a20f411ef90b2ce5 : SUCCESS in 14m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/24d56f4ac4ca470e9edabf820c98c91e : SUCCESS in 16m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9d86ab79482847998550c2a2795e7afd : SUCCESS in 17m 02s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0bd13344257e4648b2d67a601d1910d4 : SUCCESS in 14m 49s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/cfeec8e1002f479fa46561dff4b6c76b : SUCCESS in 19m 24s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/47595ac0a25a476bb308a947db9a7bfb : SUCCESS in 37m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/343b4e37740f4e239ff0789ad5e39c6e : SUCCESS in 41m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0dd7a727523948d5a67daef89f29ce09 : SUCCESS in 27m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15e20b2b965141ccb33f9892e99dddf7 : SUCCESS in 12m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30099caeaaa8446a955e64be042ba35a : SUCCESS in 5m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0710f5d0ba946038584b9f7812ee14d : SUCCESS in 40m 34s","accounts_in_message":[],"_revision_number":2},{"id":"5545987d8ccb372fc505a4674f3ec89312e40ae9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-22 08:36:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ba27dc66ce7b7d9f4141d6e49d8e9b8be38c456b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-22 08:37:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fe0d5a6be36450dbdf314dac41d344e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb6929c03c30457ca7854709da805a41 : SUCCESS in 43s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/c81b64095ceb4eec8e31b3d8edd1ecc0 : FAILURE in 30s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/7f7adebbf688423f8ded12fa4eb6e2f5 : FAILURE in 34s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c834e7a53d5a33a3fd13ffd954e6f4f4ee953dfc","revisions":{"130969be9e5a06da41f1b7f204bda380fbb1504b":{"kind":"REWORK","_number":1,"created":"2023-01-20 19:52:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/42/871242/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/871242/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/871242/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/871242/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/871242/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/871242/1"}}},"commit":{"parents":[{"commit":"4ee566aa2992bfded7d881c9ae71f3267ada8ce8","subject":"Fix stable gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ee566aa2992bfded7d881c9ae71f3267ada8ce8"}]}],"author":{"name":"Aymeric Ducroquetz","email":"aymeric.ducroquetz@ovhcloud.com","date":"2022-10-25 20:07:53.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2023-01-20 19:52:19.000000000","tz":0},"subject":"s3api: Prevent XXE injections","message":"s3api: Prevent XXE injections\n\nPreviously, clients could use XML external entities (XXEs) to read\narbitrary files from proxy-servers and inject the content into the\nrequest. Since many S3 APIs reflect request content back to the user,\nthis could be used to extract any secrets that the swift user could\nread, such as tempauth credentials, keymaster secrets, etc.\n\nNow, disable entity resolution -- any unknown entities will be replaced\nwith an empty string. Without resolving the entities, the request is\nstill processed.\n\n[CVE-2022-47950]\n\nCloses-Bug: #1998625\nCo-Authored-By: Romain de Joux \u003cromain.de-joux@ovhcloud.com\u003e\nChange-Id: I84494123cfc85e234098c554ecd3e77981f8a096\n(cherry picked from commit b8467e190f6fc67fd8fb6a8c5e32b2aa6a10fd8e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/130969be9e5a06da41f1b7f204bda380fbb1504b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/130969be9e5a06da41f1b7f204bda380fbb1504b"}]},"branch":"refs/heads/stable/victoria"},"c834e7a53d5a33a3fd13ffd954e6f4f4ee953dfc":{"kind":"REWORK","_number":2,"created":"2023-01-20 20:24:07.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/42/871242/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/871242/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/871242/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/871242/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/871242/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/871242/2"}}},"commit":{"parents":[{"commit":"4ee566aa2992bfded7d881c9ae71f3267ada8ce8","subject":"Fix stable gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ee566aa2992bfded7d881c9ae71f3267ada8ce8"}]}],"author":{"name":"Aymeric Ducroquetz","email":"aymeric.ducroquetz@ovhcloud.com","date":"2022-10-25 20:07:53.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-20 20:23:56.000000000","tz":-480},"subject":"s3api: Prevent XXE injections","message":"s3api: Prevent XXE injections\n\nPreviously, clients could use XML external entities (XXEs) to read\narbitrary files from proxy-servers and inject the content into the\nrequest. Since many S3 APIs reflect request content back to the user,\nthis could be used to extract any secrets that the swift user could\nread, such as tempauth credentials, keymaster secrets, etc.\n\nNow, disable entity resolution -- any unknown entities will be replaced\nwith an empty string. Without resolving the entities, the request is\nstill processed.\n\n[CVE-2022-47950]\n\nCloses-Bug: #1998625\nCo-Authored-By: Romain de Joux \u003cromain.de-joux@ovhcloud.com\u003e\nChange-Id: I84494123cfc85e234098c554ecd3e77981f8a096\n(cherry picked from commit b8467e190f6fc67fd8fb6a8c5e32b2aa6a10fd8e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c834e7a53d5a33a3fd13ffd954e6f4f4ee953dfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c834e7a53d5a33a3fd13ffd954e6f4f4ee953dfc"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
