)]}'
{"id":"openstack%2Fswift~871244","triplet_id":"openstack%2Fswift~stable%2Ftrain~I84494123cfc85e234098c554ecd3e77981f8a096","project":"openstack/swift","branch":"stable/train","topic":"bug/1998625-stable/train","attention_set":{},"removed_from_attention_set":{"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2023-01-25 21:03:31.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-01-25 20:22:43.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I84494123cfc85e234098c554ecd3e77981f8a096","subject":"s3api: Prevent XXE injections","status":"MERGED","created":"2023-01-21 18:09:33.000000000","updated":"2023-01-25 21:04:31.000000000","submitted":"2023-01-25 21:03:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871244-bug/1998625-stable/train","cherry_pick_of_change":870823,"cherry_pick_of_patch_set":1,"meta_rev_id":"10701d46c4300a8e80b0e1014666494af4ed63fa","_number":871244,"virtual_id_number":871244,"owner":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-25 21:03:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":1,"date":"2023-01-24 08:50:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-25 20:22:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-25 20:22:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":32333,"name":"Aymeric Ducroquetz","email":"aymeric.ducroquetz@gmail.com","username":"AymericDu"}],"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-21 18:09:33.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":32333,"name":"Aymeric Ducroquetz","email":"aymeric.ducroquetz@gmail.com","username":"AymericDu"},"state":"CC"},{"updated":"2023-01-21 18:09:50.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2023-01-21 19:19:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5f2370fd53fc2672e6b2525cc4330c939dfea4ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2023-01-21 18:09:33.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"76372be917294556ee9de38309aa776ad92f6d7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-21 19:19:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fadeaf8c6c4146d3b0ac3c4ff25a22fc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1249e286d3b84f88a662c334353126a9 : SUCCESS in 5m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/4d0fcee7e567485d8122869b55be068f : FAILURE in 1h 04m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/63ded279f8d540f9b69a4bfe9ccd1f4c : FAILURE in 28m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/398d78173cb3479f951be24040f11654 : SUCCESS in 37m 30s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ec37420c8b914fc9b288c2d54f348877 : FAILURE in 3m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6fd096111d9543e3a417874de5adf966 : FAILURE in 11m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1ecd269e70314aa298e4d9eb7a7b803d : FAILURE in 12m 00s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/00a71833a6bd4f9faecdc6d515e8d93f : FAILURE in 13m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c52c7b9c4405459dae303a0e028b16df : SUCCESS in 10m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/130dc7e084614d74a14d0b12f7524ec5 : SUCCESS in 10m 30s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/0f40571fd1874451b4bdd953cfb43e77 : FAILURE in 10m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/69305f46f5b546d68fa1d98eda536c95 : SUCCESS in 8m 53s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/248f61be02ac4c01ba91a34c972111eb : FAILURE in 12m 23s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ff9d944e934a4cb6a2981dffc25d2b98 : SUCCESS in 11m 39s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/cf726f3118dd45b58e03f33d886d4bbb : SUCCESS in 9m 35s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/4febe0b5bc3c49b0b0c0b94fc4695ca9 : SUCCESS in 9m 53s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f4c939663cf84b1dac13562476a99dd6 : SUCCESS in 9m 45s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/b3715fef0f394d86b7d768b8c46bffb1 : FAILURE in 11m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7b41cacae8374ae39796e21af63306c3 : FAILURE in 5m 47s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ba97651da63e429bbd6bae7cb56e7fb6 : SUCCESS in 46m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e9a66a8de1604b76bbd81aab78de0214 : SUCCESS in 22m 13s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/b176a94dcaed470d9da5d0d3a0ea2fe1 : SUCCESS in 22m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/acdbf0bc2fe24726b76b480c292d2b79 : SUCCESS in 21m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce764106b34f46c39be56acac4503960 : FAILURE in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e9681e0376b410496b64211430f0852 : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5ae2bb0ae240499890de6f8332ae0e19 : RETRY_LIMIT in 5m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cfdd362819524da58d25a626a2bda1b9 : FAILURE in 15m 38s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/bba20f69f71a441384f9ba10addad4c2 : FAILURE in 1h 04m 24s","accounts_in_message":[],"_revision_number":1},{"id":"c4fddfa0827de103a14a6f27f8ee444d55cc7d2a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-22 07:28:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dd1d673ad907dc260da694aa836b2ed6d66d8897","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-23 19:49:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"802d29a79af1e6cf69b8e4b332ccc23de44fd4b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-23 20:14:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"df37072f4272ae4ab74a82990009c2b257703e89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 21:01:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78de4b64069a44879dddda0d07731bd8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae2e4c9c9c284e07b4f7ba48fe15254d : SUCCESS in 5m 16s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3ba4f54b432c4d359fa42acc3c318d58 : FAILURE in 3m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e40211c83f3d46018d233c0fff53e212 : SUCCESS in 13m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9c74cd00f92a4e279f83b306ad4504f0 : SUCCESS in 12m 02s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/32e19b7d44324b1b85177fa5a86af256 : SUCCESS in 11m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0de74f8ca5ee4b5482ab14b8e460c31c : SUCCESS in 11m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9f88e85af6a54a1fac2705d6ccf1e661 : SUCCESS in 9m 04s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/5fd8287a007b4ccc8212921ae9de4b4b : SUCCESS in 10m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/38d549fbb14048e38416cb3dd4eb60d4 : SUCCESS in 10m 38s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/7c29a4e114f446bda1da1838709d8638 : SUCCESS in 12m 33s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/56cccf0f176e4d699e3f362cadfc67fd : SUCCESS in 9m 28s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e2d7d4b2320f41fb99140002682c3f54 : SUCCESS in 9m 18s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/c7680a766c6c4ae1b3e294faa5e9d114 : SUCCESS in 10m 16s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0e843c54446d445290f14b6dbc2ba1fb : SUCCESS in 9m 46s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/2175adf56076494da7bb1adad0c09060 : SUCCESS in 12m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/51f399f5b00b4b40bcef51f2c7d1d19d : FAILURE in 7m 24s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/53a901047d83428f8be8a17a9c07d1fa : SUCCESS in 39m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/baf91f85b3c3494cad38e8b0248a683e : SUCCESS in 32m 46s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/d4bc7c8a5eb54a2a8104d10be128c9bf : SUCCESS in 30m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/77c2c326eba0462dba9e82cf5211927b : SUCCESS in 30m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/366c442c7e9f41c0a377ac2737a599a7 : SUCCESS in 11m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b10e779a7b0e4bfd9e0f6fcc4c498457 : SUCCESS in 5m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9eaf07e5f198447a89db2eb1f440eabf : RETRY_LIMIT in 6m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f53d205dbd05b1c066fa489c036ace749a3873fa","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2023-01-24 08:50:21.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"70ac9630aecd1437fdbd6d673d2cd08a26d62b28","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-25 20:22:43.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"bbdad47bb5fc866bc9422f13e4f8781c7fed7120","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 20:23:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2a3ded38eb920655c303a9ce82d2d3106f139d5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 21:03:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07927b8bc3784224a2a0420fb51369ef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcba9651728d4617b9a6fe577bfc53b8 : SUCCESS in 5m 51s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/7778c1ddda0349ba93a96b71fea0658b : POST_FAILURE in 5m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/af8d5f733ef94b2a9b1c7c6cbbff3df4 : SUCCESS in 12m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/920f3b106ba043ad840086a8609f46f9 : SUCCESS in 13m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3b2461f7300f4cebbde58f167e9846cb : SUCCESS in 13m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/37bf89dc276f41aeae47adf9ab340d05 : SUCCESS in 9m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5da7124cdd9f4ec88708b2cb801577e3 : SUCCESS in 8m 55s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/357dce54dc1d4cefa4e6c5d6d5939109 : SUCCESS in 8m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6de74d61b4d7493180b826a4c1fb17cc : SUCCESS in 9m 33s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/508065d9ce054cb79dfe0ef13203ba64 : SUCCESS in 13m 05s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3659289ef4f8437c84bcfaee3fb3362d : SUCCESS in 9m 30s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/193afa33e066450c81099b024c60a52f : SUCCESS in 9m 58s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4cf8e7b6e4d64b23bdb95974ca634d75 : SUCCESS in 9m 59s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/6e27cb1501424178944f2ba225cd89de : SUCCESS in 12m 14s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d36e9087192941bcb9bb431af5d8faad : SUCCESS in 36m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3d8eb58783dc4f65b82c65a51ab13317 : SUCCESS in 30m 26s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/2d8b9fb98c69442786fed74c77e12cbe : SUCCESS in 31m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e329029a303b43d19c2ea4e29816ba35 : SUCCESS in 34m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f7ea60da6ba42d4b41d9ed9144888ed : SUCCESS in 11m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e07d7b84b79c4a8ca033fe99b0cc1831 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":3},{"id":"0b55b07445317a69713f216e5b7a02235ea0aa24","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 21:03:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"10701d46c4300a8e80b0e1014666494af4ed63fa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 21:04:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a9f71fad5094004a195572e92f2064c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/768aee0ff75a439990d7774eba6165e6 : SUCCESS in 42s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/352bada5328a491ea10dcc2e4fadd275 : FAILURE in 32s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"70b68dbf65c7bc67795a8dee65f01c14f9eb94e5","revisions":{"30dac86cbf114e74fec9825439685eebab4c8de0":{"kind":"REWORK","_number":1,"created":"2023-01-21 18:09:33.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/44/871244/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/44/871244/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/871244/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/871244/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/871244/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/44/871244/1"}}},"commit":{"parents":[{"commit":"a86270bed72cdbdf2f42eec562d9fae2d273fa64","subject":"Inline parse_request from cpython","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a86270bed72cdbdf2f42eec562d9fae2d273fa64"}]}],"author":{"name":"Aymeric Ducroquetz","email":"aymeric.ducroquetz@ovhcloud.com","date":"2022-10-25 20:07:53.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2023-01-21 18:09:33.000000000","tz":0},"subject":"s3api: Prevent XXE injections","message":"s3api: Prevent XXE injections\n\nPreviously, clients could use XML external entities (XXEs) to read\narbitrary files from proxy-servers and inject the content into the\nrequest. Since many S3 APIs reflect request content back to the user,\nthis could be used to extract any secrets that the swift user could\nread, such as tempauth credentials, keymaster secrets, etc.\n\nNow, disable entity resolution -- any unknown entities will be replaced\nwith an empty string. Without resolving the entities, the request is\nstill processed.\n\n[CVE-2022-47950]\n\nCloses-Bug: #1998625\nCo-Authored-By: Romain de Joux \u003cromain.de-joux@ovhcloud.com\u003e\nChange-Id: I84494123cfc85e234098c554ecd3e77981f8a096\n(cherry picked from commit b8467e190f6fc67fd8fb6a8c5e32b2aa6a10fd8e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/30dac86cbf114e74fec9825439685eebab4c8de0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/30dac86cbf114e74fec9825439685eebab4c8de0"}]},"branch":"refs/heads/stable/train"},"01c23d1c98f6cbcb2738c334dfcc65cd522ceb86":{"kind":"REWORK","_number":2,"created":"2023-01-23 19:49:23.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/44/871244/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/44/871244/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/871244/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/871244/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/871244/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/44/871244/2"}}},"commit":{"parents":[{"commit":"91144596c670741c2f452c864c026eff5583989c","subject":"Fix stable/train gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/91144596c670741c2f452c864c026eff5583989c"}]}],"author":{"name":"Aymeric Ducroquetz","email":"aymeric.ducroquetz@ovhcloud.com","date":"2022-10-25 20:07:53.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-23 19:48:08.000000000","tz":-480},"subject":"s3api: Prevent XXE injections","message":"s3api: Prevent XXE injections\n\nPreviously, clients could use XML external entities (XXEs) to read\narbitrary files from proxy-servers and inject the content into the\nrequest. Since many S3 APIs reflect request content back to the user,\nthis could be used to extract any secrets that the swift user could\nread, such as tempauth credentials, keymaster secrets, etc.\n\nNow, disable entity resolution -- any unknown entities will be replaced\nwith an empty string. Without resolving the entities, the request is\nstill processed.\n\n[CVE-2022-47950]\n\nCloses-Bug: #1998625\nCo-Authored-By: Romain de Joux \u003cromain.de-joux@ovhcloud.com\u003e\nChange-Id: I84494123cfc85e234098c554ecd3e77981f8a096\n(cherry picked from commit b8467e190f6fc67fd8fb6a8c5e32b2aa6a10fd8e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/01c23d1c98f6cbcb2738c334dfcc65cd522ceb86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/01c23d1c98f6cbcb2738c334dfcc65cd522ceb86"}]},"branch":"refs/heads/stable/train"},"70b68dbf65c7bc67795a8dee65f01c14f9eb94e5":{"kind":"REWORK","_number":3,"created":"2023-01-23 20:14:55.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/44/871244/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/44/871244/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/871244/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/871244/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/871244/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/44/871244/3"}}},"commit":{"parents":[{"commit":"91144596c670741c2f452c864c026eff5583989c","subject":"Fix stable/train gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/91144596c670741c2f452c864c026eff5583989c"}]}],"author":{"name":"Aymeric Ducroquetz","email":"aymeric.ducroquetz@ovhcloud.com","date":"2022-10-25 20:07:53.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-23 20:14:47.000000000","tz":-480},"subject":"s3api: Prevent XXE injections","message":"s3api: Prevent XXE injections\n\nPreviously, clients could use XML external entities (XXEs) to read\narbitrary files from proxy-servers and inject the content into the\nrequest. Since many S3 APIs reflect request content back to the user,\nthis could be used to extract any secrets that the swift user could\nread, such as tempauth credentials, keymaster secrets, etc.\n\nNow, disable entity resolution -- any unknown entities will be replaced\nwith an empty string. Without resolving the entities, the request is\nstill processed.\n\n[CVE-2022-47950]\n\nCloses-Bug: #1998625\nCo-Authored-By: Romain de Joux \u003cromain.de-joux@ovhcloud.com\u003e\nChange-Id: I84494123cfc85e234098c554ecd3e77981f8a096\n(cherry picked from commit b8467e190f6fc67fd8fb6a8c5e32b2aa6a10fd8e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/70b68dbf65c7bc67795a8dee65f01c14f9eb94e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/70b68dbf65c7bc67795a8dee65f01c14f9eb94e5"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
