)]}'
{"id":"openstack%2Fswift~871288","triplet_id":"openstack%2Fswift~master~I5c1f1be1fa5d67c6bce8be6043ddbc7cbf59978b","project":"openstack/swift","branch":"master","topic":"proxy-log","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-02-03 21:56:16.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5c1f1be1fa5d67c6bce8be6043ddbc7cbf59978b","subject":"wip: require req to iter_nodes","status":"ABANDONED","created":"2023-01-20 15:26:12.000000000","updated":"2023-02-03 21:56:16.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"28240e84ab8eb5e29660a249ce3bf7745216d517","_number":871288,"virtual_id_number":871288,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-01-20 16:48:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-20 16:48:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f9063120046a5c6889624a4549a2ef28b1ba3739","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-01-20 15:26:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc5e5d54c1b7ccd66fda21d7aa3d0f7ccc4d4ac3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 16:48:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b7cec863461e4af8bb43a8710e716a4f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09a7d65295f14a019f3d02e6c02a5872 : SUCCESS in 7m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/c1749dcd5c6f4b439cd817bbdd0e3dab : SUCCESS in 58m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5c58897111b747adbe7098a9ab78b376 : SUCCESS in 54m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/34ab7b47eb924a11a820b48f8d4126d1 : SUCCESS in 57m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/852cf8a1a7fe4e95b36986f1eacf3589 : SUCCESS in 17m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/72b4542470bd4254a9922cb6f15924d0 : SUCCESS in 17m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cc54a0a94ee9454b81bd36bac6c20af4 : SUCCESS in 19m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/44caa2e0d4be461cb81767dad46da061 : SUCCESS in 8m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/944ec34bc5f046119e7b52f51052af68 : SUCCESS in 6m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2763bcb845a74c38abb92806993f105c : FAILURE in 15m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a6f460569843496cb07a26381cd0d04a : FAILURE in 16m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/11e8f573e8684df3833a2432ed3feb88 : FAILURE in 14m 40s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/70688811eb774fa081ac214d9954fa9d : FAILURE in 17m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/77c00b4365c44ea08859ca82cbc9d79e : SUCCESS in 18m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/deffa3030a884600a5c052ad861ea271 : SUCCESS in 16m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/033697078ba842c1bdafb957e43f86a1 : SUCCESS in 19m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a2e97f0f1baf4a0c9b35fb7eafdc16a7 : SUCCESS in 15m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/344116a7105d4f138cb25a3f2eecbaf2 : SUCCESS in 16m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/507465da0fe2462094f074e8716788ff : SUCCESS in 20m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c60657c91439493c9feb0d08681e1501 : SUCCESS in 6m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/44b001d92d5547c9bd9c9fdf6904d328 : SUCCESS in 17m 17s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5cbbc06645a64dd3aef756768db33c3a : SUCCESS in 7m 46s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fe52fd357ca24bd99f1087d34c313d1c : SUCCESS in 1h 10m 39s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/bc2942481eaf499f8f651548b1f4723c : SUCCESS in 56m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0422bb9bd17a45059f50cf7be7f671d0 : SUCCESS in 1h 15m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b198df97c8774291b6d5dac0061a9c38 : SUCCESS in 49m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa9c46d92b474f0899ab9223b7597c84 : FAILURE in 18m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7164f5cfd09c44d98514a2dcba743cb2 : SUCCESS in 4m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fd5397e59065483d865d458a782ada0e : FAILURE in 28m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5215b1d688ee4568b7569a2cef1fb280 : SUCCESS in 1h 07m 51s","accounts_in_message":[],"_revision_number":1},{"id":"873a7161596ebb4a43b2f4c12e10628a67dc9372","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 17:00:34.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b15186f28fcf49538c7cb809211284fb\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b6c18942bd50407c8dc98d00b36532ca : TIMED_OUT in 1h 31m 10s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fd0958d73576469ca6ed08fa5defb71c : FAILURE in 1h 27m 34s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/077253bd692b409595938f8a4f99b137 : SUCCESS in 1h 21m 19s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/30bbe2d17d994028b937b6f464069917 : TIMED_OUT in 1h 31m 04s","accounts_in_message":[],"_revision_number":1},{"id":"28240e84ab8eb5e29660a249ce3bf7745216d517","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-02-03 21:56:16.000000000","message":"Abandoned\n\nthis got squashed into https://review.opendev.org/c/openstack/swift/+/860866","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3e6ae8b22216ec25e8cf7972d1f813fc04942428","revisions":{"3e6ae8b22216ec25e8cf7972d1f813fc04942428":{"kind":"REWORK","_number":1,"created":"2023-01-20 15:26:12.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/88/871288/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/871288/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/871288/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/871288/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/871288/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/871288/1"}}},"commit":{"parents":[{"commit":"f8e17d7859ba82804c588dcfc7b5a1dffdd3bd65","subject":"proxy-server exception logging shows replication_ip/port","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8e17d7859ba82804c588dcfc7b5a1dffdd3bd65"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-01-20 15:11:03.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-01-20 15:25:42.000000000","tz":-360},"subject":"wip: require req to iter_nodes","message":"wip: require req to iter_nodes\n\nChange-Id: I5c1f1be1fa5d67c6bce8be6043ddbc7cbf59978b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3e6ae8b22216ec25e8cf7972d1f813fc04942428"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3e6ae8b22216ec25e8cf7972d1f813fc04942428"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
