)]}'
{"id":"openstack%2Fswift~871571","triplet_id":"openstack%2Fswift~master~Iab308e8f3b7a243dd0b3132539b6388bc5c89ace","project":"openstack/swift","branch":"master","topic":"update-kota-mailmap","attention_set":{},"removed_from_attention_set":{"4608":{"account":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"last_update":"2023-01-24 17:51:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iab308e8f3b7a243dd0b3132539b6388bc5c89ace","subject":"Add Kota\u0027s private email address to mailmap","status":"MERGED","created":"2023-01-24 07:27:08.000000000","updated":"2023-01-24 17:53:00.000000000","submitted":"2023-01-24 17:51:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871571-update-kota-mailmap","meta_rev_id":"10921b4be67fc3092ce9bdbf0912c4f280635c39","_number":871571,"virtual_id_number":871571,"owner":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-24 17:51:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2023-01-24 12:57:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-24 17:10:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-24 17:10:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-24 07:51:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-24 12:57:38.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2023-01-24 17:10:27.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"5faa9dd2820c184be95d664b769920a5e4ed5eb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2023-01-24 07:27:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"80d8d12491a858a4b82b35bffc9b0c54af549337","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 07:51:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b54f59ac539e44ccb863e62dd3ac3223\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ead4d2dbe92a44ada746a3abe6d599a8 : SUCCESS in 8m 41s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a60e64e0430e4410a5afd3a4968b0c5a : SUCCESS in 9m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e7c887de705c4834b85602cc7bc8a620 : SUCCESS in 4m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bf214eb13ba74238826db06f94734bc8 : SUCCESS in 18m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f1f454c24315448f90dda964ec7a1d62 : SUCCESS in 14m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c8274b784c514389b89ee91edf2f6461 : SUCCESS in 16m 22s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5da23c60c8d24cd1ae6719181816cbc1 : SUCCESS in 17m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2cd281de352e45f7b4b501254291d678 : SUCCESS in 12m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/890ae4876e99485e8ce72a828636ce0e : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"e2a98ba22842dd830c2ee1cd9a603ac474acd6f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 08:46:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9775b0d786be4416a64561598ee1b3de\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9bc314134688429f8da71d1d4fb27551 : SUCCESS in 1h 11m 31s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4640d123f8eb4c8a82f5f6514715fb16 : SUCCESS in 1h 06m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/74e16c0b417b46079efc5141111ff80f : SUCCESS in 1h 16m 24s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e14dc7f8b6ef4334b89cfccc1c51eca7 : SUCCESS in 1h 16m 38s","accounts_in_message":[],"_revision_number":1},{"id":"93837dcea4398ab045eff33917d265f6379c1473","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-01-24 12:57:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b2127d2cfc33f9005f430809c7944fef881a5e43","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-24 17:10:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a89ed0234a2f3070a97d3569434f25ff898af756","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 17:10:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2a5aa37f0562b78c0997b940c3b1dde68c1e030b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 17:51:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8e4239f222146e4bbafcabffdda6ea5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b62fc015b2d47678fd85fd134778728 : SUCCESS in 9m 58s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/64c4124c86bd4975bfd40669c2715ba2 : SUCCESS in 9m 27s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/95c629be5c3a42a49f2a2c7117c84fc4 : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/456a415792174222b7647f39e436fc3c : SUCCESS in 17m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8dce531ae841491788d6f94baa0f79f6 : SUCCESS in 14m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3ac29ffae19a48b58b27fc322cff7ee6 : SUCCESS in 16m 50s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/01bb5104cbd94947bd26d135c31d8301 : SUCCESS in 25m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b3df988f206442c89f47badf1cad1cc4 : SUCCESS in 15m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9e06f7209b014b0dbd02334d29607f51 : SUCCESS in 15m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fd01bfbd1f9f4f579185020b5d354aee : SUCCESS in 19m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d656c719db2e4f578739a1a79187217c : SUCCESS in 15m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4eaee134d4bf4c7596fa2d2930422f53 : SUCCESS in 15m 54s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7894db79ec8449f8a3c420e74a80264b : SUCCESS in 14m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d7106e55eb6843fe89a58756b42af34a : SUCCESS in 8m 17s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/495d1e69afec42c3ace8faa788e65068 : SUCCESS in 6m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce8340d2980f4a0dabd1573923c58c0d : SUCCESS in 16m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd2e4271318048ab82394889c1145b58 : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":1},{"id":"01061f51fdd884de568ed8ff561519dbf394886c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 17:51:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"10921b4be67fc3092ce9bdbf0912c4f280635c39","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 17:53:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c73b32b766948e1ad2c30f3efa8b318\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17f5a2a4f30c42688e8fdc42bebf9ab0 : SUCCESS in 55s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/0803061128a64c0c8e10f13d77ab3606 : SUCCESS in 33s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/50442f2579c04a4181da119c1877c490 : SUCCESS in 32s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5de745c2bc3aa82049424b9c3e8744cda53a4c74","revisions":{"5de745c2bc3aa82049424b9c3e8744cda53a4c74":{"kind":"REWORK","_number":1,"created":"2023-01-24 07:27:08.000000000","uploader":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"ref":"refs/changes/71/871571/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/871571/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/871571/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/871571/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/871571/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/871571/1"}}},"commit":{"parents":[{"commit":"2d1b68b12eb4bb4e21610a6bcc28f91cb6831c8b","subject":"Merge \"Proxy: move \u0027_get_update_shard\u0027 from base class to child class.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2d1b68b12eb4bb4e21610a6bcc28f91cb6831c8b"}]}],"author":{"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","date":"2023-01-24 07:18:49.000000000","tz":540},"committer":{"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","date":"2023-01-24 07:27:00.000000000","tz":540},"subject":"Add Kota\u0027s private email address to mailmap","message":"Add Kota\u0027s private email address to mailmap\n\nChange-Id: Iab308e8f3b7a243dd0b3132539b6388bc5c89ace\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5de745c2bc3aa82049424b9c3e8744cda53a4c74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5de745c2bc3aa82049424b9c3e8744cda53a4c74"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
