)]}'
{"id":"openstack%2Fswift~871658","triplet_id":"openstack%2Fswift~stable%2Fussuri~Ie22aabb0f6bde0d7f5df71de527d3201ba605181","project":"openstack/swift","branch":"stable/ussuri","hashtags":[],"change_id":"Ie22aabb0f6bde0d7f5df71de527d3201ba605181","subject":"Drop tempest and grenade testing","status":"MERGED","created":"2023-01-24 20:10:36.000000000","updated":"2023-01-24 23:01:48.000000000","submitted":"2023-01-24 22:59:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871658","meta_rev_id":"07b9d970dbda63d533122e62987ce000f31a0953","_number":871658,"virtual_id_number":871658,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-24 22:59:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-24 21:20:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-24 21:20:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-24 20:59:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8013cc845bbe09c5da74eb5ae2445f630549b52c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-24 20:10:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"311bef3a06c2a588105865bf784b24f9d53e4f6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 20:59:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa3a528975ff4120b875005a350c2906\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b62e302ead3242f8b873f40eb1e3c6e2 : SUCCESS in 6m 09s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a64062336a894b5f85682249a9ff9d32 : FAILURE in 3m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2d940bd4421a4ba59311fbf8acd5c8d1 : FAILURE in 3m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/42176f2f9b114f51937a1a72876de90f : SUCCESS in 11m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5ef71dc7796d43e2a23288708243807f : SUCCESS in 9m 06s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cce2faa7c0c84f658ece1ffecd366638 : SUCCESS in 9m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5ebc4c67b6af4328a990888816364d4c : SUCCESS in 12m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b9817f05293e4ebbb975b0a3fd36fca7 : SUCCESS in 7m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/66ac6ebf0cbc4b7797e4972956c26f32 : SUCCESS in 10m 16s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/bf16d3db78c549ef93fc04d1010eef6e : SUCCESS in 7m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/511650ad925448499f6b531a72102429 : SUCCESS in 7m 59s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/5ce696f3d8054da6ae5784b0a8c8535c : SUCCESS in 13m 27s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0a7c821991f54dc1a5d848703de51d8c : SUCCESS in 8m 53s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ed53902168e240b78c0ce5276fd8a7a0 : SUCCESS in 15m 41s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/295b142b97d14c96ba3b5711ce719e4e : SUCCESS in 8m 34s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9b87a146630b427a853b087bd0bd04c6 : SUCCESS in 10m 01s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/ddd415d023384ca2b855dc211c4fc16f : SUCCESS in 16m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/845500814d744891a2b9f0a34a6cfa07 : FAILURE in 7m 29s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/872f790fd2d8435fab7321834b4f5d2b : SUCCESS in 44m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cde95fbfb8624808920c331d05f15dfb : SUCCESS in 27m 09s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/5917565a350544289ced8c8b8aa53025 : SUCCESS in 22m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a0d34e0dbbd9446daa03d458d855e9ab : SUCCESS in 27m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a868b73752cb4f30a1fe81d051578e63 : SUCCESS in 14m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c73e2be7c93845d7b7552c851508ebb7 : SUCCESS in 5m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ef73a75a705c446aa763506e431860b1 : RETRY_LIMIT in 5m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b60910c657e9bc0415c4e82cd7a609bfe6f910a5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-24 21:20:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e12b78825da25aa7a7b6e19d61adddba5de64574","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 21:21:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0acba327443c0ad3d64d0465c3ed6592d1b8a98c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 22:59:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33e487da70374278b8a8b81ff7f7f23d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bf887beeaf94631ad10c063844e4346 : SUCCESS in 7m 30s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b476f08e5a544b32a7425218940d4ef0 : POST_FAILURE in 5m 02s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/ab5fd78fc9c746a8a6091aa7d9279f88 : POST_FAILURE in 5m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b29cc5262c21442a9e90db7be566641f : SUCCESS in 14m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/972160ecdf874caa913239977cfd51dd : SUCCESS in 15m 13s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bf9398902d014da8a1232359f5bf7621 : SUCCESS in 14m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9b29a80c6a1941f09173891219ee2676 : SUCCESS in 15m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/376cd6650a7743f4b00578fc762d63f6 : SUCCESS in 9m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/88902b8ae25244d18ca8636e31fad67b : SUCCESS in 11m 48s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/b93a6eb5f7204fe3b98bbf2f0c432379 : SUCCESS in 10m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/642d851c8b844c71ba9c14d9a32ed21b : SUCCESS in 11m 54s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/53406c1bf14549b48dc3a7a894a2b0b5 : SUCCESS in 14m 49s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f24fa4e39cda4a03931a8ec88233a144 : SUCCESS in 12m 07s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/fde47a14f05e441290030572d30c2ecd : SUCCESS in 9m 30s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/44b5586d18164d8fafafe0784aa87085 : SUCCESS in 11m 21s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/0c9eb16b2c9a4dc289b75eece27a124f : SUCCESS in 16m 39s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/959b524ac5654e0fbe27243b2cfe43a8 : SUCCESS in 1h 01m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9d7467d51b6241f9a49d1de9cfd8261e : SUCCESS in 28m 55s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/1aaf72831dfb4cd3b0d3a3311142d056 : SUCCESS in 35m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f0f0849d70fc480a8ed12793586800ca : SUCCESS in 28m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c30b6a1f9d9548fe9c4f8783c7b509a2 : SUCCESS in 13m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91ffac6498d6431ebeaf69085b8e755c : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":1},{"id":"0018ee2ac0ded47f631305793a6d97197699f389","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 22:59:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"07b9d970dbda63d533122e62987ce000f31a0953","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 23:01:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ea0ce49d2f74169a00c1377266610f7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b99915735b5540108b1d69d897c69549 : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/954d350e721f4c7992330a6be6085c55 : FAILURE in 32s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/6c57409e4b354eed8313f03312def469 : FAILURE in 32s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"54e6afc9a98aa772a995befd29bc34908a5ea8cc","revisions":{"54e6afc9a98aa772a995befd29bc34908a5ea8cc":{"kind":"REWORK","_number":1,"created":"2023-01-24 20:10:36.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/58/871658/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/871658/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/871658/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/871658/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/871658/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/871658/1"}}},"commit":{"parents":[{"commit":"45f6d47a7c2bc13ba58f61e09892a2a1a583b9c6","subject":"Add attrs to lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/45f6d47a7c2bc13ba58f61e09892a2a1a583b9c6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-24 20:06:31.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-24 20:06:31.000000000","tz":-480},"subject":"Drop tempest and grenade testing","message":"Drop tempest and grenade testing\n\nRecently, these jobs started failing while installing tempest. The\ntrouble seems to be:\n\n* tempest 26.1.0 depends on paramiko\u003e\u003d2.7.0\n* paramiko 3.0.0 is selected which in turn depends on\n  cryptography \u003e\u003d 3.3 and bcrypt \u003e\u003d 3.2\n* cryptography 39.0.0 and bcrypt 4.0.1 are selected, neither of which\n  get installed from binary wheels and both of which require a Rust\n  toolchain to compile\n\nFor some unknown reason, this used to be OK-ish -- the stable/train\nupper-constraints file would be respected, paramiko\u003d\u003d2.6.0 would be\ninstalled with cryptography\u003d\u003d2.8 and bcrypt\u003d\u003d3.1.7, and pip would\nsimply warn\n\n    ERROR: tempest 26.1.0 has requirement paramiko\u003e\u003d2.7.0, but you\u0027ll\n    have paramiko 2.6.0 which is incompatible.\n\neven though everything would still work. Following the paramiko 3.0.0\nrelease, however, it stopped working and it\u0027s not entirely clear how to\nfix it. Given the age of the branch and its extended-maintenance state,\ndropping the testing seems best.\n\nChange-Id: Ie22aabb0f6bde0d7f5df71de527d3201ba605181\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/54e6afc9a98aa772a995befd29bc34908a5ea8cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/54e6afc9a98aa772a995befd29bc34908a5ea8cc"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
