)]}'
{"id":"openstack%2Fswift~871742","triplet_id":"openstack%2Fswift~master~Ib05f615ec3afcd6a1d9e25239cc009fb5cf62cce","project":"openstack/swift","branch":"master","topic":"sr_cache","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-01-27 18:44:46.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib05f615ec3afcd6a1d9e25239cc009fb5cf62cce","subject":"maybe squash? use a NamespaceBoundList class","status":"ABANDONED","created":"2023-01-25 16:44:57.000000000","updated":"2023-01-27 18:44:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"09d4e6564c82e27c4a1ea8784efc4868624a40e2","_number":871742,"virtual_id_number":871742,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-01-25 18:42:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-25 17:12:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-25 18:42:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ea05929181a238ee74ad2815525472bea92a8d08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-01-25 16:44:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6833a03253f5c9c357175f515a4413bfad215f2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 17:12:52.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cca77edeff4e41b290b3c2cd689b8c16\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a3b198880b63456e8ab0430590310aef : FAILURE in 17m 05s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0f1cbb0f325441ef9acfb9f7d8783f70 : FAILURE in 17m 29s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d329498bba624bb89863428bcf58937a : SUCCESS in 26m 09s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ebfbd6aeb8e941328b529034cebd5848 : SUCCESS in 20m 36s","accounts_in_message":[],"_revision_number":1},{"id":"9e7a6e0c21373a7484175ed23a91b8fa317d13ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 18:42:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b3dd153ad02f4d139f03226e42e921a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f10bcfdc7c1d418783f4d585375d041a : SUCCESS in 9m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/d2cec6e119dc4efa8618d34c2335feb3 : SUCCESS in 1h 00m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9b54dbc6940f4f10b01733dd3f941099 : SUCCESS in 49m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea3b1c8db4d84fbc9954f97c100c2812 : SUCCESS in 46m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bcf5a02132914cb38d758284784facab : SUCCESS in 19m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/59d310d7a47e43de872bcc69ee00f188 : SUCCESS in 18m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/066303c04a154f26943006d6849cdd89 : SUCCESS in 21m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fda580c46e184249a85a30f71ed03244 : SUCCESS in 8m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3227657efb5b4d80afc1544a397593a2 : SUCCESS in 4m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b5b221b7d2ab44acb2ea316f12b5f6ac : FAILURE in 18m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3b213d7c66c242d8842258a0e9400374 : FAILURE in 14m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ca24633f301d49c6a1bf7c27ea2e087a : FAILURE in 15m 09s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3f44565f84914292af92ea0d8b4766de : FAILURE in 18m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/14c42b68d2ad4334a99df11c46ccca0a : SUCCESS in 18m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f366ab84555b4d4fa5791c7276edddda : SUCCESS in 15m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1e305b956a724e18b54e8e4d9d39baa3 : SUCCESS in 18m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/27a343856549494e911fb001469008d7 : SUCCESS in 13m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a8582014f9324f9d9fd8151703b56573 : SUCCESS in 27m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3c38b86b0c3d467ba795684a657532d2 : SUCCESS in 15m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/68fb21ae668d41baa286b68723adbf3e : SUCCESS in 6m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b939c23999f4455aab647f79b659a654 : SUCCESS in 21m 03s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4e9f684937a34ee4a34c5ed0150204fa : SUCCESS in 7m 37s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d43e7d47fa9c430b953fd4a9850e92dc : FAILURE in 1h 16m 40s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c247c62eda7643d38490d75026addf17 : FAILURE in 1h 27m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4af6fb3eeac5469bb39eefb48a94f88a : SUCCESS in 1h 08m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e7ce2ea72ff248ef8a493790ec7715a5 : SUCCESS in 47m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/250e1bec14884b4fbbc94bcd2448d414 : FAILURE in 18m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54703a3898764b86862638dcc2819b0c : SUCCESS in 4m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/854ba2c82f934fa1b5852c5ca33a6186 : FAILURE in 21m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c8cf9c2379f245bba70cee2965c38bcc : SUCCESS in 48m 34s","accounts_in_message":[],"_revision_number":1},{"id":"09d4e6564c82e27c4a1ea8784efc4868624a40e2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-01-27 18:44:46.000000000","message":"Abandoned\n\nsquashed to parent","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"07053cfcea17d31de111185fc71258223efba07f","revisions":{"07053cfcea17d31de111185fc71258223efba07f":{"kind":"REWORK","_number":1,"created":"2023-01-25 16:44:57.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/42/871742/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/871742/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/871742/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/871742/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/871742/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/871742/1"}}},"commit":{"parents":[{"commit":"1a3061498d3c0bcbea572543e153779f9218f2da","subject":"Proxy: restructure cached updating shard ranges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1a3061498d3c0bcbea572543e153779f9218f2da"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-01-25 16:44:19.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-01-25 16:44:19.000000000","tz":0},"subject":"maybe squash? use a NamespaceBoundList class","message":"maybe squash? use a NamespaceBoundList class\n\nChange-Id: Ib05f615ec3afcd6a1d9e25239cc009fb5cf62cce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/07053cfcea17d31de111185fc71258223efba07f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/07053cfcea17d31de111185fc71258223efba07f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
