)]}'
{"id":"openstack%2Fswift~871843","triplet_id":"openstack%2Fswift~master~I337f4e54d1bcd4c5484fe56cfc886b16077982f5","project":"openstack/swift","branch":"master","topic":"p-misplaced-faster","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2023-02-25 07:27:11.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-02-25 07:27:11.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-02-25 07:27:11.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-02-25 06:15:03.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I337f4e54d1bcd4c5484fe56cfc886b16077982f5","subject":"sharder: make misplaced objects lookup faster","status":"MERGED","created":"2023-01-26 19:11:36.000000000","updated":"2023-02-25 07:28:45.000000000","submitted":"2023-02-25 07:27:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":56,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"871843-p-misplaced-faster","meta_rev_id":"9fac19bc9b089a5f36bea02787e930709f3b2963","_number":871843,"virtual_id_number":871843,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-25 07:27:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-25 06:15:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-25 06:15:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}],"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-26 19:32:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-26 20:23:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-27 17:34:42.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2023-02-17 05:13:45.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2023-02-18 00:28:17.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2023-02-21 19:23:11.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"27e0032cf9af7b8a51b416a39af07cc7376496b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-01-26 19:11:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe82cf3d0c3a519a049919d4d36831edf2b31c68","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 19:32:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21cf05ea6a1b441fafcf08177b7c0541\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/38d4dcb56ab04427b53f20ce0956642d : SUCCESS in 19m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cdb5cb9b41014437a0c9ca5579f03f52 : SUCCESS in 16m 02s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b06e558a8b6e457ca742a0ea99f5d2d6 : SUCCESS in 19m 41s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/47e5db10542a49e2ba9994af30a57632 : SUCCESS in 19m 28s","accounts_in_message":[],"_revision_number":1},{"id":"55ef60bcd6f202fa7d8e47b8e1f77300bc82c115","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 20:23:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8550857acb544d9dbde832fe2e59a189\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46db739dc56f49b1bb270defc80f00a2 : SUCCESS in 8m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/197278369896491596a4485ca6375403 : SUCCESS in 1h 00m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cc322457a9754a27b6f612e30116457a : SUCCESS in 55m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30cc3dc3fef84db5b108c13773fcf760 : SUCCESS in 1h 03m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5663a9ff417a4742b8937f4ed70018fd : SUCCESS in 18m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/97a3b4f3484d44de9d3a373112635a59 : SUCCESS in 19m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e01f1305c2214d57959538212c5bc59a : SUCCESS in 19m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6220539db23a48f3a18a23bd734fc6e9 : SUCCESS in 9m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f6c2f9f8ffe240f6b099ac50935879a5 : SUCCESS in 4m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/178b1cd8f5c34259be63e5c555111709 : SUCCESS in 15m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ff14c898892a4dd9a2309b3659566ee4 : SUCCESS in 14m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/966dc4954a1946b4b7e7651284204d45 : SUCCESS in 15m 41s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d94433d0e8d04dfd9ac0475ac8e165b5 : SUCCESS in 14m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5f5821c98d3b4872a5e870e018ac12cb : SUCCESS in 15m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3d43e289bdd5435388631fa086357ba7 : SUCCESS in 16m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d7ad82501139419399db95e670d4d582 : SUCCESS in 17m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7a8a001892364797965c0b39fe0a7f75 : SUCCESS in 15m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/aeb4ccd762554401b91b12a3ef18aeef : SUCCESS in 13m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a1247c80cbb2466ca73e9ffc7219a886 : SUCCESS in 16m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/146bcf23d16b49ad9531ede545a8d310 : SUCCESS in 13m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bb607840adf74ba1b4389227366b8ebe : SUCCESS in 18m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/72d7b0d0c5004f099826f96e90181711 : SUCCESS in 8m 17s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/afb1160f53c142ae8995e0fbdccd80d3 : SUCCESS in 1h 01m 14s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/92d899402813445ba1d61e9d9519f6c6 : SUCCESS in 55m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/75bb779f0dac41adab7ea7c0dd8e187e : SUCCESS in 53m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/90973dd391d14e3394e8edbd961a34ef : SUCCESS in 47m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e1a6fad6fc24bb19e9ea3e11d7d75e5 : SUCCESS in 15m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64ec09d5d4cb4ec6aab0b3a14e8bbfca : SUCCESS in 5m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f6c5a1eea92f415a95c8cff725cb5aea : FAILURE in 17m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd879d19360b402594203ef222f565b5 : SUCCESS in 58m 57s","accounts_in_message":[],"_revision_number":1},{"id":"a71153ddb4d59a5a78470325e69874ba7aae451e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-01-27 13:42:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7e9ba76b150f75f2d8d6f16204105504ff242837","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-01-27 13:42:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"91bd6601c5991ffcd8c4b64d525177a02761d899","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 14:41:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d27d4ceaff544f0e88e7171c6c1a4834\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d092699904ba470d91cfa8e8015adb6b : SUCCESS in 16m 22s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/604b13dcaa1047e6a0503e820cc249ee : SUCCESS in 15m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3d5c6f978e9c4c5dbad3359d52548a9a : SUCCESS in 20m 36s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/13db8d313762447a9afa2fca021defcf : SUCCESS in 56m 59s","accounts_in_message":[],"_revision_number":2},{"id":"7250398aed24ae37df699e05a22c81703ffad6ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 14:50:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e34de49ae5724b429fbb8a27e0b631f8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b833d9f67f704198b7dd51698e9113f3 : SUCCESS in 8m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/e1b1ae5cef87415d92ece7e4355eb2ba : SUCCESS in 51m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/714e242aa1334cfd98d8ba109e1e3d61 : SUCCESS in 44m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd441309153142d9bfee57c9806d621c : SUCCESS in 53m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c4141da529224f3ba23972826a79a94e : SUCCESS in 17m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a8fd9bab7b944faf93d9bb18da4bca99 : SUCCESS in 14m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eef5fc3e11684f53935f93ff9d47ffe5 : SUCCESS in 18m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5fb4d0bfc3b642e7b361b0771559dd4c : SUCCESS in 8m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7f358483b14c4c51bf83b392475d5119 : SUCCESS in 5m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6d6b6d3a385d4a7182eb8764e27f4fcf : SUCCESS in 13m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6ef55d47445d461596bff5350b6b1d34 : SUCCESS in 11m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7c6f07388e2f4829bd096f1a9bb4fc66 : SUCCESS in 11m 52s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1f604b29fbca4951bf9bd6894b64f959 : SUCCESS in 11m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8cd04551e6b04f2eb4290a6c7252888b : SUCCESS in 13m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ad8346471efa43b1846996479b4af1c8 : SUCCESS in 16m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/002d1b76147e4100871e5e2950166549 : SUCCESS in 16m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d54fbf63d48f499eba0e5bea4d74464e : SUCCESS in 12m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/adc2d5eedec4428eb5db60e41de77042 : SUCCESS in 13m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/27e08933e87e4918aba8b773331cc55f : SUCCESS in 28m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cd1ea9138d9c4eea9a8334f1b0a84116 : SUCCESS in 9m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b0ad410d9a094b10a2599ecddd96e1b3 : SUCCESS in 16m 56s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/059cd7c95a644a7a81490ef499b56c62 : SUCCESS in 7m 21s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9606e9b01eac40868f1f245a7167fb95 : SUCCESS in 42m 35s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/974772d0f638475b8f0c41b09cd4e233 : SUCCESS in 52m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c7ed11c961ba4e55b9576d613cb4b096 : SUCCESS in 31m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7707c5677e904d74aa02659b0a45c682 : SUCCESS in 1h 02m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32abd4dc749d4adcbc287e57cfe31631 : SUCCESS in 13m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0841a9b9ede46f3ab4a482910f5aee0 : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0b6c5afbe3d24a999561b2e1e1b821a4 : FAILURE in 21m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5afb3a216c504c24a14939f64f9100cb : SUCCESS in 32m 38s","accounts_in_message":[],"_revision_number":2},{"id":"93c719df7eb8fdcdb8b78858c0f26db2a2027d19","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-01-27 15:10:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d55e4a0be04b4f1e421a74e08467c6893c71ab69","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-01-27 17:34:42.000000000","message":"Patch Set 2:\n\n(9 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a32be7111a084d2b64972b5624cef7a9dd8e5fb6","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-01-27 18:34:16.000000000","message":"Patch Set 2:\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"cc218538c0cb87c2bba48a008cd604e255850a9c","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-02-17 05:13:45.000000000","message":"Patch Set 2: Code-Review+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3532d15bdcb66ced39ce3c0bd47c869d181474ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-17 10:24:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"60f20f26a2656e4cf114a0ca6967426ddc4e548f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 10:49:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8497b89c7e1469dbf49900b8f3cfc0d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/11f2d15fd3d24aa3b0cb138e4bb14c08 : SUCCESS in 14m 32s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f66b1bc84bda411cabc4b87d2f48527d : SUCCESS in 17m 32s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7db674ecf033453cbf9bb96ecf710a72 : SUCCESS in 21m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d52dec5660e04d19ac72843008d5fc5c : SUCCESS in 18m 40s","accounts_in_message":[],"_revision_number":3},{"id":"9ea9a5490b74b48d57d7ba7bc3faabae69f310b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 11:38:50.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fdb15c94083f4e559f94ca2e4f6197ba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5880bb3423b2495fa70ed33931704c30 : SUCCESS in 8m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/bcdb7eadf4e443a394c1646abdc0194f : FAILURE in 51m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/eedb84fe98dc4c3684f70739f3873381 : SUCCESS in 33m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/714b004975d74275b0b92856adc8063c : SUCCESS in 53m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c75bf3f5798e4465a90a1cdc1a573c7c : SUCCESS in 18m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f5846c42542d4035b4335356d4442d14 : SUCCESS in 18m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/466e21f2776a4206922d5d6a53587026 : SUCCESS in 17m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/acf46f81f01344d58979f2b3e60bb0fa : SUCCESS in 10m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f4d2dcc9728f49f287f4c1b409990e02 : SUCCESS in 4m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c0b7c957d3fe4e6a8250aca7ffd0e835 : SUCCESS in 16m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c74a506251184fa2855bc0eb67d16342 : SUCCESS in 15m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9edc3cdc69f745c7a3b7a6f201431024 : SUCCESS in 15m 56s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ff761e825e764b9a9c9df00def2e270b : SUCCESS in 14m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/703f2c81cee141b2bfb23b70c96a0612 : SUCCESS in 15m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8b6055d8f3cf4fb986b88815bf990d00 : SUCCESS in 17m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c13a7652c4824b71b721a9821300c535 : SUCCESS in 16m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5130bf1e5f5140e19e35f14748ec79d2 : SUCCESS in 15m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/32c55df88fd4479a9a6d5ca980d2d9ec : SUCCESS in 14m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a406a50969934714aa207cd58ffcaba7 : SUCCESS in 15m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1ab2f4ad586e405cb297016c3f37c71c : SUCCESS in 6m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/55eb88aa70a14ca79f8fb945ab2006da : SUCCESS in 17m 58s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5c1943f2c3cd4957a7dc758dee3c4fe0 : SUCCESS in 7m 15s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2e2cd3256f414d9fa283687dbcc7cd86 : SUCCESS in 1h 01m 45s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5e0dd5f3501748ac98063ea26e07621f : SUCCESS in 58m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a04d723f9e564651a7c15711770f1db9 : SUCCESS in 51m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7843fb2e94a247889b47821863301394 : SUCCESS in 43m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba372bb61af448a386299dcc5ac18fb6 : SUCCESS in 17m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea3594555beb4fac94e8274715f8113b : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6975590740de4b088ed90e9a0a60dfc7 : SUCCESS in 39m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9815ab14b5ae4113a3a70f95db39c122 : SUCCESS in 48m 49s","accounts_in_message":[],"_revision_number":3},{"id":"9e274bca8dd25304648b225214d008d17497a760","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-02-17 19:58:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a537bcc58a84808266cd5c244f623242dc50b3da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 20:20:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c542fbc0c6e4c069e04dd8b4cc2c111\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/15339db315c14b38bb8251ef8614123a : SUCCESS in 14m 57s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/03ab81091b9441bda1ad5e5659e35064 : SUCCESS in 15m 25s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/78d7ec33529c440b93a0afb49ebfb5f8 : SUCCESS in 18m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/905adfcc405441358191070d7beaffbd : SUCCESS in 20m 55s","accounts_in_message":[],"_revision_number":3},{"id":"ca43ab8cd471a003dff324c28016f0bfa74ab417","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-02-17 20:28:54.000000000","message":"Patch Set 3: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"5db11f5e9bc0392babb3432cf7688f4c5681be3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 21:10:08.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ad974f299134879bcf49f6313c35df6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd85ca060c4e4b60b901c336e7551730 : SUCCESS in 7m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/8104a52f9b3b4c51936c72c9271ce92b : SUCCESS in 1h 05m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7095f1cfb22640f595a21081cb11ac2d : SUCCESS in 44m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8363581e1f2449c6b978720a195400c3 : SUCCESS in 1h 02m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/576d80e8ff184eea9060007899b6ba42 : SUCCESS in 19m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/545c9e16c27c41318435a0381baf1052 : SUCCESS in 13m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0ca0ceb760884e31b8b44a4f9b24cb97 : SUCCESS in 19m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f967edf390b84eac97417da1dbda2d7c : SUCCESS in 8m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0e35f3dbbb6349cc9421df4c04d0b85b : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ed67e917dd294d33bd2838b35eae24b2 : SUCCESS in 14m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d4ad12a53bad4be9bf9a6c49fe5e7889 : SUCCESS in 15m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d68e45e88d344bc0b621e3fa2ac22648 : SUCCESS in 13m 52s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/dd743e13a8104fcd8a12005ed04baf27 : SUCCESS in 15m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a3b9d663de574556bc4a44ec9e316da8 : SUCCESS in 21m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3b81f05fcda54ab8a943451cc53abed2 : SUCCESS in 16m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c4981fc2bc034ac994bcf4ffb2d575e9 : SUCCESS in 17m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6ffbb4ae50f94940a7005ad37dde40e9 : SUCCESS in 14m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f755222c38a44ef5bcbde8abbef80e0d : SUCCESS in 15m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a5d36e606e4c402aaea6531bd7e2d2d6 : SUCCESS in 13m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/aebc723129d04a0189674eb1a6ec224f : SUCCESS in 7m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/179387863da0452c826995a8480c598e : SUCCESS in 15m 54s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/11bd7fcb759d4e459699dcd7cd6b0b00 : SUCCESS in 7m 47s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/855ec18237cf46f6b6c69ad149de65b0 : SUCCESS in 54m 46s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/df8a89448fc6469d9d7749b2c4ec575d : FAILURE in 56m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f7bc42a90e0048939b95f3af4483fe9f : SUCCESS in 53m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e9cafd12cb4d416fa9cb840a92d2e1b5 : SUCCESS in 47m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6fb9fa8235d24b7680bc4387e646ce47 : SUCCESS in 15m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf6e992b8e3d4196a674a48408a9718a : SUCCESS in 4m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e57a106a27d5482d881de27641d80b15 : SUCCESS in 26m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43392fe4756c499ea1250c88a5ed0703 : SUCCESS in 40m 44s","accounts_in_message":[],"_revision_number":3},{"id":"0b9241e7c762c7db021655804d0e41b6d5c555cb","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-02-17 21:38:02.000000000","message":"Patch Set 3: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4f190f99a2cf28e03f453d98c8ca6c4476135f94","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-18 00:28:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"240c9605ec35c86da59ce2b90368b1f447ed4f95","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 00:55:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eec61cc000144ace8e0ed37a9c444390\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/79c06ded71b244528fdd63d6dbc10ea9 : SUCCESS in 18m 30s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cb79a1e717744ae6b039b538c8109699 : SUCCESS in 22m 42s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2938e43e5e2b4d3daff7a7b6b58dca40 : SUCCESS in 26m 15s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f162c9d176514005bd7389fe9d8dfe26 : SUCCESS in 24m 10s","accounts_in_message":[],"_revision_number":3},{"id":"fb0f474300f0aba8cfe96ae8d7d256ccc3d27958","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-18 01:40:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/223262352d0a40bcbb06e69df12c5155\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01db73c623c241c9995cad91358ce9ee : SUCCESS in 8m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/b799a9cdf5fc4912b0e8ede960fac41b : SUCCESS in 1h 08m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/28127adeab6148b383f03a38dbf518c3 : SUCCESS in 33m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d2adab278e24f329ccdb32905bdf0ca : SUCCESS in 59m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7462e4b0ed9f47c4bfae86a1386a4e98 : SUCCESS in 19m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/63e5a1e90136456698765b8f085e1b81 : SUCCESS in 22m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b24236efa1534df490aa15ba7cde7563 : SUCCESS in 50m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3c2a326665af4366bcb5ea971f31ddc2 : SUCCESS in 9m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9085cf394a7645fb8d40be203063f3d4 : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9c174aa4bf514ade8e7cd9e6e012556a : SUCCESS in 17m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/90ac12492b7f42e88ffda2b157298de6 : SUCCESS in 20m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f193e9e8788b41daba6932779b1ed470 : SUCCESS in 14m 53s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/228450383bca476d9a7d20e4a819ed48 : SUCCESS in 24m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/77223e1f490b413cba795b43f0f6d769 : SUCCESS in 16m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b0f5d8dfe2f94c0aa91e0049068d8ef6 : SUCCESS in 18m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/80d34011a0f04c7c8916efd4c227075d : SUCCESS in 26m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c7a06382b8a84e898c0f8722414c1674 : SUCCESS in 15m 38s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0acddd5be4dd475faa679aba9ebd1afe : SUCCESS in 20m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/aaea5481fb804d3a8b73fdf6f30edfd0 : SUCCESS in 17m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/44ce4412a2d44dba87f1b9aca74c706e : SUCCESS in 5m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dc8758dceb304255ac855e3cbf23720d : SUCCESS in 16m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/49d8b969c3b145a7baa1cbc42ede0465 : SUCCESS in 6m 30s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7826deaf5e7d4d45a8d4cb0c347dec3d : SUCCESS in 1h 05m 52s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/97dc1de810cb47fa9dcf539a48a63274 : SUCCESS in 1h 02m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a6fe047a712e43cda00056cb29a63968 : SUCCESS in 56m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f0568586219b4a46b91a26f52f479ae4 : SUCCESS in 49m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37b411b9a8754c4784e021abbb4c92dd : SUCCESS in 18m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/654fbc0a6fa64aedb2506d118f859699 : SUCCESS in 4m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fc7f5d274bfd41c8a55fda01a35bd685 : SUCCESS in 30m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aec61e91b99941718202bde5458d2d95 : SUCCESS in 1h 01m 32s","accounts_in_message":[],"_revision_number":3},{"id":"bfd67aef540c710568c7c3cf67917edf6547017b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-21 19:23:11.000000000","message":"Patch Set 3: Code-Review+1\n\n(7 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4402c5654bcc9061b5fdd08bf096c19585b13b50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-22 13:45:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"67092f34747a88a9613ed804fb6fe1a86ef37eef","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-22 13:45:42.000000000","message":"Patch Set 3:\n\n(17 comments)","accounts_in_message":[],"_revision_number":3},{"id":"2b746b552608c1e7412c7668212313fbe49aa82e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 14:10:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ebf940fd1284b08b8e0c34374ea9d18\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/76d61a8c68c942e8a0bd17902ca4890b : SUCCESS in 20m 46s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d2a8649d09fe4c19a671a47292991e3b : SUCCESS in 19m 45s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/039b3d657f6c48c1a29be85ceee8a001 : SUCCESS in 23m 53s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fda44df6973c4cb28b7afe9d9d5b5f30 : SUCCESS in 23m 27s","accounts_in_message":[],"_revision_number":4},{"id":"63698463b38ad2335e92719c6610b4fa4e7828df","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-22 14:44:29.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"25108165ab64ee58082f6b694a28d3baf2c37f39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 15:19:27.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3fe605c6b8ba4a9f8b5fb5f863fb4825\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d196ed67e5a47a09c148e69cf523609 : FAILURE in 5m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/c13adf45e88b4e88b3cf5a3d075d1f1a : SUCCESS in 1h 00m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/55ee3e01c0e8478d8813f06e7aad6887 : SUCCESS in 43m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1dee872123b345fb9fdef6552d6f03d4 : SUCCESS in 56m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6afae9b84323407fa183c5dc2ce5340b : SUCCESS in 17m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/75e03fc48c024b8fbb9b5c2243cf9186 : SUCCESS in 20m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a04f534ec15e40f98c86a3975337d044 : SUCCESS in 15m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5f4e059b57d74d24b49d1c325fcb1693 : SUCCESS in 8m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/05543bf189b5414db5be3af00d3090bb : SUCCESS in 5m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8ce8c2a9f7ed45e4ae2eac8ec3434164 : SUCCESS in 26m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8e59396e17bf4df7ac63a2f97ef1e741 : SUCCESS in 15m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fb97d77de1c44efeac098b27908b5cf1 : SUCCESS in 14m 55s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7de5c265d6344edcb4bcfbcc61efe7a2 : SUCCESS in 25m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e423003a2c214acd94822352a44d94c6 : SUCCESS in 15m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/450632cc290c46a783a743a6a8dbd6b1 : SUCCESS in 24m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f660c2954c834f83bdcd821306676954 : SUCCESS in 16m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9d5585d3e8c848efb9fbdef543634e29 : SUCCESS in 23m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ad2a96e2bad849ae8fb1025d9394cb17 : SUCCESS in 14m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ff3dd72b94e14853b57615b66bb06412 : SUCCESS in 26m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5429c7f49906452fb857654cd976daa1 : SUCCESS in 6m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ec04a772292748ac8db31fc7ab9b79e7 : SUCCESS in 14m 33s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fdb0aee11a1946218f7575dd64b55b7a : SUCCESS in 7m 17s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/188aa3107b3c452eb030a620e8f0ca3a : SUCCESS in 50m 56s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/2261339024a54307881e4b87ade83a9a : SUCCESS in 1h 32m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2cd7236803f74525b4ff8b3370b54e73 : SUCCESS in 53m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fc8fcf463fa3463d85f3cf5168731d7a : SUCCESS in 53m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2346060e43cf4276b46306831581408e : SUCCESS in 19m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/301a715b2fcd40cf811fedf37c7fbcae : SUCCESS in 5m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/00943ea92d814f88af2d951483e13cad : SUCCESS in 20m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3b86e96e41b4914a3cdc9c23b773b6e : SUCCESS in 51m 06s","accounts_in_message":[],"_revision_number":4},{"id":"5ca8b5f3a592e7b13bd2429eff18a82fc4e7e2cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-22 16:02:39.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b137ea4a1d2603a839231621276bfb29aa2bdc86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 16:28:11.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a808a046140244b489ea8adca0c2bcc1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9dea291384064208a5c6b6d80c564f6d : SUCCESS in 20m 43s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/65e4fa5333974b5594a864e7d065ff44 : SUCCESS in 19m 21s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f57158ed93f4cd682c6eb5d7eb659cb : SUCCESS in 23m 08s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/920b694e60c24974bbf7e3a072ae415c : SUCCESS in 21m 24s","accounts_in_message":[],"_revision_number":5},{"id":"e815926093c2735846bd3bc04cce287874508798","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 20:28:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cc177148580442fbe98aa5213d2e89c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e7e59b628504847b5eecd00b5d77876 : SUCCESS in 11m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/523155ab4a78493a88ecc3ec96f15bcf : SUCCESS in 53m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1ac23e95a4d1491ea2ba0e186dbfe27b : SUCCESS in 45m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af95a4aeda044a82876fd2148f6a5422 : SUCCESS in 56m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/07bc555b58264f1c8172155ed7397b4a : SUCCESS in 20m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/845c6dfb76ea43f6906e679a9e18f7b7 : SUCCESS in 17m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b9462df43b3f4bcf84b5ea51a27daa50 : SUCCESS in 18m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b0dd65b8a1ef4d9ea16e08aa726525f0 : SUCCESS in 9m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3a0069113f4448d48e446c4cf601be19 : SUCCESS in 4m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f738b858807c4ce0ad059f619c41bba7 : SUCCESS in 17m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3d5f812e21c544b58eb09e9cf54c7e4c : SUCCESS in 17m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1e8c153bc3a943f5a54fa6f395c4994a : SUCCESS in 16m 52s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7c25c0501a57448b8ee2e8a3c6ba8623 : SUCCESS in 29m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/49e9cbce65d54fc8943fc01ddeca3e46 : SUCCESS in 16m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0293b7d18bdf4a19b38269b640ae14c4 : SUCCESS in 32m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2b0b7767abaa46c9a8e77ca56f9ab2c4 : SUCCESS in 16m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/065194598bfa49969394cc8792c66abd : SUCCESS in 15m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b5029eaf86664634b1f70f68bde1020d : SUCCESS in 27m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fd9574a15e304dd8b259d94acd1f2281 : SUCCESS in 15m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/10180c7bda4546bb827e3156e09c20eb : SUCCESS in 6m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/26ad65a5d8fc42ccbba20c6cace94eaa : SUCCESS in 16m 36s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/487149f92f5e4c868005352d34f647fc : SUCCESS in 7m 48s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2a7c6dc52cee4b1fbaecb22210c2968f : SUCCESS in 59m 45s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0bb21e98f31547a89bb18e98646a80f9 : SUCCESS in 1h 03m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/334ea88477364344bd8afaae1e6dfd11 : SUCCESS in 56m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d4f6e121704d45b99358012f7dad63d1 : SUCCESS in 52m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/254168f462cb4566b5714d7c551e00a6 : SUCCESS in 30m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cebf84af00cf4a90bcaa547ac30cd952 : SUCCESS in 4m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/518eb228230f408e96481beb18dd7988 : SUCCESS in 24m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0120629bc0724ede95c31dbea4b1fc0b : SUCCESS in 50m 31s","accounts_in_message":[],"_revision_number":5},{"id":"c4330f9b3c46b8b086cd01a23ba81b6d0ada25fe","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-23 16:57:51.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"716dbda17b3ffbb891dee413121e9590238120b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-23 18:33:53.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1a790baec6cfea430ee675b38df806f975210d58","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-23 18:34:16.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ea0f3b80a8ff471bc7e50a31d63c12c4229bf465","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 18:57:56.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7521c00824634fba93f5928e27c7bd05\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bcbdf976f8764d6d839fff3cef6a7d3a : SUCCESS in 13m 27s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f1a8441f4f544e87afda5c5067de356e : SUCCESS in 14m 48s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b440e8a366884772aff542f7b11998ad : SUCCESS in 23m 03s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/69c00b3858f641abac795a7dc5a72ccb : SUCCESS in 17m 11s","accounts_in_message":[],"_revision_number":6},{"id":"f7509bdd4d483b10ca044f5286c73f26d2be2837","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 19:43:38.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cd2e44ef6d042c1bf6ffe741cc360fd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65d78a2c69094cc394ba4365675e25eb : SUCCESS in 9m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/80454a0f661b4dfb865dc208f91071fa : SUCCESS in 1h 04m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7c77da6ed9b348b980d1c3cc9419a9ad : SUCCESS in 52m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a4f64edc6d80427cadfa75751fb1e01d : SUCCESS in 1h 04m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9fc214d81ae248f69f0c9297c0235ce5 : SUCCESS in 18m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/245c931572634577a38aa3a89758e48a : SUCCESS in 21m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3f0d5982139d4c07a12a85d0ae6bf75b : SUCCESS in 19m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4bd8561c769c4bbc96f04fd344608e69 : SUCCESS in 8m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/63c12745b05241d0a4734f52cb8c6562 : SUCCESS in 5m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/636d904c79fb4e6eb4720dfd1f683b4d : SUCCESS in 17m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/013a0a2de53d451b957dea1e01696e68 : SUCCESS in 15m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ac55ea2244764e40b39c93bdc5b29bd6 : SUCCESS in 18m 32s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/64b2a0f3b1654188b475418f910a864c : SUCCESS in 16m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/298c28d98a7d41c5bcdf2c31cb7850da : SUCCESS in 15m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7199a1f410c14c8993fc8b8147b6326a : SUCCESS in 20m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a3c99d774b364066857898d5a03de2f8 : SUCCESS in 18m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/dfcf58014ae446259c21d7ca8d40d173 : SUCCESS in 15m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a469fe56cca74c91996d9b571e61660e : SUCCESS in 15m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9b08ca354f6a463f978afcf2171855bb : SUCCESS in 15m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/93a35bb9cd1a4b67ab7a9974e1095b24 : SUCCESS in 8m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fb728f7b6ce84d06ac925734486a8813 : SUCCESS in 15m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d69723e1ca124690a190a84690cb8c1c : SUCCESS in 7m 35s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c941cb50c2ca47a0bfc8fc87222a27bb : SUCCESS in 51m 40s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/adb1ce68833a479db1cf9b770ac3473f : SUCCESS in 56m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/012ff1d1e02842968b34f5c237a09f31 : SUCCESS in 49m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/55fb9516e65147eaa578ed0073bd3243 : SUCCESS in 48m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e17bb4bcd8934af5823328c20e5bff5c : SUCCESS in 15m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b082fbf1925b440eb71ae6136bbed724 : SUCCESS in 4m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5cd7463a9ff74a998fc995ebfb29c79c : SUCCESS in 19m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2012869dd3cb46689a08f2575ef0eed6 : SUCCESS in 52m 12s","accounts_in_message":[],"_revision_number":6},{"id":"3c1530e6fe64fb4bf355c58094dbbd57a4c45ecd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-25 06:15:03.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"f5db7e31a33bf26d570d496c6038b6a18fb52af6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-25 06:15:39.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"3673bc3287712e5924583aaf7552efe2be78a615","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-25 07:27:11.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02e1642dd2d4481988956c0a098628c2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bd3e522e8354d8ea0380d4e0716cfeb : SUCCESS in 7m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/b8b6f3b63a844e74b1002066c1a9c9a0 : SUCCESS in 45m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4c43fbd400414e9b96653c4c3207b2ee : SUCCESS in 49m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e50f03767194e68a40826f4636ac8a2 : SUCCESS in 40m 54s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4f976d4b8d8341648f5c4566c98310f4 : SUCCESS in 8m 24s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/2021a1ddd90149bd8170361026a09706 : SUCCESS in 5m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/71c8a15bee0f4549bb934b20421246bd : SUCCESS in 19m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/58f39dc4498a45088bff38828b49222a : SUCCESS in 15m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7a80323a50e040389ab54ea152572f49 : SUCCESS in 15m 24s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/eb196db0926d464dbf6984f28770da1d : SUCCESS in 16m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/04d315e1fcc349c2b2b3d74bd03b1226 : SUCCESS in 17m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6054aa7829ec4ebd9c1ae5449c2a6a01 : SUCCESS in 18m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5909f3fcdd504bc6b978bd3cc1c853cf : SUCCESS in 16m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/415a4b70db484f05a2c6fccc8e708c3b : SUCCESS in 16m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3ba5ebd2bb9945e7acc8c6850f056770 : SUCCESS in 16m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d4ffd6897f0346d9be37f4954738292c : SUCCESS in 13m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/db5f082b4cac44618b83d38762a13f5c : SUCCESS in 6m 16s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d9201ce7cf2e43ab85f57b074588626e : SUCCESS in 6m 30s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/533634efefa541d89cc292c7e609c8f6 : SUCCESS in 57m 25s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b55f33d5010e4518836b43786fdbbb25 : SUCCESS in 56m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9e4c22881b2145578df6f57559d8165c : SUCCESS in 37m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ac82c13dad28455f8372dfd264a88742 : SUCCESS in 47m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3fb8eeb4968240df9a18b07fbc4716bb : SUCCESS in 17m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03f8c236f3b646639efd0ab3600e5651 : SUCCESS in 4m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/57c7386c5bbf4b5ba0343468afb2ca61 : SUCCESS in 38m 53s","accounts_in_message":[],"_revision_number":6},{"id":"b30441d68e27c96934b7a6db96d5603afc06986f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-25 07:27:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"9fac19bc9b089a5f36bea02787e930709f3b2963","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-25 07:28:45.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d26571bd688404a87997dcf21fd1d9e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6795c78b905d470cb75e002f565926d0 : SUCCESS in 39s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/4f295f81fbf64d69be2cbe9c05f705d9 : SUCCESS in 33s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/26fbca1cee85479f8d90f83fd9fc6278 : SUCCESS in 34s (non-voting)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"983879421f9a181759daacd8689a32c8afd12322","revisions":{"0e70777360cb668a46249e7c2d74ade2d85e88db":{"kind":"REWORK","_number":1,"created":"2023-01-26 19:11:36.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/43/871843/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/871843/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/871843/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/871843/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/871843/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/871843/1"}}},"commit":{"parents":[{"commit":"5de745c2bc3aa82049424b9c3e8744cda53a4c74","subject":"Add Kota\u0027s private email address to mailmap","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5de745c2bc3aa82049424b9c3e8744cda53a4c74"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-01-26 19:00:50.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-01-26 19:00:50.000000000","tz":0},"subject":"sharder: make misplaced objects lookup faster","message":"sharder: make misplaced objects lookup faster\n\nThe related change fixed a bug in the ContainerSharder\nyield_objects_to_shard_range method by replacing two DB lookups for\nmisplaced objects, one for undeleted rows and one for deleted rows,\nwith a single lookup for both deleted and undeleted rows. This\nsignificantly increased the time to make the lookup because it could\nno longer take advantage of the object table \u0027deleted\u0027 field index.\n\nThis patch reinstates the separate lookups for undeleted and deleted\nrows, and changees the yield_objects_to_shard_range implementation so\nthat both undeleted and deleted objects are consumed from the\nyield_objects generator.\n\nChange-Id: I337f4e54d1bcd4c5484fe56cfc886b16077982f5\nRelated-Change: Ie8404f0c7e84d3916f0e0fa62afc54f1f43a4d06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e70777360cb668a46249e7c2d74ade2d85e88db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e70777360cb668a46249e7c2d74ade2d85e88db"}]},"branch":"refs/heads/master"},"eef67f5319da794b7786ed0613cdbc77889551af":{"kind":"REWORK","_number":2,"created":"2023-01-27 13:42:02.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/43/871843/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/871843/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/871843/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/871843/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/871843/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/871843/2"}}},"commit":{"parents":[{"commit":"5de745c2bc3aa82049424b9c3e8744cda53a4c74","subject":"Add Kota\u0027s private email address to mailmap","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5de745c2bc3aa82049424b9c3e8744cda53a4c74"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-01-26 19:00:50.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-01-27 12:55:44.000000000","tz":0},"subject":"sharder: make misplaced objects lookup faster","message":"sharder: make misplaced objects lookup faster\n\nThe related change fixed a bug in the ContainerSharder\nyield_objects_to_shard_range method by replacing two DB lookups for\nmisplaced objects, one for undeleted rows and one for deleted rows,\nwith a single lookup for both deleted and undeleted rows. This\nsignificantly increased the time to make the lookup because it could\nno longer take advantage of the object table \u0027deleted\u0027 field index.\n\nThis patch reinstates the separate lookups for undeleted and deleted\nrows, and changees the yield_objects_to_shard_range implementation so\nthat both undeleted and deleted objects are consumed from the\nyield_objects generator.\n\nChange-Id: I337f4e54d1bcd4c5484fe56cfc886b16077982f5\nRelated-Change: Ie8404f0c7e84d3916f0e0fa62afc54f1f43a4d06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eef67f5319da794b7786ed0613cdbc77889551af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eef67f5319da794b7786ed0613cdbc77889551af"}]},"branch":"refs/heads/master"},"06166ea4dc6e7dd9ce31db17fb8ed6c5ae85e090":{"kind":"REWORK","_number":3,"created":"2023-02-17 10:24:41.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/43/871843/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/871843/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/871843/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/871843/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/871843/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/871843/3"}}},"commit":{"parents":[{"commit":"5de745c2bc3aa82049424b9c3e8744cda53a4c74","subject":"Add Kota\u0027s private email address to mailmap","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5de745c2bc3aa82049424b9c3e8744cda53a4c74"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-01-26 19:00:50.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-02-17 10:19:37.000000000","tz":0},"subject":"sharder: make misplaced objects lookup faster","message":"sharder: make misplaced objects lookup faster\n\nThe related change fixed a bug in the ContainerSharder\nyield_objects_to_shard_range method by replacing two DB lookups for\nmisplaced objects, one for undeleted rows and one for deleted rows,\nwith a single lookup for both deleted and undeleted rows. This\nsignificantly increased the time to make the lookup because it could\nno longer take advantage of the object table \u0027deleted\u0027 field index.\n\nThis patch reinstates the separate lookups for undeleted and deleted\nrows in yield_objects. In isolation that change would re-introduce the\nbug that was fixed by the Related-Change. The bug is therefore now\naddressed by changing the yield_objects_to_shard_range implementation\nso that both undeleted and deleted objects are consumed from the\nyield_objects generator.\n\nChange-Id: I337f4e54d1bcd4c5484fe56cfc886b16077982f5\nRelated-Change: Ie8404f0c7e84d3916f0e0fa62afc54f1f43a4d06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/06166ea4dc6e7dd9ce31db17fb8ed6c5ae85e090"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/06166ea4dc6e7dd9ce31db17fb8ed6c5ae85e090"}]},"branch":"refs/heads/master"},"1654da33b4de0093a5a29b29f1b480aa70ef6e82":{"kind":"REWORK","_number":4,"created":"2023-02-22 13:45:20.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/43/871843/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/871843/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/871843/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/871843/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/871843/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/871843/4"}}},"commit":{"parents":[{"commit":"5de745c2bc3aa82049424b9c3e8744cda53a4c74","subject":"Add Kota\u0027s private email address to mailmap","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5de745c2bc3aa82049424b9c3e8744cda53a4c74"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-01-26 19:00:50.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-02-22 13:45:04.000000000","tz":0},"subject":"sharder: make misplaced objects lookup faster","message":"sharder: make misplaced objects lookup faster\n\nThe related change fixed a bug in the ContainerSharder\nyield_objects_to_shard_range method by replacing two DB lookups for\nmisplaced objects, one for undeleted rows and one for deleted rows,\nwith a single lookup for both deleted and undeleted rows. This\nsignificantly increased the time to make the lookup because it could\nno longer take advantage of the object table \u0027deleted\u0027 field index.\n\nThis patch reinstates the separate lookups for undeleted and deleted\nrows in yield_objects. In isolation that change would re-introduce the\nbug that was fixed by the Related-Change. The bug is therefore now\naddressed by changing the yield_objects_to_shard_range implementation\nso that both undeleted and deleted objects are consumed from the\nyield_objects generator.\n\nChange-Id: I337f4e54d1bcd4c5484fe56cfc886b16077982f5\nRelated-Change: Ie8404f0c7e84d3916f0e0fa62afc54f1f43a4d06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1654da33b4de0093a5a29b29f1b480aa70ef6e82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1654da33b4de0093a5a29b29f1b480aa70ef6e82"}]},"branch":"refs/heads/master"},"7fb195123d2d94bb3dc838ed4ebc744882178c6d":{"kind":"REWORK","_number":5,"created":"2023-02-22 16:02:39.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/43/871843/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/871843/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/871843/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/871843/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/871843/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/871843/5"}}},"commit":{"parents":[{"commit":"5de745c2bc3aa82049424b9c3e8744cda53a4c74","subject":"Add Kota\u0027s private email address to mailmap","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5de745c2bc3aa82049424b9c3e8744cda53a4c74"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-01-26 19:00:50.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-02-22 16:00:48.000000000","tz":0},"subject":"sharder: make misplaced objects lookup faster","message":"sharder: make misplaced objects lookup faster\n\nThe related change fixed a bug in the ContainerSharder\nyield_objects_to_shard_range method by replacing two DB lookups for\nmisplaced objects, one for undeleted rows and one for deleted rows,\nwith a single lookup for both deleted and undeleted rows. This\nsignificantly increased the time to make the lookup because it could\nno longer take advantage of the object table \u0027deleted\u0027 field index.\n\nThis patch reinstates the separate lookups for undeleted and deleted\nrows in yield_objects. In isolation that change would re-introduce the\nbug that was fixed by the Related-Change. The bug is therefore now\naddressed by changing the yield_objects_to_shard_range implementation\nso that both undeleted and deleted objects are consumed from the\nyield_objects generator.\n\nChange-Id: I337f4e54d1bcd4c5484fe56cfc886b16077982f5\nRelated-Change: Ie8404f0c7e84d3916f0e0fa62afc54f1f43a4d06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7fb195123d2d94bb3dc838ed4ebc744882178c6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7fb195123d2d94bb3dc838ed4ebc744882178c6d"}]},"branch":"refs/heads/master"},"983879421f9a181759daacd8689a32c8afd12322":{"kind":"REWORK","_number":6,"created":"2023-02-23 18:33:53.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/43/871843/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/871843/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/871843/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/871843/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/871843/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/871843/6"}}},"commit":{"parents":[{"commit":"5de745c2bc3aa82049424b9c3e8744cda53a4c74","subject":"Add Kota\u0027s private email address to mailmap","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5de745c2bc3aa82049424b9c3e8744cda53a4c74"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-01-26 19:00:50.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-02-23 18:33:36.000000000","tz":0},"subject":"sharder: make misplaced objects lookup faster","message":"sharder: make misplaced objects lookup faster\n\nThe related change fixed a bug in the ContainerSharder\nyield_objects_to_shard_range method by replacing two DB lookups for\nmisplaced objects, one for undeleted rows and one for deleted rows,\nwith a single lookup for both deleted and undeleted rows. This\nsignificantly increased the time to make the lookup because it could\nno longer take advantage of the object table \u0027deleted\u0027 field index.\n\nThis patch reinstates the separate lookups for undeleted and deleted\nrows in yield_objects. In isolation that change would re-introduce the\nbug that was fixed by the Related-Change. The bug is therefore now\naddressed by changing the yield_objects_to_shard_range implementation\nso that both undeleted and deleted objects are consumed from the\nyield_objects generator.\n\nChange-Id: I337f4e54d1bcd4c5484fe56cfc886b16077982f5\nRelated-Change: Ie8404f0c7e84d3916f0e0fa62afc54f1f43a4d06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/983879421f9a181759daacd8689a32c8afd12322"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/983879421f9a181759daacd8689a32c8afd12322"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
