)]}'
{"id":"openstack%2Fswift~871854","triplet_id":"openstack%2Fswift~master~I08fb44e88534f0a00f4be1ca964334f1eb1eb59b","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-01-31 02:00:27.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-01-30 21:13:14.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I08fb44e88534f0a00f4be1ca964334f1eb1eb59b","subject":"Authors/ChangeLog for 2.31.0","status":"MERGED","created":"2023-01-26 20:50:09.000000000","updated":"2023-01-31 02:02:21.000000000","submitted":"2023-01-31 02:00:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"871854","meta_rev_id":"8f315ebce0ae3c80c5bc0fb7041fd7bd0b8e4968","_number":871854,"virtual_id_number":871854,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-31 02:00:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-31 01:36:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-31 01:36:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-26 21:12:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-26 21:22:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-30 17:34:43.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"f2e50b22b2f4a4d76d0731527647d8b4b75cec1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-26 20:50:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"980847196135095ff533f734c0e4fa4da2fed473","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-26 20:50:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"10a525eef6cbb5f405d09563b73d7cf241c3b7c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 21:12:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bc3af21b044472790d0fa2df728fc06\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2002799cd2f7427d80bc9241aff4ad8d : SUCCESS in 14m 58s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2668b6c65f4b4caf8d0834e900edb123 : SUCCESS in 14m 33s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d1ca2f4673364a44813705a4e0e616ef : SUCCESS in 20m 25s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/63fadb16e67b49efb64492ae901e1e7d : SUCCESS in 18m 08s","accounts_in_message":[],"_revision_number":1},{"id":"aeffefd260b808a520fdff4ac423ebdfed6815b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 21:22:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ed4d6f33ece4e06b17f0818a070534e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3946b4285234a38b140463df93c804f : SUCCESS in 9m 31s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ede29d8d20c3419e8573b6f29989605e : SUCCESS in 9m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5fe00758e2484c83a8eb6de0dee86164 : SUCCESS in 7m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5269467d61104bd28e157c8d37ba8515 : SUCCESS in 18m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3c99ff4bbec842f4959b43c5253d65bf : SUCCESS in 17m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3911008035094a999535be836b09debf : SUCCESS in 13m 36s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/143c2a8b827a4a8c960817017177665a : SUCCESS in 15m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1bd8d044009a4a0ea016ce114d284ed2 : SUCCESS in 16m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/272a5ac611974b7ead330814111a589b : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":1},{"id":"3ee9cd6fec30800e232a6b0771060b72bc1b2eb5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-26 21:23:39.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eab10fe2c785dd895c42231b031ce5ebb663d936","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 21:28:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/200cbb6dc48a4ce990456f910d2a511b\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/35dd015cd4a3428a962689c12be9a722 : SUCCESS in 14m 32s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/cb595d641e60450bbc8ffef3f46e7ced : SUCCESS in 12m 41s\n- swift-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/f53925178f494849847d6d0e5cfeb22f : SUCCESS in 19m 17s\n- swift-tox-func-py27-centos-7 https://zuul.opendev.org/t/openstack/build/ddea2f0a64e14d86866f427c7d6cebbb : SUCCESS in 17m 00s\n- swift-tox-func-encryption-py27-centos-7 https://zuul.opendev.org/t/openstack/build/8cd0026ea55e47278446f09e7bcdabab : SUCCESS in 22m 07s\n- swift-tox-func-ec-py27-centos-7 https://zuul.opendev.org/t/openstack/build/1c30e840ea3b4e83843ae9c6400afa6d : SUCCESS in 27m 42s\n- swift-tox-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/6c598335ed834db4ad934bb15dfb2a66 : SUCCESS in 19m 46s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/6aec80f387b844458982b7412cacbdf7 : SUCCESS in 12m 59s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a8c9061efad4220864d39a10dd20eec : SUCCESS in 15m 46s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/777f590538c74a9d9e2c2cd9b557ae0e : SUCCESS in 17m 49s\n- swift-multinode-rolling-upgrade-train https://zuul.opendev.org/t/openstack/build/0f1ff4c635964e0aad7a4069e6727db5 : SUCCESS in 19m 11s\n- swift-multinode-rolling-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/48ca03a9de4b44adac16a66d602a9cc9 : SUCCESS in 16m 02s\n- swift-multinode-rolling-upgrade-victoria https://zuul.opendev.org/t/openstack/build/24856d8157bb4941bf1890d35ab9b9a2 : SUCCESS in 18m 16s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/e82b1287d7b84a5ea3b7076497f44ab3 : SUCCESS in 21m 49s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/b96a27973edd45009a92e4358b34e4f5 : SUCCESS in 18m 13s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/ae96c2ad9ae14ebfafbc08225c914184 : SUCCESS in 18m 56s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/4fc672f239a1400cae0e30c1ad005e08 : SUCCESS in 22m 29s","accounts_in_message":[],"_revision_number":1},{"id":"a4240ca65182a3526f0ca8c0fbead3cb1da63e03","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-27 00:22:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ffa671d7be21b8a4ded072a68506efb00223e77c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-27 01:10:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"25d7bed62dd6aef832c7f9851d8e5f100dad1a4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 01:35:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec40dc072b284510a581de744b99a4f4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f84cf31c4d14d38bfc62597924a464d : SUCCESS in 7m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77e0bf377895434f8d29b8d8b04a7a33 : SUCCESS in 5m 06s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4586b00f3eb24019bbd674871f7aa327 : SUCCESS in 9m 06s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e5e4e9eca1024322894635b09a8a62c9 : SUCCESS in 4m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4c0af6c6d952446585d58e415b71f30b : SUCCESS in 16m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5ba1a4634b3048a8b8a0510bf82835f3 : SUCCESS in 15m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cdbeeaaa4e3540db81040cda3bd71518 : SUCCESS in 15m 41s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b2168a02eb504f6b827b66dda5b5850c : SUCCESS in 18m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5745268c0f9f449cb1e4bf0ec1a648fe : SUCCESS in 13m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc35603e2cac450c8f5bc953d52fbfd7 : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":2},{"id":"c9bafd1f1e3674cabba63bbf285ff11bcda4e0b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 02:42:05.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8669de48ef9b4d2cb4bff8f3a780f0a7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c97428ac5fc84da1b4c41a0b51d7828e : SUCCESS in 16m 14s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b90a553eca3445558440575ceaa9e6e0 : SUCCESS in 15m 06s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1d4805b9efd344008d09ea2789548e55 : FAILURE in 1h 15m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/595f8aa0890c458db03b6ce86260dfd4 : SUCCESS in 17m 59s","accounts_in_message":[],"_revision_number":2},{"id":"acaa242a2a73cca21b63bb54274a9b3ea9ccaf8e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-01-30 17:34:43.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"fc6f02507c0adbd7ea55b9c5e7719b80dc3ad366","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-30 21:13:14.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7b1a738f1e5f96a09a207695f03d4510f7a0c0cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-30 21:13:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"70708a61c4d7ec93bae9b113e6c3edb61cc2d937","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 21:32:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7558d468c95d4b888165ec7bde73c179\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30f55e6107544c96bf55065cc1ddc8e0 : SUCCESS in 9m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/214179208de040c78375d780d90584a8 : SUCCESS in 5m 00s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3edb5335104e4cbbbe467c3466a4d9c4 : SUCCESS in 8m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ecb470fba1ae434482b1f345e82099a9 : SUCCESS in 4m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/68ab68c879f54250922a71f6fa699fe7 : SUCCESS in 17m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fc1929883d1b4dc7bcb6c87808a5877c : SUCCESS in 16m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ea76bc6ccc9d47e69e1c81a5e5a3d95e : SUCCESS in 13m 54s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c4dd258d429a40cfbc914b46885cde7a : SUCCESS in 16m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9f8aab51fdc4af18221c99391d9633f : SUCCESS in 17m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41a06e184c9d4d01a8e6dc8247819eed : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":3},{"id":"5b7dc35fa696f2f450b950c9e090302bf1adf687","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 21:35:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9256c441f2354506885e0272d0c1cfa7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/632e18fc5b874b4aac529212b4a7ba16 : SUCCESS in 16m 05s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d767325cd8114952a60a4c2f7e341967 : SUCCESS in 15m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8a77dfafe6714204be332c25d776ab2c : SUCCESS in 20m 27s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4d7001dd580c4f118cae197ee3f8b095 : SUCCESS in 19m 02s","accounts_in_message":[],"_revision_number":3},{"id":"8933434b2727a5b468694caeda97839480ea6fef","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-30 21:38:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"31efdf78c8654be44fbd2651cf4bfed5918bd665","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-30 22:56:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"afd365031cf9cf22a1cfd3de6a84b7e40c7bf033","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 23:29:16.000000000","message":"Patch Set 3:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31fb9890745641b89fd208d303d7e744\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7929a96766274f529deb4b0e2b996b2e : SUCCESS in 19m 06s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d2cddda9deea4533ba391ef844c50155 : SUCCESS in 16m 56s\n- swift-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/74430663b23b4cfba58ab1e661bc2772 : SUCCESS in 22m 34s\n- swift-tox-func-py27-centos-7 https://zuul.opendev.org/t/openstack/build/18cf6419bcf24c4bb5abc4b8ea0d5e91 : SUCCESS in 24m 10s\n- swift-tox-func-encryption-py27-centos-7 https://zuul.opendev.org/t/openstack/build/e4be02662aba44a6af867c930f82c883 : SUCCESS in 23m 56s\n- swift-tox-func-ec-py27-centos-7 https://zuul.opendev.org/t/openstack/build/e3ba879210124558add8e2c27dcf94ea : SUCCESS in 26m 34s\n- swift-tox-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/2eeb7f0a1cfb4ca9abd7b9aa4f35e87c : SUCCESS in 23m 42s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/1e5c58107fdc4457bccb937c9507bc25 : SUCCESS in 14m 14s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/580f4622fb834601b14e0376cb7bdc37 : SUCCESS in 21m 01s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/709223c323054531bd1d1207770060dc : SUCCESS in 13m 03s\n- swift-multinode-rolling-upgrade-train https://zuul.opendev.org/t/openstack/build/698f7946c9c94617a99a92e00956c8c1 : SUCCESS in 21m 57s\n- swift-multinode-rolling-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/2faf6b1dd8c243efa7ccfa744420ab32 : SUCCESS in 16m 59s\n- swift-multinode-rolling-upgrade-victoria https://zuul.opendev.org/t/openstack/build/e3516b070c0f4d8fb26cc7b9a5215daa : SUCCESS in 17m 47s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/8610031ff48b422db5d8ccf3490a49b9 : FAILURE in 25m 40s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/e8942c34ef2344928d521d1766004d66 : SUCCESS in 24m 22s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/f9d4f2bdb3e747e0ae3405e1557293bc : SUCCESS in 30m 11s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/f0b79a675a9b4a0f9ef40b96f749c723 : SUCCESS in 15m 53s","accounts_in_message":[],"_revision_number":3},{"id":"9d2ff5ebbfc3831fac4f151ecfc35f928eccb128","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-31 01:36:03.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b44240675e218024d463a0d8afefb786133718c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 01:36:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4eea3f4b2ad3f30d606459315c46c512d7368e70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 02:00:27.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49f8dd721541428b8466dcc3890a3ca7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6846c2a515054e168833525feb3c17e4 : SUCCESS in 7m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e87671ae12b42bdab97f462a254b2ba : SUCCESS in 5m 08s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0cfa059e64cb46b48297a15ea511ba5b : SUCCESS in 8m 19s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/82df20e85fed44fcb319a86d7b783c9d : SUCCESS in 5m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/64c685fd549340efa0a46fdbc8e95304 : SUCCESS in 18m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7750c7df8b8c4cd8838c4a133b12daae : SUCCESS in 14m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/45e24273382b4abc84f2cfc41bacf398 : SUCCESS in 15m 09s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4bc0378f165e4a7db939c78b09af832b : SUCCESS in 16m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/59bb7248ce6745cba20e8e2d8df8772e : SUCCESS in 18m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5d7010d76fc74ce39c1502d729c8019f : SUCCESS in 18m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/04792b4a2654448ab4772ef6b08fffec : SUCCESS in 16m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/966cca2074d0492b8ec7097bc84c715d : SUCCESS in 12m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4b671868473a4369b7ebe32312971168 : SUCCESS in 14m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5d9fb29155c94a859210351242d59a72 : SUCCESS in 14m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/81b03f1b4f7641d6b2dd2e3557d5dda8 : SUCCESS in 7m 06s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fcd6806425944ebea991488462653e10 : SUCCESS in 6m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec3ff4ebe862462784a0875470146a6c : SUCCESS in 15m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edc7f0a0c90d459494d702b9c9e2e1cb : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":3},{"id":"7122068b303deaf0f730f7c836f858155009d25d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 02:00:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"8f315ebce0ae3c80c5bc0fb7041fd7bd0b8e4968","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 02:02:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/505eeeef7f034cfcad8a54042b461f36\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/426dcbdeb00d4c6b981e3985dd3ae0de : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d781111bdf2b4291948bd47379d83e6c : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/608b74b90eb74460b6569e3582fca004 : SUCCESS in 36s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/db97d98013d64af89b4ab6446aa1bab7 : SUCCESS in 35s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"90f9a479b67577ffa83057f5e21d464d5294fcb0","revisions":{"8755fd6cd16c96921fd31581ca272fe7cdafa249":{"kind":"REWORK","_number":1,"created":"2023-01-26 20:50:09.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/54/871854/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/871854/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/871854/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/871854/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/871854/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/871854/1"}}},"commit":{"parents":[{"commit":"5de745c2bc3aa82049424b9c3e8744cda53a4c74","subject":"Add Kota\u0027s private email address to mailmap","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5de745c2bc3aa82049424b9c3e8744cda53a4c74"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-26 20:49:52.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-26 20:49:52.000000000","tz":-480},"subject":"Authors/ChangeLog for 2.31.0","message":"Authors/ChangeLog for 2.31.0\n\nChange-Id: I08fb44e88534f0a00f4be1ca964334f1eb1eb59b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8755fd6cd16c96921fd31581ca272fe7cdafa249"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8755fd6cd16c96921fd31581ca272fe7cdafa249"}]},"branch":"refs/heads/master"},"ed48e69df0e319599f74b557a774435b0d07807c":{"kind":"REWORK","_number":2,"created":"2023-01-27 01:10:18.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/54/871854/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/871854/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/871854/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/871854/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/871854/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/871854/2"}}},"commit":{"parents":[{"commit":"5de745c2bc3aa82049424b9c3e8744cda53a4c74","subject":"Add Kota\u0027s private email address to mailmap","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5de745c2bc3aa82049424b9c3e8744cda53a4c74"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-26 20:49:52.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-27 01:10:11.000000000","tz":-480},"subject":"Authors/ChangeLog for 2.31.0","message":"Authors/ChangeLog for 2.31.0\n\nChange-Id: I08fb44e88534f0a00f4be1ca964334f1eb1eb59b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed48e69df0e319599f74b557a774435b0d07807c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed48e69df0e319599f74b557a774435b0d07807c"}]},"branch":"refs/heads/master"},"90f9a479b67577ffa83057f5e21d464d5294fcb0":{"kind":"REWORK","_number":3,"created":"2023-01-30 21:13:39.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/54/871854/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/871854/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/871854/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/871854/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/871854/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/871854/3"}}},"commit":{"parents":[{"commit":"5de745c2bc3aa82049424b9c3e8744cda53a4c74","subject":"Add Kota\u0027s private email address to mailmap","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5de745c2bc3aa82049424b9c3e8744cda53a4c74"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-26 20:49:52.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-30 21:13:23.000000000","tz":-480},"subject":"Authors/ChangeLog for 2.31.0","message":"Authors/ChangeLog for 2.31.0\n\nChange-Id: I08fb44e88534f0a00f4be1ca964334f1eb1eb59b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90f9a479b67577ffa83057f5e21d464d5294fcb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90f9a479b67577ffa83057f5e21d464d5294fcb0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
