)]}'
{"id":"openstack%2Fswift~872031","triplet_id":"openstack%2Fswift~master~Idc536034a36646de9c1880c8d0bc0a387b130ac2","project":"openstack/swift","branch":"master","topic":"bug/2003984","attention_set":{},"removed_from_attention_set":{"35733":{"account":{"_account_id":35733,"name":"Alexander Fadeev","display_name":"Alexander Fadeev","email":"fadeevab.com@gmail.com","username":"fadeevab"},"last_update":"2023-02-27 21:07:13.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-02-27 16:19:50.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"Idc536034a36646de9c1880c8d0bc0a387b130ac2","subject":"Present `pytest` steps in development guidelines","status":"MERGED","created":"2023-01-27 17:33:04.000000000","updated":"2023-02-27 21:08:36.000000000","submitted":"2023-02-27 21:07:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872031-bug/2003984","meta_rev_id":"0316d14a5f7aa3e17ad2af2d7ba7923e1ca73430","_number":872031,"virtual_id_number":872031,"owner":{"_account_id":35733,"name":"Alexander Fadeev","display_name":"Alexander Fadeev","email":"fadeevab.com@gmail.com","username":"fadeevab"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-27 21:07:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2023-02-27 16:19:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2023-02-27 16:19:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-27 17:47:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-27 18:12:45.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"68681af47491110283b15bccb80ea34890d716b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35733,"name":"Alexander Fadeev","display_name":"Alexander Fadeev","email":"fadeevab.com@gmail.com","username":"fadeevab"},"date":"2023-01-27 17:33:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c589948221c1b480ff87f57875764868e466b39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 17:47:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/657555497551408d934c48e9dd0d6c2b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f053b3f5e994444b899d1b4d60b12999 : SUCCESS in 8m 53s","accounts_in_message":[],"_revision_number":1},{"id":"3a09fa54d349bdfe396edd512eee78075f83d665","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-01-30 16:04:21.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"940572eeb50be9827a16649b8838c93e8403cec9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35733,"name":"Alexander Fadeev","display_name":"Alexander Fadeev","email":"fadeevab.com@gmail.com","username":"fadeevab"},"date":"2023-02-25 00:02:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"03019f55a8d80b9e517bbd77a72e4710214371ad","author":{"_account_id":35733,"name":"Alexander Fadeev","display_name":"Alexander Fadeev","email":"fadeevab.com@gmail.com","username":"fadeevab"},"date":"2023-02-25 00:13:52.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b1168e994d77c43be00072a73f7b40d72be52ffb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-25 00:14:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ac944a088db4914a308f559bf98e7f1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/180aab2c31924051bb0e6ddd426a6464 : SUCCESS in 10m 22s","accounts_in_message":[],"_revision_number":2},{"id":"c3f656ef31441e98f8f570b300af4f0dfd7ed74f","author":{"_account_id":35733,"name":"Alexander Fadeev","display_name":"Alexander Fadeev","email":"fadeevab.com@gmail.com","username":"fadeevab"},"date":"2023-02-27 10:55:49.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d78a8a1cc14103fae497c2d20e9fe56408d56f9e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-27 16:19:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"33bfa7683cb177d28159b69d92f7d7cbccc997c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 16:20:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8113a6074e20bdd987258bba33c423e3e8e8c268","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 21:07:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45b15281d3c14595b74014e5a766c7e5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd2c8d16491e49e8bc3a57322c19fd90 : SUCCESS in 12m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/033e417f484442f0b8846811b19e1af8 : SUCCESS in 7m 57s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d5d5a199773c4bfba54d409db14f0c4d : SUCCESS in 8m 01s","accounts_in_message":[],"_revision_number":2},{"id":"f3e086d1f7dfb79aa4ed1f75d9bd9302f51656dc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 21:07:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0316d14a5f7aa3e17ad2af2d7ba7923e1ca73430","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 21:08:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f3a616fc8d2431a9a8cd75c1a473b0a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a17f136aba3412681a62f37552a1549 : SUCCESS in 51s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/6f15137cbf634f6cab5f741bbd30be1f : FAILURE in 26s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/aa484555bedc4ee384694147302cdfb8 : FAILURE in 28s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"49ba1bc33aeeee88e8d81eef818d3c857fb3b505","revisions":{"4c74ab244d64737e88cb9b7ecd8656795a298b67":{"kind":"REWORK","_number":1,"created":"2023-01-27 17:33:04.000000000","uploader":{"_account_id":35733,"name":"Alexander Fadeev","display_name":"Alexander Fadeev","email":"fadeevab.com@gmail.com","username":"fadeevab"},"ref":"refs/changes/31/872031/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/872031/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/872031/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/872031/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/872031/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/872031/1"}}},"commit":{"parents":[{"commit":"5de745c2bc3aa82049424b9c3e8744cda53a4c74","subject":"Add Kota\u0027s private email address to mailmap","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5de745c2bc3aa82049424b9c3e8744cda53a4c74"}]}],"author":{"name":"Alexander Fadeev","email":"fadeevab.com@gmail.com","date":"2023-01-27 17:32:32.000000000","tz":120},"committer":{"name":"Alexander Fadeev","email":"fadeevab.com@gmail.com","date":"2023-01-27 17:32:32.000000000","tz":120},"subject":"Present `pytest` steps in development guidelines","message":"Present `pytest` steps in development guidelines\n\nCloses-Bug: #2003984\nChange-Id: Idc536034a36646de9c1880c8d0bc0a387b130ac2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4c74ab244d64737e88cb9b7ecd8656795a298b67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4c74ab244d64737e88cb9b7ecd8656795a298b67"}]},"branch":"refs/heads/master"},"49ba1bc33aeeee88e8d81eef818d3c857fb3b505":{"kind":"REWORK","_number":2,"created":"2023-02-25 00:02:06.000000000","uploader":{"_account_id":35733,"name":"Alexander Fadeev","display_name":"Alexander Fadeev","email":"fadeevab.com@gmail.com","username":"fadeevab"},"ref":"refs/changes/31/872031/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/872031/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/872031/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/872031/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/872031/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/872031/2"}}},"commit":{"parents":[{"commit":"5de745c2bc3aa82049424b9c3e8744cda53a4c74","subject":"Add Kota\u0027s private email address to mailmap","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5de745c2bc3aa82049424b9c3e8744cda53a4c74"}]}],"author":{"name":"Alexander Fadeev","email":"fadeevab.com@gmail.com","date":"2023-01-27 17:32:32.000000000","tz":120},"committer":{"name":"Alexander Fadeev","email":"fadeevab.com@gmail.com","date":"2023-02-25 00:01:07.000000000","tz":120},"subject":"Present `pytest` steps in development guidelines","message":"Present `pytest` steps in development guidelines\n\nExplain how to prepare venv with `tox devenv`\n\nCloses-Bug: #2003984\nChange-Id: Idc536034a36646de9c1880c8d0bc0a387b130ac2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49ba1bc33aeeee88e8d81eef818d3c857fb3b505"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49ba1bc33aeeee88e8d81eef818d3c857fb3b505"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[]}
