)]}'
{"id":"openstack%2Fswift~872224","triplet_id":"openstack%2Fswift~stable%2Fyoga~I19d7c37e2f4b63f8743560b6b6569c0bc096ee65","project":"openstack/swift","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-01-31 02:38:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I19d7c37e2f4b63f8743560b6b6569c0bc096ee65","subject":"Authors/ChangeLog for 2.29.2","status":"MERGED","created":"2023-01-30 21:40:30.000000000","updated":"2023-01-31 02:40:03.000000000","submitted":"2023-01-31 02:38:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"872224","meta_rev_id":"9a788419807dc6b2465b1079d80b1af7a4b6b398","_number":872224,"virtual_id_number":872224,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-31 02:38:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-31 02:22:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":2,"date":"2023-01-31 01:36:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":1,"date":"2023-01-31 01:36:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-30 22:45:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-31 02:22:39.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"84d9b39d00d23c41853598d11a6c4289aa877f68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-30 21:40:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7af721133eb7a6c4745a7b3f0087483b6a19dfb1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-30 21:41:11.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ce9b4a5380f680e5e8d879901939d038c798a3aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-30 22:00:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ce0529615c3c94a70406ed8a856acb2063f54527","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 22:45:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5cf2d909d47471ea89f090f1bedfad7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/261d26a1aec84fbe89a406350727a4bd : SUCCESS in 8m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/311d48bb7573487d8f4b69bbd2942922 : SUCCESS in 4m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c99a59f24143404eb03b3116dc53a94d : SUCCESS in 42m 04s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a1fc1a72e2834044ade2390891e434fd : FAILURE in 3m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/13693e3fcb4b48579a4d7b1f0e3366e8 : FAILURE in 3m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2164e460179548e085b24fa7d7629aeb : SUCCESS in 12m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/27b61dde46e64a799d02680d5f39ab0c : SUCCESS in 13m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4f8f5a84ba1e4bb0a3f66311b42bcf01 : SUCCESS in 15m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/437fa3238666428e9e74ccfd157c9fb3 : SUCCESS in 14m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5c2b76eaec94dd191baea8748972542 : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":2},{"id":"eae1040c0d1686239aa4e06d9eb77862d79e815a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-30 22:55:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0922f1122f226f49a27b8d4b7557a0814176b12f","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 23:21:11.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc8577cae31140348e1f1748b0bcd364\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/550e501906ce43eab9c9e81ea7a07c58 : SUCCESS in 14m 27s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/af26abd9801f4c288b67d0a2382a89ea : SUCCESS in 15m 50s\n- swift-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/ec2b8f0e46914535ba26c3516007c3a1 : FAILURE in 3m 38s\n- swift-tox-func-py27-centos-7 https://zuul.opendev.org/t/openstack/build/d4976e15ce924be897b045817779140c : FAILURE in 3m 50s\n- swift-tox-func-encryption-py27-centos-7 https://zuul.opendev.org/t/openstack/build/ea59eb9872604ad688a7c2abadfa3721 : FAILURE in 4m 25s\n- swift-tox-func-ec-py27-centos-7 https://zuul.opendev.org/t/openstack/build/1fc3ee7f48814ad98a68407d8bd113a4 : FAILURE in 3m 14s\n- swift-tox-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/edd1bf54206d41e1a32d5a009558dd87 : SUCCESS in 17m 58s\n- swift-tox-func-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/8f7001d173a54d8bad1b434cffe9ed84 : SUCCESS in 20m 49s\n- swift-tox-func-encryption-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/ebdf179d34db4e1098b5a7088443aab6 : SUCCESS in 18m 44s\n- swift-tox-func-ec-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/73206489aafb489e91f0dfb5e3bad254 : SUCCESS in 19m 05s\n- swift-multinode-rolling-upgrade-rocky https://zuul.opendev.org/t/openstack/build/e10f418e20a5480a912e06dbfb47d9a6 : FAILURE in 5m 50s\n- swift-multinode-rolling-upgrade-stein https://zuul.opendev.org/t/openstack/build/70b8a162484d4846a9dce94926e85d20 : FAILURE in 6m 43s\n- swift-multinode-rolling-upgrade-train https://zuul.opendev.org/t/openstack/build/06812e854b7345d39ea1e10f77644093 : FAILURE in 6m 13s\n- swift-multinode-rolling-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/69d8005de1d946e1b7f33bea049e5bc6 : FAILURE in 6m 20s\n- swift-multinode-rolling-upgrade-victoria https://zuul.opendev.org/t/openstack/build/3f3faef7063041d0b065c0c58fa587df : SUCCESS in 19m 54s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/2a74622e98354446afecc12fc7ab8d50 : FAILURE in 6m 03s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/92eae100d3d64a15a164f1b51b194149 : FAILURE in 11m 17s","accounts_in_message":[],"_revision_number":2},{"id":"57b139347d4e43c56c2e8594f3a963774884c606","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-30 23:22:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1fc4901ee8998b9cde1b6d6f9d6260c674fd80f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 00:11:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9f6acb16bfe41b6abcb80d28ea38d1f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06dbfe452cd447d3be1bfab6a8a2b29f : SUCCESS in 8m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8b727301ac9492e818ff9c9db24f87f : SUCCESS in 4m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0e90ea780d9f4a788dc0031aa7267595 : SUCCESS in 47m 41s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c20d94c8c6e144fe81c960adfe8bb9de : FAILURE in 3m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4653185f4d7b4323ba0517da6d420c76 : FAILURE in 3m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e0192b34a5d74f0eae2151b597b49e6c : SUCCESS in 19m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/03b05b7040e642bda68805b84641b20f : SUCCESS in 11m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3b1b903495ce42908c3df90f70f8c576 : SUCCESS in 13m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35b1dca7ad4641c2bd7c2ad8cdfb0045 : SUCCESS in 13m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1e12d110f3441c7a54e4204e15d3474 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":3},{"id":"7fb3e0d99c9aec228cbc555db31c467aee43bcc8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 00:54:15.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2d9ce6e529941669bf5bcf9a99c7cf4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e262565ba4324211831c8ef61d1a5af7 : TIMED_OUT in 1h 30m 43s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b0ce7375744c4c258f2853ff525e7aef : SUCCESS in 18m 18s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/41be9321f1514c8eadce883fea332789 : FAILURE in 1h 03m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7561a490d4004e56a3028a9a5e5b563a : SUCCESS in 22m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b86dff0637234aa8b56ad30a64f5f27d : SUCCESS in 20m 53s","accounts_in_message":[],"_revision_number":3},{"id":"0704aa4e795e47ba6d6ced6cac278e9c27b3b984","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-31 01:36:15.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"52aa17ee1c3ef8a03d01dc51a333290e5fecb8f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 01:36:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"58c5094246507c4f69c63148844e6dfa4a0740d6","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-01-31 02:22:39.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b5bac8568df32c4ed1dd27af533fd8cee9c047b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 02:38:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a75ff70d08044c39b6bbea9eeb2ea480\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/840b009ca9fd43e5a42d98be324374b7 : SUCCESS in 7m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6cc4a28b0204162a999ffc261615011 : SUCCESS in 5m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f6bf342d39894988965d1e293d79258d : SUCCESS in 46m 55s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/3c3ab482c11d4ee280b4ab841b34b601 : POST_FAILURE in 4m 59s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/779452093ddd428792d704c45e13e57c : POST_FAILURE in 5m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1ce0cfd167ea41cfb4322a0984b39894 : SUCCESS in 13m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/14db2ab14adc40768ee0d2967b464b0d : SUCCESS in 10m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/74949dcdb082490981b3eb9a174a4eb5 : SUCCESS in 12m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/799771cc0ffe428d991131244fca1ca1 : SUCCESS in 12m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/422ac892ac5f44aaa707bb3698577504 : SUCCESS in 15m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5007e21772a8466a9f623e2aa0361b10 : SUCCESS in 21m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2246263ab55a49bf85d5a3c90cdfd5bd : SUCCESS in 18m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9d596e922a4241d6862253a6ae6ce58b : SUCCESS in 12m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d1bc20da114849aa8a69405a137e0114 : SUCCESS in 14m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1ae62501d76b4fee9902d7228f733963 : SUCCESS in 7m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e977962eddd4cd3b543c97b127c6166 : SUCCESS in 16m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60cd76ca75694d3696c34ad18aa1a6b9 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":3},{"id":"9c17d20ff40ede66c64a45e5a25dd8d5aa9c9866","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 02:38:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"9a788419807dc6b2465b1079d80b1af7a4b6b398","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 02:40:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05480ed9ac8b4d4cb8335f550cb983a5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/268cc41682c74043ad03c536919953c2 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/625cd4290d5246ed8bd7543533eca8dc : SUCCESS in 58s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/26dd6528338440f5ae1721d7558deb54 : FAILURE in 39s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/c21a16cfb48d433dba58e54b888d4186 : FAILURE in 32s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"19879831272b37b5e49f73a4d49f1a4587ccd7c7","revisions":{"cad85fe8f5e214db5cdc81e3953fb9265186af61":{"kind":"REWORK","_number":1,"created":"2023-01-30 21:40:30.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/24/872224/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/872224/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/872224/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/872224/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/872224/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/872224/1"}}},"commit":{"parents":[{"commit":"d8d04ef43c90079d436b2e49617b4425ba39c28e","subject":"s3api: Prevent XXE injections","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d8d04ef43c90079d436b2e49617b4425ba39c28e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-30 21:24:20.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-30 21:40:11.000000000","tz":-480},"subject":"Authors/ChangeLog for 2.29.2","message":"Authors/ChangeLog for 2.29.2\n\nChange-Id: I19d7c37e2f4b63f8743560b6b6569c0bc096ee65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cad85fe8f5e214db5cdc81e3953fb9265186af61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cad85fe8f5e214db5cdc81e3953fb9265186af61"}]},"branch":"refs/heads/stable/yoga"},"785d7f074b20cf4585cfe6fc0b70f7b96c5acaae":{"kind":"REWORK","_number":2,"created":"2023-01-30 22:00:26.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/24/872224/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/872224/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/872224/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/872224/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/872224/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/872224/2"}}},"commit":{"parents":[{"commit":"d8d04ef43c90079d436b2e49617b4425ba39c28e","subject":"s3api: Prevent XXE injections","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d8d04ef43c90079d436b2e49617b4425ba39c28e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-30 21:24:20.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-30 22:00:19.000000000","tz":-480},"subject":"Authors/ChangeLog for 2.29.2","message":"Authors/ChangeLog for 2.29.2\n\nChange-Id: I19d7c37e2f4b63f8743560b6b6569c0bc096ee65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/785d7f074b20cf4585cfe6fc0b70f7b96c5acaae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/785d7f074b20cf4585cfe6fc0b70f7b96c5acaae"}]},"branch":"refs/heads/stable/yoga"},"19879831272b37b5e49f73a4d49f1a4587ccd7c7":{"kind":"REWORK","_number":3,"created":"2023-01-30 23:22:17.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/24/872224/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/872224/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/872224/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/872224/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/872224/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/872224/3"}}},"commit":{"parents":[{"commit":"d8d04ef43c90079d436b2e49617b4425ba39c28e","subject":"s3api: Prevent XXE injections","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d8d04ef43c90079d436b2e49617b4425ba39c28e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-30 21:24:20.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-30 23:22:00.000000000","tz":-480},"subject":"Authors/ChangeLog for 2.29.2","message":"Authors/ChangeLog for 2.29.2\n\nChange-Id: I19d7c37e2f4b63f8743560b6b6569c0bc096ee65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/19879831272b37b5e49f73a4d49f1a4587ccd7c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/19879831272b37b5e49f73a4d49f1a4587ccd7c7"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
