)]}'
{"id":"openstack%2Fswift~872417","triplet_id":"openstack%2Fswift~master~Ibd5614d378ec5e9ba47055ba8b67a42ab7f7453c","project":"openstack/swift","branch":"master","topic":"new_sr_cache","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-03-16 11:52:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibd5614d378ec5e9ba47055ba8b67a42ab7f7453c","subject":"Rename ShardRange*Bound to Namespace*Bound","status":"MERGED","created":"2023-02-01 15:08:01.000000000","updated":"2023-03-16 11:53:45.000000000","submitted":"2023-03-16 11:52:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872417-new_sr_cache","meta_rev_id":"8101cbfb4a4f1ddffb3fef24505f9bf542a768d8","_number":872417,"virtual_id_number":872417,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-16 11:52:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2023-03-16 09:53:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2023-03-15 16:25:20.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2023-03-16 09:53:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-01 16:19:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-01 22:49:08.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2023-03-16 09:53:32.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"a58bac73d9c3d502f14fc10eac7081f066e4f698","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-01 15:08:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2bb242883ff5787bb99199da1e50510bbac74fd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 16:19:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9da3971b54674d749bec92b490eb8413\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebefb780489c430494e0236a554e0194 : SUCCESS in 9m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/6fcbc541f8134707a258da54c5cb7d08 : SUCCESS in 1h 09m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2c951381af114046b4f3a4a070f9d747 : SUCCESS in 52m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/17f187ffe85c484ab36fa65f95389ed3 : SUCCESS in 54m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6b57e1234f694ece82a111bd43d433ce : SUCCESS in 20m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/db186a0c6ffe4cb288be812e6edae5d8 : SUCCESS in 41m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2e2e3c5fd5304fba964fc4230504d55f : SUCCESS in 22m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/84d318aa886e4de4b5a457c1286bc15c : SUCCESS in 8m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ababde56fc794e64ae13977d14dbf9ce : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/76dd8994761446c59ebb800a07e2c48f : SUCCESS in 13m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/49d15f942df449d4bf8a4976fde0ea7d : SUCCESS in 17m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/99ebf9bfae834a7ead56d86b582d4e9d : SUCCESS in 16m 25s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/95c0093a125647c2af4882533b9f22d3 : SUCCESS in 15m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2faf9537188e4dc8893d8ceaf73e0ac4 : SUCCESS in 13m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1e18cc6e19ff45598323168c6f1127b4 : SUCCESS in 21m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d40829654f4f44968ef346c3fa728786 : SUCCESS in 17m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/20ca7e0a32b04d99803b1059a10ae13b : SUCCESS in 13m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/78f06b7b704348cea2523614ed18d9e5 : SUCCESS in 15m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/679fed29637447c087f886ac75cc163b : SUCCESS in 16m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/23e906773a44441ca938730f22434536 : SUCCESS in 7m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6fd97ad8d58e426e829882fa9129f614 : SUCCESS in 15m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6e557e6f8e7242369b80797b05113d69 : SUCCESS in 7m 38s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5a44aacdde6d4b4ba8df85ea52ac5dc3 : SUCCESS in 51m 27s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a2ee3114efae43a3b9a2d1d8d7d7c176 : SUCCESS in 58m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1bf5a3f56768491e96d426e2c29b2603 : SUCCESS in 48m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5b6b74c779d84f1183af8eb2317d51a2 : SUCCESS in 1h 04m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df5a7cae35314dbc8fd008722cba9697 : SUCCESS in 16m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6116e38e83f415699d4dbfc283b910e : SUCCESS in 4m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8ada6348ea5148918c8c23fe3e24ebd4 : SUCCESS in 24m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/06d8b849745a4034a439c334cdb98926 : SUCCESS in 51m 49s","accounts_in_message":[],"_revision_number":1},{"id":"ebf406e6114787f9fe1c7579d6bb3e58db5230b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 16:19:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc60e414a17a45249a8e1cb3f912807c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/07f36d6b0f6f4406a592712e15535e3c : SUCCESS in 17m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9d25791103b346a7b5b88378d3a600ea : SUCCESS in 52m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/40d7219f8592475a8167337ed332bc02 : SUCCESS in 1h 09m 44s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ad85c8fbbe8b4dceb906d7ca41895014 : SUCCESS in 1h 04m 27s","accounts_in_message":[],"_revision_number":1},{"id":"fea6cf78576fdf44ab79c46689a798a2110442cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-02-01 22:48:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cd40471087e35881e5abdeea22d275e9f56095f4","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-02-01 22:49:08.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5725c40f142c641a53124233093441c117ef88f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 23:56:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/008388f29b7444468d1df81fc2281058\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0176b3679a4d40d097863b15a3878bf9 : SUCCESS in 9m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/bc5d764195fd44c4bdcd697f62a42bdc : SUCCESS in 52m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ca691742301a4e3d8d5408dd065d6251 : SUCCESS in 48m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/48f45e6a1f444e8f8081bd0bb0707b25 : SUCCESS in 58m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2445293e467142cbb852f97406081bc9 : SUCCESS in 17m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/584a24422e3545e49a2b1d3e6855c62a : SUCCESS in 17m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/66bbf7931ac347d48f1d70ba6129e26b : SUCCESS in 18m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/636cfd3870394fa187f060a4b8f7181d : SUCCESS in 9m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ee27025341864a00b87515a8650da05a : SUCCESS in 8m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/538a083a23874d22b1693a1d8f9872a4 : SUCCESS in 15m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4a092cb37ebd455b96164f8a42f41166 : SUCCESS in 14m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/92029e97a5e44086b8451b2c6e497d16 : SUCCESS in 13m 56s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/681f5ffd77d645f696000182ca801015 : SUCCESS in 13m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cbef7dda08ff47b1b71aad1ab5f0cc74 : SUCCESS in 15m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f60a2a3ea67542acadf60d16142a357d : SUCCESS in 24m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3951b37459e547b3b761d42e104640ef : SUCCESS in 16m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9d84b47ec1d64c8dbfda61bdea771cda : SUCCESS in 13m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a1c960bf41704bd19bced4c1bbd8d743 : SUCCESS in 22m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/91742d9bd70c424db8b1118ab43be5f8 : SUCCESS in 13m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ee60342deb2843f9810e54e266cafdbb : SUCCESS in 7m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/099f3088ad61475db73e212b4d794a47 : SUCCESS in 18m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/32a77f86768542efaf804ab784782ab3 : SUCCESS in 8m 41s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/774afc74b15b46a492c1ed41ec898573 : SUCCESS in 44m 42s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a4ae073ff28947eaaf7df265c185c707 : SUCCESS in 1h 06m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e629c71064844d9f89c6f12c9e80519b : SUCCESS in 47m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ffbad517b2c24edaa92d0845f0302913 : SUCCESS in 54m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b858da5f8fc483380b3c657446f022f : SUCCESS in 15m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d61c708123244efd985efa4a5c737006 : SUCCESS in 4m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c96e5b41c3854c67b03613a27ea87edf : SUCCESS in 26m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59971a64837248c8aae7b0f2802798d8 : SUCCESS in 45m 06s","accounts_in_message":[],"_revision_number":2},{"id":"5ed715deec13d159e381c2a0925e968d6cd4055d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 00:20:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b223fa6ea9a4369a06e1943a1804ed3\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1300229d8fdf467a99cf52342eda808e : SUCCESS in 58m 48s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f09a8b5886104297a2d5570312d569b6 : SUCCESS in 59m 33s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/904705fed069491ea130ce5dbd4840e1 : SUCCESS in 56m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7d4fbd546a184bca9fa34dbbd5f7a062 : SUCCESS in 1h 29m 49s","accounts_in_message":[],"_revision_number":2},{"id":"c3aa1bbdf82049c6e98e3c5494f2c24d67e4c4f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-03-07 01:42:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c6eedb98651c821ff6e59616325ee55d0379be95","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 02:08:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e07b65174320449cb937e112a34e12ec\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/25adb900a0ce4a0cb43b3d8a2e78e685 : SUCCESS in 16m 23s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f945e4daf8ca4ff9bfa6ebb2eee9664c : SUCCESS in 16m 33s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d4e475aedfc34eb88b660bd1b5529042 : SUCCESS in 21m 53s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7af1b89b43c54ab1a2c5b83dda8a7eb1 : SUCCESS in 24m 20s","accounts_in_message":[],"_revision_number":3},{"id":"8898686afd33b8ae8df1bff4bda0d7cb4b3a1927","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 03:10:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0a8b62ae964481b80698aeaa5f95dec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e12e29de3bd40f38e04de7920613b50 : SUCCESS in 7m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/e6e581864cfe4c73ab51b4a8288e210b : SUCCESS in 44m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a9e417c9211b44149a305967d7faa385 : SUCCESS in 35m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2323f6a29d37404eb8dc26e23b6854fa : SUCCESS in 43m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/526c64a7ed754ad09406bdace6bc93c3 : SUCCESS in 14m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5ebdd631f8fb4d55b8564b0fe82fe023 : SUCCESS in 41m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/46e01cb405d24a9e8f039ec632a8c387 : SUCCESS in 17m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ececb496f04a45f1b8259ab5bd5c804e : SUCCESS in 9m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/70b6dc3863af4ccc9c29ef43bd859e46 : SUCCESS in 5m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9d6cddd8002d4d42a08ada7b787ed3bb : SUCCESS in 15m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1e0de08822fa4e059bab43a1035342bc : SUCCESS in 37m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f1af1e98de554f63a5462e7450f55cb9 : SUCCESS in 16m 03s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/61baf7717091439ea7c88d7cb45c7d9e : SUCCESS in 14m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/03abc7c0e23147339fc37b96778089b4 : SUCCESS in 14m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8f92509e2c3d44439ecbeaf5159cad42 : SUCCESS in 13m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4e3f4e2dab0b47cdbe9e12841d9286aa : SUCCESS in 37m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b575c7e59bd64acbb7879fcd806d1569 : SUCCESS in 12m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e62ab8e3a2fb499485ca27532d98ca23 : SUCCESS in 11m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/aa0ef020b61b4c4fb2f10d955968387e : SUCCESS in 15m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/38f15badd06c4042b5744bdb56343b0b : SUCCESS in 5m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/150f0a71c7f044b0b01829bef1dfbd23 : SUCCESS in 13m 40s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/eb7316071c9741c48a71efbd999e16a5 : SUCCESS in 6m 38s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/30636afa99f440acb2ece6a97ab5dd62 : SUCCESS in 1h 00m 26s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/106bb97cfe994ac59a45032bba53ea56 : SUCCESS in 1h 04m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4c681303d1494389b516f472c228f5b2 : SUCCESS in 1h 04m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f72dfe95040d4706a7d095919811b06c : SUCCESS in 39m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/089b05f0f7004c78904361a8b358321b : SUCCESS in 14m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26a0ba6240de4f7fb4cac786a92fe6ab : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dda2b49ed9124001947dc904b8b0073f : SUCCESS in 26m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0306e9705de4d9c92f813b04f5fd8c0 : SUCCESS in 38m 23s","accounts_in_message":[],"_revision_number":3},{"id":"55b8306d1138aae479152befdae4fa3d97714ab7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-03-15 16:00:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"08ae0f0e9310df239501bdd8130aa6bf0e78d70f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 16:23:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76c1718ac72c4d84b9adbc861db885f8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7433d2106f064a9eab80003d33014760 : SUCCESS in 14m 40s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0969651cbfb548bc83592d5a55ba9124 : SUCCESS in 15m 11s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/84011ef625754623be67353256a63361 : SUCCESS in 21m 48s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c012a94a547544c2ac748dbfd2837265 : SUCCESS in 19m 22s","accounts_in_message":[],"_revision_number":4},{"id":"18247d1dc5b0ac1fb9b03ddb8b8354e9f29e7eab","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-03-15 16:25:20.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"aa82337fb598f473a67f3d724dbdad2896ae3a5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 17:18:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e848b946357046949db0855d0ef2f76d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a45aa8eb526240d7b79eddb798e8006f : SUCCESS in 8m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/6ec5275e8b4846dc94ee6169ca092fc5 : SUCCESS in 1h 05m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e66a7b7ffb624552886974584e4725b3 : SUCCESS in 48m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e977ebcf4d6844758eb0b77c27d8032e : SUCCESS in 1h 09m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/96798175cbd349fea82da34888341aae : SUCCESS in 18m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ed5e54acee4b4b9a9b6629c5166994af : SUCCESS in 18m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/01ae34723806425fa0ade25bf196e51a : SUCCESS in 19m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/742c7e1015344a5aa85bcd9160638e9e : SUCCESS in 9m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d672971e00374f329bbc8c3322f60a30 : SUCCESS in 5m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/87cddd819fd4453b91dc32a5976a9f1e : SUCCESS in 17m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/17a3b96f4ca3489897f35a3b1bde3993 : SUCCESS in 16m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0c88aa9053cf4c9aadad0fc6800c1e28 : SUCCESS in 14m 59s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b6c5772533bb4f588a5a292dd00a6adb : SUCCESS in 14m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b1f07e20aecf45458cfbf48ed153af95 : SUCCESS in 16m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7b6e3587893b46b38d954d3f4fbaada1 : SUCCESS in 17m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/23d46e5f8c9c4f01aaccac20f1931ce2 : SUCCESS in 17m 17s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/20ac537ab75a4ad9979a7c2a41284320 : SUCCESS in 14m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a30660a7fb4d47d8a0b42700569c6a7d : SUCCESS in 14m 59s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6b8630427603434babfd03bb62f98314 : SUCCESS in 14m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/477b221b5d684dc8a73a870ff9eca801 : SUCCESS in 6m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e2b942d9aac2494d95782c31c0e05345 : SUCCESS in 14m 59s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b7a5f9def4bf4f58b4083d36ab523091 : SUCCESS in 8m 58s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/61a7eab7bd294657aaeeb99c29545cef : SUCCESS in 1h 01m 25s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/16f5eeaff5434ef89b3473e2b398b6ae : SUCCESS in 1h 01m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bcd887b9ceb84d4ab873b0118701bfc3 : SUCCESS in 48m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/20da054f92164e01ae4de6205cfc3bb2 : SUCCESS in 50m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1ec095c794448658b2afcdecdeb1d0e : SUCCESS in 15m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9e31d1473534a4787f1bdbfa8609580 : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/070478f45a0e451e8e56f4170f3d5f1f : SUCCESS in 18m 02s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/049a692b13d34c4dba0d7a42f134388b : SUCCESS in 37m 05s","accounts_in_message":[],"_revision_number":4},{"id":"ee3f2f3a491000b766fa1db1f86cccdaee79e9d1","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-03-16 09:53:32.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2ab4f3450778faebda232fbccccb9c31248bbb8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 09:54:10.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"fa8ce4dd1e421f827dcb5b6f154f35d664f8f009","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 11:52:33.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/243c47b776024aaab456726d642528c0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f271c183b5c64098800b84c42030f681 : SUCCESS in 9m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/de90acea822f4867b89307eca0a012d6 : SUCCESS in 1h 00m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b60047fd480342bd8f10d1010d9d0970 : SUCCESS in 45m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/98b21f367cff47ea9e6604ad47082027 : SUCCESS in 1h 02m 47s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/402dd69aca804b869257419ea85f535b : SUCCESS in 8m 50s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/1f996c5f66484cc68c5ab2e22725f245 : SUCCESS in 5m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e6d7f46743e445beabe32f189e5cbfb1 : SUCCESS in 17m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2d68eb7df7874dbeae2d4e5ad44fdd25 : SUCCESS in 17m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b98e5f90c587495682379d8aaadea038 : SUCCESS in 16m 16s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5ff873dfecf9489fbd65b7f10fee5302 : SUCCESS in 17m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bd0a1c8a7a9641548d06889a52824b46 : SUCCESS in 16m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8b1cb64e6dd84cb2a35abf9e954dce10 : SUCCESS in 16m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cf8df384d119460db60e0ae3ba68767c : SUCCESS in 21m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0e5f3f5df8ed4979b66075a07615982e : SUCCESS in 13m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7e45dcb46c634f7baebc8cc8367f005a : SUCCESS in 13m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/74976d53a1a648cba072bca856e9af8c : SUCCESS in 14m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bbc3e8bca1dd4ccab2a5dc5515ca86b1 : SUCCESS in 8m 51s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/638ad4f9c58b40d794e39770d375a160 : SUCCESS in 8m 19s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ab5756ed2d994265843c73d0c656c240 : SUCCESS in 1h 02m 54s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/82391f04ae48430e88f7ce98068f0207 : SUCCESS in 57m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/757f64de997d49478ea7a6b9510ca110 : SUCCESS in 49m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/32100e7cb6964775972401095aa9514b : SUCCESS in 54m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e588b8f0d0944eab994d57b74c212e41 : SUCCESS in 16m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ec071873a5e466b9feb5a77292d89de : SUCCESS in 5m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce73178b05c64fc994f0164d3e74e7b4 : SUCCESS in 50m 50s","accounts_in_message":[],"_revision_number":4},{"id":"decba96fd13b7c7d49b79fa48de252b42faf0b7d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 11:52:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"8101cbfb4a4f1ddffb3fef24505f9bf542a768d8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 11:53:45.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/184698f16f4e4ca5bcbef939248db2a8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddc9f86797f44a0c96ec03559946ca12 : SUCCESS in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/7f579d50d1f447a086e9b166c33a38a5 : SUCCESS in 39s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/8012e1a2df824de288b92d3ff35745a1 : SUCCESS in 37s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"acf31a61dbd54bcc4440080900b96ac4fffe2074","revisions":{"5f0df2d51b7839033c486d83dabe4623849d66fe":{"kind":"REWORK","_number":1,"created":"2023-02-01 15:08:01.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/872417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/872417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/872417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/872417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/872417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/872417/1"}}},"commit":{"parents":[{"commit":"a56ddb39f9357d405269d52f47bd3398487b949e","subject":"Proxy: restructure cached updating shard ranges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a56ddb39f9357d405269d52f47bd3398487b949e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-02-01 15:07:20.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-02-01 15:07:20.000000000","tz":0},"subject":"follow-up: move some unit tests to TestNamespace","message":"follow-up: move some unit tests to TestNamespace\n\nplus some other fixups\n\nChange-Id: Ibd5614d378ec5e9ba47055ba8b67a42ab7f7453c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f0df2d51b7839033c486d83dabe4623849d66fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f0df2d51b7839033c486d83dabe4623849d66fe"}]},"branch":"refs/heads/master"},"9c5c93646b79334aaf2c1848f268a8d974618dc5":{"kind":"REWORK","_number":2,"created":"2023-02-01 22:48:12.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/17/872417/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/872417/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/872417/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/872417/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/872417/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/872417/2"}}},"commit":{"parents":[{"commit":"a56ddb39f9357d405269d52f47bd3398487b949e","subject":"Proxy: restructure cached updating shard ranges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a56ddb39f9357d405269d52f47bd3398487b949e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-02-01 15:07:20.000000000","tz":0},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-02-01 22:34:42.000000000","tz":-480},"subject":"follow-up: move some unit tests to TestNamespace","message":"follow-up: move some unit tests to TestNamespace\n\nplus some other fixups\n\nChange-Id: Ibd5614d378ec5e9ba47055ba8b67a42ab7f7453c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c5c93646b79334aaf2c1848f268a8d974618dc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c5c93646b79334aaf2c1848f268a8d974618dc5"}]},"branch":"refs/heads/master"},"3f471ed30ffa55e4284210088e26c3d8de7ef535":{"kind":"REWORK","_number":3,"created":"2023-03-07 01:42:51.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/17/872417/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/872417/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/872417/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/872417/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/872417/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/872417/3"}}},"commit":{"parents":[{"commit":"5da09aefdf9222dde45cf6bc80e77af7453cc611","subject":"Proxy: restructure cached updating shard ranges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5da09aefdf9222dde45cf6bc80e77af7453cc611"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-02-01 15:07:20.000000000","tz":0},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-03-07 01:41:57.000000000","tz":-480},"subject":"follow-up: move some unit tests to TestNamespace","message":"follow-up: move some unit tests to TestNamespace\n\nplus some other fixups\n\nChange-Id: Ibd5614d378ec5e9ba47055ba8b67a42ab7f7453c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3f471ed30ffa55e4284210088e26c3d8de7ef535"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3f471ed30ffa55e4284210088e26c3d8de7ef535"}]},"branch":"refs/heads/master"},"acf31a61dbd54bcc4440080900b96ac4fffe2074":{"kind":"REWORK","_number":4,"created":"2023-03-15 16:00:37.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/872417/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/872417/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/872417/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/872417/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/872417/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/872417/4"}}},"commit":{"parents":[{"commit":"6ff90ea73ebf4771efba3928ea7667b7ae731c71","subject":"Proxy: restructure cached updating shard ranges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6ff90ea73ebf4771efba3928ea7667b7ae731c71"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-02-01 15:07:20.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-03-15 15:59:14.000000000","tz":0},"subject":"Rename ShardRange*Bound to Namespace*Bound","message":"Rename ShardRange*Bound to Namespace*Bound\n\nAlso:\n\n  - move some tests to test_utils.TestNamespace.\n  - move ShardName class in file (no change to class)\n  - move end_marker method from ShardRange to Namespace\n\nRelated-Change: If98af569f99aa1ac79b9485ce9028fdd8d22576b\nChange-Id: Ibd5614d378ec5e9ba47055ba8b67a42ab7f7453c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/acf31a61dbd54bcc4440080900b96ac4fffe2074"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/acf31a61dbd54bcc4440080900b96ac4fffe2074"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
