)]}'
{"id":"openstack%2Fswift~872696","triplet_id":"openstack%2Fswift~master~I3880dfabe784f7b63dd31777ec7abfe54acf30fc","project":"openstack/swift","branch":"master","topic":"sr_cache_benchmark","attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2023-02-03 21:22:15.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I3880dfabe784f7b63dd31777ec7abfe54acf30fc","subject":"Add micro benchmark for \u0027_get_update_shard\u0027","status":"NEW","created":"2023-02-03 20:27:14.000000000","updated":"2023-02-07 10:40:54.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"238960e5c030e85cdd30334197e8a50e8ffa0930","_number":872696,"virtual_id_number":872696,"owner":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-02-07 03:50:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-1,"date":"2023-02-07 10:37:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-03 21:22:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-07 10:37:36.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"008793a82ef34be2d214dbda0bcb11907b4da813","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-02-03 20:27:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d26598ee76c4a542f24f57edb090e93a40fbb121","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 21:22:15.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea87b54e1b1b4c0db7f0ca40e45f2b28\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be07e5c517ab4eed8fd922670b63af7d : SUCCESS in 9m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bd605456c7cf468ea2253f0415e769ea : SUCCESS in 14m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/254baa8413414e55b4ca5c020a5f6d0f : SUCCESS in 19m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c7cc266e71864a939d2843e382509241 : SUCCESS in 15m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/312e3f644a094657b944322ce1175563 : SUCCESS in 8m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e280ceaa6d6647709e8f6346b08be20a : SUCCESS in 4m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3b24abe260c541a4b904755f61eb1faa : SUCCESS in 21m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a58bc694c59f4797bc56f0368d02a819 : SUCCESS in 18m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fce2c6f899b541bd9050acba05a40678 : SUCCESS in 17m 58s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7dbd4a0a5bc64b0c8a1b0c7826d925ba : SUCCESS in 21m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6cca3ca6517848b5a476369fecfcd844 : SUCCESS in 17m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f057e2bbc8ee40bdaf80d719bf5b21f4 : SUCCESS in 20m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f52e288b4c2e42bfa8d352a48b93f6c9 : SUCCESS in 17m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/39106b7be78b4a2f849e90cf5aa1e2c4 : SUCCESS in 12m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/882dd85ca52549ca8da81b3cf5a52bda : SUCCESS in 20m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4ebc9d230c734dfe8f08c7fe38d536e7 : SUCCESS in 18m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/84b7266f66054759b773fed9cddb7eca : SUCCESS in 44m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7a916cdb64ff4bc4bf4c4a8ad75f634f : SUCCESS in 45m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ae66013ebdd403080b58bd971a23dc7 : SUCCESS in 17m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f83c393fdd9496ab8f38e3f7bbd459e : FAILURE in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ff4beb99738d454dbcfe650b604fc7cc : SUCCESS in 25m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c9dc0fde62903ae6d115ce8a50aeb2f63ab9349d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 21:53:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4980ef23e12a4996b06db5444e797265\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/67b8cbf47687494ea476ba5f5fee1ed2 : SUCCESS in 16m 18s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2d79ccc3f4fe4ac08d3f8c84f2cafec9 : SUCCESS in 50m 51s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0ff7fcb3fb8d45dfa83a7c630274367f : SUCCESS in 1h 24m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/698f31a5121d456ca0adb79b5b0d2a0e : SUCCESS in 22m 11s","accounts_in_message":[],"_revision_number":1},{"id":"d07414a9309927ef81637bf017bb92a328ac070b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-02-03 23:56:53.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7c950b24574e6cd6af9f87155c92422bc8c6b25d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-04 00:19:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2af72ac9f514c6c9b577ba1693d1797\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9670db2f68e84158804e3f7af0b983ff : SUCCESS in 15m 41s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c12c3a724af549c388b322fa33fdc8d2 : SUCCESS in 16m 48s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0370addce06243049510e0fdfdb61a32 : SUCCESS in 21m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/65c926b10d334e5fa8df62dc0e49268d : SUCCESS in 18m 57s","accounts_in_message":[],"_revision_number":2},{"id":"c9ad8f2ef645c666959030198bac7d75415666d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-02-04 00:27:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"36d4ca29167917d2ab691ab67449bc2ae1dbf1b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-04 00:50:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c48aaa782281487b9551d9dbb84c3ab4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1daa5f891d484888aa7a20e394989c02 : SUCCESS in 14m 13s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/56dfcbae532f40908b2a75dcdd3d4270 : SUCCESS in 15m 02s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/27ce92a2bec749fe9b7fffa1fef7568e : SUCCESS in 22m 22s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/baeba81366744f7b9d12a8c5c467faa3 : SUCCESS in 17m 06s","accounts_in_message":[],"_revision_number":3},{"id":"1d755e5f51eb57829ffc8a5195e1436f62f78f55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-04 01:15:25.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b346a1e1a77845f28d4530ba8485f641\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4112f73f86c64caca4dbb11ebfc2ce56 : SUCCESS in 7m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8d1940dc230d4a43a5a01f7b0c19071a : SUCCESS in 14m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/01107e5ce363482c882289cb25001db8 : SUCCESS in 14m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/31d2de59c4ed40de9b6fd029c17c4f93 : SUCCESS in 15m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6e629076c76842dd9387609e69e905fe : SUCCESS in 7m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8633e7302af342fa947be3c2289e466a : SUCCESS in 5m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d2cbaf864b6945eda42b34acaba8fdad : SUCCESS in 17m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b53912c5cf2a4e859cc0ad6bbbb5abb3 : SUCCESS in 12m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3d8017f4e8aa417c85815a840bc285b9 : SUCCESS in 18m 50s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9dce99d445454917a801895ce1b6521b : SUCCESS in 14m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/637a55465d53453691070ccbe43fabd6 : SUCCESS in 17m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/801d5ea983bc47489594cb427dbdde19 : SUCCESS in 15m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3968ce8dfaa84527ad29e86ada61f4f8 : SUCCESS in 17m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/77d6406c969a4802a3996ad98233a032 : SUCCESS in 18m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/518dee987ab74281b6c4539a74ed78e0 : SUCCESS in 13m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/96ff44f3bf6c4a78aa391d37374b62bc : SUCCESS in 19m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e74d74bb357d42de9f10ee8a9febcccf : SUCCESS in 46m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/30ef1726e0f5422e8f8aec808c048324 : SUCCESS in 45m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26b2bab9cb234c94aa8bff70519d241b : SUCCESS in 16m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c9d7ea4e7b640c8b91f7c7f8420f4e8 : FAILURE in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/27a4139d4b6a4756b55c9a4660adc87a : SUCCESS in 18m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1b626473bea760e720195447f0fb36d7de94a8b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-02-07 02:09:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"28ea606753de1a451cddf96924397f828f9b9921","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 03:09:20.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/331f97ffdd9e45fa84226c2d3fe3326a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c1c07b16c924497ea64730ceefe5c0f3 : SUCCESS in 15m 40s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/336fe02234d34ef0a57a4af9d0c67ab6 : SUCCESS in 52m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f88d1bcee1047f5a6b7a1051c5cac65 : SUCCESS in 18m 44s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/04bd196c480c41a090ceea331c4e6e80 : SUCCESS in 50m 24s","accounts_in_message":[],"_revision_number":4},{"id":"5f463bbe568e9ad5ec1365ad3e3eb071c014ea73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 03:50:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/073da264556041008ec9057752688ccf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38c0caf1d0ce4114b6feb41e62312298 : SUCCESS in 9m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/80478aa084384a30a04967f7f326fe26 : SUCCESS in 16m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/285c8525a5274021942f11a5737a32e4 : SUCCESS in 16m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/449eeb383f314c4bb215802e3c76011c : SUCCESS in 17m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a96df9de932e4b94848c508d754762d7 : SUCCESS in 8m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/fc3e024f941b466e963b6eb4aaecf7b8 : SUCCESS in 4m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0cb6d00348f2462295acf48898beb627 : SUCCESS in 17m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d6fa757492eb40088b7fae7209b11665 : SUCCESS in 18m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e7b6e91d337240939f64dae6506b1fba : SUCCESS in 15m 26s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/eb28c7f80a6f4450a99ff25c53657295 : SUCCESS in 13m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/87fe6a23fe704c61a412bab0cb071df7 : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1c36b78e24064e9c93ac82ad338ad35d : SUCCESS in 17m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6c44339434f44364abb00942afc2351d : SUCCESS in 15m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3fba6ff3e03d4934a0fafce2fd0b89bf : SUCCESS in 11m 20s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/03a62c70128b4298982b533ccdc8ce7b : SUCCESS in 16m 59s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1e75e518e0af4ac7980dd987a73837db : SUCCESS in 13m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2dff8076fa9d4bdd8e99de98af843ca3 : SUCCESS in 53m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9de2b5d6536e4e81bbdbfcc5baba4286 : SUCCESS in 47m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec9d28b2951f41f5a7e1bb1ed7788cab : SUCCESS in 17m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5703a8e224b342ef9a80a71e2e4d544c : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/94ddbd1cf744413c81d540976ab697b6 : SUCCESS in 19m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c6ce1536cc98e4d6ecead78e28eeaed441888ab8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-07 10:37:36.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"238960e5c030e85cdd30334197e8a50e8ffa0930","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-07 10:40:54.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"73e3a89809c8490d96a5fe35963aa302e88a96d5","revisions":{"f9319cee9210657ab2468d4f6aba46a122a5471f":{"kind":"REWORK","_number":1,"created":"2023-02-03 20:27:14.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/96/872696/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/872696/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/872696/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/872696/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/872696/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/872696/1"}}},"commit":{"parents":[{"commit":"a56ddb39f9357d405269d52f47bd3398487b949e","subject":"Proxy: restructure cached updating shard ranges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a56ddb39f9357d405269d52f47bd3398487b949e"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-02-03 06:34:18.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-02-03 06:34:18.000000000","tz":-480},"subject":"Add micro beachmark for \u0027_get_update_shard\u0027","message":"Add micro beachmark for \u0027_get_update_shard\u0027\n\nChange-Id: I3880dfabe784f7b63dd31777ec7abfe54acf30fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9319cee9210657ab2468d4f6aba46a122a5471f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9319cee9210657ab2468d4f6aba46a122a5471f"}]},"branch":"refs/heads/master"},"043f5fc25063e15dedfa342c41a15176e5516a1a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-02-03 23:56:53.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/96/872696/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/872696/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/872696/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/872696/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/872696/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/872696/2"}}},"commit":{"parents":[{"commit":"a56ddb39f9357d405269d52f47bd3398487b949e","subject":"Proxy: restructure cached updating shard ranges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a56ddb39f9357d405269d52f47bd3398487b949e"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-02-03 06:34:18.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-02-03 23:56:30.000000000","tz":-480},"subject":"Add micro benchmark for \u0027_get_update_shard\u0027","message":"Add micro benchmark for \u0027_get_update_shard\u0027\n\nChange-Id: I3880dfabe784f7b63dd31777ec7abfe54acf30fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/043f5fc25063e15dedfa342c41a15176e5516a1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/043f5fc25063e15dedfa342c41a15176e5516a1a"}]},"branch":"refs/heads/master"},"6319d2ce64aa9768b8bbfa41b99b62504b1afbd8":{"kind":"REWORK","_number":3,"created":"2023-02-04 00:27:09.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/96/872696/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/872696/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/872696/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/872696/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/872696/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/872696/3"}}},"commit":{"parents":[{"commit":"a56ddb39f9357d405269d52f47bd3398487b949e","subject":"Proxy: restructure cached updating shard ranges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a56ddb39f9357d405269d52f47bd3398487b949e"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-02-03 06:34:18.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-02-04 00:26:59.000000000","tz":-480},"subject":"Add micro benchmark for \u0027_get_update_shard\u0027","message":"Add micro benchmark for \u0027_get_update_shard\u0027\n\nChange-Id: I3880dfabe784f7b63dd31777ec7abfe54acf30fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6319d2ce64aa9768b8bbfa41b99b62504b1afbd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6319d2ce64aa9768b8bbfa41b99b62504b1afbd8"}]},"branch":"refs/heads/master"},"73e3a89809c8490d96a5fe35963aa302e88a96d5":{"kind":"REWORK","_number":4,"created":"2023-02-07 02:09:58.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/96/872696/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/872696/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/872696/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/872696/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/872696/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/872696/4"}}},"commit":{"parents":[{"commit":"81bcd49b79db159cf362260b1f73341d81676ab8","subject":"Proxy: restructure cached updating shard ranges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81bcd49b79db159cf362260b1f73341d81676ab8"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-02-03 06:34:18.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-02-07 02:09:43.000000000","tz":-480},"subject":"Add micro benchmark for \u0027_get_update_shard\u0027","message":"Add micro benchmark for \u0027_get_update_shard\u0027\n\nChange-Id: I3880dfabe784f7b63dd31777ec7abfe54acf30fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73e3a89809c8490d96a5fe35963aa302e88a96d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73e3a89809c8490d96a5fe35963aa302e88a96d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
