)]}'
{"id":"openstack%2Fswift~872757","triplet_id":"openstack%2Fswift~master~I7dd5f3e821c60e8f1a6a21c19643c9c58081395a","project":"openstack/swift","branch":"master","topic":"bad-status","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-02-09 00:37:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7dd5f3e821c60e8f1a6a21c19643c9c58081395a","subject":"test for unmounted w/ suppression_count","status":"MERGED","created":"2023-02-06 14:52:45.000000000","updated":"2023-02-09 00:39:10.000000000","submitted":"2023-02-09 00:37:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872757-bad-status","meta_rev_id":"27ab4ca64e1ad05db4db3925d53884030125ed09","_number":872757,"virtual_id_number":872757,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-09 00:37:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-08 21:59:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-08 21:59:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-06 15:22:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-06 16:40:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-08 21:59:24.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"219d36b42c48a99627af284aaeb9e70a66d73abf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-02-06 14:52:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24a8b2053a2911fbacdd8b953c282d79108a60c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 15:22:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bff3fd171ca486bae4d70879cd30b9a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0ee373f94f4e4b9eb5ee6c2f54623f5b : SUCCESS in 22m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8bd375d889d945a8a72f46bcf7f734e9 : SUCCESS in 25m 16s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/776855728b81437a93af0cbee3fba256 : SUCCESS in 28m 03s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8ceacd06956147748116d91b85afec69 : SUCCESS in 28m 02s","accounts_in_message":[],"_revision_number":1},{"id":"c7e10f8b958dabb25ee7e6a7423b0d7ca82a225f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-02-06 15:26:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5bbd25c473bf46fcbedeadd320a00a8edff82dd1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 15:55:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b1cbba19c8f44d8a72b571cbec58103\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7045a0700ade4bfd9f0d71ee9cf10f61 : SUCCESS in 25m 45s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4cb4fa6af8f54ac1892c23314a2eccdc : SUCCESS in 20m 48s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/fc595e5de93d4ce382328e49f7e4ad37 : SUCCESS in 27m 29s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9a06d6289021413cb7ec8fdf66944ed4 : SUCCESS in 24m 45s","accounts_in_message":[],"_revision_number":2},{"id":"71f04d5098a4115c4d085f1796bd65a697731dbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 16:40:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44ee20f63077452eaa606c001d02f937\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5cf07142c844e8bb701c4132124fe08 : SUCCESS in 9m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a2acd989dcdd43388abab31041c1bd58 : SUCCESS in 25m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ad67c29c53ef45b4a066f5b4ed5294cc : SUCCESS in 18m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1b83dc13e8944df1b79639471e4b424f : SUCCESS in 20m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4fffbecb96aa4b30afc914b0450b68e4 : SUCCESS in 11m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b3a6c3e18a1b4f35ade4e34b73212a9b : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7fc06adaee1543e1a577997b1201b90e : SUCCESS in 24m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/81a65071a5f94cd49021ce090cda67f7 : SUCCESS in 17m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f108ebbee26a479dbf3d0a2b784194f4 : SUCCESS in 14m 12s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/51643e3dbe784153b0e3bfa639b26d62 : SUCCESS in 15m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6bacffb9faf74405b70a5fea32a28c40 : SUCCESS in 34m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/188ca25937e54e288ef4338297795033 : SUCCESS in 17m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2c8a33d6466047fd91a51a4035ecab7a : SUCCESS in 16m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/af7fa876021b4a31a01a2cd90fe1e599 : SUCCESS in 25m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4b91c577d5ee4a9fb749a59c0dc3b97b : SUCCESS in 16m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/83df2bb6b2044032b96b53c60ed68c09 : SUCCESS in 31m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7dff551f3d3e4cdd97b6c88ff28b0f58 : SUCCESS in 47m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b8d4f68a51f44924b8ab694c0de7b7c4 : SUCCESS in 31m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/061196e2892a4fb1ac4f70cd96860137 : SUCCESS in 18m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f41ea51ac3b94b48ad55ba7251fc5eda : SUCCESS in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/76567567ca61457bac385cceb86a446c : SUCCESS in 17m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a7d9457aaa30b8944a43e3c789dd205f562c0b79","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-08 21:59:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"13b14e568464c3b7c43a8b6a3eeeab3ee1685130","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 22:00:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"58167892146a4c60d2cfc2eeacd8eb4433707671","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 00:37:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb296cf0193c4d89994ee5d52def6a09\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99c377c638514eb29c2542c86b79ee07 : SUCCESS in 9m 07s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/769eecba73784159aa1e5d33cb5a3158 : SUCCESS in 9m 44s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/8125a81b132d477f8edf73201ba19b22 : SUCCESS in 5m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0a38853ae60c43369701faa8ac65250d : SUCCESS in 18m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d515dd0acb774c91ab8215f0827250d9 : SUCCESS in 15m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/88a01a0a64d445a6906ad1adbc898d8b : SUCCESS in 15m 09s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/cdff9611baff42409942420b9c9c7c5f : SUCCESS in 15m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0a6213ec84ae4674983694d4540e8016 : SUCCESS in 15m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/33e7518d36fa4cb3a810bd9226d156ae : SUCCESS in 17m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a096df58a7764346b85e74b0f477f4ab : SUCCESS in 19m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/52c516d523dc4533b396de81bce4be81 : SUCCESS in 13m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/dcc328de21384d98ae32415aad31f38b : SUCCESS in 14m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f43c12f5159145f1850ade3558528c1f : SUCCESS in 15m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/eccd2b09cb164b9b8d5635bd214fbb77 : SUCCESS in 7m 13s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/81bc64107a714684b1333c83c17a63df : SUCCESS in 8m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5432791f20ef48ab9bec427dfa4391aa : SUCCESS in 57m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ebf9830a36ab449c96b18e1ad97c252f : SUCCESS in 1h 02m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de451702647a4609a9a78f3585f4b67f : SUCCESS in 17m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf8062bb595d43669772e0bd76ea1114 : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":2},{"id":"4c711303227d53fbc39781c52387a98064d3d044","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 00:37:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"27ab4ca64e1ad05db4db3925d53884030125ed09","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 00:39:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c39ff8e992b411eade215dddc0b7688\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e4f8dc88b7246d3b414988f89a45275 : SUCCESS in 42s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/e60a2508847341b4b7cace042b7b9def : SUCCESS in 30s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/e070ee0f02e247ff8cd64ac58a1732c3 : SUCCESS in 30s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b02df75ed4c847f1f43b7f2142d020d6bccd8ef4","revisions":{"a72eda3554ce34095018d23d992cc85534bbf979":{"kind":"REWORK","_number":1,"created":"2023-02-06 14:52:45.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/57/872757/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/872757/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/872757/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/872757/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/872757/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/872757/1"}}},"commit":{"parents":[{"commit":"488f8c839f034172d20b3f28ca851d83b7a3bfff","subject":"tests: Fix some func tests to do with metadata maximums","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/488f8c839f034172d20b3f28ca851d83b7a3bfff"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-02-06 14:52:39.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-02-06 14:52:39.000000000","tz":-360},"subject":"wip: test for unmounted w/ suppression_count","message":"wip: test for unmounted w/ suppression_count\n\nChange-Id: I7dd5f3e821c60e8f1a6a21c19643c9c58081395a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a72eda3554ce34095018d23d992cc85534bbf979"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a72eda3554ce34095018d23d992cc85534bbf979"}]},"branch":"refs/heads/master"},"b02df75ed4c847f1f43b7f2142d020d6bccd8ef4":{"kind":"REWORK","_number":2,"created":"2023-02-06 15:26:13.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/57/872757/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/872757/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/872757/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/872757/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/872757/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/872757/2"}}},"commit":{"parents":[{"commit":"488f8c839f034172d20b3f28ca851d83b7a3bfff","subject":"tests: Fix some func tests to do with metadata maximums","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/488f8c839f034172d20b3f28ca851d83b7a3bfff"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-02-06 14:52:39.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-02-06 15:25:45.000000000","tz":-360},"subject":"test for unmounted w/ suppression_count","message":"test for unmounted w/ suppression_count\n\nChange-Id: I7dd5f3e821c60e8f1a6a21c19643c9c58081395a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b02df75ed4c847f1f43b7f2142d020d6bccd8ef4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b02df75ed4c847f1f43b7f2142d020d6bccd8ef4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
