)]}'
{"id":"openstack%2Fswift~873222","triplet_id":"openstack%2Fswift~master~I1189a60204cf96c291619f8d8ec957ed8a5be1ce","project":"openstack/swift","branch":"master","topic":"remove-netifaces","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2023-05-24 07:00:04.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2023-05-23 22:03:21.000000000","reason":"\u003cGERRIT_ACCOUNT_597\u003e replied on the change","reason_account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-05-24 09:42:55.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-05-24 11:34:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1189a60204cf96c291619f8d8ec957ed8a5be1ce","subject":"Remove hard dependency on netifaces","status":"MERGED","created":"2023-02-09 06:22:36.000000000","updated":"2023-05-24 11:35:26.000000000","submitted":"2023-05-24 11:34:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":28,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"873222-remove-netifaces","meta_rev_id":"8a0194094f179c22f58847e78215ff753bfc0f53","_number":873222,"virtual_id_number":873222,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-24 11:34:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2023-05-24 07:00:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2023-05-23 22:03:21.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":2,"date":"2023-05-24 09:43:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2023-05-24 09:43:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-09 07:28:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-09 08:04:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-09 23:03:52.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"},{"updated":"2023-05-11 21:41:34.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2023-05-19 07:13:16.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2023-05-22 17:55:29.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2023-05-24 07:00:04.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"dba51d7d518df7d36a9372a52d1154e02611dd00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-09 06:22:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7abc3c3d24c083cbbfe3e713e974145cd70dd0d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 07:28:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38798d45ac004dcdb887c04de689b366\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f1a331d2b1d64fdeaec7eebfd6d1ce5c : SUCCESS in 12m 46s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d3f2abd415b044f79b1b7342aee0e2b8 : SUCCESS in 14m 42s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d611128507264565ad22edb37e70e8b6 : SUCCESS in 22m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/172f1d6bb08c4bad9510bb82e77f5dbf : SUCCESS in 1h 04m 11s","accounts_in_message":[],"_revision_number":1},{"id":"b79d464cb033a4ad286c43d9071466560a36085f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 08:04:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/89debc4e15cc47369d19cac2235d95a5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db68e8ece7694da697a051a6f3b7a5df : SUCCESS in 11m 29s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b458ec09db3d4ef4933d37011b68034f : SUCCESS in 3m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/2eb9b7bb26c4400389901e6b0f240182 : SUCCESS in 1h 35m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ea712ececc224146993e9613345010f2 : SUCCESS in 41m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b48dec6b3b14af38be530fe6ee640b1 : SUCCESS in 1h 02m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/06610699d0ad478f9a16b3730f9fca05 : SUCCESS in 17m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/342a4164bcf84a4bb5bf2637dc4448bd : SUCCESS in 18m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c9619bf4ffec46d69d07be9d13d47ea3 : SUCCESS in 18m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c10bde8968c84e868512fb4c0685293b : SUCCESS in 9m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0e26082fbc184e33be7015aefdb40f53 : SUCCESS in 6m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3959a676fc8745e89480a0189db04db4 : FAILURE in 19m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4ab56501c6b34a1c891beb730c44f6ef : SUCCESS in 16m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cba4a12e452f4e399c21d60418c62c37 : SUCCESS in 14m 06s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b92aae7957034159ab75f217ec8d84b6 : SUCCESS in 18m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/729dc6b9ad254d8ebe65229ad5c33a21 : SUCCESS in 19m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0b533e1713524b718536f185f36fb2f3 : SUCCESS in 25m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2b2751fcf20a46d7a95844fd0ddee8bc : SUCCESS in 20m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4a64450a189c42fdb869e74be2d5555c : SUCCESS in 14m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2ff4274d1840411a9ea8367c6370d182 : SUCCESS in 23m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1e051c7f1fed4fcfae9caaac3dbb1975 : SUCCESS in 16m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c6103344bf3446fdb9782736144e390c : SUCCESS in 6m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7058f926ac9a498486176850f19ebcd1 : SUCCESS in 14m 48s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6e3924f51d544e26b4b6f28a2c99403b : SUCCESS in 6m 25s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c05d6b6479234870861e16a56e666f81 : SUCCESS in 1h 02m 13s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/841594e78b464837947731df60a43bb2 : SUCCESS in 1h 02m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c45b7658512441a2b553016f8b1af678 : SUCCESS in 53m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9f1ef1f4c41b4b79a243aab332a9a029 : SUCCESS in 52m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8b7770bef52456e9e5dfb6e3b175153 : FAILURE in 16m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f42a89edb5f1461aa7331b1febf40fdf : SUCCESS in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a581c21cd7e74435a8f36cbc46540163 : SUCCESS in 26m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6362e62121dc4865a1ffeb5acc0ebaef : SUCCESS in 46m 14s","accounts_in_message":[],"_revision_number":1},{"id":"331b85474d5669db1c9f43611bc1ff76a20b9cf5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-09 20:29:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"92291e8724b6b99856ddfc8e08cbee1db6ceb70d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-09 20:39:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8591752b8df6ded53c59ec54a50f06685e3f5e0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-09 20:49:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b8f6e1ef3e163c17026876f7c0097908840dba8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 21:14:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1c6191225ba4afc8bea1441094ad3ed\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/910074d491e948938f630e998fb4a8ea : SUCCESS in 14m 28s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9982b254bc87466b8ef793a48edb566d : SUCCESS in 14m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6644bc3f87744af287e8b78ba6bcd449 : SUCCESS in 21m 27s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2a6eca84a80343f09c2322229c7d592f : SUCCESS in 18m 44s","accounts_in_message":[],"_revision_number":3},{"id":"0f065e285828cbc677466cf67670676b07ba2a74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 22:09:06.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7dd257e12ceb4081be6c696f6a167f55\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/039bc6f60e4e4ffcbc8ec497a4b4146a : SUCCESS in 8m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/eb167959a5f54f4eb76141f31783a466 : SUCCESS in 2m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/ee903e0dc15a4b36b15edfaa32a27b91 : SUCCESS in 44m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0ad83cdc620b43edaa19e375234a2eca : SUCCESS in 41m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00736adbb013453588717c6e8f267fac : SUCCESS in 52m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/33dd7af8e62049749f54b9ffd2cedeee : SUCCESS in 18m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/723e369c00c84e9abd6245c4d77a6961 : SUCCESS in 18m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/201fa47749f644749aaaa3d8b272a319 : SUCCESS in 18m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/035c0599eb94467a99950609c57f7caf : SUCCESS in 8m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/669234d849394d9c8a6cecf5eca9273c : SUCCESS in 5m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2d20994273b6443b9db2af5b76a305b4 : FAILURE in 13m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3496bb0ad5664a72919630f9201929f1 : SUCCESS in 15m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0701e4ed6430401098c4df018fb7fab5 : SUCCESS in 12m 14s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/560492a4c9ff45c9850a10a77b990d26 : SUCCESS in 13m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3d89765cdcc64e68a230dd11b1abbca5 : SUCCESS in 18m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7059b2ae04504a41aaf2a0bbbda72fa9 : SUCCESS in 15m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d2acb39207dd4465890e8e6e37e1a1c8 : SUCCESS in 47m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ca7cfe5a551b4c818d2c841d5f7b68e8 : SUCCESS in 21m 01s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/83b2d62181124f7ab9873f050e1fbdfd : SUCCESS in 40m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6ae347b94640401a9e004c5aea6d0ac8 : SUCCESS in 13m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/072334567ec142509af21f624e74bb08 : SUCCESS in 6m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9596643eb25e4a458edd889bbefc6e1c : SUCCESS in 38m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/43b8ac1f5b46487f960171bc3dc51f9e : SUCCESS in 7m 34s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0f4fad49adff493eb5219bf317f05911 : SUCCESS in 59m 12s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/04a70a25391840099ea0fafd791f34f8 : SUCCESS in 58m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a24fc8e122464e7887a8c6dca332c01e : SUCCESS in 38m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/859700b3a75f43468ac1be66ac83ac06 : SUCCESS in 38m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f85bb8fe9ca54a199bd5384de46c75c1 : FAILURE in 13m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45a86229bfc64b7e815e84cd5c408091 : FAILURE in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7bc7e972d15c4150a68bdb15841443f1 : SUCCESS in 19m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/edbbf34f5f4a480f97b508c6e32ebe88 : SUCCESS in 55m 07s","accounts_in_message":[],"_revision_number":3},{"id":"5d409f18fcc9adde54d205215f095e68e0591bfb","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-02-09 23:03:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fd4b06477e8f5165e5322279e17f6b2f38f8065b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-11 20:25:19.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"2fc4a11861ca3bce826a8ccf9f3f364baf599eb5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-11 21:02:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b272129997a2c3948eace700d4dacf88a3b167e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 21:22:58.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bc82add7bc44186b4d45a7742176212\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aaa16a5ccb56475c8b1767b1ad48a57e : SUCCESS in 19m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cdd4978b2c8a4cd6a00a345fc046f2c5 : SUCCESS in 19m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/76069ee8a4634d3a9dabdeacb1be87ca : SUCCESS in 25m 17s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d383d4a5476a4ea1a29e2280b4d52750 : SUCCESS in 55m 19s","accounts_in_message":[],"_revision_number":4},{"id":"42247588e43200e967ed8cd9faf84998df278262","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-05-11 21:41:34.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f144bbf90eb8e8e2c07f19510c2b21ad0675fff0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 21:50:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c6431d2fdfa4a56844f4daffd752246\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb3fb7e7ada549e681096635d766f202 : SUCCESS in 7m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/40afeac9fa404b168003ec600b0b3a16 : SUCCESS in 2m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/5452230408964c8a90974ce301ec3c2a : SUCCESS in 45m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/eaa2ab9e2c7e4c10913982dfbffaf812 : SUCCESS in 37m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e2dc57a2b174b46b120fc1f33d7d69d : SUCCESS in 1h 08m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7d29a41cfe284c3baf4bc36fcdbe2b20 : SUCCESS in 24m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/73a893c8ab0f4fe0a5cb412685f3ea80 : SUCCESS in 18m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f4f9a9f1f3494c8fa97ef35adde0c799 : SUCCESS in 25m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/872f3f1e96a64f88bfda3acf97dbb8a4 : SUCCESS in 9m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e461701565d24b3b927cbd618a2d8394 : SUCCESS in 6m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dbbfce6ca1c04ef6a812d73c4d6fb6f7 : SUCCESS in 24m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/34429dc7db474d90bae727f63a637d3a : SUCCESS in 17m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3588ca77419948a6a7ffd67777e2e639 : SUCCESS in 16m 16s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2e7db4b068e34da79c052864435aa314 : SUCCESS in 16m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1d675faa12aa45759887168b2fa13017 : SUCCESS in 15m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/460e713d3eca420aa885ba426cae5d28 : SUCCESS in 16m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/96f247019a6f478bb091908f0bfd4775 : SUCCESS in 20m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1e1a5fecec9441dcb2cc213abb49df42 : SUCCESS in 15m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1624ace85a354732b710eeb181f2b1ed : SUCCESS in 16m 11s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/17fa7159ddeb4d79bf5b3d11c63f7a7f : SUCCESS in 14m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/08f7044469ab417a8a56cfa437d14f33 : SUCCESS in 9m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/757f7f2cb87b47d18581c6cc56d92d09 : SUCCESS in 16m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/59d1e404fffd427f970f77e106114754 : SUCCESS in 7m 20s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2b131495aec240acac20c9b92e1d8fd9 : SUCCESS in 1h 03m 20s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/44e49a054747496eba89315a55b509f6 : SUCCESS in 1h 06m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b4f5ad99c757443bb76c70e239c6e7e2 : SUCCESS in 51m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/542c9609fc214428a216c295438cd77d : SUCCESS in 49m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b0430d739ac446d85ee172a716e7d09 : SUCCESS in 17m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e007b644b7644b2fb0da2cab27465c7e : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/358d3528b27949548641f2fa03ca68b8 : SUCCESS in 26m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b6a78041ab374e108f18fc0f332c82ae : SUCCESS in 1h 16m 19s","accounts_in_message":[],"_revision_number":4},{"id":"23af3ee09555033ee7e4ff2b1a73001a22dd33ae","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 21:59:46.000000000","message":"Patch Set 4:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df1a5f24242f421aa3b9c0486d7d8ab3\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4710858952324237b29fd15c5ddebb78 : SUCCESS in 15m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0697debe79bb42768d1ef1c527c5afbb : SUCCESS in 12m 59s\n- swift-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/3ecb801606d74630b54e820030323e58 : SUCCESS in 20m 01s\n- swift-tox-func-py27-centos-7 https://zuul.opendev.org/t/openstack/build/d9758248d6604cc684cf73683b8a354d : SUCCESS in 21m 42s\n- swift-tox-func-encryption-py27-centos-7 https://zuul.opendev.org/t/openstack/build/7a254df759a04a6da3e40b6da82d8a92 : SUCCESS in 21m 37s\n- swift-tox-func-ec-py27-centos-7 https://zuul.opendev.org/t/openstack/build/bc358378f01e49d0b409be46c99955cd : SUCCESS in 21m 32s\n- swift-tox-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/4715a3034970478ca779d9e2544455b8 : SUCCESS in 17m 16s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/310a0835cf1c4edda900c3e179f5f8a0 : SUCCESS in 12m 40s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/ded363e478bf49489e0ee1107faa75cf : SUCCESS in 12m 42s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/33782273d6d543cf8b6f7c551c37a60f : SUCCESS in 16m 15s\n- swift-multinode-rolling-upgrade-train https://zuul.opendev.org/t/openstack/build/3dda07d0573941feb8c71c9f13ddf056 : SUCCESS in 16m 07s\n- swift-multinode-rolling-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/3d9d950e3e104e4fab1d740f3f5396c2 : SUCCESS in 18m 55s\n- swift-multinode-rolling-upgrade-victoria https://zuul.opendev.org/t/openstack/build/ffb8fd8e1e9c4c4592a8a2def0afbc0f : FAILURE in 22m 47s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/fbb651bcb1ea45edab3e5072c2b0c44b : SUCCESS in 19m 14s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/2746ad0b158c4f028c0981665b4769b7 : SUCCESS in 17m 44s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/2499c7776e2b40dbad089dcc9594e0d5 : SUCCESS in 18m 55s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/46672fcc72d44723aaa4da6db95f45ab : SUCCESS in 20m 20s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/3975e45412dd42c7a82f899b51e0574d : SUCCESS in 30m 09s","accounts_in_message":[],"_revision_number":4},{"id":"c29d91b9c7db5663c011bd5adba59804f478a1d0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-16 19:21:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"74b3cdf22d884d6ecf927efcea9b2fa30e92df22","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-05-19 07:13:16.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"21ac675cffb6f2cb9ae04cdc51d7d3d9b609e40f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-19 18:11:22.000000000","message":"Patch Set 4: Code-Review-1 Workflow-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"0c500f7c6eec53b739633ebede0de17b1255b9a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-19 18:40:34.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":5},{"id":"e0c559687cae42f07063fd0e3b2cb6dd9a38700a","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-05-19 18:55:50.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"747de0796dc56f77f86eba8ebf3321ff8e2b2cd1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 19:31:33.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5b75f87ab9d482b98d372b118cc8037\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0ba1f128471f44c9866d2db36935f3f8 : SUCCESS in 47m 17s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aed981a0da8a49eca7f7a454580b610d : SUCCESS in 45m 42s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7dead9a7814b4f89b091f3d192c30dc9 : SUCCESS in 18m 28s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fc688a6980fb4a17a461458e62f754ad : SUCCESS in 47m 51s","accounts_in_message":[],"_revision_number":5},{"id":"7e050176ffba5c6cb82c988076116f391d78fe01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 19:53:21.000000000","message":"Patch Set 5: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/01aa32ea631f49c19f782b5790415a41\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/814a0341e4d844419e738d72fc20bcfc : SUCCESS in 8m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/276e6f0db7464d4c9f5888311a1233b5 : SUCCESS in 3m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/a7dd217e36004f0bbfaa4c1a1006c5eb : SUCCESS in 1h 02m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4d70e8be67c344dd979cbbcdf5e3aa87 : SUCCESS in 43m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/963c0157853f4983b23c3dd783e7f451 : SUCCESS in 1h 05m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/97b230fc1efc4247ac3e559101cfa9b6 : SUCCESS in 14m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9fc37574796a4fa1864965c01a8ea3bb : SUCCESS in 17m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/78b90b235665481eb390ee043a9a60f2 : SUCCESS in 18m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e08b87ef7b1a4440b62f882bb9936442 : SUCCESS in 8m 43s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/582ece71476f4464b96c067dc738c258 : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7837103e6d5e4c4d8ae0e18decda6eb5 : SUCCESS in 16m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f1e7269fc20243019dcca657408a1bb1 : SUCCESS in 15m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/33428945905844c4b15a5ff67b3c55a7 : SUCCESS in 15m 21s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4d7b1e93fb344161bb05f5dbe75dcf43 : SUCCESS in 17m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b8c6aba982c944a89e51b87721d693ae : SUCCESS in 16m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b64825af6c8c4440927bde8d19203b1f : SUCCESS in 17m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dd55929bbb5f42d2bd6c903705b1f4a3 : SUCCESS in 17m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ae57081afc8849a79830153464d4ad3a : SUCCESS in 14m 10s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/91ebee0acd764247ba38d30d0fbf2559 : SUCCESS in 14m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/11fd7429d2d34c59b792161dde850a79 : SUCCESS in 14m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3502bf798e0a4dd0b33496c0d66aac85 : SUCCESS in 6m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ea46caba73b5455283f2c32644ea0da5 : SUCCESS in 15m 54s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ccc805a6a402478cb3fd8edb655acec4 : SUCCESS in 6m 57s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ccc990161fc24dbc8d9c1c77b5fee3f2 : SUCCESS in 53m 49s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/547e71527bb842c28974ba17f271ac9c : SUCCESS in 1h 03m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/47d36686f5474102922183c5764cc970 : SUCCESS in 55m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a988bbf4d36e46e999c501585ed94f58 : SUCCESS in 55m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f74e0d1eaf2046308a02cda8e522433d : SUCCESS in 18m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baf6c64a07184bfda30524de7740608d : FAILURE in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3e54f22044cc4809ae0ccd52eaf0f2ed : SUCCESS in 19m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea7cc011c2084eb98d68342dbd841a2d : SUCCESS in 1h 00m 54s","accounts_in_message":[],"_revision_number":5},{"id":"18bc00e9aee40146e48bded5b05885cc9bd8fed4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-19 20:35:34.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"0db9b7daabe704859d0732bd49476f197833f90e","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-05-19 21:09:37.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"399ed1289eb2b5c0b52d38040fff98b09d68dc67","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 21:39:04.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e643790e25cc4f61aa245ed5710e3d5a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/53d906333c3b433f9a30ad639b92c830 : SUCCESS in 15m 22s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1a53f933c16d48a4af1c5d712c89095e : SUCCESS in 13m 18s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d32431c34fbe427295bdad1cfe014657 : SUCCESS in 21m 34s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/93434b4ded6c49aa9f62676b0391d50a : SUCCESS in 1h 00m 59s","accounts_in_message":[],"_revision_number":6},{"id":"0d3a2d0e912ed83207b4e710590a4293ecbc4c88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 21:47:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f817d5a71c4d4b4894f21d0a3f85b068\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/062870b92ce1418bbffbbb427a47b7ca : SUCCESS in 8m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6fdc8e4630804b95ad57364eff727bd1 : SUCCESS in 2m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/b4f38a248adf45e59c248a89debfab1d : SUCCESS in 55m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fa7cdbb46fa742dd93b9f4c5afa48955 : SUCCESS in 49m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cc0850af66654426b414ffb77bd2905c : SUCCESS in 47m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e1a936614e9048cb9487fba3d88a01f7 : SUCCESS in 24m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/384120bb85a54a98a5d5c2b0c056f009 : SUCCESS in 24m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/21e36b21a4374014af2f48050888cba1 : SUCCESS in 24m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3d46e4063cf44867852751eb3a525caf : SUCCESS in 8m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2a618aa0aad44bd9a66b95614d87d912 : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cc774a2e71f544e8bbeef4372a283091 : SUCCESS in 23m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/768189bfefcd43aab895d62fb928642c : SUCCESS in 14m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7d1e8613e38743a7a6a4c7a1bb9cabac : SUCCESS in 14m 44s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a15d9cf74a8f493a9a2c3bac0888c50a : SUCCESS in 15m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/accab8eb8e4242439201912ba1685f69 : SUCCESS in 23m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f56ecd377c994eb99bfa3be64ad3ffc7 : SUCCESS in 23m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f2fc4a78b231495c893ab090522d60e4 : SUCCESS in 23m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a6048a2375004aa2ae38eaadf3b81466 : SUCCESS in 21m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7b3a324fe2e34968a1f70ee0eaef10fb : SUCCESS in 16m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d67948b1e12f4a88817f0753f0b5def9 : SUCCESS in 15m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/45df53502cf54b23857636b2be81ee62 : SUCCESS in 7m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d8774a1683224946aeea8257b740744e : SUCCESS in 18m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3e1f4cfbe2cb4a9aaec7a09099d6ba88 : SUCCESS in 8m 01s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f801205600544e9e874a9ae01293be79 : SUCCESS in 46m 50s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/fe3ebd09590d4abf9294c67b3e8cb410 : SUCCESS in 55m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/21622770789c475eb7ed4aa599629f46 : SUCCESS in 34m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f5512bc0dcae4671a614a7596a389b8d : SUCCESS in 43m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/85c494af1b014d20b566f44921439fdf : SUCCESS in 15m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d604c597965141ff93bcbc9972e66ff6 : SUCCESS in 4m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b0d0ca7d47a347e79ddca4f8f1b06257 : SUCCESS in 20m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/979c6dac5818488c9f45a8e0a237d1bf : SUCCESS in 49m 19s","accounts_in_message":[],"_revision_number":6},{"id":"a4580deecbe4063602b1ced4de7f6dbc5090e286","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-22 17:53:26.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"ec6d21fd683b45b6067354471781b02d21cb6d6b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-22 17:55:29.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"8d21a5cdf4910357c3733878ad931eb4f3d6c682","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 18:15:03.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ddf7de1c98874314a6ef052b2fee8fbb\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a122abe27ef84c098e3666a179f4a2d9 : SUCCESS in 14m 21s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d1b0016cedba4a6f80734e379f6cd4c4 : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f42e0e618f04a44953397de4be5233f : SUCCESS in 20m 00s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7f7726a0aef8402190784bd4acd038ce : SUCCESS in 18m 08s","accounts_in_message":[],"_revision_number":7},{"id":"48a394685a23350057be0f2a2426209f09861675","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 20:22:17.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18dad8512f3f48d5b31533f4b8e166a4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fe4083d10eb416eaf5dc77f9101b0e6 : SUCCESS in 7m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a993c428953a46cdb9095295cc9dbe9c : SUCCESS in 2m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/302116da399440bf93cc9944220a11fb : SUCCESS in 52m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/66acdb48ddf74085879b450a7f440353 : SUCCESS in 49m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/adb70d8727cc474ea30176f68372290e : SUCCESS in 58m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2fac006c9ad049ec95cb6e579868bc30 : SUCCESS in 17m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ad79435ba3934c30b15e8e870805a0b3 : SUCCESS in 15m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e13820d9bdbc4ce383b6a83578851683 : SUCCESS in 17m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/10a2396ead9c46e09174d3037e2c5616 : SUCCESS in 8m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/44f3a9d6c46242ee96f307cf9c487997 : SUCCESS in 5m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1f36dfb733cf496b9fd9e00582ce5ea3 : SUCCESS in 16m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8142fc823a7549d1983898327666af78 : SUCCESS in 15m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/883dab2efa5b495b939852e44b5cc533 : SUCCESS in 13m 03s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3ae97c0e8dd548408725596f52892e25 : SUCCESS in 17m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e089c1d4e15242b8a9624fd355a7bff4 : SUCCESS in 16m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/28c05a727e1749bba58030d28a5454f0 : SUCCESS in 18m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bd6ec24eaab14b8a88f955d926c3a8d2 : SUCCESS in 21m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/91e2fa2f7ff74b2cbc1973226d270bcd : SUCCESS in 14m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f330f1a99f6d4ad0b7a553132a33a640 : SUCCESS in 14m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d3f686215ac643a3b5adbd470a0f3ef5 : SUCCESS in 15m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/20ac1060c9124cee8199858ea2e92aad : SUCCESS in 7m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7c874fe444934dcc94429687cfceea67 : SUCCESS in 15m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6e49dafe8474423a80ec06e427a63feb : SUCCESS in 8m 02s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e5c715d233014f999ec488d7550bc741 : SUCCESS in 50m 11s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e7ebe9482dd14629bf364908bfdc39d4 : SUCCESS in 57m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/68245a4f05d24643acebf78dc4a5743d : SUCCESS in 42m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bbedf886bf6945dc91a6ff626a8d24e6 : SUCCESS in 38m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42f17804758446409badc02a65ce1357 : SUCCESS in 17m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d93b95a3d7c9474787438e359d2b5849 : SUCCESS in 4m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b73183ed7e8149d3b1cedd42c37991f6 : SUCCESS in 18m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a69bc4913eb54e989e8e7f6aefbaf481 : SUCCESS in 1h 05m 24s","accounts_in_message":[],"_revision_number":7},{"id":"02b2c7254e20fe27314b212bc3a5943a6752dfc2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-22 22:59:14.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"64521bf505459947c667a72ca9d072457e1cebf2","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-05-23 07:21:08.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"3bb7df3fd733d93193f7c5fc4a2a73529be6e6be","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-23 16:33:06.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"801d00054e4048ced2e44271b2461037506e4e78","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-23 21:35:41.000000000","message":"Patch Set 7:\n\n(6 comments)","accounts_in_message":[],"_revision_number":7},{"id":"73d4b0bff0e6c45a9d790378c143e87ad768aed9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-23 21:36:06.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"35a61af174b13bca5baa41453da2022bc5d6f249","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-05-23 22:03:21.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"c405ccb5ed9c80fd3eb45eaca7b50d93280e0edc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 22:41:23.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67971f001f40410db296bf837feade00\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fae8b114f4ef4731ae5b1e34ae6c70b3 : FAILURE in 23m 42s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/590e88b06fe44837a9e6bcb1eb2073f3 : SUCCESS in 19m 15s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9e114266ccd24e519d9ba8da578d3f58 : SUCCESS in 25m 45s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/86aef321a5af4d6ab9d78d6b90628eb9 : SUCCESS in 1h 03m 18s","accounts_in_message":[],"_revision_number":8},{"id":"d2a5b46b556b7096cfee427fefec8e050c6051ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 22:50:56.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/371852b92b3b406aa63955fac4013454\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76de604d0e8a482abda7057d6e4d0823 : SUCCESS in 7m 29s\n- requirements-check https://zuul.opendev.org/t/openstack/build/dd268167e490463fbe9a16fd00279397 : SUCCESS in 2m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/301b49878dd1446b9d02fc6cf529508a : SUCCESS in 51m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8e0f0363020847288cb307ccef09625b : SUCCESS in 52m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/54f75952c24d4d198ed21f35d901e4de : SUCCESS in 1h 06m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0b43afee35e54fdbbe21a77f897a7766 : SUCCESS in 18m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/197d8fc64cad41debeab79ac6cee042d : SUCCESS in 18m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c1819fe644714821a2ec3f8fd6b33cae : SUCCESS in 15m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dde6fe02df8c4f55a6a78ce0bfba460f : SUCCESS in 8m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/12590deb65204e3fb5f8ba5d12901c3c : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2e800aacac8a4c23ba328474f6909d0f : SUCCESS in 17m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bb15eb62cfe34c4eaeba4cc52a8789ae : SUCCESS in 17m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8e6009e20415454e9fbfb7d26b7125b5 : SUCCESS in 13m 58s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/eb5006fd1f084d67960f0bdc62a1faac : SUCCESS in 17m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8c5298de172847aba7ddc499b0e35b45 : SUCCESS in 16m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/35fbbb2e12a64544b919ed47455df8ec : SUCCESS in 17m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3460052fdb6243359c4f4c365b495695 : SUCCESS in 15m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/143e7e96e28d4a9683cdbd70d983cc70 : SUCCESS in 13m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/41059e98f3334bac9ed33ec04c2a7814 : SUCCESS in 15m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cc6c43c96b4c4ad3bab6e41823e53795 : SUCCESS in 15m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0c28e2c8a1c94d68835db9ef0b1cb202 : SUCCESS in 7m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/eb4c63b99b1e453492dce143871d76d4 : SUCCESS in 13m 59s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/196ccfa000bb40e1ae8f0f43de62ace1 : SUCCESS in 7m 42s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a0e7ba48ca5d438896cf74d0ee4f7349 : SUCCESS in 54m 34s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/64daa82730ad4079bc79b0802f9e6764 : SUCCESS in 1h 07m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2086a332ba844efcb696471e381b4d3e : SUCCESS in 49m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e79923bb823245d995c69e9241978cb6 : SUCCESS in 55m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ec65279923a462e9a508a50a6757c90 : SUCCESS in 20m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc761bf1ff0a437bbee2b2d64e1b7d5c : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8925d2295e364559a97dcc7e9b08f58c : FAILURE in 26m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b1c79676ddd4ac99aa1473d7c125627 : SUCCESS in 42m 09s","accounts_in_message":[],"_revision_number":8},{"id":"f75cf84355e6c29b35aed8a7b8df109b35a9cae6","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-05-24 07:00:04.000000000","message":"Patch Set 8: Code-Review+2\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"fea4d67be74bad017f9204c9e9820d594cab755b","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-05-24 07:03:42.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"9f8b1dcb0cfb63e8365184dc8b2c5b95af23040b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-24 09:42:55.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"dedc7cc279d4999f11f3d8166cdd9c8af35b72aa","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-24 09:43:06.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"c663dd618767f7793b4732f953433a2144e837a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 09:43:39.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"3c9dbcd4c9e04fdc3f8c3a20271c45962fd73082","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 11:34:19.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3e543937ad8446f8a38be82f7229656\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb936da70cc04bf481b2c73f6be40df8 : SUCCESS in 8m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d196ea5fffe548c49d112a3b6b5078c5 : SUCCESS in 2m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/7aa2754052c047e39aa6f6422fadb514 : SUCCESS in 41m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8751119389254487be3830ddf69b0384 : SUCCESS in 47m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b55feadd7a04fcf9403f0cddb0a1f19 : SUCCESS in 1h 01m 02s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4093236533c146b8bd230cc5dec1ee6a : SUCCESS in 9m 22s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/52374c92954140199d8d438f3b778299 : SUCCESS in 7m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/283450754dd94d4b90e2a43ee5c02992 : SUCCESS in 11m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e0035ca9a4a5410bbd9e886e39a7f7f3 : SUCCESS in 17m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8ee433d2a7d14d4194d7c6612a591e00 : SUCCESS in 12m 12s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f9f4fd5749c24d93987ecc87df0fcc12 : SUCCESS in 16m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/108d22abdd704eb4a33c311fd942f2a1 : SUCCESS in 16m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fb33c6e5d0884fb18531123de6e617fe : SUCCESS in 16m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2f0c9e40559244059ce449f735ae0b49 : SUCCESS in 17m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0e448f5e75a441b6ba229c7cc17f1102 : SUCCESS in 24m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3f628b2bd272436e8511e315b5dcc61a : SUCCESS in 23m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5c439f71a8ba44cd905a6fdf9e90987f : SUCCESS in 16m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/69ce68bd663444bbb7644dcd6924fd84 : SUCCESS in 9m 23s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7f5ebb09ffe24262b106707683acb125 : SUCCESS in 8m 56s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/496e46cdc9d94ccaba3a0052edacc709 : SUCCESS in 55m 49s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/864b7bb5f452477aab01a0ff0d843cdf : SUCCESS in 1h 05m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/732de4ca51f9457b8ef35a7efdb96cfa : SUCCESS in 57m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e93a96aa8b9d43d59f7dca40a84839ed : SUCCESS in 54m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4d9bab7578143f79835e7059a6a197c : SUCCESS in 20m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68dddbd6f24641aaa011cfa14dce864d : SUCCESS in 4m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f6ab24c7b4440c8aa5a70dc1ac29eee : SUCCESS in 1h 01m 46s","accounts_in_message":[],"_revision_number":8},{"id":"9f4ddf302b1703182a8f1bf5564bbc8438577169","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 11:34:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"8a0194094f179c22f58847e78215ff753bfc0f53","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 11:35:26.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f32a2a668db4ffe8d9198b973261682\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29dc040ca0374ba5a2d537fd3b9ffa83 : SUCCESS in 47s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/5994aed5e302422c9c586abde5d7d595 : SUCCESS in 29s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/67119d99222c43f889b39255f0cb7a37 : SUCCESS in 30s (non-voting)","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"23fa18d3021fb44d8b4dd81890d9c355eed9426f","revisions":{"55ec3df328d934b544bbf2469389f81e9a723772":{"kind":"REWORK","_number":1,"created":"2023-02-09 06:22:36.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/22/873222/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/22/873222/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/873222/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/873222/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/873222/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/22/873222/1"}}},"commit":{"parents":[{"commit":"90f9a479b67577ffa83057f5e21d464d5294fcb0","subject":"Authors/ChangeLog for 2.31.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90f9a479b67577ffa83057f5e21d464d5294fcb0"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-02-09 06:19:38.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-02-09 06:21:47.000000000","tz":-480},"subject":"Remove hard dependency on netifaces","message":"Remove hard dependency on netifaces\n\nThe project was archived in 2021, and we can fairly easily replace it\nwith a call to `ip addr`.\n\nChange-Id: I1189a60204cf96c291619f8d8ec957ed8a5be1ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55ec3df328d934b544bbf2469389f81e9a723772"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55ec3df328d934b544bbf2469389f81e9a723772"}]},"branch":"refs/heads/master"},"c131638f52a445e6561d962e20493a8bab6a618d":{"kind":"REWORK","_number":2,"created":"2023-02-09 20:29:18.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/22/873222/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/22/873222/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/873222/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/873222/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/873222/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/22/873222/2"}}},"commit":{"parents":[{"commit":"90f9a479b67577ffa83057f5e21d464d5294fcb0","subject":"Authors/ChangeLog for 2.31.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90f9a479b67577ffa83057f5e21d464d5294fcb0"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-02-09 06:19:38.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-02-09 20:28:49.000000000","tz":-480},"subject":"Remove hard dependency on netifaces","message":"Remove hard dependency on netifaces\n\nThe project was archived in 2021, and we can fairly easily replace it\nwith a call to `ip addr`.\n\nChange-Id: I1189a60204cf96c291619f8d8ec957ed8a5be1ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c131638f52a445e6561d962e20493a8bab6a618d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c131638f52a445e6561d962e20493a8bab6a618d"}]},"branch":"refs/heads/master"},"c58c67279ba35c74377aa3778fe2f2a917fc01e9":{"kind":"REWORK","_number":3,"created":"2023-02-09 20:49:56.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/22/873222/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/22/873222/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/873222/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/873222/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/873222/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/22/873222/3"}}},"commit":{"parents":[{"commit":"90f9a479b67577ffa83057f5e21d464d5294fcb0","subject":"Authors/ChangeLog for 2.31.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90f9a479b67577ffa83057f5e21d464d5294fcb0"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-02-09 06:19:38.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-02-09 20:49:51.000000000","tz":-480},"subject":"Remove hard dependency on netifaces","message":"Remove hard dependency on netifaces\n\nThe project was archived in 2021, and we can fairly easily replace it\nwith a call to `ip addr`.\n\nChange-Id: I1189a60204cf96c291619f8d8ec957ed8a5be1ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c58c67279ba35c74377aa3778fe2f2a917fc01e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c58c67279ba35c74377aa3778fe2f2a917fc01e9"}]},"branch":"refs/heads/master"},"402f493ad0d4732e4da2d227e21e351554d58b8f":{"kind":"REWORK","_number":4,"created":"2023-05-11 20:25:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/22/873222/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/22/873222/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/873222/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/873222/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/873222/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/22/873222/4"}}},"commit":{"parents":[{"commit":"5c2adc593eb20c3c52732635f5c80bbbbae304bf","subject":"Merge \"Sharder: add timing metrics for individual steps and total time spent.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5c2adc593eb20c3c52732635f5c80bbbbae304bf"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-11 05:21:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-11 20:21:16.000000000","tz":-420},"subject":"Remove hard dependency on netifaces","message":"Remove hard dependency on netifaces\n\nThe project was archived in 2021, and we can fairly easily replace it\nwith some ctypes code to call getifaddrs ourselves.\n\nBe willing to fall back to netifaces (with a warning) in case getifaddrs\nis not available, but I\u0027m fairly certain it will be for all platforms we\nsupport.\n\nWhile we\u0027re at it, pull a bunch of ip-address-related functions into a\nnew module under utils.\n\nPartial-Bug: #2019233\nChange-Id: I1189a60204cf96c291619f8d8ec957ed8a5be1ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/402f493ad0d4732e4da2d227e21e351554d58b8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/402f493ad0d4732e4da2d227e21e351554d58b8f"}]},"branch":"refs/heads/master"},"f2c9e1eb7bb74e9c0715a667d79234dde5dd2e88":{"kind":"REWORK","_number":5,"created":"2023-05-19 18:40:34.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/22/873222/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/22/873222/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/873222/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/873222/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/873222/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/22/873222/5"}}},"commit":{"parents":[{"commit":"5c2adc593eb20c3c52732635f5c80bbbbae304bf","subject":"Merge \"Sharder: add timing metrics for individual steps and total time spent.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5c2adc593eb20c3c52732635f5c80bbbbae304bf"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-11 05:21:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-19 18:33:47.000000000","tz":-420},"subject":"Remove hard dependency on netifaces","message":"Remove hard dependency on netifaces\n\nThe project was archived in 2021, and we can fairly easily replace it\nwith some ctypes code to call getifaddrs ourselves.\n\nBe willing to fall back to netifaces (with a warning) in case getifaddrs\nis not available, but I\u0027m fairly certain it will be for all platforms we\nsupport.\n\nWhile we\u0027re at it, pull a bunch of ip-address-related functions into a\nnew module under utils.\n\nPartial-Bug: #2019233\nChange-Id: I1189a60204cf96c291619f8d8ec957ed8a5be1ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2c9e1eb7bb74e9c0715a667d79234dde5dd2e88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2c9e1eb7bb74e9c0715a667d79234dde5dd2e88"}]},"branch":"refs/heads/master"},"565f6ce6f39bf451032c51458a028b9702c6646d":{"kind":"REWORK","_number":6,"created":"2023-05-19 20:35:34.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/22/873222/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/22/873222/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/873222/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/873222/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/873222/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/22/873222/6"}}},"commit":{"parents":[{"commit":"5c2adc593eb20c3c52732635f5c80bbbbae304bf","subject":"Merge \"Sharder: add timing metrics for individual steps and total time spent.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5c2adc593eb20c3c52732635f5c80bbbbae304bf"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-11 05:21:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-19 20:35:17.000000000","tz":-420},"subject":"Remove hard dependency on netifaces","message":"Remove hard dependency on netifaces\n\nThe project was archived in 2021, and we can fairly easily replace it\nwith some ctypes code to call getifaddrs ourselves.\n\nBe willing to fall back to netifaces (with a warning) in case getifaddrs\nis not available, but I\u0027m fairly certain it will be for all platforms we\nsupport.\n\nWhile we\u0027re at it, pull a bunch of ip-address-related functions into a\nnew module under utils.\n\nPartial-Bug: #2019233\nChange-Id: I1189a60204cf96c291619f8d8ec957ed8a5be1ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/565f6ce6f39bf451032c51458a028b9702c6646d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/565f6ce6f39bf451032c51458a028b9702c6646d"}]},"branch":"refs/heads/master"},"4de8516326f92c1b98756afded932fc6984a2c96":{"kind":"REWORK","_number":7,"created":"2023-05-22 17:53:26.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/22/873222/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/22/873222/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/873222/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/873222/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/873222/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/22/873222/7"}}},"commit":{"parents":[{"commit":"e29e2c3ae5fd32f6b8c3d70e6ea4db786f980151","subject":"Move IP-address-related functions out to new module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e29e2c3ae5fd32f6b8c3d70e6ea4db786f980151"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-22 17:37:12.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-22 17:51:21.000000000","tz":-420},"subject":"Remove hard dependency on netifaces","message":"Remove hard dependency on netifaces\n\nThe project was archived in 2021, and we can fairly easily replace it\nwith some ctypes code to call getifaddrs ourselves.\n\nBe willing to fall back to netifaces (with a warning) in case getifaddrs\nis not available, but I\u0027m fairly certain it will be for all platforms we\nsupport.\n\nCould maybe use some more testing on big-endian arches / BSDs, but an\nattempt was at least made at supporting them.\n\nPartial-Bug: #2019233\nChange-Id: I1189a60204cf96c291619f8d8ec957ed8a5be1ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4de8516326f92c1b98756afded932fc6984a2c96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4de8516326f92c1b98756afded932fc6984a2c96"}]},"branch":"refs/heads/master"},"23fa18d3021fb44d8b4dd81890d9c355eed9426f":{"kind":"REWORK","_number":8,"created":"2023-05-23 21:36:06.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/22/873222/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/22/873222/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/873222/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/873222/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/873222/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/22/873222/8"}}},"commit":{"parents":[{"commit":"e29e2c3ae5fd32f6b8c3d70e6ea4db786f980151","subject":"Move IP-address-related functions out to new module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e29e2c3ae5fd32f6b8c3d70e6ea4db786f980151"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-22 17:37:12.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-23 21:35:48.000000000","tz":-420},"subject":"Remove hard dependency on netifaces","message":"Remove hard dependency on netifaces\n\nThe project was archived in 2021, and we can fairly easily replace it\nwith some ctypes code to call getifaddrs ourselves.\n\nBe willing to fall back to netifaces (with a warning) in case getifaddrs\nis not available, but I\u0027m fairly certain it will be for all platforms we\nsupport.\n\nCould maybe use some more testing on big-endian arches / BSDs, but an\nattempt was at least made at supporting them.\n\nPartial-Bug: #2019233\nChange-Id: I1189a60204cf96c291619f8d8ec957ed8a5be1ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/23fa18d3021fb44d8b4dd81890d9c355eed9426f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/23fa18d3021fb44d8b4dd81890d9c355eed9426f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
