)]}'
{"id":"openstack%2Fswift~873942","triplet_id":"openstack%2Fswift~master~I4e6a2e1f5b6117ba3aba28eb748e33836fc99db2","project":"openstack/swift","branch":"master","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-02-15 17:35:53.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-02-15 21:37:36.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I4e6a2e1f5b6117ba3aba28eb748e33836fc99db2","subject":"wip: tune tcp keepalive","status":"NEW","created":"2023-02-15 15:31:14.000000000","updated":"2023-02-15 21:39:06.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":7,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"7185add57a67436c29724c8e87a53d78ff8cd603","_number":873942,"virtual_id_number":873942,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-02-15 17:10:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":-1,"date":"2023-02-15 15:34:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-15 15:52:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-15 17:10:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-15 17:35:53.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2023-02-15 21:37:36.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"}],"messages":[{"id":"8659942b4cfe906850f5121f1f1d1c4946f8c56f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-02-15 15:31:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b7eb1ef2274ab54656acfc8e5aa70b858c85233a","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-02-15 15:34:26.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cccdf7707aa0b15c87da1f6a454e88f5c8732840","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 15:52:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bc373cb6a7b43538db2b5e58cfa57b6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e8b8f5519f18458db680fb7a4153c957 : SUCCESS in 18m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c7543fd0d3574488803e307422ba8f64 : SUCCESS in 19m 39s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c3d19030bdb447ba5e7a624956226b1 : SUCCESS in 18m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/90437dc455b4403b8cbe335f970b48e7 : SUCCESS in 17m 48s","accounts_in_message":[],"_revision_number":1},{"id":"290b23546f100441e5f66a75251df70f9cb36345","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 17:10:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c4b118571254b39b1902a53c6b0d66d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/337e8376e8944699805b67587fe4310a : SUCCESS in 7m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/d2441efb76e041cdb515077dc6765b32 : SUCCESS in 47m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f6bcd038f45c4cf9bfcbc7ba74fdfd76 : SUCCESS in 38m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/329830a4188b416fafb09f625397b11f : SUCCESS in 50m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bda0fd25d4b74889817e8e6c25580005 : SUCCESS in 21m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/886f3e9bbc824ffe839ddecc4493dcc0 : SUCCESS in 20m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3daa20863c8842b19d5a55d35356b4ef : SUCCESS in 24m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/98abdff191224804b9c826e8c9bc53d3 : SUCCESS in 8m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/95602f200bf349489b8b457f6b84a774 : SUCCESS in 8m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f3e69092b906478f8a99a97d825b1e77 : SUCCESS in 16m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f5e7a5516a9448c7970bc527b4f1d4d7 : SUCCESS in 13m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/79ad9c67f12647dba215d2c85293195c : SUCCESS in 13m 45s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7d45e47d66d745a583165c8b9dfaed8d : SUCCESS in 16m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/992936c9f1e14f41b92b7c12042c360c : SUCCESS in 16m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/037e39f9132b41069ca1f3b593b2ec3a : SUCCESS in 35m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/10661ac780e94ca8a113bbaa2c66cf62 : SUCCESS in 26m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/855a394e91194d22a0a9bc5f4b4142c1 : SUCCESS in 13m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f94c69118dd5482e8882513fc5dffbfa : SUCCESS in 21m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/314621a95405404abc6ecd90141e338a : SUCCESS in 17m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4febce4bf32a4abea553495543af6891 : SUCCESS in 7m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/da0e03e3f1b34b14aff8ffa03c5df4e2 : SUCCESS in 15m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cf3b24c5e2c146da8627c51e927ca1df : SUCCESS in 9m 54s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d908c240244041b9b4bfb535b7e3ed30 : SUCCESS in 50m 02s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e44efdd93a1c4c109be379ffa7b25d25 : SUCCESS in 1h 04m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fb49576554cf4cb8986057d43dc59112 : SUCCESS in 1h 12m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dbe402124f394056b11aca3d17380cff : SUCCESS in 1h 13m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e2a438ad24c4593af140d98dd43d34b : SUCCESS in 17m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/486e7a236abe48d59aa4d7104d54ee37 : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9a9f011dc26c4beba7e32c27ae402d26 : SUCCESS in 40m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d693d98bde8347c08747cce48aba67a4 : SUCCESS in 1h 07m 32s","accounts_in_message":[],"_revision_number":1},{"id":"7677bf201c9978d183e332086d748e19af25a58f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-15 17:35:53.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1161110352c8f7f4e6636c7d70c6ea8a86303fa7","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-02-15 21:37:36.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7185add57a67436c29724c8e87a53d78ff8cd603","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-02-15 21:39:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7eb3e816ddc908978df13d44347bc2623e9518ff","revisions":{"7eb3e816ddc908978df13d44347bc2623e9518ff":{"kind":"REWORK","_number":1,"created":"2023-02-15 15:31:14.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/42/873942/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/873942/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/873942/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/873942/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/873942/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/873942/1"}}},"commit":{"parents":[{"commit":"5d67ec27ec05c68818e27002691001fe9fff6ecf","subject":"Allow wsgi keepalive configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5d67ec27ec05c68818e27002691001fe9fff6ecf"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-02-15 15:30:43.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-02-15 15:30:43.000000000","tz":-360},"subject":"wip: tune tcp keepalive","message":"wip: tune tcp keepalive\n\nChange-Id: I4e6a2e1f5b6117ba3aba28eb748e33836fc99db2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7eb3e816ddc908978df13d44347bc2623e9518ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7eb3e816ddc908978df13d44347bc2623e9518ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
