)]}'
{"id":"openstack%2Fswift~874122","triplet_id":"openstack%2Fswift~master~I74a0aac0ac29577026743f87f4b654d85e8fcc80","project":"openstack/swift","branch":"master","topic":"p-ssync-fix-meta-ts","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-04-14 18:18:09.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-04-14 18:18:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I74a0aac0ac29577026743f87f4b654d85e8fcc80","subject":"ssync: fix decoding of ts_meta when ts_data has offset","status":"MERGED","created":"2023-02-16 18:40:32.000000000","updated":"2023-04-14 18:19:19.000000000","submitted":"2023-04-14 18:18:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"874122-p-ssync-fix-meta-ts","meta_rev_id":"cd23bd6b0deea097b93cec8ddd8657446ab93fbc","_number":874122,"virtual_id_number":874122,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-14 18:18:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-14 19:17:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-14 17:14:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-16 19:01:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-16 21:26:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-17 01:20:11.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2023-02-17 15:41:48.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"2b8a0f22b81a7275b397d26ee51836c99d528552","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-16 18:40:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"845b242dc21cc93e02b8c3e04ff426a3129706cd","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-16 18:41:35.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eed4b13c8af618443b188e3aa4bd55cfa9e306ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 19:01:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b502a4b40de843f5b01803426e197d8e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f555dd818abe4a62a8b7ab665215f28f : SUCCESS in 14m 41s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c0697db7fe9a40248adcd9ef7b1d27d7 : SUCCESS in 19m 18s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e5e694e6dc354acc93c51b852724b7c0 : SUCCESS in 17m 33s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d3df519aa3de45c485b23737adbc87bf : SUCCESS in 17m 08s","accounts_in_message":[],"_revision_number":1},{"id":"4672c633cae347b2ea3e3f85132ba6cdf0f1e5b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 21:26:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4cfef137c90406a91788cea414f8b74\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92d77f2e77d242c29f16bda1c60ffb16 : SUCCESS in 8m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/171fb54659ed4cc09e661d288acff2ef : SUCCESS in 1h 03m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/008e8d9e19ce4b52959e0670eb624b2e : FAILURE in 27m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f622dfa00a546e3b68414d70849d281 : SUCCESS in 55m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3e0a138114204b03b7b9cdf155571a12 : SUCCESS in 16m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d1496209c1ea4408935ea0e637b0a6ed : SUCCESS in 25m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9024640d517749c5959d43616eab8709 : SUCCESS in 19m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/79fb2009f9614735af2f7102e14e4d99 : SUCCESS in 9m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/14650f65320546e0bdec4a2aa66eeee0 : SUCCESS in 5m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cb107a742e074e4194ca8cdb40c4ed65 : SUCCESS in 34m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/63a25e0e20494376a3ee79f37da0591e : SUCCESS in 14m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d1b1f48afc7d46b386bb65b5ddcf4a44 : SUCCESS in 15m 45s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/44f97fd80b69445cb0fdbd521231316f : SUCCESS in 16m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5386748cba0c49ceaed28c47b8d7cca4 : SUCCESS in 15m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/258327f1078c4eceaf80c99ee3a959b1 : SUCCESS in 15m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1f89651e2e8549cfbb356147935094fd : SUCCESS in 19m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9f5b77cfc15449c0b633dab8f3255358 : SUCCESS in 22m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/18d62bc5ce3f43a0b6dd2e99fd4260c7 : SUCCESS in 15m 16s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/27846333dd2644af9a045c429eaf2f49 : SUCCESS in 18m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/47bcb770415e490d91dea8d66c5a2391 : SUCCESS in 7m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7be2924a939f4d26b231e719c23d3ef7 : SUCCESS in 33m 06s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/de28eecdcf3e4c319c261032e19987c9 : SUCCESS in 8m 03s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b65b92bcc7de4ff8b69819d985779c7e : SUCCESS in 57m 28s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/59595ac33ac741fb9d29c37386f82b85 : SUCCESS in 57m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a07ff7c99f32492b9534fdc8da949b13 : SUCCESS in 52m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7b3f991bd78f41cc8edace1e28da4ccd : SUCCESS in 1h 08m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50d02426a6d644e58d8992cd1378f771 : SUCCESS in 18m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/420e3efbb694499485b4a3f2f05dc00a : SUCCESS in 4m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d7ae68095daf41a2b7b2ea898c45093e : SUCCESS in 20m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/54bfbb0e7c134cf99b4a9b2a8dda3929 : SUCCESS in 55m 27s","accounts_in_message":[],"_revision_number":1},{"id":"57de1342393bee8331590ae90248aa5ee2404221","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-17 01:20:11.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eef721f022a352acd5d0b168d2cad1260dd1485c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-17 09:46:48.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0f2aff1e7f0654c3f65887188f8b8793f36a3d4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 10:56:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a5a0efe1c9143cda75dc6cfeac14b18\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8db08a70d73e4d40aa2be9ec8c189666 : SUCCESS in 9m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/ed52d489539f483f8124617fba4151c4 : SUCCESS in 51m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0b414ff603e044cc9ff796a2d56aec92 : SUCCESS in 40m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/06d1333402764b11a9a5a3dd4d997752 : SUCCESS in 58m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/47210e4efd884bc190dc81e95fb10592 : SUCCESS in 19m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/07b3fa3a8f6c4d2d810ffde41231b201 : SUCCESS in 21m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ba7542a9316142e3a775cd05b7bb57a5 : SUCCESS in 19m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c294c95e6f8b4aeebf40cc6341134270 : SUCCESS in 9m 07s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3d6d21fa5f8646fe961e181f6b24a98b : SUCCESS in 6m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4ed297971bea445ba6d9195e2034b024 : SUCCESS in 18m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6ad5f9f1a70f47f59760f2196012c98f : SUCCESS in 16m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/11b61e31817c4552a4470b127d986728 : SUCCESS in 16m 07s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/feac4d1148914670bf2bebff2353421b : SUCCESS in 16m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2e4a87e1f62d40738ab1991ee267cbd4 : SUCCESS in 18m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6e6d80dde1ef4c17ae4c6f6a6289dd4d : SUCCESS in 18m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0920bf9940bb43f7a6d40bee9545fe68 : SUCCESS in 18m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/706723a6656e48ff86f9aa1b8f1d2cc8 : SUCCESS in 16m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/648b253f06544557a1c1d0d1e89ee00f : SUCCESS in 14m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1f61781b832d4b4f9b68262cd153da28 : SUCCESS in 17m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/26bf451166b04813a68d157116307d1a : SUCCESS in 10m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e60aaeb8d73b448084331354776fbeec : FAILURE in 20m 43s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bea68e5fca1d4a1ba8cb445e5fc0a317 : SUCCESS in 7m 28s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5640ba77874f4e289ac9ef9919f62633 : SUCCESS in 1h 00m 33s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/51489cf6dd6445e48b0d46f2e64551a7 : SUCCESS in 1h 04m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/096d2c4bf9834f0ba6836c2506031635 : SUCCESS in 54m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b6138568b3414c46855e97e7cba9fd39 : SUCCESS in 52m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c41a9db1095456e99a07b4aa837fefb : SUCCESS in 17m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df90b90794a9492b8eec29da8a2b4a06 : SUCCESS in 4m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/870d6a8a6f7e45c78ac55e3ff6ca937d : SUCCESS in 19m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29c741e207374f21b75ca20ea5af94cc : SUCCESS in 55m 59s","accounts_in_message":[],"_revision_number":2},{"id":"4f565f95465906e342b2ad1440713d6fb6a65df1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 11:20:19.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af1cf40e554b49e49eb2c5b592407cf7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e71f405699974472a3e55ed01e1f529b : TIMED_OUT in 1h 30m 58s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4b40667740e84cb899d4d62c1bc21d79 : SUCCESS in 1h 27m 21s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1372e7fb837a4a5cbc2665458b89535e : SUCCESS in 1h 25m 15s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/424ae68e1d664b9891a03a3c47bd6ec7 : TIMED_OUT in 1h 30m 56s","accounts_in_message":[],"_revision_number":2},{"id":"629e98fb840c33daf243b1f5061850492ce3ec3d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-02-17 15:41:48.000000000","message":"Patch Set 2: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6050705b1089e7db89c3664aed8bcfbeb5386363","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-02-17 20:41:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1d1eb74c98c7736de444b5933d68b5ccca826663","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-18 00:40:28.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8d00021e127e0fdfe72946c8466807c9e67acb9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-24 10:17:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"23732191b76e2a503251dbfabe064a572cb723ed","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-24 10:20:32.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1e5fc64da501ed5483c5454515dc6006703caf19","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 10:59:31.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d2ed56e73b8489cb1738fc4e704cad7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/72d84e716a504b2280f5e1db3c0ecd4f : FAILURE in 24m 54s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/993677363ce643a8974e2e2f7af9d2be : SUCCESS in 25m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/65c18c560440467ba0853756562c8465 : SUCCESS in 40m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/44d6ec964b9c4d5784866008d3a7ffef : SUCCESS in 30m 39s","accounts_in_message":[],"_revision_number":3},{"id":"20c5e7468f3f8b40f7866a8c8b4bd5e6ae993ba7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 11:29:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b6a4dbdfaaae497c983509bca1dcb9a4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2de291f1f1a04f469cdbe9b8459c89d1 : SUCCESS in 8m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/41c03759f6c84f748b7f0b20f5e12dfe : SUCCESS in 55m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/927a88bc307548c5b161ad5b6a12b44d : SUCCESS in 47m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f544cfcac85b44639f4afb11d6220532 : SUCCESS in 54m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a36fe3d4db3c468db66cb1907b24a3fc : SUCCESS in 21m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9b0cb659528542c3b41a81bc60e93bef : SUCCESS in 30m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/32bcaf583d984b5ea34c96ea000ea3fc : SUCCESS in 19m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/72e43f6caa97481386f0ff3a34720bc0 : SUCCESS in 8m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2d2d7b9e80834ad3b480d1ea748763df : SUCCESS in 5m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dd395d9cf5ec4a97a147127949c8f305 : SUCCESS in 29m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cc5103f494af470c933f581e1220cc24 : SUCCESS in 15m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1cfe1aedb61249af82180cc22f0bc2f8 : SUCCESS in 16m 04s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/66eb793822274c54b9f4b6ba57bb1ddd : SUCCESS in 29m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f95edb68276f41fc84f92f9f6178b7ae : SUCCESS in 15m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cdf07344bc2442098fc479c20a941f8d : SUCCESS in 19m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/15820d75d6ed49b78c1fc83d0bcf693e : SUCCESS in 26m 44s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a49b665887104ac2838675d6c3f3731e : SUCCESS in 13m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7863ccaeffa94338ac2fecef56306d97 : SUCCESS in 13m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2627ba0750b94025b2bc5148eac5e4ad : SUCCESS in 16m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cb0f895e51fa434eb829bc64859e0fb6 : SUCCESS in 7m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1d3c3166bad645d3a34133663d283e76 : SUCCESS in 15m 03s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/37e07f6d217c46a8bb5d907eb621b728 : SUCCESS in 7m 56s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/14549f600ec84678be34df0052108231 : FAILURE in 58m 05s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/8b3abaaf823f403384e4dcd35e408fbf : FAILURE in 57m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7d0eca2f8b1741e187d4b091420b84ba : SUCCESS in 57m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/01e40bb008e144d5b3d150f7b83504dd : SUCCESS in 59m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79de3685aff04252932deede4b0edc4c : SUCCESS in 17m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f08beb8d5f3422c8304a26b5c8f45fe : SUCCESS in 5m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fc5bf45afdaa41708c3ab470dcfa6ed0 : SUCCESS in 18m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/acbbc00894964fc3bbbdd433dc473abc : SUCCESS in 44m 25s","accounts_in_message":[],"_revision_number":3},{"id":"774e0aa2a956eefce991c8275737bfc222dd5e7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-02-24 19:31:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8320226f6880b5850fdb6141d724e736ba05284e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 20:43:39.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f2f75e6f97b044408c888c191a6d07bf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44a9e63abb4848619824cfd57083df7e : SUCCESS in 8m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/9f00d0c1ea234c898264d5192f02ec67 : SUCCESS in 45m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/677fc3b434de49b9a2febb8e65a0d564 : SUCCESS in 46m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7db31b7d20c24e99970c0a6e68d9f553 : SUCCESS in 51m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4f4afa0efb0148b28995790c84a7b33a : SUCCESS in 23m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7d0633465b8d408e9439e8ef2f6f7d28 : SUCCESS in 25m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b244b3c3eb7e472c8e9003e8e6003613 : SUCCESS in 19m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e30986742bf54cafb719c6f1e0377793 : SUCCESS in 8m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/af88f1b0cee1412db7fed5bcd7f83b26 : SUCCESS in 4m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/38f232cf244e461abfdb6601fcedccec : SUCCESS in 17m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/30d6e21da28f4b53b74fcf05c335abe6 : SUCCESS in 22m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d33aa23f6e1c4e3a9ad410050a653a48 : SUCCESS in 14m 05s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2d94e8d27e7747658aae93bbe0cefd13 : SUCCESS in 13m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ecd7805b493b42d0b2e97d2a5a71e83c : SUCCESS in 15m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a97b4796306a4df28e69ff29c5740cd6 : SUCCESS in 17m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/aecb9d2400c64934b760addfdec18833 : SUCCESS in 18m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f01d04b0a17941bc846978da9769f7fd : SUCCESS in 25m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ef67a74a9f7d45359fced9a669a9f52d : SUCCESS in 14m 16s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/368c403ff6b24924ab780ddcc3420c46 : SUCCESS in 16m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6ba65887a9b84d788904187720b7dad1 : SUCCESS in 6m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9d955243dae6478da62297ab117e15b8 : SUCCESS in 16m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e347a9201b31457a8ea5fab8bc986f8a : SUCCESS in 9m 58s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2e241d632ebc457fb46cd054df974f34 : FAILURE in 52m 11s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/464845abe1754424bf47b3055f24968d : FAILURE in 56m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/542b54421de24ff383b591d16a7ec9d2 : SUCCESS in 47m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/65c2a20d307f444c93d1ebc332606103 : SUCCESS in 46m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13c15d63e0274072aca015b293af1ee9 : SUCCESS in 16m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e304f4cf4b794e21963e5b02c81e6db7 : SUCCESS in 6m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2afabac8d2eb4be6b17ea7b4b9afb189 : SUCCESS in 28m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d0c7e41b0ff549cf8673dc66d8c9e9d1 : SUCCESS in 58m 02s","accounts_in_message":[],"_revision_number":4},{"id":"b8b61a62592fca1211fc200bead343912c039f58","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 20:47:52.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a3747958955476fb617c0da52cede58\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/31dab83804d3494aaaeb90b412af752f : SUCCESS in 14m 30s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/144b3df57b77489b851abcfd969e0c6e : SUCCESS in 17m 47s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/744ce50c41524fa48b90c24797855757 : SUCCESS in 17m 28s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6bf498f9263947b3b98f12fec0eba6bd : SUCCESS in 1h 14m 47s","accounts_in_message":[],"_revision_number":4},{"id":"92ca7962aa357b58dd0b127c1e25ee7e56b1ca28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-27 11:06:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"83cda3a672d1096f2134264ebc1fdc3172b93f5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 12:27:53.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/334749c7adc44f8faf43234c4a033d70\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9498051023c948e799ed773ab2fb4598 : SUCCESS in 9m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/2a3645bd1c394aad9c3ae3355b105dd4 : SUCCESS in 1h 02m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8d6c0d117111444cbe01346f82419d18 : FAILURE in 5m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19079c38bfed44b2a1071036aa4bcbce : SUCCESS in 1h 01m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6072f3ae888a4325afd0d0c69c00d5d6 : SUCCESS in 18m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d11be3f0d2974702825f995d5c2053c1 : RETRY_LIMIT in 2m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/05cde953b0a34a73b407e8b8b8d70219 : RETRY_LIMIT in 2m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3e21f8fc43554246b839d968c03d1e8b : SUCCESS in 9m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0924e88d3a4b4d829e54d65b44678f8b : SUCCESS in 5m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9dada794a31448f799ebfd7b04d63e29 : SUCCESS in 18m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f6b5755721e64b298f7006d31a6f89ea : RETRY_LIMIT in 1m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a6339b34f651419987d20bff8ed6ce94 : RETRY_LIMIT in 1m 54s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/da5fdb6d759a4748ba60cc3d7c105860 : SUCCESS in 15m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e9b1f87b5499468aa94750f8f112cbdc : SUCCESS in 16m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fc1a0a23f2c5467c9c22efe7d47a6d8b : SUCCESS in 18m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b62227e95adb45239a68024fd9b0deba : SUCCESS in 16m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4606b919bde648bea692a69add51998d : SUCCESS in 17m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d1af783d426f497eb1624ec06769c528 : SUCCESS in 14m 42s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1cc2195df8d4465c87a9ce784550ccd3 : SUCCESS in 16m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6ff7e8f0b04a4ce7a267e35ddb8259d4 : RETRY_LIMIT in 2m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f530a33f6e3d40aeab75347cb1dee0c0 : SUCCESS in 14m 58s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f3adcf207e78457f96200170a7779a7c : RETRY_LIMIT in 1m 53s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6073415c3c6045258ae69c73b9af92d2 : FAILURE in 59m 16s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/8034f24d43ac4be49d5bbf02682b13d3 : SUCCESS in 1h 01m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ebe1009a8e8746e9950421d9c21fca65 : SUCCESS in 49m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4f9687061d0e4189bcc3b4f15484a1a3 : SUCCESS in 37m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd7d423016d1465b93740093fed25c95 : SUCCESS in 19m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9190527bbbe24cf584c68dd0cb8a4dc5 : RETRY_LIMIT in 2m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f17b74c106f945989e3fb6121b6afb84 : SUCCESS in 19m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b31da15030864bec881fe9e1dd5ba705 : SUCCESS in 59m 12s","accounts_in_message":[],"_revision_number":5},{"id":"ed30e7453150fb059f05f37eecc51e3622da663e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 12:40:50.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/887608a89f6d427592962ab9db9db0f6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/740d00e12a1c4f5cad63d6b60a8fccd3 : TIMED_OUT in 1h 30m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/224a8252e4c24ef3a490a71b3111c47a : TIMED_OUT in 1h 31m 13s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/61ea4580be5646f8a1c86bd0eff9ec03 : SUCCESS in 1h 25m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f427cb8152a84728bd6a47738b8593dc : TIMED_OUT in 1h 31m 17s","accounts_in_message":[],"_revision_number":5},{"id":"3b9157a5ec66cea6a7e5574c1e432ebf761362ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-02-27 13:27:39.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0e3c1e7efa560650c43787e358b599e1c3496766","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 13:47:30.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05c435c0cd8243daaf4671158848d45d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9c6ff23e2b3b4396aaae068f6092fa2f : SUCCESS in 15m 28s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3566a49ed50c401ea67c53e58dd8ed99 : SUCCESS in 14m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0a438dc4a70d4a5ea61458465eaf4b43 : SUCCESS in 18m 42s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/dcd3f66e079d477a97fb9ed595472611 : SUCCESS in 17m 06s","accounts_in_message":[],"_revision_number":6},{"id":"05b14339304534d6eadee4de025422c47dfacfba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 14:55:34.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7651fd462714b2da2cada65f8d13438\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a63d0fb7f6f4383b5e38ea55ee1bd06 : SUCCESS in 8m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/f2198cf56db843c3b52dc9d8c1aa21a7 : SUCCESS in 50m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/12860c218ba841d78d2ea5bb7cad7dcb : SUCCESS in 53m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc182af5479c4203871b9c8ecdc5c0da : SUCCESS in 54m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/466ca224225c484b96e5dede989cb4fc : SUCCESS in 16m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/524329244b1b422a8d950d20b96f48dc : SUCCESS in 29m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4128d64cbf804bdabc7ee8d8ea855188 : SUCCESS in 18m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4c289435a8c444788ef12c91233e6dc5 : SUCCESS in 11m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/eef63a78ca094e4fabc03ae11fc21796 : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f575147ed1ce4e4c97bddd275b98fe62 : SUCCESS in 17m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cb2921dbcafc462eacf2c8ab7507d474 : SUCCESS in 16m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/80db62230c824acf926fa2ea5fe55c8a : SUCCESS in 14m 59s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d75c28518a5845fd9b9299d211be37c8 : SUCCESS in 15m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3d75cfbd323a468f913325b6ed7f8ffc : SUCCESS in 15m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/de52d92435934a5088c871535fcdb4c2 : SUCCESS in 17m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/394877ce80bd4c1cafab414b832dbc1e : SUCCESS in 18m 31s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3b2bfc82e13d44c9a51fc559ef0a3618 : SUCCESS in 24m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/899587f6cf0e422ea55d49a8220f810d : SUCCESS in 14m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d9934a9a2c7e40a189b4fba968680b98 : SUCCESS in 14m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/09c0519cde864fdd97b115c702997a9d : SUCCESS in 7m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fceb3c9b79c147aaa19fceb6d260e3e1 : SUCCESS in 15m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9b9fb8f0ba124cc6bc1696346c355c6f : SUCCESS in 8m 22s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f5c76a21b8814c428318bfbfee55a6ee : SUCCESS in 56m 51s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d2ea499465834e01aaaff01f6c61cda9 : SUCCESS in 1h 01m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4831ebc5c83a4a49a55b9deb2e55e3d7 : SUCCESS in 1h 20m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/43964704e8984e91b015e1965653b228 : SUCCESS in 46m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a25dc7c63d624ff4adbfeb0e7f480a9b : SUCCESS in 16m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb9b04af94854516a521ca23305187a2 : SUCCESS in 4m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4401973cf8264b8fbbd2381cf8832d6b : SUCCESS in 18m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/efa6a2d9bdef4630bd4bab826fff7dd6 : SUCCESS in 42m 07s","accounts_in_message":[],"_revision_number":6},{"id":"6457ae69de00e0c6eb816d1de3cfb3090dffaebc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-03-14 19:17:45.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"3f13e09048254d632ee800e7c459fdc543704a60","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-14 17:14:57.000000000","message":"Patch Set 6: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"aa515884aef3485fa43c8d170a3530df8a981131","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 17:15:35.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"e027a99845fbb41d62ab2231de81f2255530477a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 18:18:08.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1ee061574ef442a8c5329c1793b47e4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5396879b7e14493f87f19f88b7ff47e9 : SUCCESS in 8m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/487d7d33de754706a0e72c4ecf74bc90 : SUCCESS in 50m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/888c8c6ea23e455d90ca5c66b0b21978 : SUCCESS in 37m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fcff047142a048f394e258182801d770 : SUCCESS in 50m 44s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4c606cc8e34f4321a27902b57235af4e : SUCCESS in 8m 34s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/702d53dba20441d7b4fa9946ab863666 : SUCCESS in 5m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7a5770804cb448218cdd61466aca413a : SUCCESS in 24m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a6f18f9dcb134bad96e4afa325945385 : SUCCESS in 16m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e1c892d0e79248be9ef278e296d2d5ff : SUCCESS in 17m 48s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/58de67d129fe404694f4c5e0882c7720 : SUCCESS in 25m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/18df1d44c7df4ebe804e871d3db06f04 : SUCCESS in 16m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f5a5d035afbe42a98addde98b5ad56e9 : SUCCESS in 26m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d4dba2e270334de88def36d2d21a1110 : SUCCESS in 17m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2e599b8d6257443abed159f45793dfba : SUCCESS in 23m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1dbe03e334ad40dbb64ffeb68594be3e : SUCCESS in 17m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d7790c03ec08445599575a38edb97b84 : SUCCESS in 23m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/40144b2dd67f40ca933e2ecb3693385d : SUCCESS in 7m 19s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/439d201c8a92468db6fbae5b4d1a6aa9 : SUCCESS in 7m 02s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bf22343227354b1f8234918e5f99cd70 : SUCCESS in 57m 42s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4b3da46d2c8748ee86ee14101d962d5f : SUCCESS in 58m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/55f5565858f54be49cfed4be516c27c2 : SUCCESS in 52m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1c657069f510473c80e9c8af88ee00f3 : SUCCESS in 53m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1441a219a14247698c2b726c017269a1 : SUCCESS in 27m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93d8a2bd74da477293f3437a23ca65b5 : SUCCESS in 4m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96b39d573853439bb3168cf1d49133c1 : SUCCESS in 42m 30s","accounts_in_message":[],"_revision_number":6},{"id":"d6eb3341f5a95f39c47df9a5fbe374fdacf5e2a5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 18:18:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"cd23bd6b0deea097b93cec8ddd8657446ab93fbc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 18:19:19.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/daf4060714e746d2be518d10eb48451f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bd61aea1f8a4fcbb3d803d397101e49 : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f877745ceed24c588e966398e8224424 : SUCCESS in 30s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/2c254702b5d743f2958cc6551214f431 : SUCCESS in 29s (non-voting)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"2fe18b24cd7e709c8712fcda7369b0715a21fb89","revisions":{"83d45cd16c6ebafd784e035aa2b7d078c78df44f":{"kind":"REWORK","_number":1,"created":"2023-02-16 18:40:32.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/22/874122/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/22/874122/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/874122/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/874122/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/874122/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/22/874122/1"}}},"commit":{"parents":[{"commit":"8778b2c1328226cbcb67cd38fbb9e0771e2162fc","subject":"Merge \"Add new Kota\u0027s email entry to mailmap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8778b2c1328226cbcb67cd38fbb9e0771e2162fc"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-02-16 18:39:36.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-02-16 18:39:36.000000000","tz":0},"subject":"ssync: fix decoding of ts_meta when ts_data has offset","message":"ssync: fix decoding of ts_meta when ts_data has offset\n\nChange-Id: I74a0aac0ac29577026743f87f4b654d85e8fcc80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/83d45cd16c6ebafd784e035aa2b7d078c78df44f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/83d45cd16c6ebafd784e035aa2b7d078c78df44f"}]},"branch":"refs/heads/master"},"b5bc8d801d118c445141e0024b0cca048aa5a884":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-02-17 09:46:48.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/22/874122/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/22/874122/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/874122/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/874122/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/874122/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/22/874122/2"}}},"commit":{"parents":[{"commit":"8778b2c1328226cbcb67cd38fbb9e0771e2162fc","subject":"Merge \"Add new Kota\u0027s email entry to mailmap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8778b2c1328226cbcb67cd38fbb9e0771e2162fc"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-02-16 18:39:36.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-02-17 09:38:02.000000000","tz":0},"subject":"ssync: fix decoding of ts_meta when ts_data has offset","message":"ssync: fix decoding of ts_meta when ts_data has offset\n\nThe SsyncSender encodes object file timestamps in a compact form and\nthe SsyncReceiver decodes the timestamps and compares them to its\nobject file set.\n\nThe encoding represents the meta file timestamp as a delta from the\ndata file timestamp, NOT INCLUDING the data file timestamp offset.\n\nPreviously, the decoding was erroneously calculating the meta file\ntimestamp as the sum of the delta plus the data file timestamp\nINCLUDING the offset.\n\nFor example, if the SssyncSender has object file timestamps:\n\n  ts_data \u003d t0_1.data\n  ts_meta \u003d t1.data\n\nthen the receiver would erroneously perceive that the sender has:\n\n  ts_data \u003d t0_1.data\n  ts_meta \u003d t1_1.data\n\nAs described in the referenced bug report, this erroneous decoding\ncould cause the SsyncReceiver to request that the SsyncSender sync an\nobject that is already in sync, which results in a 409 Conflict at the\nreceiver. The 409 causes the ssync session to terminate, and the same\nprocess repeats on the next attempt.\n\nCloses-Bug: #2007643\nChange-Id: I74a0aac0ac29577026743f87f4b654d85e8fcc80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5bc8d801d118c445141e0024b0cca048aa5a884"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5bc8d801d118c445141e0024b0cca048aa5a884"}]},"branch":"refs/heads/master"},"412a9548b36375fd7728aea408504eaf31eb0788":{"kind":"REWORK","_number":3,"created":"2023-02-24 10:17:23.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/22/874122/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/22/874122/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/874122/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/874122/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/874122/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/22/874122/3"}}},"commit":{"parents":[{"commit":"8778b2c1328226cbcb67cd38fbb9e0771e2162fc","subject":"Merge \"Add new Kota\u0027s email entry to mailmap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8778b2c1328226cbcb67cd38fbb9e0771e2162fc"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-02-16 18:39:36.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-02-24 10:17:08.000000000","tz":0},"subject":"ssync: fix decoding of ts_meta when ts_data has offset","message":"ssync: fix decoding of ts_meta when ts_data has offset\n\nThe SsyncSender encodes object file timestamps in a compact form and\nthe SsyncReceiver decodes the timestamps and compares them to its\nobject file set.\n\nThe encoding represents the meta file timestamp as a delta from the\ndata file timestamp, NOT INCLUDING the data file timestamp offset.\n\nPreviously, the decoding was erroneously calculating the meta file\ntimestamp as the sum of the delta plus the data file timestamp\nINCLUDING the offset.\n\nFor example, if the SssyncSender has object file timestamps:\n\n  ts_data \u003d t0_1.data\n  ts_meta \u003d t1.data\n\nthen the receiver would erroneously perceive that the sender has:\n\n  ts_data \u003d t0_1.data\n  ts_meta \u003d t1_1.data\n\nAs described in the referenced bug report, this erroneous decoding\ncould cause the SsyncReceiver to request that the SsyncSender sync an\nobject that is already in sync, which results in a 409 Conflict at the\nreceiver. The 409 causes the ssync session to terminate, and the same\nprocess repeats on the next attempt.\n\nCloses-Bug: #2007643\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: I74a0aac0ac29577026743f87f4b654d85e8fcc80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/412a9548b36375fd7728aea408504eaf31eb0788"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/412a9548b36375fd7728aea408504eaf31eb0788"}]},"branch":"refs/heads/master"},"156287cc42e24344a6679a002a698e6f6df4a20a":{"kind":"REWORK","_number":4,"created":"2023-02-24 19:31:08.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/22/874122/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/22/874122/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/874122/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/874122/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/874122/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/22/874122/4"}}},"commit":{"parents":[{"commit":"8778b2c1328226cbcb67cd38fbb9e0771e2162fc","subject":"Merge \"Add new Kota\u0027s email entry to mailmap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8778b2c1328226cbcb67cd38fbb9e0771e2162fc"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-02-16 18:39:36.000000000","tz":0},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-02-24 19:31:01.000000000","tz":-360},"subject":"ssync: fix decoding of ts_meta when ts_data has offset","message":"ssync: fix decoding of ts_meta when ts_data has offset\n\nThe SsyncSender encodes object file timestamps in a compact form and\nthe SsyncReceiver decodes the timestamps and compares them to its\nobject file set.\n\nThe encoding represents the meta file timestamp as a delta from the\ndata file timestamp, NOT INCLUDING the data file timestamp offset.\n\nPreviously, the decoding was erroneously calculating the meta file\ntimestamp as the sum of the delta plus the data file timestamp\nINCLUDING the offset.\n\nFor example, if the SssyncSender has object file timestamps:\n\n  ts_data \u003d t0_1.data\n  ts_meta \u003d t1.data\n\nthen the receiver would erroneously perceive that the sender has:\n\n  ts_data \u003d t0_1.data\n  ts_meta \u003d t1_1.data\n\nAs described in the referenced bug report, this erroneous decoding\ncould cause the SsyncReceiver to request that the SsyncSender sync an\nobject that is already in sync, which results in a 409 Conflict at the\nreceiver. The 409 causes the ssync session to terminate, and the same\nprocess repeats on the next attempt.\n\nCloses-Bug: #2007643\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: I74a0aac0ac29577026743f87f4b654d85e8fcc80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/156287cc42e24344a6679a002a698e6f6df4a20a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/156287cc42e24344a6679a002a698e6f6df4a20a"}]},"branch":"refs/heads/master"},"c43d64bcb236bb70f3be345c73e9a4c53f0feac1":{"kind":"REWORK","_number":5,"created":"2023-02-27 11:06:43.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/22/874122/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/22/874122/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/874122/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/874122/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/874122/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/22/874122/5"}}},"commit":{"parents":[{"commit":"8778b2c1328226cbcb67cd38fbb9e0771e2162fc","subject":"Merge \"Add new Kota\u0027s email entry to mailmap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8778b2c1328226cbcb67cd38fbb9e0771e2162fc"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-02-16 18:39:36.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-02-27 11:06:34.000000000","tz":0},"subject":"ssync: fix decoding of ts_meta when ts_data has offset","message":"ssync: fix decoding of ts_meta when ts_data has offset\n\nThe SsyncSender encodes object file timestamps in a compact form and\nthe SsyncReceiver decodes the timestamps and compares them to its\nobject file set.\n\nThe encoding represents the meta file timestamp as a delta from the\ndata file timestamp, NOT INCLUDING the data file timestamp offset.\n\nPreviously, the decoding was erroneously calculating the meta file\ntimestamp as the sum of the delta plus the data file timestamp\nINCLUDING the offset.\n\nFor example, if the SssyncSender has object file timestamps:\n\n  ts_data \u003d t0_1.data\n  ts_meta \u003d t1.data\n\nthen the receiver would erroneously perceive that the sender has:\n\n  ts_data \u003d t0_1.data\n  ts_meta \u003d t1_1.data\n\nAs described in the referenced bug report, this erroneous decoding\ncould cause the SsyncReceiver to request that the SsyncSender sync an\nobject that is already in sync, which results in a 409 Conflict at the\nreceiver. The 409 causes the ssync session to terminate, and the same\nprocess repeats on the next attempt.\n\nCloses-Bug: #2007643\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: I74a0aac0ac29577026743f87f4b654d85e8fcc80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c43d64bcb236bb70f3be345c73e9a4c53f0feac1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c43d64bcb236bb70f3be345c73e9a4c53f0feac1"}]},"branch":"refs/heads/master"},"2fe18b24cd7e709c8712fcda7369b0715a21fb89":{"kind":"REWORK","_number":6,"created":"2023-02-27 13:27:39.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/22/874122/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/22/874122/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/874122/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/874122/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/874122/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/22/874122/6"}}},"commit":{"parents":[{"commit":"8778b2c1328226cbcb67cd38fbb9e0771e2162fc","subject":"Merge \"Add new Kota\u0027s email entry to mailmap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8778b2c1328226cbcb67cd38fbb9e0771e2162fc"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-02-16 18:39:36.000000000","tz":0},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-02-27 13:27:32.000000000","tz":-360},"subject":"ssync: fix decoding of ts_meta when ts_data has offset","message":"ssync: fix decoding of ts_meta when ts_data has offset\n\nThe SsyncSender encodes object file timestamps in a compact form and\nthe SsyncReceiver decodes the timestamps and compares them to its\nobject file set.\n\nThe encoding represents the meta file timestamp as a delta from the\ndata file timestamp, NOT INCLUDING the data file timestamp offset.\n\nPreviously, the decoding was erroneously calculating the meta file\ntimestamp as the sum of the delta plus the data file timestamp\nINCLUDING the offset.\n\nFor example, if the SssyncSender has object file timestamps:\n\n  ts_data \u003d t0_1.data\n  ts_meta \u003d t1.data\n\nthen the receiver would erroneously perceive that the sender has:\n\n  ts_data \u003d t0_1.data\n  ts_meta \u003d t1_1.data\n\nAs described in the referenced bug report, this erroneous decoding\ncould cause the SsyncReceiver to request that the SsyncSender sync an\nobject that is already in sync, which results in a 409 Conflict at the\nreceiver. The 409 causes the ssync session to terminate, and the same\nprocess repeats on the next attempt.\n\nCloses-Bug: #2007643\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: I74a0aac0ac29577026743f87f4b654d85e8fcc80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2fe18b24cd7e709c8712fcda7369b0715a21fb89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2fe18b24cd7e709c8712fcda7369b0715a21fb89"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
