)]}'
{"id":"openstack%2Fswift~874184","triplet_id":"openstack%2Fswift~master~Idf9b5118e9b64843e0c4dd7088b498b165f33db4","project":"openstack/swift","branch":"master","topic":"p-ssync-fix-meta-ts","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-04-18 00:01:42.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-04-17 21:33:21.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Idf9b5118e9b64843e0c4dd7088b498b165f33db4","subject":"ssync: Round-trip offsets in meta/ctype Timestamps","status":"MERGED","created":"2023-02-17 01:19:20.000000000","updated":"2023-04-18 00:02:42.000000000","submitted":"2023-04-18 00:01:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874184-p-ssync-fix-meta-ts","meta_rev_id":"5a0dd5b246b4d230eec757ac493fafd37592da80","_number":874184,"virtual_id_number":874184,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-18 00:01:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2023-04-17 12:27:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2023-04-17 12:27:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-17 02:27:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-17 10:13:46.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2023-02-24 10:33:28.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"ee6267c8df75ddf103694c9099210399a980e888","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-17 01:19:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1af25a6e449900083c051c796f7642952e3ec78f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 02:27:20.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/df068379a5b945b78eccf1c616da12b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db332fece7e14a579c3e933045b7b339 : SUCCESS in 7m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/413e55befebb481d873a59b7e7034a33 : SUCCESS in 1h 03m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bb2de9e4359c4051a8f6771bda0ae0ea : SUCCESS in 1h 01m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/16eaf24b712247638db3362b9a2017d4 : SUCCESS in 1h 02m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0b79499071c24225a2c8da30ed7c6b77 : SUCCESS in 18m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/35aea0fc73c0471293a2e4f8af917899 : SUCCESS in 19m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/77c03d477d10433a8b37d66799dd8272 : SUCCESS in 37m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4004f9206be340dda50ada40854cb106 : SUCCESS in 9m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e2e01567b32845b1af4ffeeb0d5c1393 : SUCCESS in 6m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/27d3940c847d4bde982b945e8498dadb : SUCCESS in 39m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f4211cc75af84b019418227976c76381 : SUCCESS in 15m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/885ce358a6934bd4bbabe0909d73f990 : SUCCESS in 15m 55s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9352b475b8694f3bb3cce7834bd01a5a : SUCCESS in 33m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ba0383fd50cb45a89260c996fb2e3f72 : SUCCESS in 16m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/002560fe76154a3088b488e451a47e86 : SUCCESS in 34m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4337c0a5da944e748b9932fdd3dbefed : SUCCESS in 31m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/37de400f8b994249ab9cb4d3e0421b83 : SUCCESS in 15m 55s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ebae0116d84d42c4bdb62af107b6e61a : SUCCESS in 33m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/11eb119b59ef44169c6dafc8159cff77 : SUCCESS in 15m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/aad06d7c3e8244b5b76637cf3fca2176 : SUCCESS in 5m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/93f4f18ca06b48d187262f276921ff1f : SUCCESS in 12m 24s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/95f9bc94bb8e4802bcb74f1384301764 : SUCCESS in 9m 30s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4c11387ec4a746f7ad0bae145c8b6efe : SUCCESS in 55m 09s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c5266bc506ae4a60979ee43a2d0ca68e : SUCCESS in 58m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c90e0c3ef1bb4cd8b1aa10405f4e97b8 : SUCCESS in 33m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/92a603656eaa48f9b00250e9511fa02f : SUCCESS in 59m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a6841f373f1342ab8aea01704ec85090 : SUCCESS in 37m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f047d402c718435b8f502eebe1118f6e : FAILURE in 6m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c7f5bfb51cf2453c89aebcdb625a219c : SUCCESS in 33m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/509658963a40458883f9c3b73819522e : SUCCESS in 55m 15s","accounts_in_message":[],"_revision_number":1},{"id":"41b69732961cc05507aba496aa82b3fe1e87fad9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 02:52:48.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e991e1dbdca426e9e7d74b518c3d285\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0ea9b8f49558499890ecc876a14a5f1b : SUCCESS in 15m 36s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8697a54ef2d84243841ccf70d5e3449b : SUCCESS in 1h 17m 20s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/27d745e4cff84053bcb19d2b11d963e0 : TIMED_OUT in 1h 31m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/12990434b5d74d0086d8572a7d9a48b1 : SUCCESS in 19m 47s","accounts_in_message":[],"_revision_number":1},{"id":"9d94bbbf28e69f0c7f2749f8d2171140b03ab8e3","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-17 10:13:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9e2cd59356b5000ba3f1bf2dd21b5ef0254e3319","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-17 16:38:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"36177b934c737e5c986416551ed360a159376c46","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-22 15:27:06.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4add60bb5b85332d5db1a42f0ee5663a9ac4f821","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-22 15:40:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"43dfe136a4df72a4b62b1b13a32f024868f53b7e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-22 15:41:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7b579bf455923f34fbb981b147d5c36627b70f59","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 18:34:22.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10796bbbf9cf46bdb0fccdd8d62c4548\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4583ace2aa104ff8a2982400caf16acd : SUCCESS in 1h 07m 30s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e527034a7db04272afb041a817bd05c6 : SUCCESS in 1h 12m 22s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/de3499b3aeb8417ea7cca1f5e262dbcc : SUCCESS in 1h 27m 52s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b8d89f4882c04dee85a1bf93689e68e0 : TIMED_OUT in 1h 30m 56s","accounts_in_message":[],"_revision_number":2},{"id":"433cfaf88ea57e956c61d360576f971ac76b6539","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 19:51:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f337e001a09140a68b41de05e2cae197\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e1b941083ba4259b2b062fc345bf32f : SUCCESS in 7m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/87b247914dad4e168fea6a942b5764c3 : SUCCESS in 1h 05m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7ac552a41d1f4dafa5fec53a65ad4997 : SUCCESS in 1h 07m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42cb221093dd44379866c0237e486448 : SUCCESS in 56m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/532663fdc4e3451c9d3ec2f2bdbeba43 : SUCCESS in 24m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/18440a8827b8413b80cb55c3bd70a051 : SUCCESS in 19m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/18fd547774314c1585e0e868f5174bf1 : SUCCESS in 18m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2f5c9b38e48a44ba8c0a5cbb5105ae00 : SUCCESS in 8m 46s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e9fcf63bf05744e0b137627a70fa00a6 : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/06777a9db3d446f687b5d7298ef84e6b : SUCCESS in 20m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6811a9d3d7244da989f26e72ac435e96 : SUCCESS in 36m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dcf0b5b1fa0748c7b61fbc2cc248b8bb : SUCCESS in 20m 59s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e75d380feb014f19af45d70e898b1a26 : SUCCESS in 17m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c0f061f845d2438dbe1a5b7a6e706335 : SUCCESS in 17m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3f6c5c3839574cd0b39bbbcb71d7822d : SUCCESS in 16m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/db2534ff6d8144debc67b8061dfddb5f : SUCCESS in 31m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ec36d204c6474ab3886583dd49ed2656 : SUCCESS in 26m 55s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/54916bc47da24888b12dcd06684a3dbd : SUCCESS in 15m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3eeb76f3824a40e28f27d666fbde3972 : SUCCESS in 16m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/04bbeb29e8be478eb72164c13a4ab34e : SUCCESS in 6m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cb04824531db48909b3078e09eb3d45a : SUCCESS in 14m 49s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2d8546b0fd074a49bf0e3db8b941cdb7 : SUCCESS in 7m 39s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1ef9605c788f4945a799781ad4ce4d6f : SUCCESS in 1h 22m 42s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ed8f2638ae124e8a9fbd0e18757392d6 : SUCCESS in 58m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/25d01b442f64410599bb241284a037c2 : SUCCESS in 42m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f84d9d00e3c0424a9463109cb2cf3109 : SUCCESS in 1h 01m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5eca340028404b53afa6bc0ffdcadbb2 : SUCCESS in 15m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0165770e5b24919a65fd97996715285 : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/932034d409a84e93b07099eca18f9472 : SUCCESS in 26m 41s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fdaf4754771e4e0aaad6ba357aa2faaa : SUCCESS in 49m 49s","accounts_in_message":[],"_revision_number":2},{"id":"9cbfb37914fcccad11ecb422e225261d710269fe","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-23 20:09:44.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7b9d2076a03fe01bc6feb61b48510d4675ba6009","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-24 10:19:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8eec06c47a01e0b2ad0c17904da2343f1941f5ad","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-24 10:33:28.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ce1c76408f5660038ad9dd5a3492d35c6b32955a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 10:56:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad6701f78f564bf5a4aa55f48c2d074d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/99dc915763bf426a9493e9214167e3d5 : SUCCESS in 27m 08s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f3bc93341ed94ce5839cf7dc97344b73 : SUCCESS in 23m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2611c2c7b6ec46bba6d52970dc3e7a63 : SUCCESS in 34m 51s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b3514139a74e44c0a78af09394689a59 : SUCCESS in 31m 27s","accounts_in_message":[],"_revision_number":3},{"id":"8005d5397eb50c27eff9abb9b5fe7cbe1436f113","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 11:36:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/abe39ca9805040989bbeedef74b742a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db3259916a854f75a354c78cbc5ce9ee : SUCCESS in 8m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/60797e2e79b54c7580b3938b7e231ac3 : SUCCESS in 49m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b23dbb95d60d47f4906ac7ab554a722e : SUCCESS in 45m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/23f7b55388ac451fb1b933ae2864b506 : SUCCESS in 1h 03m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7a8c0ab048704e598981293450f13d15 : SUCCESS in 20m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a3cdfae3273d4894a12c68eb27313056 : SUCCESS in 17m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1ac4cd553eda4ab4befe73200681e879 : SUCCESS in 21m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8809999fe5a24974bee664826bc865c3 : SUCCESS in 9m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8715f049b8ca4c5fa2dda9e9f664a03e : SUCCESS in 5m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e6ca476cb1cb4a1999db2ae97e060e5b : SUCCESS in 16m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/51be9f3645c242c0a8b99ef79a4cbd6a : SUCCESS in 15m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9317643365114405a781edf73b4da3e2 : SUCCESS in 15m 25s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a22aee2029dd4b5697fc8735c831a931 : SUCCESS in 13m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/477b7a436da249c68f8b1c1e22955700 : SUCCESS in 18m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f75e5fc2faa34aee99abcb8549e97fbb : SUCCESS in 14m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1a31a0b0de9e4666a23c2b6366838875 : SUCCESS in 19m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7fd3e1459e20454dbd8f418f4cf9ddf7 : SUCCESS in 14m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3b5e421bb20e40d3969c35d962171e3c : SUCCESS in 12m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5341f1e5cb51457d8e7d9a6bebb55887 : SUCCESS in 16m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/402471843f8f486aa7344fbc52cb55b3 : SUCCESS in 6m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/105c59e13c7149cea192c6a2a4ba4b5a : SUCCESS in 14m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2d2b32f34aed47f6b1871694e0a1c16a : SUCCESS in 8m 24s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6207e6e65e654ba08e0613bfb200a023 : SUCCESS in 41m 55s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4ab012ff40a548fdba34acfc04a4572d : FAILURE in 1h 01m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b7f4b61d121740f8b9e9e06ec3c2e6cf : SUCCESS in 39m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9a7640b911204a419b7ad1e400b89fcc : SUCCESS in 54m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e61f58c0d6040518c8e64438825db87 : SUCCESS in 14m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81871633c4f3422bbea32623f68808cb : SUCCESS in 5m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/92ab9b37a9224d0dba83d287fa0ae538 : SUCCESS in 18m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2960748af3e344d281e8c15f0f08cbf0 : SUCCESS in 46m 23s","accounts_in_message":[],"_revision_number":3},{"id":"549edb8497fb3e2d1cd6b7763833e57fb2c1dd38","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-03-14 19:19:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a86df42e8af1ed129c9364f871c585646c2e515c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-03-15 15:33:25.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b84501444c9b1a8f679ade184cbf58f2bb6394e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-04-17 12:26:05.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"7a545cbd06e0330590b5f609d5a2b4c1dd8ffbb1","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-04-17 12:27:13.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"1683ec38fbba295e9f754b3028a614eaf584b6c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 13:34:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02e5fe6172404a0490186f2ad2abde81\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aba6990cb576447686177efc3bf5f5c8 : SUCCESS in 1h 06m 18s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/57bbc050f44744e3ba17e856d63939c0 : SUCCESS in 14m 33s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f5751d2da81b460c875fde375656c102 : SUCCESS in 21m 53s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/84acbfb672f44988967e0ee6f4ace506 : SUCCESS in 16m 16s","accounts_in_message":[],"_revision_number":4},{"id":"7c8cf22fa0957b56deeb9e97e35fbde24f538fd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 13:45:38.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/278bad4b006e4119a593b88d022e7a9e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/414ecbbf3f794162b51d8ec296da1f56 : SUCCESS in 8m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/747788d38474480495156872fd43187f : SUCCESS in 1h 13m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8437200026204af8b02a1fa0f68e70eb : SUCCESS in 45m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/72e3e04a0004447093ab47ff9300abf9 : SUCCESS in 1h 03m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a9296c90f4df4d348cbfe7ca476e864f : SUCCESS in 28m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d6386697b06248d18f83883151ff5808 : SUCCESS in 15m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0a80557c48734d7e9fe817528be8e52e : SUCCESS in 15m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4d26d5657cb54861a11dcf19b81537bc : SUCCESS in 8m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/06e28a82898546ceb7b66b5b911f0f4d : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d2bbc632862b4e3cae22f2d117cc1cd2 : SUCCESS in 16m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bc1a644b555948fea6b33a6e7063910e : SUCCESS in 18m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/525b392c6bd247e0a2398f6a56381290 : SUCCESS in 17m 13s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4953cdf7b55145b2a44487b31a2c0a7d : SUCCESS in 15m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/356ac2c65efb4417a29224fddd14b9b6 : SUCCESS in 15m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/af2f2b0d7225401f8f885a9ea7e69f40 : SUCCESS in 16m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f95f60cb11c84198895154b9187731e8 : SUCCESS in 20m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b01853fafb5d40278c62fc3f985c62f3 : SUCCESS in 14m 44s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fba977be5ab940eaba0c38f63336d118 : SUCCESS in 15m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/250b377f5244405991aba6cf1b97f211 : SUCCESS in 14m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/65710ef8e5d94a409dd90ee5a4ae2a1f : SUCCESS in 6m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2c254fe5e49f46498275691ac96fc605 : SUCCESS in 18m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ac9b5f8a8bac4889a4c564ac88e24ce7 : SUCCESS in 7m 28s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b8505514362d48fcb92766932cb590e8 : FAILURE in 54m 01s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/414b199155c44e208955862876705d15 : SUCCESS in 1h 04m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/eef01f20953c4c35af8e98ffcccc3595 : SUCCESS in 1h 00m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/81fbc64362e54693a61cda63010197c6 : SUCCESS in 53m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/67260364e6024de7a4a5d2feeb771136 : SUCCESS in 16m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f208545d2fba44c3a5ea65e835189a46 : SUCCESS in 5m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d3fb42f86b27488ab52e8e1f001f915f : SUCCESS in 19m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0fb0e44984a4895b6b425a0eff3cb42 : SUCCESS in 59m 13s","accounts_in_message":[],"_revision_number":4},{"id":"3d4937167c331b2b3b24524618192d71a32c980c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-17 16:24:37.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"364404b5536e09771b39bf2acbade1b8d271dd6a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 17:58:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fbda9c71d2c42c083da0b0129ff3bc1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/112042e6f9674a44a2b22017aa193fc2 : SUCCESS in 16m 12s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0cbee0d21f9a4c50a8160623193437d1 : SUCCESS in 1h 31m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/74679ab4622e48e88eb68e1cddef281a : SUCCESS in 25m 15s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/04166e7d72644a189e95d58cc9612986 : SUCCESS in 21m 13s","accounts_in_message":[],"_revision_number":4},{"id":"d8b53e421ad01fb4b57dbba9be58873ed7c88ee2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 18:00:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce23e34c6d57414fa27d354efc512792\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a19a664448004380a55d4261416ba7cc : SUCCESS in 17m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/1aaee5f02d65472ebdbfc09844d925e1 : SUCCESS in 49m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/daee2764f208450090db08b2782a93a2 : SUCCESS in 47m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3ec36fd79d10492f9571bb40173f1ffe : SUCCESS in 1h 00m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2d04362af1da47e4b2fdca3085838794 : SUCCESS in 25m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/304f3c7053254293b9f27fd752f54343 : SUCCESS in 16m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6696a80ea91e4c9cb1b9a263c1f11311 : SUCCESS in 15m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/620015bb5ee2449cb73e7e29568f1189 : SUCCESS in 8m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c0b328b48c3740e98b4f706f6ba5c1af : SUCCESS in 6m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0137feca28f24e72ab9da86d4dc10bc3 : SUCCESS in 14m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f25a0b2a83674a68a6526199e3c8a0a2 : SUCCESS in 14m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e4ef6ea05aeb4fdd8e5e4788514041c3 : SUCCESS in 13m 14s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1046acae26ba4c04a537f2553f8eb607 : SUCCESS in 14m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/83d2e7ae889746a9b4713154c02ea322 : SUCCESS in 17m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f6c6a843c68344d39cbdaaf53f4491af : SUCCESS in 18m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6cc2d379b2bd457da616e09c87bab6ed : SUCCESS in 17m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/134d1e7a2b8045ba8ba6201a1a6cf5c6 : SUCCESS in 14m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b7dba1975bb14d58ba75f365dbefa1ea : SUCCESS in 13m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a2cd9a0d05be4646814d4fe6c55d6382 : SUCCESS in 11m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/643f9229c8be4886b6eee1dd0c26b00f : SUCCESS in 6m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bdf84e44f24849d2b5f37e986550db12 : SUCCESS in 15m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f95a695fc13e4900ab2582969cf2fd3a : SUCCESS in 6m 21s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cdfa80de2fe747b3975055c00104cf8e : SUCCESS in 41m 41s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/25371727c0a04cd9a3b57fe4b39b8749 : SUCCESS in 1h 06m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/84ecc076b5cf49f6bf3b4d8ff44ac3fd : SUCCESS in 39m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c5a2e67a576c49ad90b03bf5a48012bb : SUCCESS in 38m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/529074f33bb740a69999ef32e8b769ad : SUCCESS in 18m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abec1f09f60c40ecb4afa8f268110002 : SUCCESS in 4m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dbf107940b2640a88b816ddf626f7d4d : SUCCESS in 18m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a7a943ee83e427485344cc85bf9e6b5 : SUCCESS in 1h 01m 18s","accounts_in_message":[],"_revision_number":4},{"id":"ce7c517b4299d93fe94d9916733969d8cdcfb4b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 18:00:55.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9d21b4bcb2b80f0ef3fac46f9c005da3c95f378c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 19:53:50.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/38851874320f444b880be1f3c651e93d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9811b24c783d47b2a83ffec50699fe43 : SUCCESS in 11m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/4b7d82c3d7994b658f4d95160ec57468 : SUCCESS in 38m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/38c2da67c299439b817239e69dde8785 : SUCCESS in 41m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a3a6c24614224f64bb1a246824bc5f56 : SUCCESS in 56m 16s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/9b289c5c6b7f4530ae7f3d04eed3471c : SUCCESS in 9m 23s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/7fd74210ebaa45f6b69aa1f8d36bed26 : SUCCESS in 5m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b995f6d8089b4b49abd1f7d4797a7229 : SUCCESS in 16m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d230b746ce774583b32885259e64b331 : SUCCESS in 17m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/556003da5ffd4305980055c57db50435 : SUCCESS in 15m 07s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/021ef156715442eebe21301b2bf78b48 : SUCCESS in 17m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b0a6cc2c82c347b69af54c85db1c6438 : SUCCESS in 16m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ff0e697d56f845cb812d3aafa3e07716 : SUCCESS in 23m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/263e53a1e7c34bd6b3bdfd7f2f8012bf : SUCCESS in 16m 44s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3de3de828d664227ac4abf525c1f9f9d : SUCCESS in 14m 10s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d1aa101a9859474789c38ec468c1e01f : SUCCESS in 17m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/988e41b14d714879a38e31b1ad55b51e : SUCCESS in 13m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/67a80bb096ea4e73a0731adf7a5e417b : SUCCESS in 7m 28s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5a453d63cbe64a748adde2af73d74865 : SUCCESS in 6m 16s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/adb663c62133480eb739112ecb2d2a66 : FAILURE in 55m 56s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f69d60e1d3aa4c1399addddeb1c6d13c : FAILURE in 48m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2be933b5b6a2456d84604bfd4461643a : SUCCESS in 51m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a0eb47a789a74a2cb47a86200f15e123 : SUCCESS in 53m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1193112bf8fd42be8da66fa0f4f707e2 : SUCCESS in 17m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0bd073e06cc46b181760edc71da277e : SUCCESS in 4m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1f719c9066543508cb43ecec10611f2 : SUCCESS in 55m 38s","accounts_in_message":[],"_revision_number":4},{"id":"d234fd24768956bb8d88fb4864c7b521c39d00da","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-17 21:33:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"40fa018fa9d9ec3f11a06bbae07970752e27143e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 21:57:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffa75cbb5d404581be053520c6567b27\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5bcdd93eca8c4f2b869599c9ec800ee9 : SUCCESS in 14m 15s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c81b54d7e90641d2adced815acaa451b : SUCCESS in 14m 48s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3d7594502665444592c02ab871164238 : SUCCESS in 19m 23s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1f4b53ba1f0e4fcfa03e0a1ad798c2aa : SUCCESS in 23m 14s","accounts_in_message":[],"_revision_number":4},{"id":"b0837aa95084287e093dc052747493e4c3196a41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 22:45:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/311d1ee155e24f598600156dcf201dae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1a99f7e985e4f889365d24ae58dadc4 : SUCCESS in 9m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/bfd20a8837d6457685e0d11824aa3d56 : SUCCESS in 1h 03m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3d98f53cd34a4e74a2b184e1c66e66cc : SUCCESS in 55m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/487f714ce6394f0ab768947c6d76b7b6 : SUCCESS in 53m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1f794986630d4626a3e36121d8468b7c : SUCCESS in 18m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ac82dcb1c3c242a4903be9177a9db754 : SUCCESS in 17m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/782d0438eae047eda9a408c82c7a8c47 : SUCCESS in 13m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1d1ba60ca9a1424b9eb887579cd0dd84 : SUCCESS in 8m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7dc3c88b11b94fdc95eb4f3461fa52b8 : SUCCESS in 5m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/814d865fdf774841b4502e8f90b917cf : SUCCESS in 18m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a12a6bef99c7479e83f1702b04836aa4 : SUCCESS in 15m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0d99f8d291e4429783ec257f035059ba : SUCCESS in 14m 53s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8d9de32ad1e24a8ea0d26f13d1fd199b : SUCCESS in 34m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a55cfa0cfe4b44049b8e13f3b61eed71 : SUCCESS in 37m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/dde2dfa6f71640d78c02b5690dc3f8bd : SUCCESS in 15m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5321654067b64a858d5c8b789c6e4541 : SUCCESS in 17m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/63f0d23511914ff29f9ad5a4df220d14 : SUCCESS in 31m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/729beb3df6df48b2b694ada06efd42ad : SUCCESS in 33m 29s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b50d710caacb412e98dcf94ea7f883c7 : SUCCESS in 14m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6c97307b9e0646339bf784ac1d20e68d : SUCCESS in 6m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ed6d0ee174c943c99a8b5cc848113e2a : SUCCESS in 30m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d329337ed2a3404eab59c22d1045adac : SUCCESS in 7m 31s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/43404bde2839464eae7053d4335174a6 : SUCCESS in 54m 24s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e7fb3714b7ca48c2b5e91c3b1cc7ac8d : SUCCESS in 1h 05m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7e4398d44546434f9de9becf3957bd98 : SUCCESS in 53m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6327b1ec9b984202b36a67e852e69adc : SUCCESS in 50m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/206b73d319404b76a7e60598d4dfa993 : SUCCESS in 16m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e541641417d4ad9ab720220bd22ef80 : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2153f0639edd48b08b7d49312e2b3886 : SUCCESS in 17m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a545795810634b538d75cfe7d61d9bee : SUCCESS in 51m 04s","accounts_in_message":[],"_revision_number":4},{"id":"b5993f2d5c6ccaf1d7be9853283e7e41f7d952cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 22:45:46.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ae67dd7b16253de77bc2030be010be20bfe7f8cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 00:01:42.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/730e172e783242138aa918fec351002d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcd7c0f867454b7b934704d6ac3dfcd5 : SUCCESS in 7m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/a86f3d4e65684e609516738421647135 : SUCCESS in 55m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/09286bae5aaf44dc8781710d4f2ee239 : SUCCESS in 39m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/133a9f79917748bda001c63bb6d791ce : SUCCESS in 52m 39s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f79d9bb02ac84e15be0b414091714037 : SUCCESS in 8m 59s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/7653981bcb46489d9e1322f600118c9c : SUCCESS in 5m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f5c638df9d6a4cf6ad104df9e83c8b69 : SUCCESS in 27m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e583a51210704f6cbd4c6a8c7485ced4 : SUCCESS in 24m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d7d99634b8254b28bf57a457276033f6 : SUCCESS in 14m 27s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8ba76dcbdf434dc28fc516ee203ca18f : SUCCESS in 15m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6a24818f655246c7a6ba2ed4ed5d18d7 : SUCCESS in 24m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9ed2a6c1681b4867aaad1981d0f65fd1 : SUCCESS in 26m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8ae61520a638425eaced51462a69f9b7 : SUCCESS in 18m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/89dec902eca941109386ee892b60f501 : SUCCESS in 14m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/481ea5b88015499aa31f963a7d1fa6ac : SUCCESS in 25m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d6804254aafc4fb3b6b88daea938b20a : SUCCESS in 24m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6b22f8d23bd64cb6852f38020ee81971 : SUCCESS in 6m 58s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8f8d4f692120466cba7249b97e62bdc8 : SUCCESS in 7m 15s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bb2fb7df55584e6dbe6a4d80bac83f6c : SUCCESS in 56m 01s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/035ef810e2dd4916a26c6c3c004cb481 : SUCCESS in 1h 08m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7d72d9c36f4b4d5596cfc3295a5f7fe6 : SUCCESS in 56m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e5689b21377e4a01b748d533adbcb951 : SUCCESS in 1h 08m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4a379ee8f5a4f079c232238467e7e59 : SUCCESS in 15m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b89f8adb1c05417bb12105fe40e7ccd3 : SUCCESS in 5m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/121bee8b77904e0583c7fbce10dce099 : SUCCESS in 47m 07s","accounts_in_message":[],"_revision_number":4},{"id":"978e91aa7a3bd721dc44c38b875217f19eb5c5d1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 00:01:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"5a0dd5b246b4d230eec757ac493fafd37592da80","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 00:02:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33cea57c5d09474cb85b2f6e1823ba1c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56085ea8ba3f481a9deb9c6f3c3e1572 : SUCCESS in 41s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/68447b33a1d644fea2dce11d3c043128 : SUCCESS in 35s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/37dc4b6192334952b9c73d5e4dfc8019 : SUCCESS in 31s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4e74e7f558c44068a45556be02c184a753476c56","revisions":{"822a0a0bf88fef9939c3232d1dd9b356f0c65a66":{"kind":"REWORK","_number":1,"created":"2023-02-17 01:19:20.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/84/874184/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/874184/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/874184/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/874184/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/874184/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/874184/1"}}},"commit":{"parents":[{"commit":"83d45cd16c6ebafd784e035aa2b7d078c78df44f","subject":"ssync: fix decoding of ts_meta when ts_data has offset","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/83d45cd16c6ebafd784e035aa2b7d078c78df44f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-02-17 09:16:54.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-02-17 09:18:25.000000000","tz":-480},"subject":"ssync: Round-trip offsets in meta/ctype Timestamps","message":"ssync: Round-trip offsets in meta/ctype Timestamps\n\nUse double-underscore to separate to ensure old code blows up rather\nthan misinterpret encoded offsets.\n\nChange-Id: Idf9b5118e9b64843e0c4dd7088b498b165f33db4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/822a0a0bf88fef9939c3232d1dd9b356f0c65a66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/822a0a0bf88fef9939c3232d1dd9b356f0c65a66"}]},"branch":"refs/heads/master"},"8fb7f35d5e126f3ecccf478a82017a1cebcb55e1":{"kind":"REWORK","_number":2,"created":"2023-02-22 15:40:54.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/84/874184/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/874184/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/874184/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/874184/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/874184/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/874184/2"}}},"commit":{"parents":[{"commit":"b5bc8d801d118c445141e0024b0cca048aa5a884","subject":"ssync: fix decoding of ts_meta when ts_data has offset","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5bc8d801d118c445141e0024b0cca048aa5a884"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-02-17 09:16:54.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-02-22 15:40:31.000000000","tz":0},"subject":"ssync: Round-trip offsets in meta/ctype Timestamps","message":"ssync: Round-trip offsets in meta/ctype Timestamps\n\nUse double-underscore to separate to ensure old code blows up rather\nthan misinterpret encoded offsets.\n\nChange-Id: Idf9b5118e9b64843e0c4dd7088b498b165f33db4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8fb7f35d5e126f3ecccf478a82017a1cebcb55e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8fb7f35d5e126f3ecccf478a82017a1cebcb55e1"}]},"branch":"refs/heads/master"},"5f7e0df126715d82a33fb025e656098fa7234b56":{"kind":"REWORK","_number":3,"created":"2023-02-24 10:19:51.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/84/874184/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/874184/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/874184/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/874184/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/874184/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/874184/3"}}},"commit":{"parents":[{"commit":"412a9548b36375fd7728aea408504eaf31eb0788","subject":"ssync: fix decoding of ts_meta when ts_data has offset","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/412a9548b36375fd7728aea408504eaf31eb0788"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-02-17 09:16:54.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-02-24 10:19:27.000000000","tz":0},"subject":"ssync: Round-trip offsets in meta/ctype Timestamps","message":"ssync: Round-trip offsets in meta/ctype Timestamps\n\nUse double-underscore to separate to ensure old code blows up rather\nthan misinterpret encoded offsets.\n\nChange-Id: Idf9b5118e9b64843e0c4dd7088b498b165f33db4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f7e0df126715d82a33fb025e656098fa7234b56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f7e0df126715d82a33fb025e656098fa7234b56"}]},"branch":"refs/heads/master"},"4e74e7f558c44068a45556be02c184a753476c56":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-04-17 12:26:05.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/84/874184/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/874184/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/874184/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/874184/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/874184/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/874184/4"}}},"commit":{"parents":[{"commit":"2e89e92cb77bd5341a1a290bdfc9402b02f71ab6","subject":"Merge \"ssync: fix decoding of ts_meta when ts_data has offset\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e89e92cb77bd5341a1a290bdfc9402b02f71ab6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-02-17 09:16:54.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-04-17 12:25:56.000000000","tz":60},"subject":"ssync: Round-trip offsets in meta/ctype Timestamps","message":"ssync: Round-trip offsets in meta/ctype Timestamps\n\nUse double-underscore to separate to ensure old code blows up rather\nthan misinterpret encoded offsets.\n\nChange-Id: Idf9b5118e9b64843e0c4dd7088b498b165f33db4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e74e7f558c44068a45556be02c184a753476c56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e74e7f558c44068a45556be02c184a753476c56"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
