)]}'
{"id":"openstack%2Fswift~874806","triplet_id":"openstack%2Fswift~master~I106103438c4162a561486ac73a09436e998ae1f0","project":"openstack/swift","branch":"master","topic":"open-expire","attention_set":{},"removed_from_attention_set":{"4608":{"account":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"last_update":"2024-04-26 10:28:47.000000000","reason":"Change was submitted"},"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2024-04-26 10:28:47.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-04-26 09:19:50.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"36763":{"account":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"last_update":"2024-04-26 10:28:47.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-04-25 22:00:26.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"396":{"account":{"_account_id":396,"name":"Mandell","email":"mdegerness@nvidia.com","username":"mdegerne"},"last_update":"2024-04-26 10:28:47.000000000","reason":"Change was submitted"},"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2024-03-13 19:12:28.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-04-26 10:28:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I106103438c4162a561486ac73a09436e998ae1f0","subject":"expirer: account and container level delay_reaping","status":"MERGED","created":"2023-02-22 20:14:59.000000000","updated":"2024-04-26 10:29:46.000000000","submitted":"2024-04-26 10:28:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":167,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"874806-open-expire","meta_rev_id":"c1024a8455e90bf329eef25f0e653ddaa8b146ee","_number":874806,"virtual_id_number":874806,"owner":{"_account_id":396,"name":"Mandell","email":"mdegerness@nvidia.com","username":"mdegerne"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-26 10:28:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":2,"date":"2024-04-25 22:00:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-25 22:28:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-04-25 22:28:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"reviewers":{"CC":[{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"}],"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-22 20:31:02.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2023-02-22 20:31:19.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"REVIEWER"},{"updated":"2023-02-22 21:53:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-27 19:17:30.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2023-05-04 20:17:36.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2024-02-14 21:14:49.000000000","updated_by":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"reviewer":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"state":"CC"},{"updated":"2024-03-08 05:33:54.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2024-03-08 05:38:04.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2024-03-13 19:12:28.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"CC"},{"updated":"2024-03-13 19:12:28.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"state":"CC"},{"updated":"2024-03-14 14:27:27.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2024-04-25 19:47:15.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"cc396106954d77421eda2de20c6c861569b0ec96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":396,"name":"Mandell","email":"mdegerness@nvidia.com","username":"mdegerne"},"date":"2023-02-22 20:14:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f501b54be83fd1e0dbff5943a9cdcc40d9a15fa7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-22 20:31:02.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"9b8a35bed347cc64a66c365c097b9afce8673fe8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":396,"name":"Mandell","email":"mdegerness@nvidia.com","username":"mdegerne"},"date":"2023-02-22 20:38:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7dc441433b2f710f2d57d7c87391915468d5c2c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 21:53:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c54cd42270e4e318e4297d89332058e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1af69c9b4ab242a3b9ed2f1236086db3 : SUCCESS in 8m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/5843fa80bc034cedaa43670c165ab705 : SUCCESS in 1h 05m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a9c34093904c4100a155a7ab78dce334 : SUCCESS in 42m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a48ea7543624fab80eb362787f9dc37 : SUCCESS in 1h 01m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f23995398ec40779c0b9f5ef5e52a63 : SUCCESS in 19m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/31d01690331c4f318ebb838a98821e3b : SUCCESS in 17m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9af40c8d098c45cea85c305e1f929460 : SUCCESS in 19m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/14e76dd20d1b4ea9b2f7f9361726bdac : SUCCESS in 8m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/34fb91e41af54292b52b0491aa4a0125 : SUCCESS in 5m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b163629365474d70a3e72ff4307806cd : SUCCESS in 16m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3d27874dbdd14e39b8864ddc5c43f20a : SUCCESS in 16m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5f995a8a84324c51821ca02ce24be9db : SUCCESS in 18m 27s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1a8a4255a7e049b89f16af77725642ca : SUCCESS in 16m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6d63a2f73fb74b9b9d14a7971ac3ffcd : SUCCESS in 15m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/97b8078179a94bab8c4bf32afedc75e2 : SUCCESS in 18m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b1730034858e4e11ad029c3065a082b4 : SUCCESS in 17m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cf4f7519310c443491af7b871899cb74 : SUCCESS in 21m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9e5d196b60ea4d658655bbce9c56569e : SUCCESS in 22m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ab985ffdbcab472eb0e95cdf95c45b05 : SUCCESS in 22m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8be6887ed0a64bd5bdf8d59bcd60d9ae : SUCCESS in 7m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/39b7c1821c54475dbc551b420a69b95d : SUCCESS in 22m 18s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/69c5b62062ce4831ae7950148a3eb415 : SUCCESS in 8m 17s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d26a29059f244a36a9056c1770203784 : SUCCESS in 1h 12m 39s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/1aace12dd0864ee585df00f060e70947 : SUCCESS in 1h 10m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8df5e5545e7a42678cacbce365753bd9 : SUCCESS in 47m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/007e07cdb8624b358522c8eebf5ffbdf : SUCCESS in 54m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34e87db61d424e21873513a6b74e3a19 : SUCCESS in 23m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5ad5bf676924543b916c5c22dfab76d : SUCCESS in 5m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fc0b099020f549bfbd6948e265772bd1 : SUCCESS in 31m 41s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/610ce65b477340bb86c5941cca9554f7 : SUCCESS in 49m 05s","accounts_in_message":[],"_revision_number":2},{"id":"aaee30cba573c7446eff4be4b30c1280084d9e6d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 22:08:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd22825e685a49c89bd20981d195e896\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/28d8a211ab4441689953e04c9adae600 : SUCCESS in 1h 15m 33s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1b8f495b11a6469a983e864c27b48688 : SUCCESS in 1h 11m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9d0984bebfde471ab1995ff09b76b76f : SUCCESS in 1h 29m 29s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9a7719a68a2441619f709e3528e1730a : SUCCESS in 56m 58s","accounts_in_message":[],"_revision_number":2},{"id":"f91dc24aa83edd252d224f5933c1dbafc398c2e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":396,"name":"Mandell","email":"mdegerness@nvidia.com","username":"mdegerne"},"date":"2023-03-06 17:17:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5ced5129c6aa14fb0c1e77809703ddb6069f8790","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 18:23:05.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1412dc70bc8c4c0cb12103787fffab2a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/085aff8daa5843aeb814d116eeb6f9eb : FAILURE in 1h 01m 43s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a320534e1e31413a96881775f554c263 : FAILURE in 55m 15s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/191a94f9e86c42fdb78460e984f8b009 : SUCCESS in 1h 02m 38s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/da773ca0fae74d35b9f50be006d18465 : SUCCESS in 54m 49s","accounts_in_message":[],"_revision_number":3},{"id":"22663da6d756f41ea54121f8b8b13f501c27b253","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 18:50:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c6ea87c0fb24b3cbc9c65ad41420f5f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12ee614bcb934ce89886ab2728cf4f87 : SUCCESS in 8m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/2b6a944175be42b2b5a6f2317e2045bd : SUCCESS in 1h 09m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/24ef1037e14543f0a3dd248c73d2344f : SUCCESS in 49m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8bde1bcb33b7488fa3e70b084971b0e4 : SUCCESS in 57m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3661341abb684823b7f6284aaa13361c : SUCCESS in 24m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/03c3b2dcff4646bfaefb116b75b70b83 : SUCCESS in 18m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e148fa83ca404c35ac7bec36daa24295 : SUCCESS in 23m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/38881e9ea3614d3e8de70bf1772e047f : SUCCESS in 8m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/51c9618e62ff4fd4b930fcbff929a8d3 : SUCCESS in 5m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/34455960e56f4ccd80ccada0146f5c65 : FAILURE in 16m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ce573ea2c3d846e19da6dae5a4370ec5 : FAILURE in 19m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1d4d2ce47138496680d70e15824528fc : FAILURE in 21m 24s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/137792d9de9b473da48b9080a9b1772c : FAILURE in 18m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/37f7fe4cedac41cd83d161e79a81702c : SUCCESS in 16m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1dcc0227369e4267b5c4c4e7a4ab6855 : SUCCESS in 19m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/05587e1e90114b5a817fca8ddfd4dbf3 : SUCCESS in 17m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/375ffecf6b164a6e96baaf310d8ff52e : SUCCESS in 16m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/47ac42830df2425097ac2a73527230ab : SUCCESS in 14m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0a3cd3c336fd498a8d6aa6f11e110995 : SUCCESS in 14m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6bba08d5369249ffab58c0be9b0b7ee4 : SUCCESS in 7m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dd02ebe29655404ea44b30414e13de88 : SUCCESS in 17m 14s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c188c208db524378bd29f38202f932ac : SUCCESS in 7m 31s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2cc840ac90964b83ac3a12b559eab165 : FAILURE in 44m 54s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/68ebdba36b1b40d795238bad2b24d89d : FAILURE in 57m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ed2c60f658594d9eb62a5ce911404337 : SUCCESS in 49m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f1454b2cd62e4635b9188974a34596d6 : SUCCESS in 59m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d425e974f1b4f9ea63d69200952cf61 : FAILURE in 18m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59f653766f2c40bc9f258528afa5c331 : SUCCESS in 4m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/50a2145c93cb45da9e8b1b3f9c60628f : SUCCESS in 19m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/864dab92f8274c04aece3f7ae7c2d5e6 : SUCCESS in 58m 02s","accounts_in_message":[],"_revision_number":3},{"id":"e1d497e16175d55a460122452d1c4d06c12bb3a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-03-06 20:29:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c476ee07b1a50724d24878459cfa47e6ed4e41e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 20:53:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/910eb10413cf408b82b988b5829ecd8a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bba30b98c0c7401e8a8a04e4c257c572 : SUCCESS in 15m 41s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4375286e262546a49ceebf0fcae504ba : SUCCESS in 16m 30s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/59b95d94e1064818ba53bcca776d5676 : SUCCESS in 20m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/458b1c11e60545b9bbfa713f43335bf6 : SUCCESS in 22m 05s","accounts_in_message":[],"_revision_number":4},{"id":"8c940f983a1481a5bd71cdc41e1addb27846a1b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 22:05:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b0438d94e2b418d8d3575c0224db9e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59a6d24a837240928d21b30f403fe9e0 : SUCCESS in 9m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/28ed2b54c7ab4c2fbcc1f3875c685c9d : SUCCESS in 1h 02m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a5c1f39ae01c44c49fa0f97b26d3d013 : SUCCESS in 52m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f3b5652c837846b6ad6cc89c53afa33e : SUCCESS in 1h 07m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/578db7f0651e4efeadac82ec21b3ae85 : SUCCESS in 20m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3a768e4582a84629a9ee2d90cfbac493 : SUCCESS in 20m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0bd6abf1d07249e9a69316b49fde74fa : SUCCESS in 22m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/04477146c70a471698f05f48d43d0484 : SUCCESS in 8m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/146c2feb52a249a0a5aa095dcfd5b5c9 : SUCCESS in 5m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5ab0c3a358f3424186d9866ba725c5dc : SUCCESS in 17m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a89bcbca63bf48b3847dd2029b551652 : SUCCESS in 16m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1c8a51c1bd7c4994a6bb818830a55e88 : SUCCESS in 15m 15s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/df74ba1f42a0442896be7b167b4bf532 : SUCCESS in 15m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/50bf36f483a8480ca6a873ea27ad98ae : SUCCESS in 17m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c6cf1b3f6a204c7eae2b683b53ad8764 : SUCCESS in 15m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/958b506c3f7b466f8bd1b9af8d2ba140 : SUCCESS in 18m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b79af3ac04a544b7b6ae0ebc5953e322 : SUCCESS in 15m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/908c35adf7fb48e08f1df1ec45d93a36 : SUCCESS in 15m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/dea07595b5564dbab194a226c1857f7d : SUCCESS in 18m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6fb0838a8d0e426f9f21649e947e558e : SUCCESS in 8m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1f0b0a1f009543a1b1f15d3cfcd43ccc : SUCCESS in 14m 56s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/df78aafc7ea44437abbf7b119440b5f4 : SUCCESS in 8m 23s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/08864b9d33634999b86762d09884f16b : SUCCESS in 1h 19m 26s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a957b175321b4a30a1e63e09d92b6958 : SUCCESS in 1h 01m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/19a29485da0b4422aa5755c626a0bc74 : SUCCESS in 46m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/210a6a84f38344cc8c33cb2731b958e9 : SUCCESS in 59m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/944756e9f4d548d29f97d89830bf5b26 : SUCCESS in 16m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edf9a17df668407ba7ec2abff6faa513 : SUCCESS in 4m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/722048dea1d64e69b54bfc78d24ab0be : SUCCESS in 21m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/df3c0fa27dcf4410b08ce5821ff03535 : SUCCESS in 47m 50s","accounts_in_message":[],"_revision_number":4},{"id":"f95ca0e7866a10d58dab670a39b39050c99a3aa9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-03-21 18:34:15.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"a7c176b1f8f1caf54bb37d7dc7695bbf494517c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 19:34:48.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d612cb6d20a42dd9c35d3f09f3d3c3f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5d0af8142a9544a0ac133c0f5185969a : SUCCESS in 46m 27s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2b43752746e54b93bb0d1d0d6b541e50 : SUCCESS in 16m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d5852b66c92a459494079adc64353405 : SUCCESS in 58m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/64e4bd6b1b7c4c938388ba6976bd635e : SUCCESS in 18m 29s","accounts_in_message":[],"_revision_number":5},{"id":"4860b6d578e8c52992528b2fffa4eb46d359dd1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 19:45:48.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cade3e69bdff43c298d9c5ffbe17a7a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/658f340de3114634b1ed2ba10ce7b906 : SUCCESS in 8m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/ade72d8d675f4853b7858682ff5f5f95 : SUCCESS in 1h 05m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/484e2d92ec2a4577a7ed9d3e25628d89 : SUCCESS in 35m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec9185dc50a24d10bfc97e6465812153 : SUCCESS in 57m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c138874f12854f9793b8f29d12c2d6a7 : SUCCESS in 16m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ca7b8c3f4f3749249d62257e6d02c1fa : SUCCESS in 19m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/822e7802c1844ec6bb178d6053c086ed : SUCCESS in 13m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/03d084fd6a7d41e4aad67a4b271e1aab : SUCCESS in 10m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a66429e9f2ba43838568f898b5c5ef8d : SUCCESS in 6m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ee9052b8cdc249dd867b44bb361ed295 : SUCCESS in 12m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9969aefdf8fc4b098ebedcb51685c8e3 : SUCCESS in 25m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5b13cb777a92478284a43b40aae12283 : SUCCESS in 15m 31s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/25a4c76bb73c4fe694abee4e9ceeab7d : SUCCESS in 16m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0d3784cb2920413bbd0ce45ebe7911b0 : SUCCESS in 17m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f4e41d6f604b4a55b7955b38b13c3732 : SUCCESS in 13m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/be73438062084bb9974597258c02738f : SUCCESS in 18m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/723a942b1cf44bdb9f214d784db37c75 : SUCCESS in 14m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e6a1109dd29e4e61a6e0eba05b1ac12b : SUCCESS in 17m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/10c311a2c7684021b42a4c83e94b7fa2 : SUCCESS in 16m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e31601339fbb41c2bc8b76809ffae40e : SUCCESS in 6m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/489e7696cd354d0a8ed3d6a4bcf4be92 : SUCCESS in 16m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4d31d5c9708747108c435f9b4dea0231 : SUCCESS in 8m 56s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/83327ac55e604954929717e80ec9386d : SUCCESS in 55m 02s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4a9f00184598463bac23f47f89926674 : SUCCESS in 54m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/85bda9b2fc7347f1863799e879261856 : SUCCESS in 46m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/22bca7dd524d49a2af232d2c714047c9 : SUCCESS in 41m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cff7ddd3bb3d4034898407de3d3fd1a4 : SUCCESS in 17m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65ca8fdcc47048a5825b7d77c85bfd2c : SUCCESS in 4m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e96b64d51f1e45dcae27ec606fad2af4 : SUCCESS in 33m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/748b6453544b41839954dd2c3619ded1 : FAILURE in 46m 07s","accounts_in_message":[],"_revision_number":5},{"id":"25eb489f6e8ab2d2ea18b2ad9cbfa7bf3db8f991","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-03-27 19:15:18.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"44585cc05c3de34aa31101585d15872fa9d5f223","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-03-27 19:17:30.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"9e3ac0e1059a674da48dcdd58559aca714dcedd6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 20:08:25.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42ffaf05e3b5470a9a6d6b417d517e94\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6fbffc85a12b4008b0eb1ada67fe6ac6 : SUCCESS in 14m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5fa66ceb041f46c08652ad2d6593cdc3 : SUCCESS in 14m 16s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/21005d0f1b8644288387a5d41f614383 : SUCCESS in 50m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/089c25b7e8b54b6c8c86bf1038cd0b4c : SUCCESS in 23m 06s","accounts_in_message":[],"_revision_number":6},{"id":"915909ba1bdda5d32499ece5525fa83a19a57080","author":{"_account_id":396,"name":"Mandell","email":"mdegerness@nvidia.com","username":"mdegerne"},"date":"2023-03-27 20:24:40.000000000","message":"Patch Set 6:\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"762848e0cc1ea990bbf860068d73af8bb7d1815a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 20:29:56.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8145e204ea54f9abc7f8313a282e476\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16875d9393b74913976d945ee31c5fc8 : SUCCESS in 10m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/43c4c6d644874c1aba53cc93282ea647 : SUCCESS in 54m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/69ef8d6d31d34a1096fc5dec8337ca37 : SUCCESS in 55m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aaf2885fac8c4f83bd1e287c9b8970c8 : SUCCESS in 1h 08m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0a4da1b960e6453eb5643887cbfe5693 : SUCCESS in 31m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ad0a1dd4e42a4a5db028dbd9747e9f7b : SUCCESS in 17m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8ea40d9ac3ff4947aad5e7940f1c770a : SUCCESS in 19m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3c907e10839d403ab86c80b88749f092 : SUCCESS in 8m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/786c3269b6134e4792897426a6e61e14 : SUCCESS in 5m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/858318bdab514bd3a979e3f606d21217 : SUCCESS in 32m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8dc9e112be3b4e04bd2ab63eddd3e81f : SUCCESS in 34m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/19fce2061d5d404fbe77495caaeccbf1 : SUCCESS in 14m 59s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3e7dacbedba04e72956c6d73555fdcb8 : SUCCESS in 13m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c4b938a90e75485e8bc0840ad617e18c : SUCCESS in 33m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/17f8bdae742c4f46a8c974e0d4a7653b : SUCCESS in 16m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7b81e36963904654b8b379cf341cdb48 : SUCCESS in 31m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4d0e8c0287434a5bad97eac61108b86c : SUCCESS in 15m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2baa66c25e8f4dd3981325da589e091d : SUCCESS in 30m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fcf2fabc64f54676b0f8c4e7fd453e1e : SUCCESS in 13m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fc22b57d39e44e97b39a4c4ac7dc3dd9 : SUCCESS in 6m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c14be0ad27da44e38717ab6873a19fae : SUCCESS in 26m 40s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/45470e6d5b1a415f873830a7aac98453 : SUCCESS in 7m 38s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/de9a9e47d6b94574b0aacf1804bf41df : SUCCESS in 1h 02m 45s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/42da226af08f4e59a80d7c05b6cb9374 : SUCCESS in 50m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/06a66100741443faa048c5873ccec01c : SUCCESS in 54m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4b3a838b4e164eb3966d79df0650d7c7 : SUCCESS in 49m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66623e899952454a882fa52ea41a2525 : SUCCESS in 36m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc22a75152b14247810f420adf403323 : SUCCESS in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7da05ae3cd574db5868f6dd6e75ee58d : SUCCESS in 34m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c08ef042ae604a6c8179bb8c6f99aae4 : SUCCESS in 45m 29s","accounts_in_message":[],"_revision_number":6},{"id":"659a3ab10c546453c1591b245ae5c3f48b823192","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-03-27 21:08:33.000000000","message":"Patch Set 6:\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"ceadb0d90d63dc190ffed8ccfbff54af4f4551e7","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-04-04 21:12:20.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"eedd90e2165f55189c677041dfce0a701d0b3155","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":396,"name":"Mandell","email":"mdegerness@nvidia.com","username":"mdegerne"},"date":"2023-04-18 19:01:50.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"8e28582d1bb7650fc2662f561d80fb97221c48f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 19:19:21.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10f8daa53b3548178907bed7abc0e20a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/47620608175546d0ac8d1514455c6c9a : NODE_FAILURE Node request 300-0021003873 failed in 0s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/11747d94478143b5a41040c195873206 : SUCCESS in 16m 29s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/49a02074034a458396cbd62166dae59a : NODE_FAILURE Node request 300-0021003875 failed in 0s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9edb4f03d8234a9bb3de0042598a9220 : NODE_FAILURE Node request 300-0021003876 failed in 0s","accounts_in_message":[],"_revision_number":7},{"id":"89105f694b4f02c516e637ad3447c18705019211","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 20:16:36.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d96130ad4aa4005badb2f5288140fcd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c3e1c27849b4f34a620f95b55b37b2c : SUCCESS in 9m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/34f53860c07b4b37b4c14994a12bf9dc : SUCCESS in 1h 05m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/16d4b7a3eadd47f79af5cc884360f8bd : SUCCESS in 36m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c853752ad5c4419eb2a55531a5011948 : SUCCESS in 1h 06m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c873f36ad7e940cea916f241b3b8144c : SUCCESS in 16m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1f423b383d8840698abd3c107ae68b29 : SUCCESS in 19m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7d115c38f8fb4bb39cca0d3c9949eafe : SUCCESS in 43m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e1953e796ade41b78e3d61c2ed113b2a : SUCCESS in 7m 40s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f212595c0af6489782430123dfe99a57 : SUCCESS in 5m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8ee75577d7fc4a2b9e9e0c79a18cbdea : FAILURE in 4m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a732ca8fe7e342548dcdcef14ecf1869 : SUCCESS in 16m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/11abe90bdda94042953c6f568005ef6f : SUCCESS in 15m 26s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2317a7951822433aa8fb32f34913bc6c : SUCCESS in 16m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d9faa02005234b1f919396dd5c31ae9a : SUCCESS in 15m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/313d4287fa9a4f108a441ac9b2fcbf7d : SUCCESS in 17m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/601328a3c0d442f5b7fcae9262672552 : SUCCESS in 15m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/eb1adbb95d874b54a444c610e71237d6 : SUCCESS in 13m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/658c48b51d0b4ca88aabba76e6039d64 : SUCCESS in 14m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2cfab7c8342b4f149c9259e56df9a137 : SUCCESS in 14m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bdb0c773410f4ad4a76130d7a98b6b06 : SUCCESS in 6m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/257b24e7ea5d4b31bda0b03f36d637cf : SUCCESS in 16m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0061975059ff4815b76d54aca9cffb4f : SUCCESS in 7m 28s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c8b637682b154c7d8399d1c7b236aa1d : SUCCESS in 58m 11s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/eb63197393c54e88b7e1f01dc130a9fc : SUCCESS in 1h 01m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f8a83d40e20f45ffab6d3f1a924365e4 : SUCCESS in 40m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2dec799ed3494fc8a6c37196991f37a7 : SUCCESS in 51m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a8593dc470042aa9a4b0f897a42e6de : FAILURE in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9e0941bf1064d15b319e0226c752d52 : SUCCESS in 5m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fc4fa2d45ee7467f89480b4c616a33b8 : SUCCESS in 19m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d9323d6cf3445ac91e9e4978138c84b : SUCCESS in 36m 02s","accounts_in_message":[],"_revision_number":7},{"id":"fb8b1418d2005d9dc9dfc463a1f12e3d3ac6af1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":396,"name":"Mandell","email":"mdegerness@nvidia.com","username":"mdegerne"},"date":"2023-04-20 20:21:02.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"8fdb32557537371fe0e641af8f4930d9101513dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 20:40:44.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f3efb0358fb4395af635cddd9dcc5cc\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/509b6b1239d64df8a04d9e4d917b8b89 : SUCCESS in 14m 18s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f6909cc0f199418499571e1a50787c08 : SUCCESS in 14m 42s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/89d0146d2f3b43b49388b26ce9c0b74e : SUCCESS in 18m 27s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/604bf40ae6874286be209ec1922c8066 : SUCCESS in 17m 45s","accounts_in_message":[],"_revision_number":8},{"id":"2aae2fd81b4c8e9f07fcbbd00a656f788d0adfe8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 21:26:34.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ab29e2827a64a28a3425c0c5ff52b96\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e867bf0f443c40e2b21ff8fa42809fee : SUCCESS in 8m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/1a816247494f4f4686a37311ed3391b9 : SUCCESS in 50m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1f0e5dca6a7341f29f773514ed6a7a6f : SUCCESS in 37m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a630b2788a9437586bd12c9f0fc6c51 : SUCCESS in 56m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/00a3e700b1bb4e09b19dce2abc0fa2f6 : SUCCESS in 25m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9eb8de08c1a54455a2c230ea40664778 : SUCCESS in 26m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b7d51aebe36045858d0a6c6d45567138 : SUCCESS in 29m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/076427eee58e4267a8fbc5c8ca2dc8da : SUCCESS in 8m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9ffcd2f5be5c4dfab392e7fe87fedf37 : SUCCESS in 5m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b00e96a108cf4b0b853eac4efa8abbc8 : FAILURE in 4m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e053faa04b2248bca158e7b478c385fc : SUCCESS in 17m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/88900d65b07946e3a1016c3b5751343c : SUCCESS in 24m 55s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b884cf7286a345ec8cb2874eba200163 : SUCCESS in 16m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0111cb3a998a4e40bfc50bb15fa01c51 : SUCCESS in 16m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/82aaf85e3a5e43b29eecbfbf8307b7e8 : SUCCESS in 17m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0c0aa2e5a6ff43ca89b599b08db429ad : SUCCESS in 16m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b2d23bd5086247a0b34b8bc21c7daad4 : SUCCESS in 13m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/10d9b1dc882e4246846fbd13af3cf34b : SUCCESS in 15m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3b7310871fbf42fe84b88e677ff661f4 : SUCCESS in 15m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0b479c6266064eaa984882b85bb048df : SUCCESS in 6m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/20bbaf656c524d25aa55a75c663de198 : SUCCESS in 17m 43s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e5d934722d22465a9834262ffaa4a9f7 : SUCCESS in 6m 34s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/aa948fb67f4f423882b1a55a14d73f51 : SUCCESS in 56m 32s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/256f154f828c43a6a788c4cf41e0a6e2 : FAILURE in 1h 00m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/18d4dcbd2d4143e1a711b27fe38f920e : SUCCESS in 56m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b722b10648484e8a978c104ea9466734 : SUCCESS in 55m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06a9f92665174013ba5b8919dfc7b585 : FAILURE in 2m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d32b9da6c3f5402389be19cfc38b22f6 : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/403a3838f28f4fd68be8003ca5f9a0bc : SUCCESS in 21m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4202c55861294bf5b16e4e6dc1a4aced : SUCCESS in 54m 02s","accounts_in_message":[],"_revision_number":8},{"id":"4e486f8622cbff25124c3de26ccc9e67fe33a898","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":396,"name":"Mandell","email":"mdegerness@nvidia.com","username":"mdegerne"},"date":"2023-04-20 22:22:06.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"8bc78a418ab13df43f00dd2dbb167c7382730340","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 23:30:21.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd7528fdf6544122833173ed2a9137c5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1b88a6bf85d419d8d3bbfa9c5a8bd34 : SUCCESS in 8m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/ac235f5d605c43a589a1570c74c92234 : SUCCESS in 36m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/37c506c6147740d0a4667920a6b50987 : SUCCESS in 31m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf202df14b1445a3ad08961d4a5fc40a : SUCCESS in 1h 00m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/60f34013750c438086ad3af191d1e90c : SUCCESS in 16m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/255a3ff6fd104736b51ec6857cccc811 : SUCCESS in 20m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/63a89719567549739b6388edd18495b3 : SUCCESS in 16m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f17bfa94f3a24449a2e7425d3016a1b4 : SUCCESS in 9m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/defa2ce04bff45d78be43568362a83ef : SUCCESS in 5m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9969000ce86241739839bd694f84e860 : SUCCESS in 19m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4a772003dfe14ed59aba40e1a799fb21 : SUCCESS in 12m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/15dddec80e644b77b76d57ccfd93a0f9 : SUCCESS in 17m 25s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0c8c607bf20b4689a8319ef5def5b1ca : SUCCESS in 16m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d590b343162c4c518750471adc0e9d6c : SUCCESS in 16m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c3c73dfcc5944a9c8ffd9f92d43906a1 : SUCCESS in 17m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/171dfec6aa43406d8ed7196180d9a935 : SUCCESS in 14m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f6488a79cf084a90a8e8557218e4c4d7 : SUCCESS in 16m 55s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e484c31e5c6a421fb1570615fbc07355 : SUCCESS in 14m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5969ffbc99e44868a9f4f25e3a5ded71 : SUCCESS in 13m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2a64832f929649aa9453312aba9018ac : SUCCESS in 6m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c3129a936c6e455583973896ace99117 : SUCCESS in 16m 24s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/96fc3c4ab3b34c879e29c4295550a379 : SUCCESS in 9m 31s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f46a6c6837a04635bce30f15a87f261a : FAILURE in 53m 35s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/24350376aaa44f8e88312c59261a4498 : SUCCESS in 1h 00m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/884877d97d654d2fb0497ad997d56ed2 : SUCCESS in 49m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9faa506050794b4390837846409ff55c : SUCCESS in 51m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d645df6e721478fa0ef109f5534d0e3 : SUCCESS in 12m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01be732ba950478bb98b3c25aa0887a9 : SUCCESS in 4m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2f37f6278c1441139f175d07ca0a7c8d : SUCCESS in 19m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd0f074c7d38436db4735b04b7679d29 : SUCCESS in 58m 30s","accounts_in_message":[],"_revision_number":9},{"id":"25f336ce0dd440f6209b9f38272ad0958a0c0da1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 23:51:03.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/056cc77543714d698ec246f742fbc54a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/933003b6289f4a399ab9c0f9e1544dea : SUCCESS in 15m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/36b0f51343e542929cfe5c57748fa0ab : SUCCESS in 1h 27m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a9b3c7cd4450478eb2172a1fbf4c5444 : SUCCESS in 17m 43s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8d1fba8ec426454f918617332e39f05a : SUCCESS in 1h 07m 04s","accounts_in_message":[],"_revision_number":9},{"id":"2d8bbb9a801919376f06d10c7853821e3feb2e7d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-05-04 20:17:36.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"5100d643476f92ae9eac0a477b171cdd5c9305d7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-24 00:10:47.000000000","message":"Patch Set 9:\n\n(6 comments)","accounts_in_message":[],"_revision_number":9},{"id":"80b53d32971e2987d0aed8c62e81160e35ab91f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-10-19 21:29:52.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"4f76397fdfdd806cd800f9959b297103b5628bac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 22:43:42.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d47593a4936c4cbdbc1e3b00037f6ff5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87c8824e00dd4a8091ffb365dad37578 : SUCCESS in 7m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/040aea0302ad4afdb49d3f74aa0bb8f7 : SUCCESS in 40m 22s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/cc31d65d9875401d9b8bab88d3478f16 : SUCCESS in 40m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/414fe23e33684ff5b85614a9a1766812 : SUCCESS in 31m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/96ca5e7dd9a24db5af55b06da01838ce : SUCCESS in 1h 06m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/246f4b6c11444cd9a39b580facb447e2 : SUCCESS in 18m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f5d05e90c4a5424b8c34a4d354ac6f39 : SUCCESS in 19m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1ee104773c5345788cbea65267b0a0b3 : SUCCESS in 22m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b065df788d4c48c09216c9e20884b0bc : SUCCESS in 7m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/fae36b819ab1464db9bfb3a087642745 : SUCCESS in 5m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/695a9ba7f2fb499989a8c7e8da7bf4e4 : SUCCESS in 17m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5965b9f965644d1fbb428324d46cbb08 : SUCCESS in 18m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ef653780cd494094a2bc01ca5e47639b : SUCCESS in 16m 37s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2c1902a96e5043f9a5eb1c8fba818e19 : SUCCESS in 15m 40s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c15896eb32b6489b9678b989cc4de66d : SUCCESS in 18m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9bde19d5390143db9615befa397f20f1 : SUCCESS in 17m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/90b86b71aba5405087712cbdff846f6c : SUCCESS in 18m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/937b94730f294514831b0692e74dc4ff : SUCCESS in 15m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0f9865ff75c8447e8f66b8f876a4cea7 : SUCCESS in 19m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/65dfe5197b08434489b3b2c8ac0d94f0 : SUCCESS in 15m 54s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/56796a2e7890413dbbca42bbb8c6e3a2 : SUCCESS in 18m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6cb373405d7442c9be3a56eebf93914d : SUCCESS in 7m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/434f0225a22b4bc4a297e499b27708ab : SUCCESS in 13m 47s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/265bb8e3e4b8427297e9f6d7aab6a4a9 : SUCCESS in 7m 58s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1458dfe67ebb49b4a5e77aefbad0d31d : SUCCESS in 58m 11s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b509f37c67c746e8b154aced9acd9045 : SUCCESS in 54m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/706520dda8174394bcad19366b901a8e : SUCCESS in 36m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/70c9cb7a273548a28335ed01c9d86c87 : SUCCESS in 58m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4441681d5b6d43a7b38bef15503a9f8d : SUCCESS in 12m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb45ab326a1d4351b43a29d0aa04baa1 : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5edbdf5598cf4ed985685d95383ccdad : SUCCESS in 20m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3261998578634f86a8111833fa23edf2 : SUCCESS in 1h 00m 03s","accounts_in_message":[],"_revision_number":10},{"id":"d5966dd111b37ee2878396428fa5bd4b46592b55","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 23:29:45.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcc61c4b5d194ad6810a5d6f1e60d6d7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b4169ef44a047baaef4467b4ee88b14 : SUCCESS in 1h 22m 19s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/35b206aeb83f4e1abce3d546b74246ca : SUCCESS in 1h 03m 49s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4d9953dd04ed4384ab09776b3b8510a4 : SUCCESS in 1h 24m 20s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ead9843328b4422b874d2c12f3cb49fd : SUCCESS in 1h 08m 06s","accounts_in_message":[],"_revision_number":10},{"id":"12134b7f11cc97dc1ba63d1c87c64e229e13735e","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2024-02-14 21:14:49.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"f43096d7623414ed35d64063f68f9fb551552783","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-03-05 19:24:20.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"8ccddd1a418751ae55119514bd362ea7a21287f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 20:33:46.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdc7000fe8714a7a9ba9e312f9328640\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5d31c070e204431aa4feeae2fec1487 : SUCCESS in 9m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/418e406ebf854e10879479813e2c07d8 : SUCCESS in 49m 58s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/cd8ab2ef003b43d3a9365efe311679a8 : SUCCESS in 57m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/890aa523b31d43ca8c71c6332b30c77c : SUCCESS in 51m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/391415febb584b43bc9b8e559513cbb4 : SUCCESS in 55m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/018c8d6ac5b94dd8a683ca935d503b05 : SUCCESS in 25m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/19aeef20620a4078b493658d6be617dc : SUCCESS in 26m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/664f185c2f524554a4839c75b4a6aaa9 : SUCCESS in 19m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9c79bbc02da84e8c91b343f7afa46949 : SUCCESS in 10m 21s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5a88a7d867404a89939a25f05560f29f : SUCCESS in 5m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ff22ec28321b444e976414048b2f274a : SUCCESS in 26m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1f60b2d3a67543de96c5e139379f4d41 : SUCCESS in 25m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/20940599c12346c8828bda4aca470b41 : SUCCESS in 17m 16s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0d90a035b7e04d89a99178f400376775 : SUCCESS in 16m 47s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/951050ac2e584898905f4f77fff92a49 : SUCCESS in 24m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0826be09b10c40879a75a79a9bbd722a : SUCCESS in 16m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7999885ec068435b920a93099a87cb57 : SUCCESS in 16m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c947acfe83414c2a8dd85c4b18114f3f : SUCCESS in 17m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3fbd5b4749e743429696338e81f87cae : SUCCESS in 15m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f1181e1f0eb646479738dacda12f4c42 : SUCCESS in 15m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a0bf65fe017a49418372e8300cab4e38 : SUCCESS in 15m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/260e3bceb60740dd879886733744853b : SUCCESS in 6m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6928c9e782574b218dea1c8064bf9336 : SUCCESS in 15m 04s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b420035480764442987d3138360b038a : SUCCESS in 7m 12s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/bc069736cc5c4245a9576e10ada83050 : SUCCESS in 58m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c87a4c8c8e064c5d8a82b447c4a116eb : SUCCESS in 58m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4e4f32fd88f5445cad8604c6f1967961 : SUCCESS in 51m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2bb0bb8d74714378a4a203ec35dbb615 : SUCCESS in 20m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f17ba9165f5b4f2592941624163fe3a5 : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f3e0fc5924b247a99c27fb900b9974f7 : SUCCESS in 20m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72cd688e0579400bbdcf65300e665e1e : SUCCESS in 53m 27s","accounts_in_message":[],"_revision_number":11},{"id":"d219db93b9c05402d04d1b203dd5e27dd538f7ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 20:50:40.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f9bbadeed514ff085bf11dabdd80329\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0b20b73a48ff4582a5b7b9966cbbead4 : SUCCESS in 15m 57s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ab554616e1a843b1b46dad7eb2010bae : SUCCESS in 1h 18m 39s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/115fe87444024c51b792061d2bbf43e9 : SUCCESS in 23m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6ef159278c8640f9852f5acded56f77d : SUCCESS in 1h 23m 31s","accounts_in_message":[],"_revision_number":11},{"id":"2ecf398df80549fd0f05ae7ca2dc31ea8004f437","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-03-06 18:21:59.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"100776e60320354960d3c37e722079b15e17a7e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 19:33:35.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28cd0fe30b0d428c9987e9a9d5ab30a1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f6a42b4364f74ed4bf6507978ebed571 : SUCCESS in 15m 54s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ee23500e74e248a88e42993fd2e7c4f7 : SUCCESS in 16m 50s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/fda9fa36fba54c97ae9589c768c1e04e : SUCCESS in 19m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a9997c6b214a432eb902771c7e360d18 : SUCCESS in 1h 09m 42s","accounts_in_message":[],"_revision_number":12},{"id":"dbfc730ce32ad1add7880d58da87a044f67f42f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 19:42:27.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/661e285b6d654041b9836e06d1c731f8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d11f66b0fbdc4ebeaa784b0b078e86b0 : SUCCESS in 8m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/9adbfc55a24242c69e223ec5aa99dbe5 : SUCCESS in 1h 02m 38s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c71ea571940a47dc93654889ebb6c638 : SUCCESS in 48m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/094d75d5e54d4a089460b537ab90156e : SUCCESS in 48m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c206c7048974231a6ed7560414ce66e : SUCCESS in 1h 01m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a2cf21441f894dadb74e0e37f8a713a8 : SUCCESS in 20m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/59d442a0579d44c2a63ab36e092ed6e4 : SUCCESS in 20m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9278ae57fbd1419b911fbab6d5569b5b : SUCCESS in 19m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/18949ce175b543bc9f38471e79257937 : SUCCESS in 12m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b6334f378d50464d8bedb75fb43da67a : SUCCESS in 6m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6822651847b447cca29eecfaf7e80f46 : SUCCESS in 20m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7861cf165be641719f4c0754965b7c2c : SUCCESS in 20m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/06f5479b98634ab28b7d247db3fe7bcc : SUCCESS in 19m 23s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9af9ad2259224a7991170ab5e6e2f947 : SUCCESS in 18m 36s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/7d4bba70d0824bc0812a96d09f1239c9 : SUCCESS in 16m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e735652ea1554d4eb7a89eded8dd859f : SUCCESS in 17m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e5e7135ff0ea4234ac7b8803cfa22a79 : SUCCESS in 20m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/681b04f2a75b4187a202709cc953bc91 : SUCCESS in 25m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/63b7676dda8948c6b1508e89fa221ac7 : SUCCESS in 21m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/93bf25d7151f47c381fe55f597af00db : SUCCESS in 17m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/362a4cddfdfb4f03b7308fe6ca4c44ea : SUCCESS in 21m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/76152034eb0a440ebc0d7e083859429e : SUCCESS in 7m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/241baa51737b4784a3f4e7c3603b8b4e : SUCCESS in 17m 58s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fd547877e19e4c3185a3277678b7f11d : SUCCESS in 7m 52s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6dcbb60e532d41c498efcd73a54b9a0c : SUCCESS in 1h 05m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d0cb0e5294d8409c8e926730c695f592 : SUCCESS in 50m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b7264a0462f94717860f91c8369624ef : SUCCESS in 43m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a03db3015c954aa0989229772caaaf9e : SUCCESS in 21m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/640288aa65d44f8c8e950d9efe468f9a : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5a0eb95d520b4a3da4a1b68bc2d55f00 : SUCCESS in 24m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75c5ba56c0a14b6e9d1184272fe2a920 : SUCCESS in 1h 03m 04s","accounts_in_message":[],"_revision_number":12},{"id":"59332633212350457a08021ddb53e0bdd85cde33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-03-07 22:56:15.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"f46db4d215bd18fac39aa56ed90154c8e8c0eeb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-03-07 22:59:04.000000000","message":"Uploaded patch set 14: Commit message was updated.","accounts_in_message":[],"_revision_number":14},{"id":"8509f1587b629267e65efd80bd7a4ac0e7aff9f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 00:08:25.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15bfbc965b73431e95a30d307f0c7f2d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e33f1450782d446ba7310fb7c321a100 : SUCCESS in 9m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/8da7c1bfb1d241358179a2df1117b2e5 : SUCCESS in 46m 34s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/193ea9b1d3d740bdb1278f168df967f4 : SUCCESS in 1h 03m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c410a9ec2af2440da746e32feaffbf27 : SUCCESS in 33m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e445c1b2d57a4e859e88ed55ddd614e6 : SUCCESS in 1h 02m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8e6d292b59d54878b2c9b0c066625573 : SUCCESS in 18m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8ffe20dc8ef246aca119fa56955cf85c : SUCCESS in 15m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/43fa76511f8f4f1a81ea637a948bfa02 : SUCCESS in 19m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b140d651dc7d475da405f767f0cf3672 : SUCCESS in 8m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/21e0488514f6455ca4a84f648ce77496 : SUCCESS in 6m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d7027d8f988a4e9b91b06edde3571138 : SUCCESS in 17m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/769d310dba9b4d5d9d17a32ac30b25b7 : SUCCESS in 19m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/aec0657758244f7695e9757cecf532e0 : SUCCESS in 21m 14s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7d3a6cbd40a54cbd80333852570870db : SUCCESS in 18m 37s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3dca8bd927194d7a9c766ec232fcb102 : SUCCESS in 16m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ee8bb52dc3964d2f896f1fffb9066913 : SUCCESS in 13m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a1a1e67705d140248a0a96ae1e26cf33 : SUCCESS in 18m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a30921f838f24a2ab1f55a14979d15f7 : SUCCESS in 19m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3866504129a746c49bb388b347d24852 : SUCCESS in 14m 40s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3aaca6a613454e13b724c63b3ac9ad26 : SUCCESS in 15m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/86fa7d7b0eef4dc7ad6ff34228278641 : SUCCESS in 15m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7943d448f19f406f98eaba543b31e1a6 : SUCCESS in 5m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/95ff9181088a4603a3f6a10c8998a84d : SUCCESS in 16m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e6aa91f224774db9a66a81d1943c77e8 : SUCCESS in 7m 04s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ba3bbadb41e94c6baec362950f1e48f2 : SUCCESS in 1h 02m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0c49223f6a734b0daa1c095af2b59532 : SUCCESS in 54m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7fbc366578494eea8c75fdba02e4f979 : SUCCESS in 37m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/748829ef0ec042089e821dbb33bce916 : SUCCESS in 19m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d6c4498f3d842ea97acdf24bd4ec2dc : SUCCESS in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b262a3ef89c14cbdbde9d11ea5a99a93 : SUCCESS in 19m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/559ddc09221d496c9833c9f5db16cafa : SUCCESS in 59m 03s","accounts_in_message":[],"_revision_number":14},{"id":"c7baff63ce270173e67287b4cfb65cb1c1fc2879","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 00:16:37.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02be5701f4aa4b9787a1a22421d64335\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c69c2b955d984f30ba9baac558d146c8 : SUCCESS in 16m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/850556834ba74eac88fc2bd508bafb62 : SUCCESS in 1h 11m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/cfc9efa3e21748e39dec713b63e784c4 : SUCCESS in 17m 41s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/919c12945adf4cd18eab538a629a90d8 : SUCCESS in 1h 15m 57s","accounts_in_message":[],"_revision_number":14},{"id":"6e0fa030403260d51e4986a3df4c910ad36df073","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-03-08 05:33:54.000000000","message":"Patch Set 14:\n\n(5 comments)","accounts_in_message":[],"_revision_number":14},{"id":"c9b2aeebb031cb40e07f1a76de631f52641a3129","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-03-08 05:34:50.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"724938bca5d080010f7816148cef66a5179199e1","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-03-08 05:38:04.000000000","message":"Patch Set 14: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"0d8b545e63c8175daae7a4cd9ce7059e0c0a94f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-03-08 16:58:37.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"41c84bdb0500842ff91b8477bd4e6ae5cfb4f850","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 17:20:20.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1549b677c854eaeb15fb1114c7b719e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/36a56a09741049a2affd864b0bce944a : SUCCESS in 16m 17s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fb9c0b8cd465401184553df57c1b169c : SUCCESS in 16m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/adb554f8a3254544b614890411312488 : SUCCESS in 19m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/bf69c2662bd045b99a1e057174f9412b : SUCCESS in 17m 09s","accounts_in_message":[],"_revision_number":15},{"id":"872a004f803bcac73fa544dd8a837d323109c8dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 18:07:22.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd940979dbd44c3a9556b4daf63c452a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c8e358975e04cb1823de7c456ac57e8 : SUCCESS in 8m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/cbcb7706e8a64ddbba140b6f7cec864f : SUCCESS in 1h 03m 00s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/336c20c0166b444a92e6e868bbad76a9 : SUCCESS in 56m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/de63328dfc774c3290f5e97019a733d2 : SUCCESS in 48m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f245eb0e87394eb3bb2ba38e81384980 : SUCCESS in 58m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3074633db69c413a81da446bd7094299 : SUCCESS in 17m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5ed5a7f354784346a1001ca020bddc50 : SUCCESS in 19m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/479733f4b033456b97693a342dd92e89 : SUCCESS in 18m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ce389e9ffb1547ab9584b560bf22c86c : SUCCESS in 11m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/aba300c5c270490aada128471b597a42 : SUCCESS in 5m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3ff8bd77b13e4a1c85eee10035a435e9 : SUCCESS in 20m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5140f698c889491fa3feb40529ed53b3 : SUCCESS in 19m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/468e2961064c4bf6b5a88b811f5ab0eb : SUCCESS in 14m 11s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/394ab7e07cd040008eb05ae19e841855 : SUCCESS in 20m 09s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/56e6f42438d54e4381e0ac8cca619786 : SUCCESS in 12m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/eff0670a92954c1eb8ed298342f6abb4 : SUCCESS in 20m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e140253dcb1d48b9a0ae02b235387244 : SUCCESS in 21m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c89b7e49e0c44371a25f1e356c77a136 : SUCCESS in 18m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/38f77f64089246adac4c04297b7ea067 : SUCCESS in 19m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2eebcb6218da40afa6067d4587db9592 : SUCCESS in 18m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e892236ba6e447f0b7706abc85496b02 : SUCCESS in 16m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b377ddc846dc4aa0896d202ca0e35d5d : SUCCESS in 6m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4a1d471e5d534f8f9411a0587058689a : SUCCESS in 17m 03s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a4209e97f262434d8c80f88cad6f1ed3 : SUCCESS in 7m 03s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/62ae23c0782e4a0b867d528b40a30edd : SUCCESS in 56m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7764b1265e234563b00fe2a2dbfb668d : SUCCESS in 54m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7fce2f49755a41f795a6bdc841dfa5f9 : SUCCESS in 47m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/184b14191f3a482780a854628f6f5c92 : SUCCESS in 19m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fe45d2cad97450990fba518f7ecc40a : SUCCESS in 4m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dbc0d44f48d740a4a30773fd3d63136c : SUCCESS in 17m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cded2e3953a64d5f85a73c6c88d09f60 : SUCCESS in 53m 49s","accounts_in_message":[],"_revision_number":15},{"id":"a730ce0d227dfac5339ea008f303b16764aca4dc","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-03-13 19:12:28.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"1e4a137280c7d4a3a65ac04bd1b5edcf4032b764","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-03-13 21:35:47.000000000","message":"Patch Set 16: Patch Set 15 was rebased on behalf of \u003cGERRIT_ACCOUNT_1179\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}],"_revision_number":16},{"id":"f95b17e91ce5b6847e8950656b5ccdf8d570b011","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 22:44:24.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1be875da187342719769d24f95881f49\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38f319523b8c49cfabe7b876ef24eeba : SUCCESS in 8m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/9a973c98866b431fb7d2a7b708ed4872 : SUCCESS in 1h 02m 49s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/edbee51927ae411ba7687334003dcb2b : SUCCESS in 42m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2108454915ff479f8b76547a60d7b491 : SUCCESS in 34m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c1ef2a2d3995474590fe8aab6c8644dd : SUCCESS in 58m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/972b8bb83ce7493887fb199d1e2ba1af : SUCCESS in 21m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9a7ea5f6217a440e96bb2671806bb5f0 : SUCCESS in 17m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/593c4a2dd3704b46af03e8d6fca3f9c7 : SUCCESS in 17m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5050a65d60aa4321b18851b04c148c6d : SUCCESS in 10m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a9aee067cf2540eb987d84342aef895a : SUCCESS in 5m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a34cfbfd4ca1408993db257cc98be80c : SUCCESS in 19m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c94bcb55fcac4b55baa720bb7942a87d : SUCCESS in 17m 47s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/48f0a7f3aac946759f811913d8276594 : SUCCESS in 24m 40s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/97ec506eea0f41c0bd1b28c3769e48d0 : SUCCESS in 18m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8909462a91644645a64970c3a27fb44c : SUCCESS in 23m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7d780a275853416980d190530ec36808 : SUCCESS in 17m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/547d68ab77984909a4c2a504f49709c1 : SUCCESS in 27m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0a8a2030399141b1b33b587a85ebd12b : SUCCESS in 15m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3c1dd92240cc4d09be1f2291702839d2 : SUCCESS in 19m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c77da33b1ca9441180e06c4822e98b37 : SUCCESS in 21m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bcbc2ba4ea2249d2a350441f630abba7 : SUCCESS in 6m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5528532da643404f826d1de011583464 : SUCCESS in 18m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/77c62f6f70f044d982f541122bc35b14 : SUCCESS in 7m 19s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/1d7d6452c4084ab6a73461be53517ca6 : SUCCESS in 59m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0244260216d1419e9ad1dbd5e0722407 : SUCCESS in 36m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/334a202f9a56428eaa6016b6cf7c269b : SUCCESS in 58m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17391b08dbc947dda54a432e0f85fa86 : SUCCESS in 20m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d210ce2e48a141a9aa54d36894fcf312 : SUCCESS in 3m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d817e4fdcfe441b491d6641f71eb2fb6 : SUCCESS in 20m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c0cc1d886b2485ba770e044326f3874 : SUCCESS in 57m 40s","accounts_in_message":[],"_revision_number":16},{"id":"f1f9d4dfd4d75759d56c88130f5655bc1603354a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 22:57:18.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ba5da08e6ad4565b2becb863f667644\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8727338219e04fc09adefb67e08b8866 : SUCCESS in 16m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/41e42c80542c40949b040c99ef3b652b : SUCCESS in 16m 21s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b6c2fe7771f04425931b8eb024120c90 : SUCCESS in 22m 19s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/69ec05856f9d4c769e0b1cd6fc155316 : SUCCESS in 1h 19m 40s","accounts_in_message":[],"_revision_number":16},{"id":"a79043e569a789197d3d3a377073f0dc29cf740d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-14 14:27:27.000000000","message":"Patch Set 16: Code-Review-1\n\n(13 comments)","accounts_in_message":[],"_revision_number":16},{"id":"a0e6178669056e2b4bab4b99d2d21551b9bc61e5","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-03-15 01:38:23.000000000","message":"Patch Set 16:\n\n(16 comments)","accounts_in_message":[],"_revision_number":16},{"id":"f9cc4ad4e8d8d1477568afbd55672e76953cc6a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-03-15 01:38:24.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"6dfbbfc0fc6e6cf89d244a57c888756e6a3185d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 02:54:59.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe212b7ffefd4d259de60e35dc906ad1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/680c9aef2f12418d8d1ea9a32d4e814c : SUCCESS in 9m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/fd0d77145ca84f5990745d0a7bac22df : SUCCESS in 1h 04m 35s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/931c9b43719a430eba6f3a1c96071939 : SUCCESS in 1h 08m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e50a8dcdc02046e8a6d1cef4e95ab7ce : SUCCESS in 52m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a518ec307c794f6ea4607cc047b6e19d : SUCCESS in 48m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5d960498ffe74e1780e66f1cf0718c45 : SUCCESS in 21m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9c90ffbf7ce641d7b6081a551484ac53 : SUCCESS in 18m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0ac57968e3554caba1d709b30cf24f76 : SUCCESS in 20m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9967ae1a32b44606854718e49bfe88d2 : SUCCESS in 10m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7ac548e403e1403592567a884f3967b7 : SUCCESS in 5m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2425dfb9a8354a7cbf2d062926101312 : SUCCESS in 13m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/79e85441cfa042518713c2307c6adc15 : SUCCESS in 12m 12s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/885c053fcce5436c9c058bb078a00c4e : SUCCESS in 13m 31s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9afa7c9067c34fd481dbcafd061b4737 : SUCCESS in 13m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/42212df9e8e447a59ac04c45831c8e21 : SUCCESS in 14m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ef2ce13ed5af4916a8070853856cb772 : SUCCESS in 20m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5cd6a3aabb024c829f622dcb97b657cc : SUCCESS in 18m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c1160bb1c25e4391aaac9f48523e7247 : SUCCESS in 15m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/22eefcc3b3e34cbb99ed7036b485cd99 : SUCCESS in 16m 26s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f9364925fde644b1866bae045b486493 : SUCCESS in 15m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/17bc209bfc044561828b508044df3906 : SUCCESS in 7m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/99e8f44fd3c24604bdcfdf6f9efa2467 : SUCCESS in 16m 59s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/65849747dd9149bbb015bca0f6c45ed0 : SUCCESS in 8m 41s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/70a1b137df3749b2988e1fc247b8abb7 : SUCCESS in 1h 07m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/33f394bc5674488c9a3e392d04d1e711 : SUCCESS in 54m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2af4bd163c3f496899d0cb4e7cc590e1 : SUCCESS in 45m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7300f8c19264961b882830a272ad663 : SUCCESS in 19m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae9ceadfea2f4b8bb01a486b9de1627c : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6016eb75a6f447138a38d031d83424ae : SUCCESS in 18m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb4363d321f142b7a4656d7e5bcc9a99 : SUCCESS in 1h 04m 15s","accounts_in_message":[],"_revision_number":17},{"id":"d3b96bbc6f019eb2fda6cbcba7b9d169f995e4a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 03:12:20.000000000","message":"Patch Set 17:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aba03613f5964d33adc46d2417d8686d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9c472f803b7043549fd0eeac532bd7ec : SUCCESS in 16m 17s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3ac28ecbf1ba4887acbd24dbea3cdd01 : SUCCESS in 17m 36s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b21753d566b841f69eb4b2f1d24e419c : SUCCESS in 20m 07s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/012f576780b6420981571e65aaba10fe : TIMED_OUT in 1h 31m 08s","accounts_in_message":[],"_revision_number":17},{"id":"bb2ac41ab492b4a2f08b5a5e158a8df78563c143","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-03-19 20:17:11.000000000","message":"Patch Set 17: Code-Review+2\n\n(24 comments)","accounts_in_message":[],"_revision_number":17},{"id":"5171b4ae2bd6ce0225359af7d6b037bbe7abd4b7","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-20 10:52:09.000000000","message":"Patch Set 17: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":17},{"id":"f39dba0a5f35cf4c1cbbf3e7300656860c5348c0","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-03-20 22:17:08.000000000","message":"Patch Set 17:\n\n(21 comments)","accounts_in_message":[],"_revision_number":17},{"id":"d6c39e286417cf7bf296f1d5195c2d0820839681","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-03-20 22:18:08.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"2ccd0a3078f564026c97a918449f2a3bc773ef3f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 23:38:42.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89ea0e9717ac4e8f92d4a060a4979998\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c163a6cde8e4b08ab7acd9ac479877c : SUCCESS in 16m 44s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f4b71f0e3a414cafb51946f3b3de31a4 : SUCCESS in 1h 18m 48s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/81aff98d0a0c4d249f9826adf51e81c6 : SUCCESS in 18m 11s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/32c198b573f54c7ba675f93a4e809431 : SUCCESS in 53m 42s","accounts_in_message":[],"_revision_number":18},{"id":"95a5bb59aa2e775347d6170c07a26851879b5746","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 23:44:19.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43ec1da6750b4582a2587b2ded377232\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8193af73aa24b768b9555ad8e432ef4 : SUCCESS in 11m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/8f3d147bca8b45b0a3ad3c229c7adb38 : SUCCESS in 54m 37s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6fa3f22d671a4541b7d150ac2ab6188a : SUCCESS in 48m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3b7440e6c17a4a8f8afe8d2783827c7c : SUCCESS in 54m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/40e879afcd2541fb872f8639bf6ead33 : SUCCESS in 59m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9fb5a3f91a774188baf776db2a0ab62f : SUCCESS in 17m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bd95ac472c6c46d2acdb37d12470a207 : SUCCESS in 16m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4d60faacc5384fe6820a55e65492768c : SUCCESS in 15m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4154fafdc228419ba3502865556ab1d5 : RETRY_LIMIT in 3m 33s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9fd6e0ac5608482bb2ec0a6e6f942107 : RETRY_LIMIT in 2m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7405831520f64a5e8af41c08706916bd : SUCCESS in 19m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b4df2b71685148359dea445e85490bed : SUCCESS in 20m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/edb8b0f82e254891a594516c21bcfc50 : SUCCESS in 14m 37s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f090ec22f6cd43a5ac85a942a8861d23 : SUCCESS in 17m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e7b7ff5247f1497f8d719e942c08f370 : SUCCESS in 20m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4287e079824649e7913b486454943f6b : SUCCESS in 18m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d91103c92df6488f9a481762ae698ec6 : SUCCESS in 17m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6b8100425f1045518d39584ad190e703 : SUCCESS in 16m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d9b20cfa8bf04c24a9ed91913f7145b3 : SUCCESS in 16m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6270cb9532de4519b0221c4ad66235b3 : SUCCESS in 17m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/58e1adc432c346a19f482178f511b543 : SUCCESS in 8m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d6e502942daf4ae0a97d859a850c0985 : SUCCESS in 16m 15s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/db5dbd43a613477fab5df86585404c38 : SUCCESS in 7m 49s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7c07c7810d39402aa750acf12014873b : SUCCESS in 1h 19m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e3f9b9fbf4b1435aa9211f60b13a0442 : SUCCESS in 50m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8d367192a6044508b88ee138525ec356 : SUCCESS in 49m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa94110368884ecabd9513c293a74391 : SUCCESS in 25m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe63d8db7749495289f65ce03744e816 : SUCCESS in 3m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b48ca71eebd441c6ad8981044710f98c : SUCCESS in 18m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e89080be276d44baabcf6e2a99fb1875 : SUCCESS in 1h 00m 57s","accounts_in_message":[],"_revision_number":18},{"id":"d3fa1288016e101c084197cd9f0e283b2570d7ea","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-01 19:43:12.000000000","message":"Patch Set 18: Code-Review+2\n\n(7 comments)","accounts_in_message":[],"_revision_number":18},{"id":"400ab710b516aa4f5e0b16b56a73bd0e34338242","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-02 09:27:57.000000000","message":"Patch Set 18: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"7eae2548ca70bc25f8565a934de2174bd33d7610","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-04-08 16:37:31.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"ede40f39546ed4000cfc3dba94739ccab044b741","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-08 17:50:52.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08488f09c5ee444fac479eb93f785a9a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/645842065f9141bfa8343a2128f1ba41 : SUCCESS in 8m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/3cd2c85f5dab4be0b922f5683798ba22 : SUCCESS in 45m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e9936f91a9ad4b24b448303c15d1c059 : SUCCESS in 38m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7680a933fda3455a8960d64e8ca42c1e : SUCCESS in 58m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3b6708ed09be4b24ba8d68e62d15bc28 : SUCCESS in 20m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3afffe91be1f484f8872ab78612edbee : SUCCESS in 23m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a0f6990f8a1b43a9adc2ea2536ea80ed : SUCCESS in 18m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/621ac6c183c947829ac8e7bacc43aa0d : SUCCESS in 10m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1c079204cc384be49a14aeeb76018bda : SUCCESS in 5m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2b45074643304aa9aadb071389b7ac20 : SUCCESS in 20m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3f828e490baf483cb0ea595b60682fd8 : SUCCESS in 16m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/01d103f06b9743c0a4fd8c72819469fd : SUCCESS in 17m 16s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d73990687d6942a6bb9b9d518c4b3dd3 : SUCCESS in 17m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/923aab5166dc4618900f8de188c95840 : SUCCESS in 22m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2106abdf273b4ad4bf1982e33074e5ce : SUCCESS in 14m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/18d4aea743a445acab28a41bfbdba777 : SUCCESS in 18m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/04f1f7277f0246a881068a297d8ebe98 : SUCCESS in 15m 10s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3e261df6f4244afd87ddf222327751c6 : SUCCESS in 16m 29s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/09a4fc671c8349c49f43c905cc754621 : SUCCESS in 12m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/271eb3eeaf3b40a982c5bc0de9086feb : SUCCESS in 5m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/df6a55e34fea43cba8626c75439d3bc7 : SUCCESS in 16m 52s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5b77c85cb78644bb896a62a2d748eea9 : SUCCESS in 7m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/48d86d9de3d74ec9bcb4166fcbf80fed : SUCCESS in 53m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fd311d16e64545918ee501415a22d84c : SUCCESS in 40m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4de33b5b60b347bbb93ac0f8b4df9e26 : SUCCESS in 41m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80d61a47f772436a877a68d30b43d374 : SUCCESS in 17m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f648ec75749a4fcaa102f0f6eb6ec5c0 : SUCCESS in 3m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d52fe104132a4dfebf178ec513f708b2 : SUCCESS in 19m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/acc1343bd6b848b684057e4af08f26aa : SUCCESS in 1h 02m 04s","accounts_in_message":[],"_revision_number":19},{"id":"0850e34dc8f87521a910e683460cebe3ae52d0f1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-08 18:13:48.000000000","message":"Patch Set 19:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5a90f3460524ca797e83b0507d77753\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/581cd1c1e7634d65a95c7ebfddf8f6b3 : SUCCESS in 1h 10m 35s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5dae3670335b44c8b4411c68fcdc49a3 : SUCCESS in 14m 04s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3987671e216642cb9eb09cc3eca2ee7c : SUCCESS in 53m 15s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/dd257753c20e4529bd6ccf620ff68ca6 : TIMED_OUT in 1h 30m 59s","accounts_in_message":[],"_revision_number":19},{"id":"a7d11d55556695f07bc2713ab84d9856c59ef748","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-09 11:04:39.000000000","message":"Patch Set 19: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"42d4c8f469bb0116aa2a333c18a3a72568f15660","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-16 18:14:01.000000000","message":"Patch Set 19: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":19},{"id":"ad1608ac3a9e6eae762da8f5208b0e6f1671e1f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-04-17 20:04:07.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"1092af33b3baa80286c6c87b3c16d1c3841fc3b7","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-04-17 20:05:11.000000000","message":"Patch Set 19:\n\n(3 comments)","accounts_in_message":[],"_revision_number":19},{"id":"c029a0997e2b60e454684bb56a46003bb5afab80","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 20:25:04.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab79d3b711dc4718b87d8bc48a7ec95f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a04c50b2a956417d80b5d998416d3066 : SUCCESS in 15m 13s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2f09d95ae1b34e29a6dfa2e219da70ca : SUCCESS in 17m 36s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b2db9be363dd4d408e147b36f69b2c69 : SUCCESS in 18m 31s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/751a9018dc6b45e5a05e495f6021e3d4 : SUCCESS in 19m 15s","accounts_in_message":[],"_revision_number":20},{"id":"2f03613e232d141d7a94107085a9bea934c84073","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 21:23:33.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d748bae3156a4b9c9dbfa11aa809a041\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac9fad09f3d24116ac019fd3b2ba0e62 : SUCCESS in 8m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/7884b8a95ad84c8db79d8a22c098666b : SUCCESS in 1h 02m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/09d309e9e5d74f909dd026ef2577ad35 : SUCCESS in 51m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/523db2e7793e44bb936a3935961693da : SUCCESS in 1h 00m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/abb0e29b455c464fbb5e1778d503aca8 : SUCCESS in 21m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/764eeda40a84438cb1f7933c82944016 : SUCCESS in 19m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fa2e0acb46e04aa28d481c9a24f4cf8c : SUCCESS in 19m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/94a1c28f74a34f038970e834aa5ca3c9 : SUCCESS in 11m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/887452e8f0bf46e49d13cf581906c125 : SUCCESS in 5m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ea4854d09d0d4babaf9bc8cb0aeee83b : SUCCESS in 19m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/23c4397cc18a4d01b1a3039e656c4ec3 : SUCCESS in 21m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e58d115cfa5f43c4bd8e20a314914eb6 : SUCCESS in 17m 47s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b6e6ee4ab27748bfa3b245b9228a5bfe : SUCCESS in 19m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3361f6c799444c93bb8277b6553aa207 : SUCCESS in 13m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/030139a733e14991ba4e420ddbd72a85 : SUCCESS in 20m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dd8903b8a125472588839187710f719e : SUCCESS in 22m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/47a11233239b4e32a7ba58d548286ef6 : SUCCESS in 18m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6d185a2b2ed0420cb77f84dd04eff153 : SUCCESS in 18m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e3b9f0fa86b147beba5bef577a48f23b : SUCCESS in 21m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d8a642b317654a4e9522ff4f97e25494 : SUCCESS in 5m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/47fc08d514d64f02ba4cb1276f9bed86 : SUCCESS in 17m 37s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c5a23b82a67b458e8e79b735c40a9f84 : SUCCESS in 7m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b2bab432546c4d84aa783aed8988d548 : SUCCESS in 53m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c6e8aa5971344428a8618786b1e3b01c : SUCCESS in 55m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c7585225af8f4492a2b18acee1635ca8 : SUCCESS in 58m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/763cc30c99764cfdae963686640e47d5 : SUCCESS in 20m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a523cb6517b441883fe66b42c4def81 : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9f70234c935c4b1abf32b3d00b805837 : SUCCESS in 19m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/859973103efe4eadbb854b4347e2ea30 : SUCCESS in 44m 19s","accounts_in_message":[],"_revision_number":20},{"id":"9bfa35a2708b8ac0f90a56523284457677874aa4","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-18 08:00:46.000000000","message":"Patch Set 20: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":20},{"id":"ea8ec77cc12ccd1cf94d9dea4f6eb82014f8522c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-04-18 16:10:56.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"b2617d896106f8d5fb1b6df6be7a69de91d59631","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-04-18 16:11:29.000000000","message":"Patch Set 20:\n\n(2 comments)","accounts_in_message":[],"_revision_number":20},{"id":"7d313a4d7e501b9121c11d50794ccc16c7f4d71d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 17:21:57.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8dfe408f29c497c95a4c59b338eda66\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37ca62996d1843968afc73542e3cc79a : SUCCESS in 7m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/8f3391397e42441191c2df740340d05e : SUCCESS in 55m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a97251f260de426a9a712f03a5d141f2 : SUCCESS in 57m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19513f84c54c47bca12d68492a016348 : SUCCESS in 1h 02m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5a8b795bfc8a4385bfef4e7f8a85b2e0 : SUCCESS in 19m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2810eedfa43d4f5ea7403846efb72f3d : SUCCESS in 19m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/855363cb1bf0478f97c5d770f58c9dbf : SUCCESS in 21m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fbe1e2cc02e347de833ddf4f657576ef : SUCCESS in 11m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e09defe27f0644cbb1b8ea7c300ef042 : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a9b318b7dc2a4f96b505b21e831f0df8 : SUCCESS in 19m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6e29a47563454b87b70b6fe8c770c70d : SUCCESS in 18m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1c3437d87fb64b4689894c0aeb759fb4 : SUCCESS in 17m 30s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b02991b8331d425cabed5e7d816848b8 : SUCCESS in 17m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/889e893fdbfc4559b8e2ce21a5181cca : SUCCESS in 17m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1278cd9b03a6478fa333542f2ca76c60 : SUCCESS in 22m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/057934511eaa4a268e8f5f7d58e37a84 : SUCCESS in 21m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5060389533484dd2b5fbe89e716807ef : SUCCESS in 18m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fe93ab0ae36848e0b453a8414e6a4879 : SUCCESS in 20m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/da4464fe051f47c5afe97bad2ad8f895 : SUCCESS in 17m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/11bde9c866cf41f097d94bf204300f4e : SUCCESS in 8m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8961dd833187429c90d792e02472092d : SUCCESS in 18m 42s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/79a17475e46e459ebd934f465b3fd971 : SUCCESS in 8m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/45cc1b84eb3b4bbd8c4c442f81374df7 : SUCCESS in 1h 04m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7fcc72a15bb241b0aa8846c291897c47 : SUCCESS in 51m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d040f924b03d4aba817d9d3d8ed9841c : SUCCESS in 42m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3386738e1a404e3e93f8a9fc4c6175ef : SUCCESS in 23m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35c6c5c379bf4b2790ca2d72f610505e : SUCCESS in 5m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8eb86267aef345798ebb127d25881634 : SUCCESS in 20m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c61dac5fc7848c79ae129be3b1e37c0 : SUCCESS in 43m 44s","accounts_in_message":[],"_revision_number":21},{"id":"8cd7c9fe85df82e099acb612d0935cf2454fc740","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 17:25:12.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79f93bcdbd0047ed9b7c98c835418b8b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2918fc6c22ff47adb6c18e3d99080250 : SUCCESS in 59m 24s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5b3118dd58874ed8872fc8c34c0b29c0 : SUCCESS in 18m 18s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c491c7a9db7e4773834d5ed7c1ceade9 : SUCCESS in 1h 12m 36s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6273a8a73e9e44ca80e26369bb70406e : SUCCESS in 18m 32s","accounts_in_message":[],"_revision_number":21},{"id":"63dc6ad992976da09ba7926745c90e4b75e9638a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-19 10:35:48.000000000","message":"Patch Set 21: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"57a01146a3965cb166bf3dd0ce267ced030dc5b3","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-22 18:40:51.000000000","message":"Patch Set 21: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":21},{"id":"34a158cade6e3149ba96b72056a1562194f31b9f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-22 20:15:17.000000000","message":"Patch Set 21:\n\n(4 comments)","accounts_in_message":[],"_revision_number":21},{"id":"6415f27c2c9cff12d6e7aed26a3d71fcd41baa62","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-23 00:02:42.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"bd450a5a8c81e890b63aefbadd633983db5aca38","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-24 16:37:37.000000000","message":"Patch Set 21:\n\n(2 comments)","accounts_in_message":[],"_revision_number":21},{"id":"978374dfd1075745647f4bd1883685221bb39115","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-04-24 21:08:27.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"e859efbab069a688bdb0d231c81b1d7052a3bc0f","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-04-24 21:15:22.000000000","message":"Patch Set 22:\n\n(7 comments)","accounts_in_message":[],"_revision_number":22},{"id":"7389ae4dbbd9d75a11c450539fd4f11cdb424342","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 22:16:31.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ddd48192169f44d491ce87805d876608\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c10db926095a45ed9beebeda0b85632c : SUCCESS in 8m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/43d5ce6436a0441db9ad7ffb445888d3 : SUCCESS in 53m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ae2e15ddea2d4a10a2c8daa04400a665 : SUCCESS in 1h 00m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec54c84f52ad413c8c9906baaf42e3a8 : SUCCESS in 1h 02m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e9df27e741ca4aca998341a8e3afd783 : SUCCESS in 21m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ad5d8531b91f4cfbb40543bc044158d2 : SUCCESS in 23m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/14769ba10d7b4212aa566b22b4266844 : SUCCESS in 21m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/123f856f3a114c5fa59733680695f531 : SUCCESS in 9m 13s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b92d47c62a054026b32a5a1a9a10ea78 : SUCCESS in 5m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/aaebe32929e34e39b4cfaaf9f69a3dde : SUCCESS in 19m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/68d8211ba9fb4c848f28d5aa8073c5be : SUCCESS in 20m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/748aecb020804f8aa14570c00e04a2ab : SUCCESS in 14m 59s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d4007be64e8e44e0a7e0a0dc7967814e : SUCCESS in 15m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1c99fcbcbfac4d849bb96c7c74c57307 : SUCCESS in 22m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f7700dd960fb4729b7d6ad1a52654228 : SUCCESS in 18m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0633c55ef8fc43a3b87a020f1abc8ca0 : SUCCESS in 18m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7faa8f779bdc4b4c8dea816613b86810 : SUCCESS in 18m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6dd23a7165e14871aa8cb81ba424b8cb : SUCCESS in 20m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2fd08b5c7a2b4d45961233490870cf52 : SUCCESS in 21m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/19342fb7a37e43959ed50021bb1db966 : SUCCESS in 7m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8b538d102ca444808ff709dfd317c7f2 : SUCCESS in 13m 18s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f24e8c0a5a1042b18bc5398ac13b7612 : SUCCESS in 7m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4533f0091fc4a45a5a653a3da347859 : SUCCESS in 54m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4f8c9bce7a2741238babb2fad32c0fce : SUCCESS in 49m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/41ef3aa92caa4291bd5d139faf669b49 : SUCCESS in 55m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb5b9ff586114fecb9fd96a26c3adc7f : SUCCESS in 20m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5071d97d826a4ba5aef9cc69b846c710 : SUCCESS in 3m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/13b1da527fa741caa75826ee6c7e4f95 : SUCCESS in 21m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ff044b8fd2d4ea5b7ee04b075ccad78 : SUCCESS in 1h 00m 36s","accounts_in_message":[],"_revision_number":22},{"id":"92e4946dc5e998bb4f95f74e2315b208c0a92e5f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 22:39:43.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da3ddb36f69e41258c4a34beedd37ef7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b5952cb04cad44878892af7baf08e172 : SUCCESS in 17m 48s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/105768c5b7df4da0b437a6d2bca48f19 : SUCCESS in 1h 10m 34s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8a9580a17cf64454a56799155127caf4 : SUCCESS in 20m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b4fa71fc2fae4011b03bdfe8fc12e02e : SUCCESS in 1h 29m 26s","accounts_in_message":[],"_revision_number":22},{"id":"01bce868600f37e421ab91ae1210cc263374b0e9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-24 23:07:41.000000000","message":"Patch Set 22:\n\n(2 comments)","accounts_in_message":[],"_revision_number":22},{"id":"e08b342b6db95701c35576f32004871fc00ea46b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-04-25 18:05:46.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":23},{"id":"20ba874544f66649f542817bdc5117cd1babbb79","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-04-25 18:07:13.000000000","message":"Patch Set 22:\n\n(2 comments)","accounts_in_message":[],"_revision_number":22},{"id":"dd9436b80d55dbe8d63735b3b6ffa4b6d6c8db38","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 18:59:59.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff82f9e5d06b4049b16fa1a2d81ee4dd\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb564391b8bc425f9ea2c5a736965a19 : SUCCESS in 16m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a67fabcebed045aa8278f6f39b7fd3e6 : SUCCESS in 52m 32s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5431ceb29d3f47bbaaf3b26ed1e0d6d8 : SUCCESS in 18m 31s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/93e22c8440a14076a7e7fb090afcdce4 : SUCCESS in 49m 30s","accounts_in_message":[],"_revision_number":23},{"id":"ba31ecfe07e5a7ddf48cb80a6a6797b1b865687e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-25 19:47:15.000000000","message":"Patch Set 23: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"28876f7b6bc463e3954987bae310475b03f32027","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 20:09:35.000000000","message":"Patch Set 23: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/30ed6c607ced47218ae96308daa8635c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d640aaed20d346e687e4916001e77a21 : SUCCESS in 8m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/3a502c1a11c249b3bc77bd627be4dedb : SUCCESS in 1h 03m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e548955440ae40fab91f58f9f6a66983 : SUCCESS in 43m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a5ac1b26ddd34efb80f804d98024c768 : SUCCESS in 53m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dc62536902024d8f89da76c45f8f77d4 : SUCCESS in 24m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eea5ed320fc141b7964a8ce49ae0f9d3 : SUCCESS in 20m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/868286b25f0c4498bd4df17b05675442 : SUCCESS in 19m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f00be46a241047199ae6de6e70b352bc : SUCCESS in 11m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d589122da59e4048b4112f2d22985cd7 : SUCCESS in 4m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bb3d0e14623a4e55919e4312bf337092 : SUCCESS in 18m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/32bcbeba8b24461ea443a802703da1f0 : SUCCESS in 18m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c61dda809f424102823eb0da3fb987d0 : SUCCESS in 20m 14s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/7ba01228d57f4ea69e233c2e1214c1c9 : SUCCESS in 18m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8700fe8fd9c94663962710a9cfdd89ea : SUCCESS in 15m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bf9a6425d57f419ab42b7d1078aedd5c : SUCCESS in 18m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/51684f11a0bd4b8ca3f6978ed28ae66e : SUCCESS in 18m 06s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c454caf07fd04ea1afeb524fb511f720 : SUCCESS in 15m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d2107dbd836f4124a03561aeaf3662c2 : SUCCESS in 19m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c4bf6cbe00a3498d884d8530a7a355a4 : SUCCESS in 13m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e8ea8b0f490e45b7b9f7f7e23d4a93c2 : SUCCESS in 6m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e22daf2140ab4455bfa72e60cc420111 : SUCCESS in 14m 21s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/816dd0536e02415a858337f0ee4f9c1b : SUCCESS in 9m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ddc315ea96d046848286baf3650a4a06 : TIMED_OUT in 2h 01m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cf4250e7c4f34a088a5cd67955fc4bab : SUCCESS in 52m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/36b57899f153444591a5e54ef90cf38d : SUCCESS in 34m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2dabdb450f3c4748b4953b49b330d783 : SUCCESS in 22m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b518a6f3ebf04361b906e9a421058774 : FAILURE in 3m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cb6a7cee0e9840cbbedd490f2f55fae9 : SUCCESS in 19m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/edf85cdb24dd4a4a8cd0516221114b94 : SUCCESS in 53m 31s","accounts_in_message":[],"_revision_number":23},{"id":"4338b7c662f3231531d013e5a8c5b161720ac556","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-04-25 20:59:43.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":24},{"id":"5c2ab37b1df13091170cf38662bddaebb01ecc63","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 21:46:27.000000000","message":"Patch Set 24:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c907df4f52ca4ee68293bcbbe993cc52\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0b947f29c9a747e7b4c63479aeeaf17a : SUCCESS in 16m 15s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c3f8b20d0e7445d598f9e8dcc17df0f2 : SUCCESS in 16m 06s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/73a0dabdeb664e44b794170276365524 : SUCCESS in 23m 20s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/912a348bdde44137bad90b9359a05ed2 : FAILURE in 45m 07s","accounts_in_message":[],"_revision_number":24},{"id":"102bbd9cfafa9498248a92979321670650def8b0","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-25 22:00:26.000000000","message":"Patch Set 24: Code-Review+2\n\n(5 comments)","accounts_in_message":[],"_revision_number":24},{"id":"65963dd1ed19062ac074abe150caf57361af5c25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 22:13:41.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/650d3b09317347c7abe6bd991931290a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/466248d2af4d4ef599b98e6aa510c00e : SUCCESS in 9m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/ec0dd87b4b8a41f4b6c1fe16e627de4c : SUCCESS in 41m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bc89b055345a423ba744d54a4e2853ea : SUCCESS in 52m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b362f183f9514bf0a7e4b09b6be817bd : SUCCESS in 1h 03m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3500721a86214fdaa628a41f864cc768 : SUCCESS in 18m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8db3bef86dea4b359c22b1ca82bd2ca6 : SUCCESS in 15m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7daf4d574ce34d0298a5a5c59377c52e : SUCCESS in 20m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a3ade5dc69cd469182e9c51d5d23fb8c : SUCCESS in 10m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4c23f054178f4b38abd854dc013b8d02 : SUCCESS in 5m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/48f2309f681d495390f01753f6c838fb : SUCCESS in 19m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6043a7711141479ebe83e8c20d40126b : SUCCESS in 17m 56s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fae0aaa5d15d42c1bcd4854c0bc9c4db : SUCCESS in 16m 50s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5459034313d848a880ae53fcb870e045 : SUCCESS in 17m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/32856d0a7a16477ab3c16679f6fb6e1b : SUCCESS in 19m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fd7ae77de41847e6a806b5d0d5405501 : SUCCESS in 18m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/65f397e8c37f477987899abe98c87ed5 : SUCCESS in 18m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2c0e1dd50e0344ce8b97248dad441d2c : SUCCESS in 17m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1e830a2c2612408c9bc1c670f2daaf9f : SUCCESS in 17m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4f33e7707e65479d9c81e1538a12823f : SUCCESS in 16m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/50f8a3327e6944799b241dcbebd47e73 : SUCCESS in 6m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4ac0ebafb35b4228b8edef409c52cf6f : SUCCESS in 18m 47s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/16797786c9644c06ba4257b9c3cf9033 : SUCCESS in 7m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6755c1048fde4c478718d87e4fc90156 : SUCCESS in 57m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2de32721faed40b486b3adba09b24639 : SUCCESS in 49m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/eb810d9922e045b8aec58de532cb204f : SUCCESS in 39m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c209fdea1da4825a0c1b6e3f505b468 : SUCCESS in 19m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ebf1cb71efa40e98e521495361957ed : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e8d80591e67745d891a13c22f157b85f : SUCCESS in 18m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5fba9e9a73c41f896f8d1b29bcf9029 : SUCCESS in 47m 00s","accounts_in_message":[],"_revision_number":24},{"id":"28bf172d812f5b5751bee1a88d8f591fcae96cbc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-25 22:28:46.000000000","message":"Patch Set 24: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":24},{"id":"6bfc40a4ab23e90e0d18c33c74806a379046874d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 22:29:31.000000000","message":"Patch Set 24: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":24},{"id":"46ae181aa33ce43b6e0d5590f051c4c58e1c9d5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 00:37:23.000000000","message":"Patch Set 24: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bee1ab3e7d624157beef9e523c375423\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0af0369415c4163b5179805168ff971 : SUCCESS in 8m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/b4a885c2667946589e4f0163478d69da : SUCCESS in 1h 02m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3a3b0beca1b049269687b9e53f7551b4 : SUCCESS in 52m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d5770c977ecd48f6b8e341a5a1c00963 : SUCCESS in 57m 00s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ce51c922b3b74ed185c649379edc5a8d : SUCCESS in 10m 45s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/e2d80622b00940028a18ea6f50a14715 : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cf3080bcbe094526afbd9e35ecaacea5 : SUCCESS in 17m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/833a777bcf1d4d8f9a5b8677241002e6 : SUCCESS in 17m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f41c185ed48146f28f3bc6be512b5fba : SUCCESS in 17m 37s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0ceb86c92b1442f99a82a1586d4da2df : SUCCESS in 16m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cd63e86895b4421dacd87c98cb34cdb6 : SUCCESS in 17m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/83d6b410dc834e98baa7c025ebffe219 : SUCCESS in 18m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f290c77e5fdf4ad588ed53f84732d7e2 : SUCCESS in 19m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3b406386d7974d5d87e5c6579d6d7552 : SUCCESS in 15m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/38eab891aa3e477d97d55f892eff254e : SUCCESS in 16m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b7572c53f1734bfc88ec55ab0c18607d : SUCCESS in 13m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a530b56f8b8c48e385e0cd516a6719e9 : SUCCESS in 6m 08s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/38b23e3ce6c6478aa63c8f1de2369e90 : SUCCESS in 7m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/247be126bc434e1f96d001fb45b13c1c : TIMED_OUT in 2h 00m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/26d712b03a424d04aefcb704cc8cdd26 : SUCCESS in 1h 03m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/34b55e326d9540e7bf0af735b73f22ce : SUCCESS in 38m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/427977d2ec8a423a81f16fa5902c1df1 : SUCCESS in 19m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/847858c9e70d426a8c8b1b2369554850 : SUCCESS in 3m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6b041dd89996442cb941b87cde64c496 : SUCCESS in 41m 53s","accounts_in_message":[],"_revision_number":24},{"id":"d7388f52791539a1e3b06cc1066ac172e44f6e78","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-04-26 07:42:49.000000000","message":"Patch Set 24:\n\n(2 comments)","accounts_in_message":[],"_revision_number":24},{"id":"00140f8b52a9cbffc6817fa7d8ab67e7d586f92d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 08:27:42.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a02dffc16ab64aa888acdee14d67eb4e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6a868b463fc241ecad50e1aee5866f68 : SUCCESS in 16m 47s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/517783959a494cdb9eec04a9f092ed7e : SUCCESS in 24m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8059f540272949e095063fc7a8955f5e : SUCCESS in 43m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9c4cb29522e24361868841ca3a887df6 : SUCCESS in 20m 22s","accounts_in_message":[],"_revision_number":24},{"id":"1e85bd8a36efdd616f8482b71033971b74f8e810","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 09:13:11.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4f103297ad7471b8eae50942f271599\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7be61437acd4e83a94319c86a8c3808 : SUCCESS in 8m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/a16bb4e0e6bb49da868837cd68e5b48b : SUCCESS in 56m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1a346426431c493d94e52f7c2d82731b : SUCCESS in 58m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/24acbed7c8504e82ae8222862944c4cb : SUCCESS in 1h 04m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1fa4799509604c46bd791960a8337506 : SUCCESS in 21m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b34d600c4aa64bafa0c4344d87ccb285 : SUCCESS in 19m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2555a2e87a4b44a2a7694783d15c51b2 : SUCCESS in 22m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/05c2ce9973eb4631a4b3b0adc1b7d2a1 : SUCCESS in 11m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b504168db3bb439eb73ead47509e362e : SUCCESS in 6m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3e7f0b090cf64614977aa10f3eba24cf : SUCCESS in 19m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/216ba167d27b43a382036c1ace4323b6 : SUCCESS in 16m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e01865fd991b4b7dba3beea35563a7d3 : SUCCESS in 17m 19s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ef4bcfbec4eb4c81ad6acf52eb51db07 : SUCCESS in 16m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/765084bdeaf24218bca865206cb8629e : SUCCESS in 19m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/54c918949ecb40b498f407e31ae517d8 : SUCCESS in 24m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/60c15b675b114442a97b13f6b61dbb86 : SUCCESS in 18m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7b25c0b582a345dda03299c1d7c898ba : SUCCESS in 14m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/bd6d06a95b974ff1a7edd06413dfcdaf : SUCCESS in 18m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e08e640726904d2db4dc852dc975195f : SUCCESS in 16m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/141d88fb25714951ac792efe82892e69 : SUCCESS in 8m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6dae6734351a4c09ab4631be73bfed40 : SUCCESS in 17m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/58d2e079fd9e46c099f6c483ddfdb1fe : SUCCESS in 7m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9874d073dc454d7b88e70c9fd56c446e : SUCCESS in 1h 08m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/104da9a5eb614ebaba80c32931fa82b4 : SUCCESS in 50m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/619ea6cb02654d42993aed9293e4a2ac : SUCCESS in 56m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1df9b77e7d9940069a0fe70867cdb201 : SUCCESS in 22m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/becc2ce1c2d546c38e5a955aa3edc681 : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4a96e4bacb1349c9969eca3b41fbe8cd : SUCCESS in 26m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74577a5279884330a3455166f1cd18a7 : SUCCESS in 58m 02s","accounts_in_message":[],"_revision_number":24},{"id":"19bfdd45d1bb217714b034b8d7485685be5efa0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 09:13:44.000000000","message":"Patch Set 24: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":24},{"id":"6666e57b021ef828ad08294c1c6034301663b130","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-26 09:19:50.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"928acbd6d6f3c74a2ac91a77cc586683312a4d7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 10:28:46.000000000","message":"Patch Set 24: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fa43eb72f00433b96643b6c911200a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ffe14c43f934bae8bc1a0e5e50fad8c : SUCCESS in 8m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/476cb505c7ab42a8aaa6072438386ee1 : SUCCESS in 54m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a38046892b144c12b0e4a170483c7431 : SUCCESS in 40m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f1f13894c2774cb292bf44758db65dba : SUCCESS in 45m 35s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/2a83c9bce64c4db8bc15bc7eaca2a30a : SUCCESS in 11m 50s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/7dddfbfc244b4e8f99fac68d6204cbbc : SUCCESS in 5m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d725571ec6d0425a861fd10f04df533e : SUCCESS in 20m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6c35835141c04726bfadcbe3987e16bb : SUCCESS in 13m 55s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0c9ec736e8064715865ca03e091227a6 : SUCCESS in 14m 26s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c6165654b59943ce8b09ea03d14f29fa : SUCCESS in 17m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2206e3453adf40f18c6cf0c02d579fb3 : SUCCESS in 17m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/748f1f86c4364d2fa4be113f57dd6c0f : SUCCESS in 21m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/89c0ef810fcf488dbee96f51ccde9698 : SUCCESS in 20m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/329dba6895b2453b86a75960cee21da1 : SUCCESS in 20m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f3239d7b212740279690a16adea7543c : SUCCESS in 20m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8bb3e3e0cca3403482cc16b247f27377 : SUCCESS in 19m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b0c9282972a64edeb08859438816647b : SUCCESS in 8m 08s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/548787ceb15147b0a2bd1eb6a717a31d : SUCCESS in 6m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3150b6ce67364bb6b48f6599c85c3946 : SUCCESS in 1h 09m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/851639a5683f4b26994509e08ca07a5f : SUCCESS in 52m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8d6bd1e190ac449da8bc5b55f40447b3 : SUCCESS in 54m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c2cdb646ff64015a0dfbb6d1744db74 : SUCCESS in 21m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78643247e73e4ee6892695ea49ff7b02 : SUCCESS in 4m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0dc93fe2260e4e87af270da99df8a7c7 : SUCCESS in 42m 05s","accounts_in_message":[],"_revision_number":24},{"id":"c669feb6901911020bde13462f4b6c5d420e353e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 10:28:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":24},{"id":"c1024a8455e90bf329eef25f0e653ddaa8b146ee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 10:29:46.000000000","message":"Patch Set 24:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf4b6af8d8434bbeb76a7a72ce84b576\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/769274ec9ce94eb4846df253d44abc07 : SUCCESS in 41s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/380d69de22d74bd19106a9443ed96a1f : SUCCESS in 34s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/1f9ddf3061b041eb80b4a607c0c7d2c0 : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":24}],"current_revision_number":24,"current_revision":"5961ba0ca7ddcfe296627fbef01ec4b32d538290","revisions":{"639625dda626c68fabd4bc0a98c69cb90daeb03f":{"kind":"REWORK","_number":1,"created":"2023-02-22 20:14:59.000000000","uploader":{"_account_id":396,"name":"Mandell","email":"mdegerness@nvidia.com","username":"mdegerne"},"ref":"refs/changes/06/874806/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/874806/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/874806/1"}}},"commit":{"parents":[{"commit":"8778b2c1328226cbcb67cd38fbb9e0771e2162fc","subject":"Merge \"Add new Kota\u0027s email entry to mailmap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8778b2c1328226cbcb67cd38fbb9e0771e2162fc"}]}],"author":{"name":"Mandell Degerness","email":"mdegerness@nvidia.com","date":"2023-02-22 20:14:05.000000000","tz":-480},"committer":{"name":"Mandell Degerness","email":"mdegerness@nvidia.com","date":"2023-02-22 20:14:05.000000000","tz":-480},"subject":"WIP Add grace period to object expirer","message":"WIP Add grace period to object expirer\n\nChange-Id: I106103438c4162a561486ac73a09436e998ae1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/639625dda626c68fabd4bc0a98c69cb90daeb03f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/639625dda626c68fabd4bc0a98c69cb90daeb03f"}]},"branch":"refs/heads/master"},"32fff1bb6f5442bede86c70d526286792e1139f2":{"kind":"REWORK","_number":2,"created":"2023-02-22 20:38:45.000000000","uploader":{"_account_id":396,"name":"Mandell","email":"mdegerness@nvidia.com","username":"mdegerne"},"ref":"refs/changes/06/874806/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/874806/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/874806/2"}}},"commit":{"parents":[{"commit":"8778b2c1328226cbcb67cd38fbb9e0771e2162fc","subject":"Merge \"Add new Kota\u0027s email entry to mailmap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8778b2c1328226cbcb67cd38fbb9e0771e2162fc"}]}],"author":{"name":"Mandell Degerness","email":"mdegerness@nvidia.com","date":"2023-02-22 20:14:05.000000000","tz":-480},"committer":{"name":"Mandell Degerness","email":"mdegerness@nvidia.com","date":"2023-02-22 20:38:31.000000000","tz":-480},"subject":"WIP Add grace period to object expirer","message":"WIP Add grace period to object expirer\n\nChange-Id: I106103438c4162a561486ac73a09436e998ae1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32fff1bb6f5442bede86c70d526286792e1139f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32fff1bb6f5442bede86c70d526286792e1139f2"}]},"branch":"refs/heads/master"},"827c79c892c56155c5d94181ce497aa9af0fdfa5":{"kind":"REWORK","_number":3,"created":"2023-03-06 17:17:25.000000000","uploader":{"_account_id":396,"name":"Mandell","email":"mdegerness@nvidia.com","username":"mdegerne"},"ref":"refs/changes/06/874806/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/874806/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/874806/3"}}},"commit":{"parents":[{"commit":"8778b2c1328226cbcb67cd38fbb9e0771e2162fc","subject":"Merge \"Add new Kota\u0027s email entry to mailmap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8778b2c1328226cbcb67cd38fbb9e0771e2162fc"}]}],"author":{"name":"Mandell Degerness","email":"mdegerness@nvidia.com","date":"2023-02-22 20:14:05.000000000","tz":-480},"committer":{"name":"Mandell Degerness","email":"mdegerness@nvidia.com","date":"2023-03-06 17:14:40.000000000","tz":-480},"subject":"WIP Add grace period to object expirer","message":"WIP Add grace period to object expirer\n\nAdded the per-account grace period\n\nChange-Id: I106103438c4162a561486ac73a09436e998ae1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/827c79c892c56155c5d94181ce497aa9af0fdfa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/827c79c892c56155c5d94181ce497aa9af0fdfa5"}]},"branch":"refs/heads/master"},"89c1fc5cc46e7272e8a95c9d15334373d58f8ff3":{"kind":"REWORK","_number":4,"created":"2023-03-06 20:29:47.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/06/874806/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/874806/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/874806/4"}}},"commit":{"parents":[{"commit":"8778b2c1328226cbcb67cd38fbb9e0771e2162fc","subject":"Merge \"Add new Kota\u0027s email entry to mailmap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8778b2c1328226cbcb67cd38fbb9e0771e2162fc"}]}],"author":{"name":"Mandell Degerness","email":"mdegerness@nvidia.com","date":"2023-02-22 20:14:05.000000000","tz":-480},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-03-06 20:29:38.000000000","tz":-360},"subject":"Add per account grace period to object expirer","message":"Add per account grace period to object expirer\n\nChange-Id: I106103438c4162a561486ac73a09436e998ae1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89c1fc5cc46e7272e8a95c9d15334373d58f8ff3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89c1fc5cc46e7272e8a95c9d15334373d58f8ff3"}]},"branch":"refs/heads/master"},"ee0f6a180b98308db3f1dc8b48002d034f673436":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-03-21 18:34:15.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/06/874806/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/874806/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/874806/5"}}},"commit":{"parents":[{"commit":"9a1bfb8975fb1315fb417161a2865b6bb4760415","subject":"proxy: add tests for listing with shard gaps","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a1bfb8975fb1315fb417161a2865b6bb4760415"}]}],"author":{"name":"Mandell Degerness","email":"mdegerness@nvidia.com","date":"2023-02-22 20:14:05.000000000","tz":-480},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-03-21 18:34:15.000000000","tz":0},"subject":"Add per account grace period to object expirer","message":"Add per account grace period to object expirer\n\nChange-Id: I106103438c4162a561486ac73a09436e998ae1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee0f6a180b98308db3f1dc8b48002d034f673436"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee0f6a180b98308db3f1dc8b48002d034f673436"}]},"branch":"refs/heads/master","description":"Rebase"},"083cde333c61c988d89816e642f474b238d68721":{"kind":"REWORK","_number":6,"created":"2023-03-27 19:15:18.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/06/874806/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/874806/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/874806/6"}}},"commit":{"parents":[{"commit":"9a1bfb8975fb1315fb417161a2865b6bb4760415","subject":"proxy: add tests for listing with shard gaps","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a1bfb8975fb1315fb417161a2865b6bb4760415"}]}],"author":{"name":"Mandell Degerness","email":"mdegerness@nvidia.com","date":"2023-02-22 20:14:05.000000000","tz":-480},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-03-27 19:15:08.000000000","tz":-300},"subject":"Add per account grace period to object expirer","message":"Add per account grace period to object expirer\n\nChange-Id: I106103438c4162a561486ac73a09436e998ae1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/083cde333c61c988d89816e642f474b238d68721"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/083cde333c61c988d89816e642f474b238d68721"}]},"branch":"refs/heads/master"},"4c68d9ed7546e11a3558846adbb7d7a296f0a943":{"kind":"REWORK","_number":7,"created":"2023-04-18 19:01:50.000000000","uploader":{"_account_id":396,"name":"Mandell","email":"mdegerness@nvidia.com","username":"mdegerne"},"ref":"refs/changes/06/874806/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/874806/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/874806/7"}}},"commit":{"parents":[{"commit":"4b6f54d063c33e13ced7d6dd5208921680f56ca6","subject":"Merge \"ssync: Round-trip offsets in meta/ctype Timestamps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b6f54d063c33e13ced7d6dd5208921680f56ca6"}]}],"author":{"name":"Mandell Degerness","email":"mdegerness@nvidia.com","date":"2023-02-22 20:14:05.000000000","tz":-480},"committer":{"name":"Mandell Degerness","email":"mdegerness@nvidia.com","date":"2023-04-18 19:01:42.000000000","tz":-420},"subject":"Add per account grace period to object expirer","message":"Add per account grace period to object expirer\n\nChange-Id: I106103438c4162a561486ac73a09436e998ae1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4c68d9ed7546e11a3558846adbb7d7a296f0a943"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4c68d9ed7546e11a3558846adbb7d7a296f0a943"}]},"branch":"refs/heads/master"},"63d2c9c58d4b37f75d3924f0719a917b51e698b6":{"kind":"REWORK","_number":8,"created":"2023-04-20 20:21:02.000000000","uploader":{"_account_id":396,"name":"Mandell","email":"mdegerness@nvidia.com","username":"mdegerne"},"ref":"refs/changes/06/874806/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/874806/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/874806/8"}}},"commit":{"parents":[{"commit":"4b6f54d063c33e13ced7d6dd5208921680f56ca6","subject":"Merge \"ssync: Round-trip offsets in meta/ctype Timestamps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b6f54d063c33e13ced7d6dd5208921680f56ca6"}]}],"author":{"name":"Mandell Degerness","email":"mdegerness@nvidia.com","date":"2023-02-22 20:14:05.000000000","tz":-480},"committer":{"name":"Mandell Degerness","email":"mdegerness@nvidia.com","date":"2023-04-20 20:20:41.000000000","tz":-420},"subject":"Add per account grace period to object expirer","message":"Add per account grace period to object expirer\n\nChange-Id: I106103438c4162a561486ac73a09436e998ae1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63d2c9c58d4b37f75d3924f0719a917b51e698b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63d2c9c58d4b37f75d3924f0719a917b51e698b6"}]},"branch":"refs/heads/master"},"e1b8c4b68fb58efded42166b866c64b60b3dd9ea":{"kind":"REWORK","_number":9,"created":"2023-04-20 22:22:06.000000000","uploader":{"_account_id":396,"name":"Mandell","email":"mdegerness@nvidia.com","username":"mdegerne"},"ref":"refs/changes/06/874806/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/874806/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/874806/9"}}},"commit":{"parents":[{"commit":"4b6f54d063c33e13ced7d6dd5208921680f56ca6","subject":"Merge \"ssync: Round-trip offsets in meta/ctype Timestamps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b6f54d063c33e13ced7d6dd5208921680f56ca6"}]}],"author":{"name":"Mandell Degerness","email":"mdegerness@nvidia.com","date":"2023-02-22 20:14:05.000000000","tz":-480},"committer":{"name":"Mandell Degerness","email":"mdegerness@nvidia.com","date":"2023-04-20 22:21:58.000000000","tz":-420},"subject":"Add per account grace period to object expirer","message":"Add per account grace period to object expirer\n\nChange-Id: I106103438c4162a561486ac73a09436e998ae1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1b8c4b68fb58efded42166b866c64b60b3dd9ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1b8c4b68fb58efded42166b866c64b60b3dd9ea"}]},"branch":"refs/heads/master"},"e22691b7aa2d5036280a58ddc9d4fefe64c4b273":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2023-10-19 21:29:52.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/06/874806/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/874806/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/874806/10"}}},"commit":{"parents":[{"commit":"195f54ecb5c7520b7ae19157be77047a0f4e68ea","subject":"Merge \"Add a swift-reload command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/195f54ecb5c7520b7ae19157be77047a0f4e68ea"}]}],"author":{"name":"Mandell Degerness","email":"mdegerness@nvidia.com","date":"2023-02-22 20:14:05.000000000","tz":-480},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-10-19 21:29:52.000000000","tz":0},"subject":"Add per account grace period to object expirer","message":"Add per account grace period to object expirer\n\nChange-Id: I106103438c4162a561486ac73a09436e998ae1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e22691b7aa2d5036280a58ddc9d4fefe64c4b273"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e22691b7aa2d5036280a58ddc9d4fefe64c4b273"}]},"branch":"refs/heads/master","description":"Rebase"},"a342856a1c5bf85679a9ae32c6e5ddce8f94f4ae":{"kind":"REWORK","_number":11,"created":"2024-03-05 19:24:20.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"ref":"refs/changes/06/874806/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/874806/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/874806/11"}}},"commit":{"parents":[{"commit":"195f54ecb5c7520b7ae19157be77047a0f4e68ea","subject":"Merge \"Add a swift-reload command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/195f54ecb5c7520b7ae19157be77047a0f4e68ea"}]}],"author":{"name":"Mandell Degerness","email":"mdegerness@nvidia.com","date":"2023-02-22 20:14:05.000000000","tz":-480},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-05 19:21:11.000000000","tz":-300},"subject":"expirer: per-account and per-container grace period","message":"expirer: per-account and per-container grace period\n\nAdd config parsing to register account and container level grace\nperiods.\n\nAdd expirer functionality to expire accounts and containers after\ngrace period.\n\nAdd config sample for grace period config keys and values.\n\nAdd tests for bad config.\n\nAdd documentation for account and container level grace periods.\n\nChange-Id: I106103438c4162a561486ac73a09436e998ae1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a342856a1c5bf85679a9ae32c6e5ddce8f94f4ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a342856a1c5bf85679a9ae32c6e5ddce8f94f4ae"}]},"branch":"refs/heads/master"},"0e32dd691ce00bad6dbdfc769716fc6e69c08f4b":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":12,"created":"2024-03-06 18:21:59.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"ref":"refs/changes/06/874806/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/874806/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/874806/12"}}},"commit":{"parents":[{"commit":"627448362abc1f83d9385165de232f47545f18e7","subject":"Merge \"CI: Remove centos-7 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/627448362abc1f83d9385165de232f47545f18e7"}]}],"author":{"name":"Mandell Degerness","email":"mdegerness@nvidia.com","date":"2023-02-22 20:14:05.000000000","tz":-480},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-06 18:16:05.000000000","tz":-300},"subject":"expirer: per account and container grace period","message":"expirer: per account and container grace period\n\nAdd config parsing to register account and container level grace\nperiods.\n\nAdd expirer functionality to expire accounts and containers after\ngrace period.\n\nAdd config sample for grace period config keys and values.\n\nAdd tests for bad config.\n\nAdd documentation for account and container level grace periods.\n\nChange-Id: I106103438c4162a561486ac73a09436e998ae1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e32dd691ce00bad6dbdfc769716fc6e69c08f4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e32dd691ce00bad6dbdfc769716fc6e69c08f4b"}]},"branch":"refs/heads/master"},"bd43b32832fa9de427fe8165cc60958a61eccb27":{"kind":"REWORK","_number":13,"created":"2024-03-07 22:56:15.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/06/874806/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/874806/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/874806/13"}}},"commit":{"parents":[{"commit":"627448362abc1f83d9385165de232f47545f18e7","subject":"Merge \"CI: Remove centos-7 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/627448362abc1f83d9385165de232f47545f18e7"}]}],"author":{"name":"Mandell Degerness","email":"mdegerness@nvidia.com","date":"2023-02-22 20:14:05.000000000","tz":-480},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2024-03-07 22:56:07.000000000","tz":-480},"subject":"expirer: per account and container grace period","message":"expirer: per account and container grace period\n\nAdd config parsing to register account and container level grace\nperiods.\n\nAdd expirer functionality to expire accounts and containers after\ngrace period.\n\nAdd config sample for grace period config keys and values.\n\nAdd tests for bad config.\n\nAdd documentation for account and container level grace periods.\n\nChange-Id: I106103438c4162a561486ac73a09436e998ae1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd43b32832fa9de427fe8165cc60958a61eccb27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd43b32832fa9de427fe8165cc60958a61eccb27"}]},"branch":"refs/heads/master"},"5f753b64956a8bb7292ae034f44c674a9872fb99":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2024-03-07 22:59:04.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/06/874806/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/874806/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/874806/14"}}},"commit":{"parents":[{"commit":"627448362abc1f83d9385165de232f47545f18e7","subject":"Merge \"CI: Remove centos-7 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/627448362abc1f83d9385165de232f47545f18e7"}]}],"author":{"name":"Mandell Degerness","email":"mdegerness@nvidia.com","date":"2023-02-22 20:14:05.000000000","tz":-480},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2024-03-07 22:56:47.000000000","tz":-480},"subject":"grace period and x-open-expired for expired objects","message":"grace period and x-open-expired for expired objects\n\nAdd config parsing to register account and container level grace\nperiods. Add expirer functionality to expire accounts and containers\nafter grace period. Add config sample for grace period config keys\nand values. Add tests for bad config. Add documentation for account\nand container level grace periods.\n\nIf the global configuration option \u0027enable_open_expired\u0027 is set\nto true in the config, then the client will be able to make a\nrequest with the header \u0027x-open-expired\u0027 set to true in order\nto retrieve an object that has expired (it is in its grace period).\nIf this config flag is set to false, the client will not be able\nto access any expired objects, even with the header, which is the\ndefault behavior unless the flag is set.\n\nWhen a client sets a \u0027x-open-expired\u0027 header to a true value for a\nGET/HEAD/POST request the proxy will forward x-backend-open-expired to\nstorage server. The storage server will allow clients that set\nx-backend-open-expired to open and read an object even after the\nx-delete-at time as passed\n\nChange-Id: I106103438c4162a561486ac73a09436e998ae1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f753b64956a8bb7292ae034f44c674a9872fb99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f753b64956a8bb7292ae034f44c674a9872fb99"}]},"branch":"refs/heads/master"},"a70d230f7cdf7a4c202d727e11454b798d2fda9c":{"kind":"REWORK","_number":15,"created":"2024-03-08 16:58:37.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/06/874806/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/874806/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/874806/15"}}},"commit":{"parents":[{"commit":"627448362abc1f83d9385165de232f47545f18e7","subject":"Merge \"CI: Remove centos-7 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/627448362abc1f83d9385165de232f47545f18e7"}]}],"author":{"name":"Mandell Degerness","email":"mdegerness@nvidia.com","date":"2023-02-22 20:14:05.000000000","tz":-480},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-03-08 16:55:02.000000000","tz":-360},"subject":"expirer: per account and container grace period","message":"expirer: per account and container grace period\n\nAdd config parsing to register account and container level grace\nperiods.\n\nAdd expirer functionality to expire accounts and containers after\ngrace period.\n\nAdd config sample for grace period config keys and values.\n\nAdd tests for bad config.\n\nAdd documentation for account and container level grace periods.\n\nChange-Id: I106103438c4162a561486ac73a09436e998ae1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a70d230f7cdf7a4c202d727e11454b798d2fda9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a70d230f7cdf7a4c202d727e11454b798d2fda9c"}]},"branch":"refs/heads/master"},"8e7d26f249249f324bf9953b577cd47a1bddf85c":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2024-03-13 21:35:47.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/06/874806/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/874806/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/874806/16"}}},"commit":{"parents":[{"commit":"b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7","subject":"Merge \"s3api test for zero byte mpu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7"}]}],"author":{"name":"Mandell Degerness","email":"mdegerness@nvidia.com","date":"2023-02-22 20:14:05.000000000","tz":-480},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-03-13 21:35:47.000000000","tz":0},"subject":"expirer: per account and container grace period","message":"expirer: per account and container grace period\n\nAdd config parsing to register account and container level grace\nperiods.\n\nAdd expirer functionality to expire accounts and containers after\ngrace period.\n\nAdd config sample for grace period config keys and values.\n\nAdd tests for bad config.\n\nAdd documentation for account and container level grace periods.\n\nChange-Id: I106103438c4162a561486ac73a09436e998ae1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e7d26f249249f324bf9953b577cd47a1bddf85c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e7d26f249249f324bf9953b577cd47a1bddf85c"}]},"branch":"refs/heads/master","description":"Rebase"},"0c534050516b5923cf80aefdc287f8f1ebac2243":{"kind":"REWORK","_number":17,"created":"2024-03-15 01:38:24.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"ref":"refs/changes/06/874806/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/874806/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/874806/17"}}},"commit":{"parents":[{"commit":"b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7","subject":"Merge \"s3api test for zero byte mpu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7"}]}],"author":{"name":"Mandell Degerness","email":"mdegerness@nvidia.com","date":"2023-02-22 20:14:05.000000000","tz":-480},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-15 01:38:14.000000000","tz":-240},"subject":"expirer: per account and container grace period","message":"expirer: per account and container grace period\n\nAdd config parsing to register account and container level grace\nperiods.\n\nAdd expirer functionality to expire accounts and containers after\ngrace period.\n\nAdd config sample for grace period config keys and values.\n\nAdd tests for bad config.\n\nAdd documentation for account and container level grace periods.\n\nCo-Authored-By: Anish Kachinthaya \u003cakachinthaya@nvidia.com\u003e\nChange-Id: I106103438c4162a561486ac73a09436e998ae1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c534050516b5923cf80aefdc287f8f1ebac2243"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c534050516b5923cf80aefdc287f8f1ebac2243"}]},"branch":"refs/heads/master"},"7bffb150fd2281e125cdb5ac99a627aa035a5ead":{"kind":"REWORK","_number":18,"created":"2024-03-20 22:18:08.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"ref":"refs/changes/06/874806/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/874806/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/874806/18"}}},"commit":{"parents":[{"commit":"b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7","subject":"Merge \"s3api test for zero byte mpu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7"}]}],"author":{"name":"Mandell Degerness","email":"mdegerness@nvidia.com","date":"2023-02-22 20:14:05.000000000","tz":-480},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-20 22:17:56.000000000","tz":-240},"subject":"expirer: per account and container grace period","message":"expirer: per account and container grace period\n\nThe object expirer can be configured to allow access to objects\nin specified accounts and containers even after their expiration\ntime using per-account and per-container grace periods. The grace\nperiod of accounts and containers in seconds is configured in\nthe object server config. The object expirer references these\nconfigured values to only reap objects from specified accounts and\ncontainers after their corresponding grace periods.\n\nThe goal of the grace period feature is to prevent accidental or\npremature data loss if an object marked for deletion with the\n\u0027x-delete-at\u0027 feature should not be deleted immediately, for\nwhatever reason.\n\nConfiguring the grace period in a granular per-account and\nper-container manner is beneficial for being able to keep\nstorage capacity consumption in control while maintaining a\ndesired data recovery window.\n\nThis patch also adds a sample configuration, documentation, and\ntests for bad configurations and grace period functionality.\n\nCo-Authored-By: Anish Kachinthaya \u003cakachinthaya@nvidia.com\u003e\nChange-Id: I106103438c4162a561486ac73a09436e998ae1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7bffb150fd2281e125cdb5ac99a627aa035a5ead"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7bffb150fd2281e125cdb5ac99a627aa035a5ead"}]},"branch":"refs/heads/master"},"d51168cc493a7326f5af30e15550efa272dd571e":{"kind":"REWORK","_number":19,"created":"2024-04-08 16:37:31.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"ref":"refs/changes/06/874806/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/874806/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/874806/19"}}},"commit":{"parents":[{"commit":"b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7","subject":"Merge \"s3api test for zero byte mpu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7"}]}],"author":{"name":"Mandell Degerness","email":"mdegerness@nvidia.com","date":"2023-02-22 20:14:05.000000000","tz":-480},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-08 16:36:34.000000000","tz":-240},"subject":"expirer: account and container level delay_reaping","message":"expirer: account and container level delay_reaping\n\nThe object expirer can be configured to allow access to objects\nin specified accounts and containers even after their expiration\ntime using account and container level delay_reaping values. The\ndelay_reaping value of accounts and containers in seconds is\nconfigured in the object server config. The object expirer\nreferences these configured values to only reap objects from\nspecified accounts and containers after their corresponding delays.\n\nThe goal of the delay_reaping feature is to prevent accidental or\npremature data loss if an object marked for deletion with the\n\u0027x-delete-at\u0027 feature should not be reaped immediately, for\nwhatever reason.\n\nConfiguring the delay_reaping value at a granular account and\ncontainer level is beneficial for being able to keep storage\ncapacity consumption in control while maintaining a desired\ndata recovery window.\n\nThis patch also adds a sample configuration, documentation, and\ntests for bad configurations and grace period functionality.\n\nCo-Authored-By: Anish Kachinthaya \u003cakachinthaya@nvidia.com\u003e\nChange-Id: I106103438c4162a561486ac73a09436e998ae1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d51168cc493a7326f5af30e15550efa272dd571e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d51168cc493a7326f5af30e15550efa272dd571e"}]},"branch":"refs/heads/master"},"d875bee2f4458f4e47652c147335c6611566a684":{"kind":"REWORK","_number":20,"created":"2024-04-17 20:04:07.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"ref":"refs/changes/06/874806/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/874806/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/874806/20"}}},"commit":{"parents":[{"commit":"b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7","subject":"Merge \"s3api test for zero byte mpu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7"}]}],"author":{"name":"Mandell Degerness","email":"mdegerness@nvidia.com","date":"2023-02-22 20:14:05.000000000","tz":-480},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-17 20:03:58.000000000","tz":-240},"subject":"expirer: account and container level delay_reaping","message":"expirer: account and container level delay_reaping\n\nThe object expirer can be configured to delay the reaping of\nobjects from disk after their expiration time using account\nand container level delay_reaping values. The delay_reaping\nvalue of accounts and containers in seconds is configured in\nthe object server config. The object expirer references these\nconfigured values to only reap objects from specified accounts\nand containers after their corresponding delays.\n\nThe goal of the delay_reaping feature is to prevent accidental or\npremature data loss if an object marked for deletion with the\n\u0027x-delete-at\u0027 feature should not be reaped immediately, for\nwhatever reason.\n\nConfiguring the delay_reaping value at a granular account and\ncontainer level is beneficial for being able to keep storage\ncapacity consumption in control while maintaining a desired\ndata recovery window.\n\nThis patch also adds a sample configuration, documentation, and\ntests for bad configurations and grace period functionality.\n\nCo-Authored-By: Anish Kachinthaya \u003cakachinthaya@nvidia.com\u003e\nChange-Id: I106103438c4162a561486ac73a09436e998ae1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d875bee2f4458f4e47652c147335c6611566a684"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d875bee2f4458f4e47652c147335c6611566a684"}]},"branch":"refs/heads/master"},"6f7afa83927ea9fb4d06a255112616c5bf9e51df":{"kind":"REWORK","_number":21,"created":"2024-04-18 16:10:56.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"ref":"refs/changes/06/874806/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/874806/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/874806/21"}}},"commit":{"parents":[{"commit":"b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7","subject":"Merge \"s3api test for zero byte mpu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7"}]}],"author":{"name":"Mandell Degerness","email":"mdegerness@nvidia.com","date":"2023-02-22 20:14:05.000000000","tz":-480},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-18 16:10:26.000000000","tz":0},"subject":"expirer: account and container level delay_reaping","message":"expirer: account and container level delay_reaping\n\nThe object expirer can be configured to delay the reaping of\nobjects from disk after their expiration time using account\nand container level delay_reaping values. The delay_reaping\nvalue of accounts and containers in seconds is configured in\nthe object server config. The object expirer references these\nconfigured values to only reap objects from specified accounts\nand containers after their corresponding delays.\n\nThe goal of the delay_reaping feature is to prevent accidental or\npremature data loss if an object marked for deletion with the\n\u0027x-delete-at\u0027 feature should not be reaped immediately, for\nwhatever reason.\n\nConfiguring the delay_reaping value at a granular account and\ncontainer level is beneficial for being able to keep storage\ncapacity consumption in control while maintaining a desired\ndata recovery window.\n\nThis patch also adds a sample configuration, documentation, and\ntests for bad configurations and grace period functionality.\n\nCo-Authored-By: Anish Kachinthaya \u003cakachinthaya@nvidia.com\u003e\nChange-Id: I106103438c4162a561486ac73a09436e998ae1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f7afa83927ea9fb4d06a255112616c5bf9e51df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f7afa83927ea9fb4d06a255112616c5bf9e51df"}]},"branch":"refs/heads/master"},"1c8b1e86343c1c4caf9d75d9359b2126b4b37788":{"kind":"REWORK","_number":22,"created":"2024-04-24 21:08:27.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"ref":"refs/changes/06/874806/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/874806/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/874806/22"}}},"commit":{"parents":[{"commit":"b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7","subject":"Merge \"s3api test for zero byte mpu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7"}]}],"author":{"name":"Mandell Degerness","email":"mdegerness@nvidia.com","date":"2023-02-22 20:14:05.000000000","tz":-480},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-24 21:07:24.000000000","tz":-420},"subject":"expirer: account and container level delay_reaping","message":"expirer: account and container level delay_reaping\n\nThe object expirer can be configured to delay the reaping of\nobjects from disk after their expiration time using account\nand container level delay_reaping values. The delay_reaping\nvalue of accounts and containers in seconds is configured in\nthe object server config. The object expirer references these\nconfigured values to only reap objects from specified accounts\nand containers after their corresponding delays.\n\nThe goal of the delay_reaping feature is to prevent accidental or\npremature data loss if an object marked for deletion with the\n\u0027x-delete-at\u0027 feature should not be reaped immediately, for\nwhatever reason.\n\nConfiguring the delay_reaping value at a granular account and\ncontainer level is beneficial for being able to keep storage\ncapacity consumption in control while maintaining a desired\ndata recovery window.\n\nThis patch also adds a sample configuration, documentation, and\ntests for bad configurations and grace period functionality.\n\nCo-Authored-By: Anish Kachinthaya \u003cakachinthaya@nvidia.com\u003e\nChange-Id: I106103438c4162a561486ac73a09436e998ae1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1c8b1e86343c1c4caf9d75d9359b2126b4b37788"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1c8b1e86343c1c4caf9d75d9359b2126b4b37788"}]},"branch":"refs/heads/master"},"b0d21d3a5d40fa98448d45d9e2bf9a087b3abc7b":{"kind":"REWORK","_number":23,"created":"2024-04-25 18:05:46.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"ref":"refs/changes/06/874806/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/874806/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/874806/23"}}},"commit":{"parents":[{"commit":"b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7","subject":"Merge \"s3api test for zero byte mpu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7"}]}],"author":{"name":"Mandell Degerness","email":"mdegerness@nvidia.com","date":"2023-02-22 20:14:05.000000000","tz":-480},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-25 18:05:32.000000000","tz":-420},"subject":"expirer: account and container level delay_reaping","message":"expirer: account and container level delay_reaping\n\nThe object expirer can be configured to delay the reaping of\nobjects from disk after their expiration time using account\nand container level delay_reaping values. The delay_reaping\nvalue of accounts and containers in seconds is configured in\nthe object server config. The object expirer references these\nconfigured values to only reap objects from specified accounts\nand containers after their corresponding delays.\n\nThe goal of the delay_reaping feature is to prevent accidental or\npremature data loss if an object marked for deletion with the\n\u0027x-delete-at\u0027 feature should not be reaped immediately, for\nwhatever reason.\n\nConfiguring the delay_reaping value at a granular account and\ncontainer level is beneficial for being able to keep storage\ncapacity consumption in control while maintaining a desired\ndata recovery window.\n\nThis patch also adds a sample configuration, documentation, and\ntests for bad configurations and grace period functionality.\n\nCo-Authored-By: Anish Kachinthaya \u003cakachinthaya@nvidia.com\u003e\nChange-Id: I106103438c4162a561486ac73a09436e998ae1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b0d21d3a5d40fa98448d45d9e2bf9a087b3abc7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b0d21d3a5d40fa98448d45d9e2bf9a087b3abc7b"}]},"branch":"refs/heads/master"},"5961ba0ca7ddcfe296627fbef01ec4b32d538290":{"kind":"REWORK","_number":24,"created":"2024-04-25 20:59:43.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"ref":"refs/changes/06/874806/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/874806/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/874806/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/874806/24"}}},"commit":{"parents":[{"commit":"b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7","subject":"Merge \"s3api test for zero byte mpu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7"}]}],"author":{"name":"Mandell Degerness","email":"mdegerness@nvidia.com","date":"2023-02-22 20:14:05.000000000","tz":-480},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-25 20:59:36.000000000","tz":-420},"subject":"expirer: account and container level delay_reaping","message":"expirer: account and container level delay_reaping\n\nThe object expirer can be configured to delay the reaping of\nobjects from disk after their expiration time using account\nand container level delay_reaping values. The delay_reaping\nvalue of accounts and containers in seconds is configured in\nthe object server config. The object expirer references these\nconfigured values to only reap objects from specified accounts\nand containers after their corresponding delays.\n\nThe goal of the delay_reaping feature is to prevent accidental or\npremature data loss if an object marked for deletion with the\n\u0027x-delete-at\u0027 feature should not be reaped immediately, for\nwhatever reason.\n\nConfiguring the delay_reaping value at a granular account and\ncontainer level is beneficial for being able to keep storage\ncapacity consumption in control while maintaining a desired\ndata recovery window.\n\nThis patch also adds a sample configuration, documentation, and\ntests for bad configurations and grace period functionality.\n\nCo-Authored-By: Anish Kachinthaya \u003cakachinthaya@nvidia.com\u003e\nChange-Id: I106103438c4162a561486ac73a09436e998ae1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5961ba0ca7ddcfe296627fbef01ec4b32d538290"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5961ba0ca7ddcfe296627fbef01ec4b32d538290"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
