)]}'
{"id":"openstack%2Fswift~875220","triplet_id":"openstack%2Fswift~master~I7d2fe064175f002055054a72f348b87dc396772b","project":"openstack/swift","branch":"master","topic":"p-sharder-uniform-log","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-03-02 08:48:25.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-03-02 08:48:25.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-03-01 23:10:34.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I7d2fe064175f002055054a72f348b87dc396772b","subject":"sharder: show path and db file in logs","status":"MERGED","created":"2023-02-24 16:23:44.000000000","updated":"2025-08-08 20:05:45.000000000","submitted":"2023-03-02 08:48:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"875220-p-sharder-uniform-log","meta_rev_id":"50b042b8b1d2e6753dc4f7ef73754efdcf3acd22","_number":875220,"virtual_id_number":875220,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-02 08:48:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-01 23:10:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-01 23:10:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-24 17:22:09.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2023-02-24 17:27:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-24 20:20:24.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2023-02-28 21:41:04.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2023-03-01 23:10:34.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2023-03-02 05:12:17.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"},{"updated":"2025-08-08 20:05:45.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"}],"messages":[{"id":"0ede3280147ebc887baa425f2eb183a12fc9c00a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-24 16:23:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dee8ebc0235c275d5087941f43fd50b74522c1e5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-24 17:22:09.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"02c6454beb94717a3a86e55995d7143d3ea7efda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 17:27:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d29c374f7b84a499a34d85275ce9caf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2beb0a32e08b4977b2b0ee91610b3961 : SUCCESS in 8m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/8e08674e45234b4a84ca007d034ce1f9 : SUCCESS in 42m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/95423739fb9b4051a844cf8f1019b3b5 : SUCCESS in 40m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e1fe3caa2b44976b7cb05ab48f14e90 : SUCCESS in 52m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ffef70864f8b48578200f7406bc4838a : SUCCESS in 23m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7033394e0153489ab49befb99b5d4527 : SUCCESS in 18m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/35d54810a24747b3891203136cbb3c5b : SUCCESS in 19m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d7b70a82dea84468867512065f236c5b : SUCCESS in 8m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e7130e2be4f5404dbf676382e12d409b : SUCCESS in 4m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/43fbc72e6f2b45ee99796f780803b22f : SUCCESS in 16m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/537f12801a1e4f84a542ce56f3d5deac : SUCCESS in 15m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ee3bf83f9170432cb5abf91f42c6ad92 : SUCCESS in 14m 00s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6be8655cf4fe48c5a12bd23478ae7a39 : SUCCESS in 15m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/61936dd408344fc089a04d4dc3f0726f : SUCCESS in 18m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/22a9d5c199374502aa7563c75743b48e : SUCCESS in 17m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e982bcab7d2d4644aaf7ee8d335e72fe : SUCCESS in 18m 24s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2242453eb41a472289ea9fe289818a94 : SUCCESS in 13m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7f117bf5b84243c8954caf192049af21 : SUCCESS in 19m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d24a0b80ff364f98a31ab6fb05692860 : SUCCESS in 14m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a2021cb129a34268ae1d1011c42ed085 : SUCCESS in 6m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0a882235098546be83b99f4970f70716 : SUCCESS in 14m 03s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/73b85d78bd4546dfa5d72aa9befd381d : SUCCESS in 8m 14s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5d7f5870d1ed4cb7a2a2313ce43e128d : FAILURE in 49m 01s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c83358d4c3f1455d8befdfa0480778c5 : FAILURE in 56m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a81048c980884308a516b362aaf00b0f : SUCCESS in 36m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4d8596a3810647aabee8c84ada482561 : SUCCESS in 35m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d25665df37a478fb3f86045394dcee0 : SUCCESS in 17m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7df0106667ca42eea5f4a3f1837072f7 : SUCCESS in 4m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6e47ac6c86514c42a019f8b829c6fee0 : SUCCESS in 30m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/92b8b824e11d496f83f624ace9b6b097 : SUCCESS in 50m 59s","accounts_in_message":[],"_revision_number":1},{"id":"46d183d4206072616f5361d416098ca4a96bb635","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 17:57:26.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa51cd97ba5341c9b24b789f6bc2c2c7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6abcb85217ec4fa0b0ff28cbb84867de : SUCCESS in 14m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b63e95e97a614701b635df79fe9b2055 : SUCCESS in 15m 25s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/440d73e11cbb49228aaed75ed88fc68b : SUCCESS in 19m 10s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/54239e85c3844d5db2af6f7e589d5381 : TIMED_OUT in 1h 31m 00s","accounts_in_message":[],"_revision_number":1},{"id":"a427c6a1e16cd9582d7ef471816c91fcef8f4cbe","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-24 18:33:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c2bc1cbcebd91d22c321ddabd37959af1ae36cee","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-02-24 20:20:24.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f7aa5b6de260a59913b4863400eb200e66dd22a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-27 14:39:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"00e101e897ba1f22b40af7c49603808cd00e8729","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 14:58:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96b50112b5174802aeac3480bbe8ca3c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e35dd6020df48a1bcee11f293f79e3f : SUCCESS in 14m 46s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/489f90ad298343d4b169afe07b844890 : SUCCESS in 16m 26s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5a633eb5a37348708ec382c107134b27 : SUCCESS in 18m 00s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c81fe41960d47a984ab8fc23736aed7 : SUCCESS in 18m 04s","accounts_in_message":[],"_revision_number":2},{"id":"a4212dc8e613ff93f1d76c1e78dbf432f2eba5f0","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-27 15:25:45.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"de4811770753f8b11ab7e269ac2193d7ab130236","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 15:59:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78d07f5f45c34b7c9101cd0d853bf6f5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e68fd44680341e68dd97d0190671a80 : SUCCESS in 9m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/d32c7e1e04af433cadd076f0da7d3222 : SUCCESS in 1h 03m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b5c22b201d60407c83c19d7a13f6cfaa : SUCCESS in 36m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef6a3ea05b0c457d962d44810292da67 : SUCCESS in 45m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ef9c88575e84408082b40ca606f75a3d : SUCCESS in 16m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/07caa96bb7734e4a8b6ee9aa85fa1fcb : SUCCESS in 18m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fffb2e0300e24f0ca06ee6cb3fbfae28 : SUCCESS in 17m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/80bf85bb22c8425c8a6fb3d0024c8bd0 : SUCCESS in 8m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/08ac10668da14181b72c1a43a93b04f4 : SUCCESS in 5m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7c47214ddfed4ef28b9bf04f8984149d : SUCCESS in 19m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f95399726f294c9db55dd78db9623c6a : SUCCESS in 15m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/510d76ecdeb54f4dacb211aafb3a92ab : SUCCESS in 14m 50s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4c45f2b910f944c3aa9569d14b045c9b : SUCCESS in 14m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f13d4d3f387340639f9cbd7ba2ed09c8 : SUCCESS in 15m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/050ed35e8ade49d48a0f293790178042 : SUCCESS in 17m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/49111253067b45eeb89cdc3cbb06e68b : SUCCESS in 15m 44s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/edb322ad963d462d9f9f94c2fbbd77e0 : SUCCESS in 13m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3b7e4d20a1a64a1aa9183e5841e5c4cc : SUCCESS in 18m 11s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b2f2275fbc85413895dadc65137e1b51 : SUCCESS in 14m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0b17ec9659934e7d8e7640ed44de9c4d : SUCCESS in 6m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ea65f9001b9b45949483dd9af841a537 : SUCCESS in 15m 16s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f624639864c1473791e756dd6dfa0599 : SUCCESS in 7m 57s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1ce76b1824354ae6ad52728fc83cc69a : SUCCESS in 54m 43s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/297dffbc539945e0811afd04ac751f83 : SUCCESS in 1h 01m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c5774e82e87f459a9036769dfe6dc5d5 : SUCCESS in 41m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c2a1c88f6f7a4437b19a0f28cd0b01dc : SUCCESS in 49m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f1cd8e475d084466b02ef2a9b9e64dc5 : SUCCESS in 14m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9c71058b6e745e4950e91b75e470415 : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c137e9b8e9294cfbbe55b9d2a939d7fc : SUCCESS in 21m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b096836e732346e79b913935f5325874 : SUCCESS in 57m 27s","accounts_in_message":[],"_revision_number":2},{"id":"e28bf1ce9bcb1fb2885515d205ff6060eb91d734","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-27 17:03:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"22d9adddcda74bfaa4f2db9b7daf598a1234275c","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-02-28 21:41:04.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b8ea63fa8f63e305ff4474f68ae7269a23bb7fd7","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-03-01 13:10:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ce6248ee959036ef2b462ce619de021c937f14a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-03-01 15:02:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"500ae116b1b9c7b630deee82c34637f5c021e55f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 16:36:54.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a8fbb39d0814ab2bcf6630c240fd15d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/55f11b212274457aaeada869d6edf213 : TIMED_OUT in 1h 30m 57s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/727d74d854e54ec08769c4cdbfff3600 : TIMED_OUT in 1h 31m 06s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/83f999c848344d4185c582e3a378ae51 : SUCCESS in 1h 27m 05s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4fce5a9c2b9444a0a5e8975b6a2d0c0a : TIMED_OUT in 1h 30m 55s","accounts_in_message":[],"_revision_number":4},{"id":"16e5d37416ca14b274be5368c4adacc62d1ed762","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 18:32:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/549e762e8b0f459b96153c296b8071a2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dbdc45072ef4790a2e2ca2d8a1624bb : SUCCESS in 9m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/90ec646b60a843a0a0d69d35570aeb92 : SUCCESS in 1h 03m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f451156259d74422a07e963fabd9c2ae : SUCCESS in 46m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bce71f5b2de94df69beea4bfe53ef8ce : SUCCESS in 59m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cec0737340e54f9ca787c9a8e5fd12c7 : SUCCESS in 18m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/38b52e6544184f0c89ffef1ffd3a920c : SUCCESS in 19m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8880e50cc82c4c6296a3e33ce3d9df45 : SUCCESS in 17m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c5bc5238e3964ee7a2b9e3b91588db91 : SUCCESS in 9m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b9b952ea15bd4c84a5378f15d09680c8 : SUCCESS in 5m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/33dd7d874a6e45079314401f4703aa5f : SUCCESS in 14m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d631b939aed842dca6f4199a70ac6c85 : SUCCESS in 18m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/33cbcf76b9b64255831a825399ddf822 : SUCCESS in 15m 07s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0bca8208cda441f5b386248dc0434b9f : SUCCESS in 15m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c0b0a562dd8340eb8b8b5a3e333bd01f : SUCCESS in 27m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5c4c21630c484505a621a4ebcc7f3083 : SUCCESS in 58m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5b1f0163dc0c449e884ab2a09b0969b7 : SUCCESS in 57m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/741790d313c3438aae05e8b981e33770 : SUCCESS in 13m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/de1df09c241c46a2a4c135f9063389af : SUCCESS in 37m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e51df77f41f541dc87d8d61e38d491e5 : SUCCESS in 49m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d2274001a07746d1b7207db3da2b23eb : SUCCESS in 7m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0ba41653f9c64b4e8121adc7746158d9 : SUCCESS in 55m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/908cd940bc43480893f33e4a4baa8d58 : SUCCESS in 6m 28s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/133c893ead1b496191539eecb2b7ea2b : SUCCESS in 1h 01m 27s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c16f9c0a0c2c4593a450c47521f3427b : SUCCESS in 58m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/db4e27fc6cbd4e0bad2e1772e3522bd1 : SUCCESS in 57m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0453e06601f64fb7b4b910f9ac52669d : SUCCESS in 53m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9488a1814e8c49c1ba4fdd3f7e7dd88d : SUCCESS in 17m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8f8cc2c2fc5438f8469757e9a572b83 : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6f7582f0170e4ceab589e0dcd5e32e68 : SUCCESS in 21m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/592270a5985f476485e83a0833e45381 : SUCCESS in 53m 08s","accounts_in_message":[],"_revision_number":4},{"id":"967b310b6693c3dee36948af56ff682628de59ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-03-01 21:53:29.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"dd47a3a3f3c93202b64a9c722cbc4db5aa44aea9","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-03-01 21:55:52.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"da9b2abc61014281cf123dcbfe9966afe93c291a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-03-01 23:10:34.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"dfff5664e8ca2b3352cf12d06f3e8c4c456c3e58","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 23:11:57.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e0c4f491aaf4e148fdd37b613d91f9a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/970915546a654512bccf721295cd1ade : SUCCESS in 1h 16m 26s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ebb6e29d46cb42a0935cbb9c45bd6e51 : SUCCESS in 18m 08s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/99d6df3d3a6d408da51690af2ea25bc3 : SUCCESS in 1h 09m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a5ce1ee4042b463788b834d2fa29e0b8 : SUCCESS in 18m 14s","accounts_in_message":[],"_revision_number":5},{"id":"c36bc76f590c87d6d20c9ea8cf712f7f1c990b50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 23:28:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ad19ff9bf6943bfbff1ea17b31c4663\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57423500197b4afd873c53c220ffae9b : SUCCESS in 9m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/99368f2422224ab7991d13ebb6f45bdd : SUCCESS in 49m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/75135ba9c0e34a11a7d1ad5b7faf9336 : SUCCESS in 36m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9884d340b3db480d9ed3633caf85c1d7 : SUCCESS in 1h 15m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fdac7f23ffad43c598c9fc079656b7a2 : SUCCESS in 20m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cc154df7aa784edd8350d6f50b106b16 : SUCCESS in 19m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3b6d66ae065a49f2b41c479eb7147bf8 : SUCCESS in 16m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/03b7668bc24647d6b5ba28b0fae5e48b : SUCCESS in 8m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3fd97f5afc094056b46ae918572e718a : SUCCESS in 4m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/26a1e620dc7544d9b805883dd6ed9091 : SUCCESS in 17m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/957491a0af314e908394509b674912ca : SUCCESS in 15m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8ecbcdf0e36e45e299911b773f7c8133 : SUCCESS in 15m 11s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1653c512004b4375ad9d23915e351865 : SUCCESS in 16m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6084bc846bad4abfbdbfec23b62ca36d : SUCCESS in 14m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6a2520df5744428b8131401f7e20fa5c : SUCCESS in 17m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a4a53f52eb054f87b745bd1f0fd95d37 : SUCCESS in 14m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/47aee801e85e4402934a9229ef0f3672 : SUCCESS in 39m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/24ee17fda3c044a7957282ebe2500e4d : SUCCESS in 14m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3c012b767c144c339364d7e2cc91c337 : SUCCESS in 19m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ea52fa9200154c56a918faf24b32b55e : SUCCESS in 6m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4ec9ad3011eb430f902dcbfb35baa362 : SUCCESS in 19m 44s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/83b62d19e305467fbebea03d8c65923e : SUCCESS in 7m 38s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e9835dac4a804d90b573ca853d8602ee : SUCCESS in 56m 35s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/db0b318e458c411d98a6f32dd59a81c1 : SUCCESS in 1h 20m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/411aa8593aef4e188d2a645ded18dce8 : SUCCESS in 57m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c668a9d1bb1e44469452980aac6599b5 : SUCCESS in 43m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23212f845dc94259825f8c299a7e69bd : SUCCESS in 16m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4236e791599c4e6c80d2e7990ce52c93 : SUCCESS in 4m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d2121d65f48a481495b844b108806f4c : SUCCESS in 20m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fcfb06bb882d409da0e317c71ddb1513 : SUCCESS in 39m 49s","accounts_in_message":[],"_revision_number":5},{"id":"955a2100c9617a53d42c5d1afdcca28750202aab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 23:29:31.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"27af4849d70dafc7c862be7c834b81baddc1e4a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 03:19:33.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3049917b2b354d30954700f307fbcaa3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0a2f37abb794cb3bbf460feb99d2526 : SUCCESS in 8m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/29aeed863a2048988f116653ccc40924 : FAILURE in 33m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b145c9d880d846dbb7d1756d5615041f : SUCCESS in 33m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/32f8954f21454704a638ed0a3ebb3e13 : SUCCESS in 1h 02m 25s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/464a5749ecfb4978abb838aa643bfe69 : SUCCESS in 10m 40s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/3ce617968f714317b17fb4fa78ded38a : SUCCESS in 5m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1050c81b7e9543c0adab78912545cb40 : SUCCESS in 19m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/867f4fbf5a82421a87cb285cf640dc34 : SUCCESS in 18m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6da776a604bf4a6aaaefb98ee06beb31 : SUCCESS in 15m 13s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/79f6a5d892964507bb6748413ffb3baf : SUCCESS in 15m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ab6c2f81a70c4a86b655b8e55222fc9b : SUCCESS in 16m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/08a4c219fc684eecb89b947ca5b5f355 : SUCCESS in 17m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/93a6955541114029a3af3d8ebf932176 : SUCCESS in 18m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c799995d28984160b39db0e9efb63985 : SUCCESS in 14m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/db7441fe891b426aa7cc759feb8953c8 : SUCCESS in 18m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/22e41a91ccca48648343546e43c902fb : SUCCESS in 15m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/54d14001af6a442c98abac26d35b990f : SUCCESS in 10m 04s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/670a5c4eaeae4eefb8fced7e1806b15a : SUCCESS in 7m 54s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/29d7b5cf3d2a42ce8b4ecb95580f0351 : SUCCESS in 52m 27s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f2fa63273f6a46e79fb4a659e8e9e907 : SUCCESS in 1h 09m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/72b7e9d8efdf4631b0a72670343575c7 : SUCCESS in 53m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/717067bc44fe49719789832f59d5ca44 : SUCCESS in 53m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/095b546d6ab34050b612ed3f888a35d8 : SUCCESS in 17m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7465bf3c25f44b76aecca3c475113221 : SUCCESS in 4m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9271b40b513b4d4f858c2dbb40763b85 : SUCCESS in 48m 35s","accounts_in_message":[],"_revision_number":5},{"id":"11b030848b750f3c0d9e5c888eef8231b1aeb32c","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-03-02 05:12:17.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d6dc91396e2e5e6c55ae5c09e6484e4b19913ed4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-03-02 06:03:41.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"023a4a2fb16139f5291bf5f5c7736cdc7de61fa8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 06:24:10.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/474f7fee88574db792b100bfe275e600\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/11c7bd034ee2460a8eeb7c8be5f5ac41 : SUCCESS in 17m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1966203770d04552a4b4b377b66e61bb : SUCCESS in 15m 50s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/826c0ef0aac948ddbb27e6937e9ec96f : SUCCESS in 18m 48s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5f8137a63b67428f9c89421bf312b3e6 : SUCCESS in 18m 27s","accounts_in_message":[],"_revision_number":5},{"id":"65cb53edbe8acd170176f84ec33a2de2d17dbf31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 07:39:09.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59cc401e4da345268878e96f87e43ab9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bc74dc14f1c44cdb2bd30424cbad914 : SUCCESS in 7m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/18e5a84d976f4bc8970119ec9fc6686b : SUCCESS in 48m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6078a94b78874a95861291bb5f6948eb : SUCCESS in 46m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ab4cfd5fc864851bcc4b7e9078f3371 : SUCCESS in 43m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/55580d7a8c1b41d9937f44a558233cde : SUCCESS in 14m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d81bd94396bb498b9f5999ba831033c4 : SUCCESS in 17m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9eed395844c24191ba4a1fa11c601433 : SUCCESS in 17m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/32e085fb52d04d09b96674256c1650dc : SUCCESS in 8m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4ac69670084e47e6a294028ad6502b25 : SUCCESS in 6m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f8d58400f402442aa2c412dc3077977b : SUCCESS in 16m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b71d0ed97aab4337a9891107f783de3c : SUCCESS in 27m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3a39b8ef51fa4537b6a5bd0f42fbfc0b : SUCCESS in 26m 06s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c026e393a96f4c70b25a20a76b71be91 : SUCCESS in 12m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d9bce5d35f5c482ca0eb68feb26a8cff : SUCCESS in 16m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4a4ed5412e83490892a745387bb52f41 : SUCCESS in 26m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bf5911ed139146d989b806a291537b64 : SUCCESS in 19m 44s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cb5e4295c42e4dbba41be79e521639fb : SUCCESS in 17m 53s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0ed809a9412a456ca20f0f05c3fc9c1e : SUCCESS in 14m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e4abde56377d426085ae43bdb53ebfa7 : SUCCESS in 14m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e70573047a074f9cbe9f630e91548246 : SUCCESS in 7m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/edb7e80ea8bc4d1fbb4a6d0489f16ef0 : SUCCESS in 17m 18s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cc4e9a1d9880483c8d3d1dfcac7f7cbd : SUCCESS in 7m 47s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ba8d9d7ab73649b08efaa36526d4fd8e : SUCCESS in 1h 18m 01s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e1500beb80c446d28d36626264010593 : SUCCESS in 1h 23m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8dcfe2f1795a4a88a8f7cf20e33cdf91 : SUCCESS in 51m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e2e5914e17a24b5785d5ee9a60102418 : SUCCESS in 55m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f20bb6cce6004fe481499cc5c7fd9566 : SUCCESS in 18m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc4fe7de15a44d939b134ded5a766152 : SUCCESS in 5m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d12dc4c858c8457dbd345e27897cdba1 : SUCCESS in 19m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/02637730b5df4d44b0b62a063ebff09e : SUCCESS in 43m 35s","accounts_in_message":[],"_revision_number":5},{"id":"be22df8ea59624f52c0b391b1fd22aed9ad1467a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 07:39:41.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"08f0bfe0093a4bb97bc4feb6182c7240aff018ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 08:48:25.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad844bd2969e4d43b7398d9e3047be1e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6afe54a3045e4414af51bc107226f440 : SUCCESS in 8m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/91b44d1d424842baaae1d1f97171bc24 : SUCCESS in 1h 08m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c1b8ba879f424431852a0751617dbdfe : SUCCESS in 33m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e20e293e92b452bbfef112f56622a1c : SUCCESS in 47m 43s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/2c88d2bc38c242b1bada144101155363 : SUCCESS in 8m 21s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/6fed05b633024c059d986abc0d646b9d : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/074801a99a19454297ea0fefdebf6697 : SUCCESS in 15m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/19491f5ac3f446d2b7b8c1dfb46206cc : SUCCESS in 12m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d1f899ca54ca4e358bda4a2d2a84cb06 : SUCCESS in 38m 57s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4dc8b508b93547dfb34f6719a12c5e4e : SUCCESS in 13m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d3641c07785747c8adc2b264a5a1c6fe : SUCCESS in 16m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ad26c4106be14bc785b384660a10266b : SUCCESS in 16m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a7bba54a1fa84a47940ea0e3e0020560 : SUCCESS in 19m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c687b52c0da1473082c3cd9c754c4128 : SUCCESS in 15m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/250289327066490f9f0347f2875240be : SUCCESS in 13m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1b10a8ff4eb44e0a882db09f04f1e67b : SUCCESS in 15m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8288723499e946b5bc660897e711d38f : SUCCESS in 7m 00s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/240c1e8906af41e79ceacce01ce44efb : SUCCESS in 6m 39s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ea87ba3dbf2c4c8fbd546bc062828d70 : SUCCESS in 56m 02s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ac6304ee2a2b43dea5ff798cffa0d6e7 : SUCCESS in 59m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1a755b32a53d4143b14b4e99ffccc66d : SUCCESS in 53m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9c71740f6bad46388d651a667451a74d : SUCCESS in 49m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b83af3b8bed148ed9ce648f83be54fd1 : SUCCESS in 17m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c764d3266e645f2bcd5002f21e64b4d : SUCCESS in 4m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/53762f67d3f04b37a36b8340a824043b : SUCCESS in 42m 12s","accounts_in_message":[],"_revision_number":5},{"id":"59042459b82a158fe1b02f41a73c2f64134f0068","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 08:48:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"990b1cc48df7a47998181836ddd7607ec108b123","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 08:49:31.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45876b1da75a427baac5babae54e4964\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b76b3e1a156342998a4e65688f88f127 : SUCCESS in 40s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/2ab4ac55e9064e4b91a53ba810a36a6f : SUCCESS in 35s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/b01b4ffbc81b46c98195facff33c8c04 : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5e2029f84ce9133cb969f6389a17ab4bef6ad4c8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-03-02 17:08:49.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"8814cde6813f92abee5f2285789e86caaac781f3","revisions":{"9d875497750a1121e8fe9a956c86ece23b016ab7":{"kind":"REWORK","_number":1,"created":"2023-02-24 16:23:44.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/20/875220/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/20/875220/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/875220/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/875220/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/875220/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/20/875220/1"}}},"commit":{"parents":[{"commit":"3bf7cf60b9a812d73f430870bef1203be99df8dd","subject":"Fix docstring regarding private method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3bf7cf60b9a812d73f430870bef1203be99df8dd"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-02-24 12:20:19.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-02-24 16:09:45.000000000","tz":0},"subject":"sharder: show path and db file in info and debug logs","message":"sharder: show path and db file in info and debug logs\n\nMake all debug and info level logs that are associated with a\nContainerBroker include the path to the DB file and the namespace path\nto the container.\n\nChange-Id: I7d2fe064175f002055054a72f348b87dc396772b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d875497750a1121e8fe9a956c86ece23b016ab7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d875497750a1121e8fe9a956c86ece23b016ab7"}]},"branch":"refs/heads/master"},"7001a9278d4b5f013b3b8f2195966a5e6b48f95a":{"kind":"REWORK","_number":2,"created":"2023-02-27 14:39:08.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/20/875220/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/20/875220/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/875220/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/875220/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/875220/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/20/875220/2"}}},"commit":{"parents":[{"commit":"c0483c5b940d559b4a9efd2b5c64bf5b5528fa11","subject":"Merge \"sharder: make misplaced objects lookup faster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0483c5b940d559b4a9efd2b5c64bf5b5528fa11"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-02-24 12:20:19.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-02-27 14:38:43.000000000","tz":0},"subject":"sharder: show path and db file in info and debug logs","message":"sharder: show path and db file in info and debug logs\n\nMake all debug and info level logs that are associated with a\nContainerBroker include the path to the DB file and the namespace path\nto the container.\n\nChange-Id: I7d2fe064175f002055054a72f348b87dc396772b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7001a9278d4b5f013b3b8f2195966a5e6b48f95a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7001a9278d4b5f013b3b8f2195966a5e6b48f95a"}]},"branch":"refs/heads/master"},"f26bbe43cedc417405ed29c758ebb81b87f3e160":{"kind":"REWORK","_number":3,"created":"2023-02-27 17:03:27.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/20/875220/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/20/875220/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/875220/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/875220/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/875220/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/20/875220/3"}}},"commit":{"parents":[{"commit":"c0483c5b940d559b4a9efd2b5c64bf5b5528fa11","subject":"Merge \"sharder: make misplaced objects lookup faster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0483c5b940d559b4a9efd2b5c64bf5b5528fa11"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-02-24 12:20:19.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-02-27 17:01:01.000000000","tz":0},"subject":"sharder: show path and db file in info and debug logs","message":"sharder: show path and db file in info and debug logs\n\nMake all debug and info level logs that are associated with a\nContainerBroker include the path to the DB file and the namespace path\nto the container.\n\nChange-Id: I7d2fe064175f002055054a72f348b87dc396772b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f26bbe43cedc417405ed29c758ebb81b87f3e160"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f26bbe43cedc417405ed29c758ebb81b87f3e160"}]},"branch":"refs/heads/master"},"d1b2ee12870153d611a31ff32d7bc4688f07f780":{"kind":"REWORK","_number":4,"created":"2023-03-01 15:02:46.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/20/875220/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/20/875220/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/875220/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/875220/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/875220/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/20/875220/4"}}},"commit":{"parents":[{"commit":"c0483c5b940d559b4a9efd2b5c64bf5b5528fa11","subject":"Merge \"sharder: make misplaced objects lookup faster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0483c5b940d559b4a9efd2b5c64bf5b5528fa11"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-02-24 12:20:19.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-03-01 13:29:59.000000000","tz":0},"subject":"sharder: show path and db file in info and debug logs","message":"sharder: show path and db file in info and debug logs\n\nMake all debug and info level logs that are associated with a\nContainerBroker include the path to the DB file and the namespace path\nto the container.\n\nChange-Id: I7d2fe064175f002055054a72f348b87dc396772b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1b2ee12870153d611a31ff32d7bc4688f07f780"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1b2ee12870153d611a31ff32d7bc4688f07f780"}]},"branch":"refs/heads/master"},"8814cde6813f92abee5f2285789e86caaac781f3":{"kind":"REWORK","_number":5,"created":"2023-03-01 21:53:29.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/20/875220/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/20/875220/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/875220/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/875220/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/875220/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/20/875220/5"}}},"commit":{"parents":[{"commit":"c0483c5b940d559b4a9efd2b5c64bf5b5528fa11","subject":"Merge \"sharder: make misplaced objects lookup faster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0483c5b940d559b4a9efd2b5c64bf5b5528fa11"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-02-24 12:20:19.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-03-01 21:51:18.000000000","tz":0},"subject":"sharder: show path and db file in logs","message":"sharder: show path and db file in logs\n\nMake all logs that are associated with a ContainerBroker include the\npath to the DB file and the namespace path to the container.\n\nUpgradeImpact:\nThere is a change in the format of sharder log messages, including\nsome warning and error level logs.\n\nChange-Id: I7d2fe064175f002055054a72f348b87dc396772b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8814cde6813f92abee5f2285789e86caaac781f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8814cde6813f92abee5f2285789e86caaac781f3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
