)]}'
{"id":"openstack%2Fswift~875612","triplet_id":"openstack%2Fswift~master~I71ad4de0ab3af8e7e865cb924f96e5c415935654","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-04-12 08:37:23.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-04-12 09:50:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I71ad4de0ab3af8e7e865cb924f96e5c415935654","subject":"Test nested (Metrics)PrefixLoggerAdapters","status":"MERGED","created":"2023-02-28 05:40:52.000000000","updated":"2023-04-12 09:52:08.000000000","submitted":"2023-04-12 09:50:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875612","meta_rev_id":"8bf363c008fe88cff984518574ddd743a8fcf7b1","_number":875612,"virtual_id_number":875612,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-12 09:50:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2023-04-12 08:37:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2023-04-12 08:37:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-28 07:14:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-28 07:29:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-28 10:11:39.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"4a8a1a5955cf53b2e482c2a0c3830e6940e6bca4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-28 05:40:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"110b27abfb921babcf7436ae4c7272cc58e7fa5e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-28 06:14:36.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e3e626a7aea32f3f2a9f3781e9460c7948656673","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 07:14:06.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1fd351d3fdf41fe9d28f6ee6efe1e60\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cefee58d7858472282b486bb0cc9e219 : SUCCESS in 13m 29s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c6d69287b43f4aff988e341f6b077a89 : SUCCESS in 16m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e967c5b2062a461d88391b13c00eb352 : FAILURE in 1h 21m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/69cc30370e6a462f8cc48ecc3e07587a : TIMED_OUT in 1h 31m 00s","accounts_in_message":[],"_revision_number":1},{"id":"b0be1f4a3d607b16cce04e2b0fee76fb81e36951","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 07:29:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c270e0f3f2e04d8ba3a06c16b38c49ab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc94d14f5c4e4474abe0e09315d25ecb : SUCCESS in 8m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d3c0b125236045d786d45401e8c45bbb : SUCCESS in 16m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6a927397d2574f91bf9a5c277346b3ff : SUCCESS in 18m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/65ebab4d57804ebebd3b0df0e85a5e02 : SUCCESS in 19m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1ccde4d0e5ae4278906caa5a62463a37 : SUCCESS in 9m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a52b6d0d17e24425b91fbf2855062020 : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e1052507d0a64bb1a189527c2ded2b0e : FAILURE in 17m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/152e14b90a504f7cbfa69fc148f197d9 : SUCCESS in 17m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9674ed28e49942cfbc41c6ab16065372 : SUCCESS in 18m 21s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1609cf38cc7247ad9041a98eacc89f03 : SUCCESS in 16m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/259caf3189394a12b356f01f49f2b61f : SUCCESS in 18m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1fe98663fc1a4025939ad81a6e922d18 : SUCCESS in 17m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d92eb1bbaf8348bbadc32f9fd7fde3a5 : SUCCESS in 22m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e78ea84d0e554a858fcfe6cfe4f5fcef : SUCCESS in 18m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/eef99fcf649e4629b400c153ce46305e : SUCCESS in 18m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b302a19b26bc46eba4a19bf3543360ed : SUCCESS in 35m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dc35972307cb44f0925bd7846ededf75 : SUCCESS in 1h 01m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/35558a4bcf7c421bb9e51cc91d52e676 : TIMED_OUT in 1h 38m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5403bbc8a3e543ff827dbd0a3abd4371 : FAILURE in 18m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a553db0247b4ef6a3ffb9580cd1c078 : SUCCESS in 4m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/292f2f8354924d36bc1b09f477fc53ad : SUCCESS in 22m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aaeb30e30e30f76f6b673fc5f5e56a303fd45af2","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-02-28 10:11:39.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e94d65fa1bd0a65eba7a851569569afe74457c32","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-28 19:31:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d672b9fa1f77a783de6e43c5b7c7cf73c45b2a50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-28 19:32:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8a121cf5d1a628fab79ede3b243bff3dc8d9335c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 21:06:41.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74741cd4100342c6998282b6c60ab52d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e99b5c6dc99544ea82c6023ea463a7ad : SUCCESS in 49m 13s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/428c341422524d9f807a118870fad3e2 : SUCCESS in 1h 14m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/13f13d855d2a495a885795ace85c2d09 : TIMED_OUT in 1h 31m 53s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/07ca72dbb22f478da7124374df639e7a : SUCCESS in 45m 35s","accounts_in_message":[],"_revision_number":2},{"id":"70d71352e0a8ffac3eeae7350556846dcac68cf0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 21:58:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd9be28d5c3e4592bd8aa70069570315\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32ca65e321d542b1bc7d66b0b4212ece : SUCCESS in 8m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/a43edb8a3f4347a1827378716d57c340 : SUCCESS in 1h 11m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/29a7300e445745d5964d0a87555ef200 : SUCCESS in 48m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9df8257457ce42c29d51e19d7387d973 : SUCCESS in 59m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/35430aa6d34c4432bdd6bf6ba28272fd : SUCCESS in 18m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f0fe5985c1c94d998a10e88b20f81e0a : SUCCESS in 18m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/09f9aeccd0bf4fd1b8aed55af4100955 : SUCCESS in 20m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0c1369d11e704237ae460d5354514f06 : SUCCESS in 8m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c95acf027396454290b74b6bd17247ab : SUCCESS in 5m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/83ffe7fad09c429d9b06763844ed3604 : SUCCESS in 16m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/035d5f93b68c42669784980fedf5f8d2 : SUCCESS in 16m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4c34b4c4032a4ba2915963d67d50fc08 : SUCCESS in 15m 46s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/57f26e84aedc4bc09cac5c981d627eb7 : SUCCESS in 16m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/32ae1f7d5aa54dbe916fc0fe43ac64c2 : SUCCESS in 16m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ab8c52ed3a3741b69b29d2324f175cac : SUCCESS in 17m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4cc0cf7925f444b980aed367ce6d7d8b : SUCCESS in 17m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4bbd26aebc204fed93b749229a326263 : SUCCESS in 15m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b8f84051b2d049909f760742e9b9da92 : SUCCESS in 15m 59s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/149b5662dd734e6087a792c1f87e620b : SUCCESS in 17m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8c0e9cae82804224bd070c80e6ec8e66 : SUCCESS in 11m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4c5eac64cd8e4f6493e8fc96e0e2870b : SUCCESS in 16m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/abe19cad48b540c586733d55fa2bcb40 : SUCCESS in 7m 41s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4fa4c940dff64a519c6a55f56e78ed56 : SUCCESS in 52m 02s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a9a829e041be4850b38272732ff8e5e5 : SUCCESS in 58m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c11e7b8bea5841cda24edcc357edeb0c : SUCCESS in 55m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b4a83e60d15042dd9779ce2db5511e4d : SUCCESS in 47m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99f730679fbb48948b8f93e6edd7c927 : SUCCESS in 15m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21970e92de6a41a68a3ea7fbac193306 : SUCCESS in 4m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ff71477e8136491a9214cc59c52f4a64 : SUCCESS in 19m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6656ed2957ee4266a9e6878dbfb2f7ba : SUCCESS in 43m 44s","accounts_in_message":[],"_revision_number":2},{"id":"edbb75910defc3e5fccc23c17c8e543622aece92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-06 03:57:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ae2ab0e721e7efcfac2a9ae0f05f89811ee0aca7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 04:20:11.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07ddc9dd667c4bd4a86be5f0292e1524\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ab5ae23c5b6c410f900fe60a6ba6467d : NODE_FAILURE Node request 300-0020917252 failed in 0s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ccdf31a210af42a7a45806ddfd8f749a : NODE_FAILURE Node request 300-0020917253 failed in 0s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9cfb6dbe4a2f4daeb711805e0ee12677 : NODE_FAILURE Node request 300-0020917254 failed in 0s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/435f04212d2c4860a26368bd16bbd0dc : NODE_FAILURE Node request 300-0020917255 failed in 0s","accounts_in_message":[],"_revision_number":3},{"id":"2a1f7d89ba45e37d7f7b8ced01020a39b7326fb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 05:12:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d20ea0403952487990c5c3e83c2ee101\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3388b2e6f694651b654c02f77a1cd9b : SUCCESS in 7m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/290c628490714719a6df6f4e7567b5f1 : SUCCESS in 47m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3138839be5d947519784f6b2361be27d : SUCCESS in 46m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b9818d2de064457985f6ad4437d6110d : SUCCESS in 50m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e30656d65f554079b9722e863e4e63c3 : RETRY_LIMIT in 5m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/41c2119795a64e1da21159ccc5c1bc07 : RETRY_LIMIT in 4m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a79f07f16fba4ddbaa90d8bb8a8e42fb : RETRY_LIMIT in 5m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e9e6c3519cc14bfc8c623df1c2ac28a5 : SUCCESS in 8m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d537cb9cc34a4549a7749e0bed61ac7f : SUCCESS in 5m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9d8df7c100a84cfaa5b246ff8b41991e : SUCCESS in 17m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0c6050b70acd48639f86df27d631050f : SUCCESS in 16m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3be6d6e3c90d4d3fa3ba26a8a8f0689b : SUCCESS in 14m 49s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/03c85aec32714585be7c74f9c4675992 : SUCCESS in 15m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e3137dbc371a44078f95bec438a9f576 : SUCCESS in 22m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2b6b1103c8cb48e58be2a369b3058635 : POST_FAILURE in 17m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b85500aa160b4e8da1c06d3e29570dcf : SUCCESS in 25m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0932e95214d3465fb116afbc159137e8 : SUCCESS in 14m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/823a1e2e931a4b6d878b6c67b12df511 : SUCCESS in 14m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c65a558837fe4dc9bb4a9bd11c6af120 : SUCCESS in 11m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f2311b9bc52a41c98ad6d4202308b852 : SUCCESS in 6m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9383b10b91f2431aa02b53ae38f900cd : SUCCESS in 15m 55s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/776c9928b1f648d09be8202c7f6761b2 : SUCCESS in 7m 43s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3c56937cda514d498a65b36802109525 : FAILURE in 50m 35s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e2bd26806f334065bd09ed6a4a04015c : SUCCESS in 56m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/96f8dcbab4af427dba286696e125062d : SUCCESS in 53m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d3609073899747fdaad9a7aba406683f : SUCCESS in 43m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76944167189f4f639434f6c5fab7f098 : SUCCESS in 14m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cccb5c50d6f4628948f41b52c63dd6d : SUCCESS in 5m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/33d82ab1465441d7891a8f83d7315bf4 : SUCCESS in 17m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/456de8efe9a745cab48df92288d3337d : SUCCESS in 57m 16s","accounts_in_message":[],"_revision_number":3},{"id":"067b86b89ac4e619e60cfb07d74472d05ce223f1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-06 15:39:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8cf5f3ff7d4e56060f5fa71ab0c98d1522374760","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 16:24:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec4f9c6260c246ed874a87b3d571f7d5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/18397ab083084e94aadc9917bc928f1c : SUCCESS in 17m 35s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6f61ce26862642e286d1381725ceb4a5 : SUCCESS in 17m 06s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9ed3422bd4074c7eadf75c1b6d550376 : SUCCESS in 22m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2d303ed3bf294b7b93d6b06d62da0ba7 : SUCCESS in 43m 05s","accounts_in_message":[],"_revision_number":3},{"id":"b1c766285072cdfe1791ac154c515c979587fedc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 16:46:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f634d6fa08254fdd944a0fa6a33c5d0f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4da06b45fae14611b028f0e4ac186d9d : SUCCESS in 7m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/d3e01c01260f4f07965a1412d6607cad : SUCCESS in 51m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2a44552b14834518bf285ba28f974897 : SUCCESS in 32m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87a61c6fd0d14ac29726e7f0c711e638 : SUCCESS in 47m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b3d1251f962d4d7c868d49ee2177b8ba : RETRY_LIMIT in 4m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/11121214f6cb40848f72338ca0309df7 : RETRY_LIMIT in 4m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0b1576b4b1774afa9e37732db843d456 : RETRY_LIMIT in 4m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9d3a111e1edb4aeea8404bcda3c0cf5d : SUCCESS in 7m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/61fc8c86a67f449e8eec02ed50670b74 : SUCCESS in 5m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/59eb4d1914e44f2f9ede35232be69b35 : SUCCESS in 16m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d53c329c26fb4e8cb4a623dc18761d59 : SUCCESS in 34m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7132b367c3224666a9e5b997986b2376 : SUCCESS in 15m 27s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/af3e2cf9b08c4e7691bbeabbef7fe09a : SUCCESS in 36m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/add1895712994262a5523951d3936d29 : SUCCESS in 16m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ad466e60e8c34022a7e4cd12677c57a3 : SUCCESS in 18m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c737cd279905471f9ef6ccfbd15c004d : SUCCESS in 19m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8ed801fbd5a24be19ad04cef14757919 : SUCCESS in 14m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c46c52999ade43ab8c14c3e6d551839f : SUCCESS in 13m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/55c750a8af2c43699d7e6474d35a5a7f : SUCCESS in 14m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/60952aed14a54934aca04c5183682eb9 : SUCCESS in 7m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2658f95f1cd844e099166923075ee510 : SUCCESS in 29m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5bd96f49101d4fae903751f99a6d1d3f : SUCCESS in 7m 27s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9fb989fc0bfc4c85a47e39b287a34440 : SUCCESS in 53m 30s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/dcb1e9d96afc43e586fabfb1179fac37 : SUCCESS in 58m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3d0af92e8d4b4f6c9972c7efa5d09cb8 : SUCCESS in 1h 00m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5e131476e22541fe80e4ce6aaa77204b : SUCCESS in 54m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be393862f1b94a9c926907dafb93ab52 : SUCCESS in 13m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/794f353601624da097fc5d0a2fc46e3b : SUCCESS in 5m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/57d9271d74cc4963b6870643e96f0de5 : SUCCESS in 20m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/541866d5845846dd8a6f03178fe7f1aa : SUCCESS in 51m 48s","accounts_in_message":[],"_revision_number":3},{"id":"5f698d695715d6f7e1eb7cf708621659256df28e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-04-12 08:37:23.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7d11a1909440b8ddbfb2d73036e23c7ba0b3d2f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 08:37:56.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4b862309f32fba811c2d7970b7d8ec74c8eafd09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 09:50:58.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/497d4d0e85844474896058551a005990\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4233a295e254014a2613c5e706d3cc2 : SUCCESS in 8m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/f0a1cc163139449cb767cdc180540245 : SUCCESS in 58m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bb114252209445e0877274b846b4f20d : SUCCESS in 48m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eeb1f7610c84497997f0e017ac2b6ca5 : SUCCESS in 1h 05m 34s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/2ac501bf55b0439d930f40a8b69e1f05 : SUCCESS in 9m 15s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/d6c737a2db714fa9841382741e7a68f5 : SUCCESS in 5m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5712714db9c242a0b5947a92cbcc9c25 : SUCCESS in 22m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6ac58600b2074ffea0702236b07a692a : SUCCESS in 16m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d6e289bd16e84ee39d8b6226705bd7ca : SUCCESS in 16m 02s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3eeb6a9f0de449cf8b6c9e80dc061e44 : SUCCESS in 14m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/aa6e88f3c8254ae59c317e934a3c06ef : SUCCESS in 16m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/38f3fd1a648a4860be5cfda23ed95d63 : SUCCESS in 18m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/616d582162bf41a88653f63d1f8a7639 : SUCCESS in 17m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/13077612102744eb8a87dc6dea32ae45 : SUCCESS in 13m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/11def0fe9a15456a891948fa41f4c703 : SUCCESS in 13m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3e1d6130f360451ab9a580973d7128c2 : SUCCESS in 15m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/acc2a6497301470997070426a764eb3f : SUCCESS in 6m 45s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ef758bd5d8f4473db71589caa9324199 : SUCCESS in 7m 56s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bad924a6e3754daca8b1256115ebf826 : SUCCESS in 56m 36s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/095ebedf7a97451da8232c09b401b870 : SUCCESS in 54m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7c260c9018894e3a925c19bf67a04934 : SUCCESS in 54m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2a5db34c64d44937af6447fa5619edaf : SUCCESS in 46m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ffb15623d49a4c58adedecac6dcc96dc : SUCCESS in 17m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28478ac5809d48ba84474e091d2d0898 : SUCCESS in 4m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41707c474c514ef7bab4d5547e09f5ba : SUCCESS in 40m 32s","accounts_in_message":[],"_revision_number":3},{"id":"25f10aec676162735a794ce194b3615246f9064d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 09:50:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"8bf363c008fe88cff984518574ddd743a8fcf7b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 09:52:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e5d2e63474b447db6e2fd810d08afbe\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a4026b339204f5b8f17d68fd8a1258d : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/a866d3b2c0f646f198d70f785bdf664d : SUCCESS in 35s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/5f52de0dc19d40a6a213999c8a9a2ced : SUCCESS in 31s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"85d68ee491a61317bffb2b9be85d17c00dbabccb","revisions":{"0b52d461c929f5100b3975063dfafdac3520cacf":{"kind":"REWORK","_number":1,"created":"2023-02-28 05:40:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/12/875612/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/875612/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/875612/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/875612/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/875612/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/875612/1"}}},"commit":{"parents":[{"commit":"bba3a3145dc73a1fc7a4c77cdafc3c40e1ea1548","subject":"Merge \"tests: Get rid of test.unit.SkipTest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bba3a3145dc73a1fc7a4c77cdafc3c40e1ea1548"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-02-28 05:40:42.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-02-28 05:40:42.000000000","tz":-480},"subject":"Test nested MetricsPrefixLoggerAdapters","message":"Test nested MetricsPrefixLoggerAdapters\n\nChange-Id: I71ad4de0ab3af8e7e865cb924f96e5c415935654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0b52d461c929f5100b3975063dfafdac3520cacf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0b52d461c929f5100b3975063dfafdac3520cacf"}]},"branch":"refs/heads/master"},"27247e358ce884032912228647ffd46e15586896":{"kind":"REWORK","_number":2,"created":"2023-02-28 19:32:37.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/12/875612/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/875612/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/875612/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/875612/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/875612/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/875612/2"}}},"commit":{"parents":[{"commit":"bba3a3145dc73a1fc7a4c77cdafc3c40e1ea1548","subject":"Merge \"tests: Get rid of test.unit.SkipTest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bba3a3145dc73a1fc7a4c77cdafc3c40e1ea1548"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-02-28 05:40:42.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-02-28 19:31:26.000000000","tz":-480},"subject":"Test nested (Metrics)PrefixLoggerAdapters","message":"Test nested (Metrics)PrefixLoggerAdapters\n\nChange-Id: I71ad4de0ab3af8e7e865cb924f96e5c415935654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27247e358ce884032912228647ffd46e15586896"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27247e358ce884032912228647ffd46e15586896"}]},"branch":"refs/heads/master"},"85d68ee491a61317bffb2b9be85d17c00dbabccb":{"kind":"REWORK","_number":3,"created":"2023-04-06 03:57:40.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/12/875612/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/875612/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/875612/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/875612/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/875612/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/875612/3"}}},"commit":{"parents":[{"commit":"57ce156a7f5cecd58fb4d83acbe1416d583275d1","subject":"Merge \"Start common.utils refactor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57ce156a7f5cecd58fb4d83acbe1416d583275d1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-02-28 05:40:42.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-06 03:57:26.000000000","tz":-420},"subject":"Test nested (Metrics)PrefixLoggerAdapters","message":"Test nested (Metrics)PrefixLoggerAdapters\n\nChange-Id: I71ad4de0ab3af8e7e865cb924f96e5c415935654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/85d68ee491a61317bffb2b9be85d17c00dbabccb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/85d68ee491a61317bffb2b9be85d17c00dbabccb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
