)]}'
{"id":"openstack%2Fswift~875806","triplet_id":"openstack%2Fswift~master~Ib77c564ec2ec4e5d84e403887cd1179343c0c454","project":"openstack/swift","branch":"master","topic":"object_updater_with_memcache","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2023-03-06 08:46:07.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib77c564ec2ec4e5d84e403887cd1179343c0c454","subject":"POC: updater: only memcache lookup deferred updates","status":"ABANDONED","created":"2023-03-01 05:32:37.000000000","updated":"2023-03-06 08:46:07.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ec804a3e36d2f3f44af4582d5d07a07716d72cd9","_number":875806,"virtual_id_number":875806,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-01 06:49:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-01 06:49:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e516ccceb822b8bf97913726b6402fe8d691dc07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-03-01 05:32:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8594926779290533c85fe2277acc45002f47e33c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 06:49:47.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8679aaa23952499398a51a454cf57622\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0b6409fac7af47eb8037a3857afe93bf : FAILURE in 1h 11m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d4a4b09fd8ef48e38cb1362af490de14 : FAILURE in 1h 14m 26s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7cc5b3336e0d4297944f1a097d9df121 : SUCCESS in 56m 05s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d1199435e9ce497e94bb0e21cef28bb7 : SUCCESS in 1h 11m 52s","accounts_in_message":[],"_revision_number":1},{"id":"1d7e339f02de83e43d74b58d2f971e74330e0edb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 06:49:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/419d7a6e606b44348ba2e23ea092938c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df7e91dfffa34342ac63fde46edd5a5b : SUCCESS in 9m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/8c9aa0892c1c4b21a7a915b2cf241b80 : SUCCESS in 52m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2d2f6c58d94648f2abdc02f9b5c750b9 : SUCCESS in 53m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/54ae1921b6a940fcb86bdc1ed8e40f21 : SUCCESS in 59m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f13e832842b04469a60abe8e54612418 : SUCCESS in 22m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3e0ae977da0f45f3a7b22cd3bc15462f : SUCCESS in 20m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fd7261df68f94b169083ed34d1c204f3 : SUCCESS in 18m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/50d4fc3ee5564d5687a9493d9a6d4f35 : SUCCESS in 8m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6828a6a844284c48ac72ab95244e2339 : SUCCESS in 6m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3b54c19378c547d28e27d0e140d6fbe4 : FAILURE in 51m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7ad22591a3df416198e5b6db8afc7d2f : FAILURE in 17m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b74a9095b9124aa79f8c778357b0d3a2 : FAILURE in 12m 20s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8d4c2cd36de84434b1857c8a915411b3 : FAILURE in 12m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/74146d1f19a74ab8894e619b0a0f863b : SUCCESS in 17m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/376f185e75ab42fe9e162a2760d5695c : SUCCESS in 17m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e9b08f1a777241a28dec094a176ff5e4 : SUCCESS in 17m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/05edb20616974968808ba80f35115140 : SUCCESS in 14m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6bed74a198464fdd8b2697c0c2248202 : SUCCESS in 13m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/48d1501dedb64c77b4ec6e5ea00ef791 : SUCCESS in 15m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ba58b15243514f4b912fd3a53948f87b : SUCCESS in 8m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/eb61204e1e774b21a23b48710d51e88f : SUCCESS in 14m 58s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cb6056f69fb044cd9c203916db9fd4c8 : SUCCESS in 8m 11s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/83300f8a700a4b62afd05e1015a1202f : SUCCESS in 55m 07s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b5abdace84ec45d39643a0e2afe625da : SUCCESS in 1h 08m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1bd727c052ee4831aa48d21e279afeef : SUCCESS in 45m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/542e3f22650e4cfe834161b1afb24182 : SUCCESS in 36m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/748be9723cbd4bab9c3c3f26d311a6f7 : FAILURE in 17m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52aef70d42b242178a1c76ed588a6e31 : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6baab714cd484b9fbbd040e01c561690 : SUCCESS in 22m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77b11225e40248648c9903980e24497a : SUCCESS in 39m 22s","accounts_in_message":[],"_revision_number":1},{"id":"fafd89d59d246b61849e6c5806eb44c9a7035537","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-03-06 08:45:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ec804a3e36d2f3f44af4582d5d07a07716d72cd9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-03-06 08:46:07.000000000","message":"Abandoned\n\nNah this doesn\u0027t really work. I don\u0027t like it.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"67a6913b0f413ccbb60b7cc800d1855e5bc7e73c","revisions":{"87f673f184f53a2dcbe83b6598edc3a6991b15be":{"kind":"REWORK","_number":1,"created":"2023-03-01 05:32:37.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/06/875806/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/875806/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/875806/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/875806/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/875806/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/875806/1"}}},"commit":{"parents":[{"commit":"6ede2f3e364aadd9d2a7cea27112687872b0c01d","subject":"updater: add memcache shard update lookup support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6ede2f3e364aadd9d2a7cea27112687872b0c01d"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-03-01 05:22:34.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-03-01 05:32:14.000000000","tz":660},"subject":"POC: updater: only memcache lookup deferred updates","message":"POC: updater: only memcache lookup deferred updates\n\nIn case a memcache lookup on every non shard container is too much,\nthis patch instead only uses memcache on deferred updates.\nThe down side to this is it means under load to a single root there will\nstill be hit by initial updates but once they get ratelimited they\u0027ll\nfall back into the queue and then memcache will be used.\n\nNot convinced about this\n\nChange-Id: Ib77c564ec2ec4e5d84e403887cd1179343c0c454\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87f673f184f53a2dcbe83b6598edc3a6991b15be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87f673f184f53a2dcbe83b6598edc3a6991b15be"}]},"branch":"refs/heads/master"},"67a6913b0f413ccbb60b7cc800d1855e5bc7e73c":{"kind":"REWORK","_number":2,"created":"2023-03-06 08:45:00.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/06/875806/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/875806/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/875806/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/875806/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/875806/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/875806/2"}}},"commit":{"parents":[{"commit":"cdc66d846c53f8eb1bc24424831af74f2f84cef3","subject":"updater: add memcache shard update lookup support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cdc66d846c53f8eb1bc24424831af74f2f84cef3"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-03-01 05:22:34.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-03-06 08:44:39.000000000","tz":660},"subject":"POC: updater: only memcache lookup deferred updates","message":"POC: updater: only memcache lookup deferred updates\n\nIn case a memcache lookup on every non shard container is too much,\nthis patch instead only uses memcache on deferred updates.\nThe down side to this is it means under load to a single root there will\nstill be hit by initial updates but once they get ratelimited they\u0027ll\nfall back into the queue and then memcache will be used.\n\nNot convinced about this\n\nChange-Id: Ib77c564ec2ec4e5d84e403887cd1179343c0c454\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67a6913b0f413ccbb60b7cc800d1855e5bc7e73c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67a6913b0f413ccbb60b7cc800d1855e5bc7e73c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
