)]}'
{"id":"openstack%2Fswift~876662","triplet_id":"openstack%2Fswift~master~I227ca4f16c4314fc91663a25f2bb4623c6d761ca","project":"openstack/swift","branch":"master","attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2023-04-28 20:55:13.000000000","reason":"Vote got outdated and was removed: Code-Review+1"}},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-03-10 20:39:18.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I227ca4f16c4314fc91663a25f2bb4623c6d761ca","subject":"Use signals for timeout on os.wait()","status":"NEW","created":"2023-03-07 03:26:11.000000000","updated":"2026-07-09 03:31:22.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":11,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"97b310a7644b937b29d042d1d0e69d27da755fca","_number":876662,"virtual_id_number":876662,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-09 03:31:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-07 04:42:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-07 04:57:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-10 06:04:51.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"b815b8b4f74c37d8389444e592cb311869732952","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-03-07 03:26:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ecb33b49b0bc6baefd993429563e849eeb84df9d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 04:42:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e141b41d8da94b4491b16607378f85fb\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e863f55622d149a3a144467eb46cf8c6 : SUCCESS in 17m 19s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/189a25bd44904623929f1279469221f5 : SUCCESS in 45m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ceac16e1a71c46449bf1125a78a704d9 : SUCCESS in 21m 32s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2ccce8504ab3491bb9df356f40027aa3 : SUCCESS in 1h 13m 34s","accounts_in_message":[],"_revision_number":1},{"id":"b0c3b2417237c6e2416c6cf5157c6ad67b9de32c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 04:57:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79ce8dc181d746fca9819846eadf8b94\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8a3f728de3e451eaf474870064e4d7c : SUCCESS in 7m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/bf87e97a234240d398edc61d0425bc77 : SUCCESS in 1h 17m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0de913e0fc664d0c88b4b03ab7386470 : SUCCESS in 42m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/446dab9dc6934f1b98044bc572efb8db : SUCCESS in 47m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e1280b3c06574e40ba56cf2facf8ead9 : SUCCESS in 16m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6fcd6951aa6740839d28aefe14ce24a1 : SUCCESS in 15m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9b20cb260d7040c78b5d60c7ebfb1986 : SUCCESS in 18m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c6802bf347494a429b2b5b72d8d2f78e : SUCCESS in 9m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e607f11245f141e29e2e3516dfb36932 : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4f8dd248d57341ab93d4e15e8d0a5226 : SUCCESS in 25m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/583bbe6f557b4f92bf671da9a3c6542e : SUCCESS in 16m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d630a315ae754a7c8af04bb56680feb9 : SUCCESS in 14m 39s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f79bd1b1f23b4309b08d5973d7a68acd : SUCCESS in 14m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a9a82f69cf1d42d19620dffb2bc6a8f7 : SUCCESS in 16m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d1fdca5923f94b68891be1f1cee9d16c : SUCCESS in 15m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/77c3449cbfba4b32acc4287fc78511f4 : SUCCESS in 15m 13s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/aa666ad718794fb5baf821aede35bd80 : SUCCESS in 12m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/eadfc238bd7543ff9a0fcdc02f96069b : SUCCESS in 12m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/de9d5cce61594a51858f17932cd7fdac : SUCCESS in 13m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/87687e52583e4488aeb79c6d79fb111c : SUCCESS in 6m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/83d9295e46c1476faf31424f73430b3a : SUCCESS in 15m 11s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/17c1019b13d6428f82479d79e91c49b2 : SUCCESS in 7m 00s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b4a9afe5481c4fa983c7b646f731e796 : SUCCESS in 49m 12s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0940fb09c24949b9a9f21d795a306087 : SUCCESS in 54m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3c27fac5a3654f1084c716fd6363fea8 : SUCCESS in 40m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e92ceae166ba4d53a79494f57229cb73 : SUCCESS in 40m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3fd6b477dcd042208209f42587b7ffc6 : SUCCESS in 15m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9630c879026046fc806023ebfaea66af : SUCCESS in 4m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9162f8b4dff14d13b119a86cebeaa9c8 : SUCCESS in 18m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29a101c67bff4c19889cc32f7d6c71c6 : SUCCESS in 1h 13m 57s","accounts_in_message":[],"_revision_number":1},{"id":"a5ccafa0c3c79a62669dbda75cb752e4b5da0e30","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-03-10 06:04:51.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eb241b84f0d4bf426a15217ea5cf696c525171b0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-03-10 20:39:18.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"371fbd3d00866be12703f8afd9ed520bafd78bf0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-28 19:49:33.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d9e60ecb6ef1602712b6dc14054c78a95995ac46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-28 20:55:13.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"45338827d8c2ef707623902cc3b25b7a1ec1391a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 22:06:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d19fc693e7b49f09263d8b63b35282c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5b4616a654e48a5ac2cbf33e83dfd7a : SUCCESS in 8m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/79a5d500ebaa4d29a188566bfda31c3b : SUCCESS in 43m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/95d30e5f91d04df0966276b977d898d2 : SUCCESS in 50m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c0959763848e45c6b198f1135e1bff73 : SUCCESS in 55m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6abcaddcbe5045b0b9a58dbbe7c64c7c : SUCCESS in 14m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c5bb855bed0f4ac7837d569e91be0cad : SUCCESS in 40m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f0d833bfda0645cab4de4292cd811b15 : SUCCESS in 15m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0c14a4ec40514deca88e73dad94d83e8 : SUCCESS in 9m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f25e67e7d67d4675b169fa787da1d8d2 : SUCCESS in 7m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4564c26fb3404f53ad7a4ab5491d0a2a : SUCCESS in 18m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3d8dff10cedc42fe963bbc6a2362e49d : SUCCESS in 17m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/99283c1f1aff42cf97e36b8b9715fcfb : SUCCESS in 15m 17s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/dc3f1a28a1c6483b85e72aa283bc17a3 : SUCCESS in 14m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a641cd7b909b4e3aa3d7314f465e0f98 : SUCCESS in 17m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cf4c4eb349584d5096505b06dcf9f9f9 : SUCCESS in 19m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a2b5eb293cf346cf8c154477cf9166d9 : SUCCESS in 17m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fb8c22137bc245ca8afaad7278a17c30 : SUCCESS in 14m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5664d6c8fda74b28a37cbab8f4e49e2e : SUCCESS in 14m 29s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f1fef996d05140a5bba1a64dc2a313bf : SUCCESS in 15m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/410a21c0d8b442958120517cc74a7ca6 : SUCCESS in 10m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f0692614d59d4d968821df8c8c54a780 : SUCCESS in 16m 57s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2aafc2c329594398b3f68624084a5e89 : SUCCESS in 7m 23s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7dc6304741ac4afebc47f690989901c2 : SUCCESS in 50m 20s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b5ff114f41f74ec48974c3562a7f6df2 : SUCCESS in 1h 03m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cad1750695774071bee51c6d6be4832a : SUCCESS in 39m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/92a0ea2940d24e098c0b33bb54e49ab6 : SUCCESS in 50m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15ef1de82df542f78b3d05fca6cd00c1 : SUCCESS in 18m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/537c7d0f99d14aba8bb6c23602e20f54 : SUCCESS in 4m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e2e9cd29682f4379af673c61693fb5eb : SUCCESS in 20m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a1b66fe6a454858bd02df5511f496a7 : SUCCESS in 56m 46s","accounts_in_message":[],"_revision_number":2},{"id":"311d5321870a2bdde0a21df078f4dce034932e87","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 22:28:53.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10c7768e4d454a088ee3b458feec7f95\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3e20e27fd452462aaf60d56fd58e202f : SUCCESS in 1h 10m 50s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/00868d63722b420cabaf88d5a40e05dc : TIMED_OUT in 1h 30m 51s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e1a7e50d1e0b42ad96fe9908722f2c08 : SUCCESS in 1h 05m 09s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/500a3099d0ea42149c25d8a53c964db0 : NODE_FAILURE Node request 300-0021094196 failed in 0s","accounts_in_message":[],"_revision_number":2},{"id":"7893c7feaa9b155d38b4e28de4d715ae3561301f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-10-20 21:03:41.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"8952c4b883f05c44189336207a42582aa280719f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 22:13:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff8ec38529094010b46a7c479a6f4fc8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0cb32c1ef4146ec8f052aa976615425 : SUCCESS in 9m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/89e31000a81340fa8227a14a06247a5a : SUCCESS in 39m 17s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/921f0271f5ff4984ad3462b59a76dcba : SUCCESS in 49m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3e0a350bd25f49a78311aedc44626d61 : SUCCESS in 52m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b55c404dc8524f71a5cfd215b2968fd5 : SUCCESS in 1h 01m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a56fa4b0086045a9b9a6cb7158ed0a4d : SUCCESS in 18m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a115a65f37d7442183c7d549c7a957ff : SUCCESS in 19m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9755fc6796d84342ba01b0f6a88962a1 : SUCCESS in 14m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2a971bd1e5f747d5850760318974ae0e : SUCCESS in 9m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9377b4dc0d0a4dfbb4efdb72a0aba351 : SUCCESS in 5m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0a2c6d8705d940279b7def7f56f57244 : SUCCESS in 17m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2411c6ebed4e4c3c8ed75b327da0ab8e : SUCCESS in 11m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4a6d8c434fba4f42b8a6226161a99722 : SUCCESS in 17m 25s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1996f0663afe4e89a0cd65566c58c612 : SUCCESS in 18m 17s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1248085669c34b4aaf9ea6b32914ee4c : SUCCESS in 20m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/27210aa99e5a4f00a33cd43297d443d2 : SUCCESS in 19m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fc1dc7644c0549d9aee5b09aca236be8 : SUCCESS in 20m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/027f9454428c444db18d33c855d35a33 : SUCCESS in 17m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8585bb889e084aedbd919e7ca8dc3608 : SUCCESS in 17m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d9b4d7d4a44c48fe9be338852c9eed79 : SUCCESS in 14m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/39041f4ae1b6423e95337f432a9c3792 : SUCCESS in 15m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/676d17c65b6345469d49b3794dd7b3b3 : SUCCESS in 7m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/95845ff8941742898313e12324133798 : SUCCESS in 18m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/102d199706a1485eb7de32ebb4f6c87e : SUCCESS in 7m 32s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b573b47fb8a04fcab0c8cea6eb25d4a6 : SUCCESS in 50m 13s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/56d86922a1bc4c5f9f55a8d9f8efb780 : SUCCESS in 1h 02m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/014d4b3e43564555b0d1053afb6db868 : SUCCESS in 55m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/68da9f3c883c4e20bb84fdf76825be44 : SUCCESS in 41m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/24f0b7e23ec3495792ee83d5bebdcee2 : SUCCESS in 17m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e72d1d8f39a34f73b872bb38f111b786 : SUCCESS in 5m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8c81845e100e4f45a5ac001b0323552a : SUCCESS in 23m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72c4ae171dfc48fab43c2f9f85415a61 : SUCCESS in 1h 02m 26s","accounts_in_message":[],"_revision_number":3},{"id":"5ecb8fe8b83cee3902970a6748797e70d9933e24","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 23:23:22.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d137eaa4294944beb3fb56b45421ddb2\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/62a2bde2a4524f26b5dfc265171c01c9 : SUCCESS in 1h 17m 39s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6eee7ef5dad84be3a60fcb696ec50757 : SUCCESS in 1h 21m 21s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/09c398b02da246f3849991827a69d2fb : TIMED_OUT in 1h 31m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fe7a2e4622814bd1a3742251d7df1475 : SUCCESS in 1h 24m 49s","accounts_in_message":[],"_revision_number":3},{"id":"332280c579b78bddeca15bf473f89fb23302ed12","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-11-08 23:52:47.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"bdae8595515073a7566281ec7458f9452933d484","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-22 22:24:14.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"44c8b00c95b73157c9f4ceea44fa15252fccf799","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 23:00:04.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cd4021a4f454c1daaee3ac070745a50\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c90cb57cbd7a4e7bb29a07b2b9d731e8 : SUCCESS in 33m 50s","accounts_in_message":[],"_revision_number":4},{"id":"2ed0e576d9b2ba6bf69959ed51cd538b42544c8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 23:24:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c648249ccc84c81a7db06413ccda07c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9c07995197a4bd9a4593867a9faf32b : SUCCESS in 12m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/b4767093474a479b9230566138ca3f43 : SUCCESS in 28m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e2517e96a6054e69aaf4e90afd17919b : SUCCESS in 31m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bcc741c79b7844d2aa768001773c540a : SUCCESS in 39m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc324f4b1b32417a8145447101339782 : SUCCESS in 47m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2fed4c3601784f7bac546520c9753504 : SUCCESS in 20m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/feda54f848d6486681e7e93ebec4cf21 : SUCCESS in 20m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/35e549c0fbd245edbb8b5f78386fd4de : SUCCESS in 23m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/24f44a5cc8ed49538497ba4ba1ff0700 : SUCCESS in 3m 33s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/89b6fa1fd5b9406abf2473471f8c7ad5 : SUCCESS in 22m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8188406564c847deb9d7faa798221612 : SUCCESS in 21m 06s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6b156bc126d543bfb6be3405dab5c263 : SUCCESS in 19m 01s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6912568a541e4dd2aa4d49c93e64dc57 : SUCCESS in 24m 38s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5627060ec8f8430b91cf08f34b1de44d : SUCCESS in 8m 01s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/48a323b427634232ae46a18b4552880c : SUCCESS in 18m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c639e629a89f4f9287618e5ce8026b4b : SUCCESS in 19m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a594fac3a8db4f549d009d17aba6e136 : SUCCESS in 6m 18s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c5637c93d85541d9863bd839829b054a : SUCCESS in 14m 08s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/38df161cf8e0478aba60f7c8de2ea591 : SUCCESS in 6m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/14791a1bcb084e02a8869a6847dac687 : SUCCESS in 53m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/78c17631f6f1480bbe824146459b7526 : SUCCESS in 46m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/565cb972eac34f7e9a679b55cfc44e8a : SUCCESS in 18m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5f096864fcb4e9b8058aa485a50affb : SUCCESS in 21m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28a1357f1b504339b0d41d31100f9588 : SUCCESS in 2m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1e7e7277f89c45a7927192f6eac31a6a : SUCCESS in 19m 02s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b185b7cd8d6b4989922b92bc3c5fc02f : SUCCESS in 50m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad69da64fa8041d79d17f5547730df00 : SUCCESS in 56m 56s","accounts_in_message":[],"_revision_number":4},{"id":"9fdf4605e09edb2d7430f7c7b8d09832df6b9f78","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-23 20:06:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"784cafbefe2b749cc264ca89b900e7e84edd2405","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-09 01:53:11.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"98e4d529b6c1b1d3d59a3284d71ebc2565b58f5d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 03:24:49.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99207fc2f03e4346951d4b1ea0f5522a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4e0cd79b4a6a4ef8bb956cbdfde86040 : SUCCESS in 1h 29m 08s","accounts_in_message":[],"_revision_number":5},{"id":"97b310a7644b937b29d042d1d0e69d27da755fca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 03:31:22.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85595c98fa4c4542a894321ef9f6513b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9d0e99d5b7c4452aca49aa138639c32 : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9269085683894a6690b7ba6aeadd3a94 : SUCCESS in 4m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/f32663d2d6de4c7ca8a2a1999e0156e2 : SUCCESS in 1h 08m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bc6e1b72d1f241b794ba9675ca41e11a : SUCCESS in 1h 19m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5cb9bdc2985d49b18f413fef1dbd07a4 : SUCCESS in 56m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3ff5c6142fa54629a816bcbde24b08ad : SUCCESS in 1h 30m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2664977fed4642a99239fb3e89431788 : SUCCESS in 21m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/82554deb790f4f26bec8340ebff2a819 : SUCCESS in 11m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e502dbb2667445628732605c76a1b083 : SUCCESS in 21m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/863b77d761b842cf99b4c5d6a56f6ba4 : SUCCESS in 4m 20s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a0d4776af46642e982dd53bee3e430ad : SUCCESS in 12m 30s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2165c0c754d34839a5a1dcbe3cd864a4 : SUCCESS in 22m 01s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f3ecb58a4d354003814a990e69b973f6 : SUCCESS in 30m 41s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/c76262a8aed04a32b8b7bb1b03d297ba : SUCCESS in 9m 48s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e4abd2a58b0b433495defe63e24774bd : SUCCESS in 17m 22s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/45ca927091db45398d027858e86aa9e1 : SUCCESS in 19m 04s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7122dacac9bb45da80cdaf923545a020 : SUCCESS in 16m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c34d5e61634e4dc28c429a7c608b2e81 : RETRY_LIMIT in 9m 17s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/61e922d2213e4324a4a7e2dea21b70f0 : SUCCESS in 19m 25s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6911dc56586e4ab0833939fee9e456e7 : SUCCESS in 9m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/30758790f9c744b7812d99a82b74ff65 : SUCCESS in 1h 05m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7597eb84ea9d4ea899990b9d21a46c67 : SUCCESS in 24m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8dd9641333c34437a8bac9d73b226dc8 : SUCCESS in 56m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ac4fef7e53b048b399d8c2615dbd7687 : SUCCESS in 29m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e32bee105dac42d39efa5a0113404f8d : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/662e7844134147c7a6ced562faffa4f2 : SUCCESS in 22m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e8cf59d443b479db6eaa2154f1f8374 : SUCCESS in 49m 46s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e5af3a97ea5fa85ded34154c082cb3eaf73c96a9","revisions":{"625c37a86abeb6315515a0b156acce8ada1e7a64":{"kind":"REWORK","_number":1,"created":"2023-03-07 03:26:11.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/62/876662/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/876662/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/876662/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/876662/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/876662/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/876662/1"}}},"commit":{"parents":[{"commit":"a2952962d2f451d63b482646e23f515be9b00b86","subject":"Merge \"proxy: Reduce round-trips to memcache and backend on info misses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a2952962d2f451d63b482646e23f515be9b00b86"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-03-07 03:26:00.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-03-07 03:26:09.000000000","tz":-480},"subject":"Use signals for timeout on os.wait()","message":"Use signals for timeout on os.wait()\n\nChange-Id: I227ca4f16c4314fc91663a25f2bb4623c6d761ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/625c37a86abeb6315515a0b156acce8ada1e7a64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/625c37a86abeb6315515a0b156acce8ada1e7a64"}]},"branch":"refs/heads/master"},"c0829f0fdc3dc87cc8c824f7aeea8a1ea48a2b50":{"kind":"REWORK","_number":2,"created":"2023-04-28 20:55:13.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/62/876662/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/876662/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/876662/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/876662/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/876662/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/876662/2"}}},"commit":{"parents":[{"commit":"aaab0e0ad449b3efb5dcd511462e419c9f20c246","subject":"Merge \"Use instance attr instead of non-local\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aaab0e0ad449b3efb5dcd511462e419c9f20c246"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-03-07 03:26:00.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-28 20:42:44.000000000","tz":-420},"subject":"Use signals for timeout on os.wait()","message":"Use signals for timeout on os.wait()\n\nEventlet is pretty good at waiting on I/O, but pretty bad at waiting\non processes -- it enters a poll loop with a non-configurable 10ms\nsleep between checks. This busy-wait approach is tolerable when there\u0027s\n*also* I/O to wait on, but the WSGI server management processes are\nonly waiting on subprocesses. As a result, I regularly see 0.5-1% CPU\nusage per manager process despite there being no changes in child\nprocesses or rings.\n\nNow, use the real, blocking os.wait() instead of eventlet -- but set a\nsignal timer to break us out of it periodically so we can still check\nfor ring changes.\n\nChange-Id: I227ca4f16c4314fc91663a25f2bb4623c6d761ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0829f0fdc3dc87cc8c824f7aeea8a1ea48a2b50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0829f0fdc3dc87cc8c824f7aeea8a1ea48a2b50"}]},"branch":"refs/heads/master"},"899c38aa7e6cf4f1d846e3fd55cb1d2242fe70ac":{"kind":"REWORK","_number":3,"created":"2023-10-20 21:03:41.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/62/876662/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/876662/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/876662/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/876662/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/876662/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/876662/3"}}},"commit":{"parents":[{"commit":"195f54ecb5c7520b7ae19157be77047a0f4e68ea","subject":"Merge \"Add a swift-reload command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/195f54ecb5c7520b7ae19157be77047a0f4e68ea"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-03-07 03:26:00.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-10-20 21:01:09.000000000","tz":-420},"subject":"Use signals for timeout on os.wait()","message":"Use signals for timeout on os.wait()\n\nEventlet is pretty good at waiting on I/O, but pretty bad at waiting\non processes -- it enters a poll loop with a non-configurable 10ms\nsleep between checks. This busy-wait approach is tolerable when there\u0027s\n*also* I/O to wait on, but the WSGI server management processes are\nonly waiting on subprocesses. As a result, I regularly see 0.5-1% CPU\nusage per manager process despite there being no changes in child\nprocesses or rings.\n\nNow, use the real, blocking os.wait() instead of eventlet -- but set a\nsignal timer to break us out of it periodically so we can still check\nfor ring changes.\n\nChange-Id: I227ca4f16c4314fc91663a25f2bb4623c6d761ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/899c38aa7e6cf4f1d846e3fd55cb1d2242fe70ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/899c38aa7e6cf4f1d846e3fd55cb1d2242fe70ac"}]},"branch":"refs/heads/master"},"a6808aa5a8ab1bef3b8148d33a5b4530764f827c":{"kind":"REWORK","_number":4,"created":"2025-05-22 22:24:14.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/62/876662/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/876662/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/876662/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/876662/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/876662/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/876662/4"}}},"commit":{"parents":[{"commit":"f64269b981753c21c87c3ec7c2b4bbcd5978292c","subject":"Merge \"Object-server: add labeled timing metrics for other object server requests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f64269b981753c21c87c3ec7c2b4bbcd5978292c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-03-07 03:26:00.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-05-22 22:23:58.000000000","tz":-420},"subject":"Use signals for timeout on os.wait()","message":"Use signals for timeout on os.wait()\n\nEventlet is pretty good at waiting on I/O, but pretty bad at waiting\non processes -- it enters a poll loop with a non-configurable 10ms\nsleep between checks. This busy-wait approach is tolerable when there\u0027s\n*also* I/O to wait on, but the WSGI server management processes are\nonly waiting on subprocesses. As a result, I regularly see 0.5-1% CPU\nusage per manager process despite there being no changes in child\nprocesses or rings.\n\nNow, use the real, blocking os.wait() instead of eventlet -- but set a\nsignal timer to break us out of it periodically so we can still check\nfor ring changes.\n\nChange-Id: I227ca4f16c4314fc91663a25f2bb4623c6d761ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a6808aa5a8ab1bef3b8148d33a5b4530764f827c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a6808aa5a8ab1bef3b8148d33a5b4530764f827c"}]},"branch":"refs/heads/master"},"e5af3a97ea5fa85ded34154c082cb3eaf73c96a9":{"kind":"REWORK","_number":5,"created":"2026-07-09 01:53:11.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/62/876662/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/876662/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/876662/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/876662/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/876662/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/876662/5"}}},"commit":{"parents":[{"commit":"9f5881d5143fd06a328bde2713cd9c61df579289","subject":"AUTHORS/CHANGELOG for 2.38.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f5881d5143fd06a328bde2713cd9c61df579289"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-03-07 03:26:00.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-07-09 01:53:03.000000000","tz":-420},"subject":"Use signals for timeout on os.wait()","message":"Use signals for timeout on os.wait()\n\nEventlet is pretty good at waiting on I/O, but pretty bad at waiting\non processes -- it enters a poll loop with a non-configurable 10ms\nsleep between checks. This busy-wait approach is tolerable when there\u0027s\n*also* I/O to wait on, but the WSGI server management processes are\nonly waiting on subprocesses. As a result, I regularly see 0.5-1% CPU\nusage per manager process despite there being no changes in child\nprocesses or rings.\n\nNow, use the real, blocking os.wait() instead of eventlet -- but set a\nsignal timer to break us out of it periodically so we can still check\nfor ring changes.\n\nChange-Id: I227ca4f16c4314fc91663a25f2bb4623c6d761ca\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e5af3a97ea5fa85ded34154c082cb3eaf73c96a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e5af3a97ea5fa85ded34154c082cb3eaf73c96a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
