)]}'
{"id":"openstack%2Fswift~877025","triplet_id":"openstack%2Fswift~stable%2F2023.1~I52c535bd2078f25bb57c42a2c2c966247cd3094b","project":"openstack/swift","branch":"stable/2023.1","hashtags":[],"change_id":"I52c535bd2078f25bb57c42a2c2c966247cd3094b","subject":"Drop bandit check B309","status":"MERGED","created":"2023-03-11 00:23:00.000000000","updated":"2023-03-13 18:05:43.000000000","submitted":"2023-03-13 18:04:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877025","cherry_pick_of_change":877131,"cherry_pick_of_patch_set":1,"meta_rev_id":"5f97ff35f6cbaf1c47ebb73ae26defd23491bae9","_number":877025,"virtual_id_number":877025,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-13 18:04:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-13 16:11:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-13 16:11:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-11 01:22:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ed28e2d4a9643a20abb92638c9eef948ab7bce1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-03-11 00:23:00.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"517cf2821a8e3760811f57a4392ac196b79bb73c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-11 01:22:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e1f512b68394e0fbeb734bb9c5e4162\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a3938099c7d4ecca63af74d068fd827 : SUCCESS in 7m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9f18950369874bb7867d83ebbf8dcc70 : SUCCESS in 19m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c2efde53d10041a79d64cab2fd162439 : SUCCESS in 18m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8644b1a579b94303ad2948030d57df6c : SUCCESS in 21m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1303f7fa2949457398bc2adda6c3d4a7 : SUCCESS in 7m 59s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1f4fd3bc78204a4fabc1e326c6b29117 : SUCCESS in 4m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f4b38acc290f45e1b3d5fba05e00c0dc : SUCCESS in 21m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bddbb96bfe2d4caab50090e32f498c43 : SUCCESS in 18m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5af2b2a29b9440f1bbfad22bf3d7ee62 : SUCCESS in 13m 58s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/bd6de28c54a8406ea823e7f231af1697 : SUCCESS in 14m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/67ab27c2bb1146cb9f269454e45ef8ef : SUCCESS in 18m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4598a3cc2fe642c4b67dfd9801f4683c : SUCCESS in 21m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/91570cfeb63c4129867967af3ef7f5da : SUCCESS in 19m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4c6921cefc4d4bcbb784e5a8049474df : SUCCESS in 18m 01s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8a2b209f062048039ca48957f9d183cd : SUCCESS in 16m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b3baa71cdd654056b531a79ab1216b4a : SUCCESS in 15m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8e662b5107584181811014f603e569cb : SUCCESS in 6m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/748eb537750648d5aedc941d2a2f5859 : SUCCESS in 16m 28s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ba83c3115da84e449119566357b04872 : SUCCESS in 5m 55s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/368b9381c45346e5a8868c99805a4428 : SUCCESS in 58m 43s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/579a103427854695b048b5e593af8c17 : SUCCESS in 55m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d96541e21634239bfd8495e706cc092 : SUCCESS in 17m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa5d105ce4e54adab2963977e4d47765 : SUCCESS in 4m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/17ff2f1e711e4d698088ac6c0aa5fa25 : SUCCESS in 21m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"617a4f02ea3ecb92507037bc40a40ee1c5a3fbbe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-11 01:26:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4150739aacc44d1cb2aeda66825cc079\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/60950338b5dc4ec7b179d860827a9f6d : SUCCESS in 1h 01m 21s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/238a574348b34c1f9bcaaf76be99f4ae : SUCCESS in 1h 00m 46s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/221919da227f4964b0c0a69ce177a1b6 : SUCCESS in 53m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ae11f806d5c1472fa274fa5bd8836455 : SUCCESS in 43m 18s","accounts_in_message":[],"_revision_number":1},{"id":"1162d475da81c34116f99382f5d9ef976622b4c2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-03-13 16:11:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3e61833c7d1cff72000572ee5fda107410f1e9c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 16:11:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"53b86a5e267af0d71ae9def62a9c261640a4fbcb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 18:04:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/343ea974dbdd4ad5a5026c711057ef87\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4c878053bfb471fb58553ba2e74adb6 : SUCCESS in 9m 01s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4d4e76e2bf7e4682a666ae4f451c6b1c : SUCCESS in 8m 59s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/63fb2648ebba42568408af7fc2a6ebbf : SUCCESS in 5m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3b38742cfc0b4eeab300508373bace64 : SUCCESS in 15m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cdf9420f86a84a9897c9fed4624e4f53 : SUCCESS in 16m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3afa2205a0454782a4a139181617c42f : SUCCESS in 15m 44s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d2be1a25c6a94ea68099d3fca67efa88 : SUCCESS in 14m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/11dd3a29afcd4df79859eb25e9265915 : SUCCESS in 16m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9a1eee9eadb34ccb8d21ddae901aa27a : SUCCESS in 16m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/42d8242b409645ab9c1c93449cbeda6e : SUCCESS in 16m 44s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/04bbb4f51db441fca6b3b3a0e17bc19f : SUCCESS in 15m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1b2edec8069943a3b87af25c1fe42d5e : SUCCESS in 15m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8fef7c7508634012b0bc49a3818abf34 : SUCCESS in 14m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/744786afde1e43d88947d0c794578c17 : SUCCESS in 6m 24s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0f7149a0284947de908f5cc8926d3c5d : SUCCESS in 8m 57s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1052f60f46584640810859948a37ae5a : SUCCESS in 51m 03s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d95c307558f846038233f9770443e31c : SUCCESS in 1h 05m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e94ef57a1a8246b0be80c6a0fd1971a3 : SUCCESS in 15m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d170797fbaf14b7f9493ec5154912eb6 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":1},{"id":"183a046d063ad6275b81b57ef02dfa25f425ca7d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 18:04:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5f97ff35f6cbaf1c47ebb73ae26defd23491bae9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 18:05:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3ea9dd7da5d412a8bc00374e0e7920d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98094990cac84b3fb043ff839df4b902 : SUCCESS in 43s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/72fee2a3d262458d8eb33de19f9066bb : SUCCESS in 31s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/3cd502290876439895ccb3e116ebcb74 : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"81454ab1e7463c23fe216636a2a3ce9b04f480e4","revisions":{"81454ab1e7463c23fe216636a2a3ce9b04f480e4":{"kind":"REWORK","_number":1,"created":"2023-03-11 00:23:00.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/25/877025/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/877025/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/877025/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/877025/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/877025/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/877025/1"}}},"commit":{"parents":[{"commit":"049ffd5a89f9e9b9de3ac51c5d34cf81e9d9bd06","subject":"Authors/ChangeLog for 2.31.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/049ffd5a89f9e9b9de3ac51c5d34cf81e9d9bd06"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-03-10 20:33:28.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2023-03-11 00:23:00.000000000","tz":0},"subject":"Drop bandit check B309","message":"Drop bandit check B309\n\nIt was retired upstream: https://github.com/PyCQA/bandit/pull/858\n\nContinuing to list it causes CI failures like\n\n    ERROR\tUnknown test found in profile: B309\n\nChange-Id: I52c535bd2078f25bb57c42a2c2c966247cd3094b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81454ab1e7463c23fe216636a2a3ce9b04f480e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81454ab1e7463c23fe216636a2a3ce9b04f480e4"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
