)]}'
{"id":"openstack%2Fswift~877038","triplet_id":"openstack%2Fswift~master~I7ee960dee1329aee6ea535c6c64634cc05ca679a","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I7ee960dee1329aee6ea535c6c64634cc05ca679a","subject":"Authors/ChangeLog for 2.31.1","status":"MERGED","created":"2023-03-09 20:20:39.000000000","updated":"2023-03-09 22:09:52.000000000","submitted":"2023-03-09 21:34:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877038","meta_rev_id":"230f6069a04ebd16654badb1c09b031f2afd9262","_number":877038,"virtual_id_number":877038,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-09 21:34:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-09 20:36:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-09 20:36:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-09 20:59:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"32bd2ae29edf6137fc854b83617fba081e30fa16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-03-09 20:20:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba1a4c8de2d799ce7f727fe9e6cce67e4649f94d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-03-09 20:36:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1f35510857f93c93e36627a9737856bbb0e943e4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-03-09 20:36:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c8043f1c1ebcb2fbe0c8c151ad384f0d76d18294","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 20:59:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d169d65a035f4736ae15ed7b136ff996\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/087c7f7063f0423fac499ac3e981e49b : SUCCESS in 8m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8cafd801738e4ae3a51836e486784f4e : SUCCESS in 5m 10s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1c9aa3f1c1c24a138ce40c39cff206c6 : SUCCESS in 8m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d4a44519656a46d1b588288d1ce48cff : SUCCESS in 5m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/07fb2710329f434fad93863b11371e46 : SUCCESS in 17m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/de7e9ac679c34c718db286ae5cf35a66 : SUCCESS in 16m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dd60f54a1a7e4867a47a21ee0f10d42c : SUCCESS in 15m 24s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/65fd5d88e22f4e00b0c249d2c526010d : SUCCESS in 15m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/654a6477bf364e14bd1f616bd037317b : SUCCESS in 16m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a622af38174c4d4783a5a59778b36c70 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":2},{"id":"69c775e1ca23c5042016c77c2b482b2ba278e082","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 20:59:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4b257e1afed74a4a7aa88ee58eccff1087abb314","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 21:34:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7401c2d8dac403982b9f28f543bd9f7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f22def13422845a381166bfe24ed553e : SUCCESS in 9m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3938684393374ce6a214d899ec7ed08d : SUCCESS in 5m 25s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/e9f7b0bbd62c46a7aa005070a5d89410 : SUCCESS in 10m 23s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/b777d62e7f5546508453f08cbb4d1960 : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/24ea0274ef1243e69e84cd9fdd06184d : SUCCESS in 16m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bdee300744f147f38003a97d090968d6 : SUCCESS in 12m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4fcbb0bd905846e38029b19ff6d76d68 : SUCCESS in 17m 38s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d36dd765f6044fcbadbc1397d24b6216 : SUCCESS in 17m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cdb71b476b514aab88b889c32729ed56 : SUCCESS in 16m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bb508a76e193483ca069110886b2a1e2 : SUCCESS in 19m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e3a1bee8396442b8a74a88bc10f49f1b : SUCCESS in 19m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/58f4027df08c4f47a99edcfe75964c05 : SUCCESS in 14m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e839ad7c3c7e4b4db931d6a352904423 : SUCCESS in 15m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1e27ead2ac714fe398ab1b599e871d36 : SUCCESS in 16m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/370bbcacf8eb4766b28fbb003b798f76 : SUCCESS in 6m 37s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/431084709e2040898f72699502a07a1c : SUCCESS in 8m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/518645f96e754aee8acf86984fea63b2 : SUCCESS in 17m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eae415b7ff554f52a0dbfcfb33304e61 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":2},{"id":"d77374e1e1f0efeff021310a0ff9f521abc2913e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 21:34:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a56349350d12febcd4e9faa12c52f6f175ca97fb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 21:35:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/703c583667f5453a9c766907df9e3bab\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ca980da96a342d3a8e34eab864bb8f2 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b385256589474e09b4cf3f427b138368 : SUCCESS in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/043919f671144acdb0fcaed0ab8d4b96 : SUCCESS in 32s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/d06ddc2a2dba4762b0c69eb67ba17674 : SUCCESS in 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"230f6069a04ebd16654badb1c09b031f2afd9262","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 22:09:52.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5483bef6b7e41aab86508d0247146f8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/064a20934a0941a4ba228b117933f108 : TIMED_OUT in 1h 30m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/23bdbe1827b14a458edc6f6aa5a48a53 : SUCCESS in 1h 25m 48s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/502afa5432ac4c9ba13f513bec92aec2 : TIMED_OUT in 1h 30m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fa25e349bf3c4d37a4db918f0f192f88 : TIMED_OUT in 1h 30m 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"049ffd5a89f9e9b9de3ac51c5d34cf81e9d9bd06","revisions":{"361aaad60b8d7a21ec94207af8c4d7e1ce39fe4f":{"kind":"REWORK","_number":1,"created":"2023-03-09 20:20:39.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/38/877038/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/877038/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/877038/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/877038/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/877038/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/877038/1"}}},"commit":{"parents":[{"commit":"28cb0c30d90c4e227d9880fa9492729d2c08fbfd","subject":"Merge \"quotas: Move account-level handling to a separate function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28cb0c30d90c4e227d9880fa9492729d2c08fbfd"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-03-09 20:16:38.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-03-09 20:20:25.000000000","tz":-480},"subject":"Authors/ChangeLog for 2.31.1","message":"Authors/ChangeLog for 2.31.1\n\nChange-Id: I7ee960dee1329aee6ea535c6c64634cc05ca679a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/361aaad60b8d7a21ec94207af8c4d7e1ce39fe4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/361aaad60b8d7a21ec94207af8c4d7e1ce39fe4f"}]},"branch":"refs/heads/master"},"049ffd5a89f9e9b9de3ac51c5d34cf81e9d9bd06":{"kind":"REWORK","_number":2,"created":"2023-03-09 20:36:20.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/38/877038/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/877038/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/877038/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/877038/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/877038/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/877038/2"}}},"commit":{"parents":[{"commit":"28cb0c30d90c4e227d9880fa9492729d2c08fbfd","subject":"Merge \"quotas: Move account-level handling to a separate function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28cb0c30d90c4e227d9880fa9492729d2c08fbfd"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-03-09 20:16:38.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-03-09 20:36:12.000000000","tz":-480},"subject":"Authors/ChangeLog for 2.31.1","message":"Authors/ChangeLog for 2.31.1\n\nChange-Id: I7ee960dee1329aee6ea535c6c64634cc05ca679a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/049ffd5a89f9e9b9de3ac51c5d34cf81e9d9bd06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/049ffd5a89f9e9b9de3ac51c5d34cf81e9d9bd06"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
