)]}'
{"id":"openstack%2Fswift~877703","triplet_id":"openstack%2Fswift~master~Icd62528e5c19d5f84c79360ea766336999b93e6e","project":"openstack/swift","branch":"master","topic":"listing_sr_cache","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-03-17 14:53:51.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Icd62528e5c19d5f84c79360ea766336999b93e6e","subject":"sq: refactor code into NamespaceBoundsList","status":"ABANDONED","created":"2023-03-16 17:54:48.000000000","updated":"2023-03-17 14:53:51.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9f61b38926f2ded8a6b01529d8481d14ce4303ee","_number":877703,"virtual_id_number":877703,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-03-17 13:55:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-16 17:55:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-16 17:55:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-16 17:55:11.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"bd1d8a987fe534e77f49de66f52a402b008f1397","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-03-16 17:54:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2052e9dd1b93e024c966d343afde5ba3fbe74030","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 17:55:05.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 876939,3","accounts_in_message":[],"_revision_number":1},{"id":"30187b960620cdc9d5387b0505e180ad9476a841","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 17:55:06.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 876939,3","accounts_in_message":[],"_revision_number":1},{"id":"144a307e12823444b288843ac11096c016ce940f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-03-16 17:55:11.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/swift for 876939,3","accounts_in_message":[],"_revision_number":1},{"id":"bcdc658db4d42ee98152f806fb78f01e2c561295","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-03-17 12:37:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a8a68663d30e5ce59a662998e012948e1acd8080","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 13:46:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/525db9aecbe84feeb71b1cba46e9a46e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/61717427f5494f118daeee9842e8e8c2 : SUCCESS in 17m 19s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/faa04a35b1c24ce88d197ed4bce75332 : SUCCESS in 1h 06m 29s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0eb8b3f34e784f24a885eb0c03e790c5 : SUCCESS in 28m 34s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5f58d9610d7642b798c6c491aeec8061 : SUCCESS in 1h 02m 16s","accounts_in_message":[],"_revision_number":2},{"id":"d9d9d88751a145db901e18d770c079f9959354bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 13:55:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d536f359aaf94ff59cd49f7858b765b9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e67276fa3ab24437bed5e8f2f1b985fa : SUCCESS in 8m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/f817840c0aef47fd81fe17f66c7d2e8d : SUCCESS in 1h 02m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/06222b1f27c24cc889e2f582edf81052 : SUCCESS in 1h 00m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3cb2f9d257894861bfb3dd9e928ad854 : SUCCESS in 57m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c27ed263cb4d4c8fb07be0c05018550f : SUCCESS in 20m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b5c7a0b0069940aab72a327ed8daa309 : SUCCESS in 19m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/824be86e0f0c402c8ba5a1f8af59a508 : SUCCESS in 19m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/794bb447dd6c489eaa7b78b9dbc32d16 : SUCCESS in 9m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f1c0a28cbb3d4812af15a8f5e4c911a1 : SUCCESS in 6m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/23034db31f1f4c289f2130feb7b54358 : FAILURE in 19m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/903d2ae9a693468588d8621ba7a820a3 : SUCCESS in 19m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/988f0a0888d64f6594d0a15d97629a68 : SUCCESS in 15m 44s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c8e050affccf426b904950a7e6143f95 : SUCCESS in 17m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2aa27cb705b2421a9b5e6c928eacf83b : SUCCESS in 14m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/defdc73e6718475b8f8bba521ad99441 : SUCCESS in 17m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c74332c7bd374267bf998b8a9f4766a6 : SUCCESS in 17m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/22aaccbfb82b4b858dd43dd37cec9856 : SUCCESS in 14m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/aef5a47db24e4a518f0cfd965c23e5f9 : SUCCESS in 15m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/80ba60c38cae4258a2dabe892056b67b : SUCCESS in 14m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/38e50ebe07c24a3bade5498608ef2b94 : SUCCESS in 6m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e43802269eaf46769720075ef981f4e6 : SUCCESS in 16m 54s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/048f3a1fa4fa48958e6696774563be48 : SUCCESS in 7m 22s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b213425c49e14d19abbcda27b3462438 : FAILURE in 1h 02m 57s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/720cf045e49b45bdbaa6e92e93b9ad56 : SUCCESS in 1h 06m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9815d58e15a244f082521f5c589d1181 : SUCCESS in 55m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c1d2d10575864a9585296112d1b4e1a4 : SUCCESS in 55m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d45f2a67725499e9ee550f6f23a0ac3 : FAILURE in 19m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bab311bcc1543f88b741643b88a6d06 : SUCCESS in 4m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/eddb202c5242485d9f0cdee394e1a160 : SUCCESS in 18m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77907c4c9e6645bdbd5ba50ba933edce : SUCCESS in 54m 41s","accounts_in_message":[],"_revision_number":2},{"id":"9f61b38926f2ded8a6b01529d8481d14ce4303ee","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-03-17 14:53:51.000000000","message":"Abandoned\n\nsquashed docstrings into parent","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f197e4641b35b777a7cde6ad7032413584093988","revisions":{"f5ab38f552517a7ec0dffdc90ac4a90f3acf10d6":{"kind":"REWORK","_number":1,"created":"2023-03-16 17:54:48.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/03/877703/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/877703/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/877703/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/877703/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/877703/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/877703/1"}}},"commit":{"parents":[{"commit":"483de79aeffd91ea13919eeb1a0b1c49bc9ce31b","subject":"Proxy: restructure cached listing shard ranges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/483de79aeffd91ea13919eeb1a0b1c49bc9ce31b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-03-16 17:53:43.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-03-16 17:53:43.000000000","tz":0},"subject":"sq: refactor code into NamespaceBoundsList","message":"sq: refactor code into NamespaceBoundsList\n\nplus some docstrings\n\nChange-Id: Icd62528e5c19d5f84c79360ea766336999b93e6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5ab38f552517a7ec0dffdc90ac4a90f3acf10d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5ab38f552517a7ec0dffdc90ac4a90f3acf10d6"}]},"branch":"refs/heads/master"},"f197e4641b35b777a7cde6ad7032413584093988":{"kind":"REWORK","_number":2,"created":"2023-03-17 12:37:42.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/03/877703/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/877703/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/877703/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/877703/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/877703/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/877703/2"}}},"commit":{"parents":[{"commit":"1bce37fdecee84f5ddaec6a4a5b7e307f7701c6e","subject":"Proxy: restructure cached listing shard ranges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1bce37fdecee84f5ddaec6a4a5b7e307f7701c6e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-03-16 17:53:43.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-03-17 12:36:48.000000000","tz":0},"subject":"sq: refactor code into NamespaceBoundsList","message":"sq: refactor code into NamespaceBoundsList\n\nplus some docstrings\n\nChange-Id: Icd62528e5c19d5f84c79360ea766336999b93e6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f197e4641b35b777a7cde6ad7032413584093988"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f197e4641b35b777a7cde6ad7032413584093988"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
