)]}'
{"id":"openstack%2Fswift~878476","triplet_id":"openstack%2Fswift~master~I155bd0d64b9662fe79323c9a5d3dbfe8e9109d79","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I155bd0d64b9662fe79323c9a5d3dbfe8e9109d79","subject":"Forward-port stable-release CHANGELOG entries","status":"MERGED","created":"2023-03-23 18:07:46.000000000","updated":"2023-03-23 20:02:51.000000000","submitted":"2023-03-23 20:00:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878476","meta_rev_id":"0059bbe9cd395442e7f3e44fed5f3259ac4b7302","_number":878476,"virtual_id_number":878476,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-23 20:00:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-23 18:34:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-23 18:34:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-23 18:29:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-23 18:39:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a65d71233b51bde1aec87d69a8bd7883b464a6fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-03-23 18:07:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"574e9ce1f1793484d436c0624da04b60448ff0e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 18:29:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a1542b918f74836880bb74e6fb0ea1c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c4841f954638482da2e69a2235ff88ab : SUCCESS in 14m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/47c2be2a8b5544a5b45ef1862d6bd4c5 : SUCCESS in 14m 04s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/02c0e8fed890458dbb56b1893c951750 : SUCCESS in 20m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c0c17eb0c2046f7968cc65f760df5b8 : SUCCESS in 19m 42s","accounts_in_message":[],"_revision_number":1},{"id":"8b68116482050e5daccf55e9d3abb1658138b137","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-03-23 18:34:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"411debf815aa3b2a1c93de780f3fe185ee49598d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 18:39:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c78f26486a71498cb0fa69f04ff58aab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df834d0578e64387a82cc985720e0770 : SUCCESS in 8m 16s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1273d23678b84d209a680a21c798ab2b : SUCCESS in 8m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ca95471f5c354679aaca0e3603af8afc : SUCCESS in 5m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/96c379ac117f4199bfe9bedbc46e99cb : SUCCESS in 17m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ca4b40ce64de4f8683a2e481b3026a4d : SUCCESS in 28m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d919ae8c2eae4e23bfcb98ba46390c30 : SUCCESS in 14m 50s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e046cc552fd04054bc12c840f666af18 : SUCCESS in 30m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/030a21a3f1cf4dfcab6263db884bd271 : SUCCESS in 17m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c75f3ab5489495895bd744884660a44 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":1},{"id":"7a2ef6061338a511c5951a1ac907e6a95bd43bcf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 18:40:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"10ad70066c001c47cc7ead941d9ac799a5cd9f65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 20:00:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81d9280e471840ba9cc0867425270b25\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3094bdd018dc407b92c012418d470a83 : SUCCESS in 9m 24s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/77aa0a0448534bd1b5c00e709ab37dd2 : SUCCESS in 9m 10s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/2c9aa2db71024e0e85ca1bbde926c73e : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6e4c283b9479400caee53a915f99a155 : SUCCESS in 16m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/083866e30f854507ae202575e1def5c3 : SUCCESS in 16m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/942be8b7a75c4be7aa81f7b74509042f : SUCCESS in 16m 26s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6da43dbb9271487c91b3e7a13222f865 : SUCCESS in 15m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6d01fff8be0d4d6284c0830246922782 : SUCCESS in 16m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/370e151a61f74446b65873252538d25a : SUCCESS in 17m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7e2d54372d6b483685bc37630695ffd7 : SUCCESS in 17m 15s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/afd4f3ac93924593bac6a5c0004bf4b8 : SUCCESS in 14m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ef380fe9e63d4b27983d75e1bf5f0341 : SUCCESS in 15m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d4809e24c95e4267b62a8168ae3d8a46 : SUCCESS in 15m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ce8f195727bb495a9ee2d9232b3910f4 : SUCCESS in 6m 57s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a7fab5fc9aa7487a8cccca12273f8f28 : SUCCESS in 7m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5af36b4a3ee14bea95ccf6d72751b283 : SUCCESS in 17m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/875ae9d5752c413d91e6ead82eb1a27a : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":1},{"id":"4378c2cbcf4ba025bce8f1240e1432a48ec154ee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 20:00:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0059bbe9cd395442e7f3e44fed5f3259ac4b7302","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 20:02:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0242dc5ccc6946a3878d0736d4889f28\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c38f1a9f8c6549e4bbc37addcffc23e6 : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/eed37c1979b54450b2a3758f5249bf75 : SUCCESS in 35s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/03579e787c2240759d11de4055318f99 : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"56a9f72ec7d2720cc213a17c1f12fa3511e29576","revisions":{"56a9f72ec7d2720cc213a17c1f12fa3511e29576":{"kind":"REWORK","_number":1,"created":"2023-03-23 18:07:46.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/76/878476/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/878476/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/878476/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/878476/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/878476/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/878476/1"}}},"commit":{"parents":[{"commit":"2dc9899b147663225540a71a5a591587c2284f6d","subject":"Merge \"Allow internal container POSTs to not update put_timestamp\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2dc9899b147663225540a71a5a591587c2284f6d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-03-23 18:07:37.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-03-23 18:07:37.000000000","tz":-420},"subject":"Forward-port stable-release CHANGELOG entries","message":"Forward-port stable-release CHANGELOG entries\n\nChange-Id: I155bd0d64b9662fe79323c9a5d3dbfe8e9109d79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/56a9f72ec7d2720cc213a17c1f12fa3511e29576"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/56a9f72ec7d2720cc213a17c1f12fa3511e29576"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
