)]}'
{"id":"openstack%2Fswift~878593","triplet_id":"openstack%2Fswift~stable%2Fyoga~I650a80cb45febc457c42360061faf3a9799e6131","project":"openstack/swift","branch":"stable/yoga","topic":"update-url-stable/yoga","attention_set":{},"removed_from_attention_set":{"33787":{"account":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"last_update":"2023-05-04 23:03:44.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-04-24 17:57:04.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I650a80cb45febc457c42360061faf3a9799e6131","subject":"Update url","status":"MERGED","created":"2023-03-27 02:00:08.000000000","updated":"2023-05-04 23:04:51.000000000","submitted":"2023-05-04 23:03:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"878593-update-url-stable/yoga","cherry_pick_of_change":878513,"cherry_pick_of_patch_set":1,"meta_rev_id":"076850127dad9b74d0c4ae2eba3782254f04ea02","_number":878593,"virtual_id_number":878593,"owner":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-04 23:03:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-24 17:57:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-24 17:57:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-27 03:07:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-28 10:01:00.000000000","updated_by":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"80a34330eea85572de12f9b2919cd74633c80e3e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"date":"2023-03-27 02:00:08.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* doc/source/getting_started.rst","accounts_in_message":[],"_revision_number":1},{"id":"42952a2e706e10500cfd2ffbd9d4eb121a217e2f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"date":"2023-03-27 02:06:07.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"1ab51588f943be8828b24db605728b694fd24b10","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"date":"2023-03-27 02:07:09.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"2f9f4cf44db3ce1e762e7c589360d09479bcbbf8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 03:07:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6c9f253ae0874265b6a52b3d28a12218\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/061e3c046ec14f10a8803af9a06b5e72 : SUCCESS in 2m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c0b425bd84a42ce9d175a69a73279ba : SUCCESS in 9m 45s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6b2671e3f36546c1b67eb2563e865f87 : FAILURE in 4m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/69af6f9b499a409fbb2e4f4a22d6eed3 : FAILURE in 4m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0432bfafc2f6463c90671239205e43c0 : SUCCESS in 14m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/70a8d8ce3ad948d4ae00dd28f992f154 : SUCCESS in 14m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e04597c2401d488587a9e608067849ef : SUCCESS in 18m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7fe724d62feb4de49c97803ef90e2351 : SUCCESS in 8m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/045802878b6e485f8e2388e90c8025ed : SUCCESS in 16m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/626ab68ad5744c7ea2896e51d79d0efe : FAILURE in 5m 09s","accounts_in_message":[],"_revision_number":2},{"id":"6af7b94e8c2d8cc27f5ba963162adca2e02153fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 05:07:00.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46fe6fa50cc140bab89ecbe3aa959dfd\n\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/a65cfe2005a542e39377c5f96f80aaad : FAILURE in 2h 53m 55s","accounts_in_message":[],"_revision_number":2},{"id":"1dd1df810d101d2e5ac5521f059bb3ff080590c5","author":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"date":"2023-03-28 02:06:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3723dae81a1672d9a1751dfd35ad9a1bab74a2f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 03:06:39.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/29662e86a53f4a659f8c7de9bff0eabc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c21719f44a7f4991b0a343786887a18b : SUCCESS in 2m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d37fe13da15348058c7015a2e02c33cf : SUCCESS in 8m 39s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7d3358d4e3c04b73a421767673318866 : FAILURE in 4m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8fc54f3cd3914809991ca4e12397e97b : FAILURE in 3m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a19644c79e6f49afaa4869e090c3dd24 : SUCCESS in 26m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1a1165ae4fd54b1c88e43a61cf6a70d0 : SUCCESS in 16m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bbda9f10ac7b47f68034d44df5c90678 : SUCCESS in 16m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/70e9fec929e54d74a0d5759091edab9e : SUCCESS in 7m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aeaf3b85ef564a4fbed6a17a0be8eb21 : SUCCESS in 16m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9006b8f5bb1645d19734785bef83b975 : FAILURE in 5m 17s","accounts_in_message":[],"_revision_number":2},{"id":"f48ae76552b1cf6e6fe6280ce151b5c6efb3321d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 03:41:22.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fac920977b0f4859a98cb2ac1e12e436\n\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/2457364d3d2a470bbe37087ec0515bfa : FAILURE in 1h 28m 56s","accounts_in_message":[],"_revision_number":2},{"id":"1df6485ffd3dae49e3493d64b5bd02d9f725a2e7","author":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"date":"2023-03-28 06:20:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"78e9d06650d9fb84cc82b08233b5a5fb09e65eb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 06:37:24.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/29bcff9ac6784ee6b734242fe50f5ac8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ae60976820364a46853fd47d0b2286ed : SUCCESS in 2m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/140305f467d2464cb44a449a18189a20 : SUCCESS in 9m 03s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c4651d2eea234ebfb745db511962c470 : FAILURE in 4m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3b7c80660cf24697b0f6a8ff4f907c74 : FAILURE in 4m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8501a409fb64435b865cf08dbf34ba6f : SUCCESS in 15m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8a75a243a8474f3db996c5f398c3f6ee : SUCCESS in 13m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/25b696907e7248e8be9aeb093e551ec3 : SUCCESS in 13m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/476e5764743a4377a9d78abaa3e7da93 : SUCCESS in 6m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9920f0c6720441328bfeff93168efcb5 : SUCCESS in 14m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab1c2c70a7af46dfa2de33962deaf5ba : FAILURE in 5m 01s","accounts_in_message":[],"_revision_number":2},{"id":"4be20dd46515a0762bd11518c943ff9a5ee621e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 07:27:45.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8239f6442524aefb749babf1764b4e7\n\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/a8277f8a82484752966b45531e2d8c80 : FAILURE in 1h 06m 08s","accounts_in_message":[],"_revision_number":2},{"id":"717118a04ee4da4899799dccab834e13e552658e","author":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"date":"2023-03-28 10:01:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"58a9e4b61b642578ad6a41af667b60be25426a58","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-24 17:57:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a2d3070cd67f1e1b7d91a5057105372a029f7202","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 18:24:53.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/03ad9b2b5c94465895b5b3ccbe8ea3b3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/96adc6def373454c8d8916cb092b0f8f : SUCCESS in 1m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74e191520b96492c95e66afa3ac04846 : SUCCESS in 8m 13s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/63bb9da780d542e6b961f1b1df941b7b : FAILURE in 3m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/fe765507691c49b887d8954642f3ae1e : FAILURE in 3m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c92baa4832c64d898f1df8e267a494d2 : SUCCESS in 17m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8f3a193c4e944cccae89a33492adce30 : SUCCESS in 14m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b58573a80a324463ac2cd2e56df3e739 : SUCCESS in 13m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f3b7b88ed57441c5920c7ea04815f808 : SUCCESS in 12m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94e6e19312004393b8ae1fd01b34f87d : FAILURE in 2m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/019d84c0af2b44a7b40e5acbcf0c00aa : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":2},{"id":"8fa85826c7882ff7c66dfe7f08679ca79d67cb13","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 18:58:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ceb184f73994664943604bee6dd563c\n\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/36094421203049f0a53075e69afe6f81 : SUCCESS in 59m 29s","accounts_in_message":[],"_revision_number":2},{"id":"fe51dc28231ff385e069e0717dc54e56c63a99a1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-04 21:27:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b8364aec63c6cfe323f6e1bc2696c69dd4a47714","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 21:54:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2231fe9c47de4922846999a05d27e39c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0090725914144e23b3e993e64cd48d1a : SUCCESS in 1m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6726896b09f64251a1cf0c6535cac44e : SUCCESS in 8m 08s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7eb770c9cab845438109ea1027783fd2 : FAILURE in 3m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5155268a6a7445fba0f20696774f3c3d : FAILURE in 3m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e770af67c2794bf3af420e3d90a0de47 : SUCCESS in 27m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ad4e705f7c394b8295ea33eb9a81eedc : SUCCESS in 15m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d00884fec7eb4be9bf0eeb3f6ec29654 : SUCCESS in 13m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/14e0ddad66874c958e8559e8186c55d6 : SUCCESS in 7m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b768a8ee812041c0899cab3de0836f91 : SUCCESS in 25m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56ad5faeb6b549fcb571d649e355352a : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":2},{"id":"738df5bd32ff422f342e7b8a6a3e4f96b42dccc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 21:55:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6c91c27f6c756f99548c9a2f9fe22df648fefee9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 22:34:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94904630cbb746f3ac51f7f66a13a777\n\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/306a54953c454081bab9fee2a6dd2948 : SUCCESS in 1h 06m 20s","accounts_in_message":[],"_revision_number":2},{"id":"d4384273ba6668bbb629350cb86e3b6cfd6e52a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 23:03:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c73a03305d124641b96f0cdff40c76c1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/502d63577fcd4d229fb7bf7315437818 : SUCCESS in 2m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bde1706203a47d989ce5358be4c1448 : SUCCESS in 8m 00s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/6ca6d0919f764ee3915d413e1d171112 : POST_FAILURE in 5m 19s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/36f5714714fa4094844f4084ac52a611 : POST_FAILURE in 5m 31s (non-voting)\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9b00b83806b3447da8d8d117a2d8106b : SUCCESS in 8m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3143dd420d044c568bd5235870197a04 : SUCCESS in 17m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b47872a10aea4362b669be065d17f052 : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":2},{"id":"e89984fe31fe2340d4c120e882f2ffab26a2c74b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 23:03:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"076850127dad9b74d0c4ae2eba3782254f04ea02","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 23:04:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4b7ad6fdd2f48be87b008dffb976a5b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61a1686afbad4812a3d64f94469cc9b8 : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/23090b559bc44464a560fb6e6c9cbe2c : FAILURE in 30s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/c6ac055bf92b4befb579a448d4e14697 : FAILURE in 34s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d2800e8e97d65faf5b0d3881846049c08650bb82","revisions":{"56e3dcc60056d2e9c1297a6bb7c087482212d364":{"kind":"REWORK","_number":1,"created":"2023-03-27 02:00:08.000000000","uploader":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"ref":"refs/changes/93/878593/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/878593/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/878593/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/878593/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/878593/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/878593/1"}}},"commit":{"parents":[{"commit":"19879831272b37b5e49f73a4d49f1a4587ccd7c7","subject":"Authors/ChangeLog for 2.29.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/19879831272b37b5e49f73a4d49f1a4587ccd7c7"}]}],"author":{"name":"Wei LingFei","email":"weilingfei@uniontech.com","date":"2023-03-24 06:11:44.000000000","tz":480},"committer":{"name":"weilingfei","email":"weilingfei@uniontech.com","date":"2023-03-27 02:00:08.000000000","tz":0},"subject":"Update url","message":"Update url\n\nThe OpenStack project is currently maintained on opendev.org, with github.com serving as a mirror repository.\n\nReplace the source code repository address for the python-swiftclient project from github.com to opendev.org.\n\nChange-Id: I650a80cb45febc457c42360061faf3a9799e6131\n(cherry picked from commit 5032b42b10548bda6d27dba81cfec2f95a67d91b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/56e3dcc60056d2e9c1297a6bb7c087482212d364"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/56e3dcc60056d2e9c1297a6bb7c087482212d364"}]},"branch":"refs/heads/stable/yoga"},"d2800e8e97d65faf5b0d3881846049c08650bb82":{"kind":"REWORK","_number":2,"created":"2023-03-27 02:06:07.000000000","uploader":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"ref":"refs/changes/93/878593/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/878593/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/878593/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/878593/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/878593/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/878593/2"}}},"commit":{"parents":[{"commit":"19879831272b37b5e49f73a4d49f1a4587ccd7c7","subject":"Authors/ChangeLog for 2.29.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/19879831272b37b5e49f73a4d49f1a4587ccd7c7"}]}],"author":{"name":"Wei LingFei","email":"weilingfei@uniontech.com","date":"2023-03-24 06:11:44.000000000","tz":480},"committer":{"name":"weilingfei","email":"weilingfei@uniontech.com","date":"2023-03-27 02:05:50.000000000","tz":0},"subject":"Update url","message":"Update url\n\nThe OpenStack project is currently maintained on opendev.org, with github.com serving as a mirror repository.\n\nReplace the source code repository address for the python-swiftclient project from github.com to opendev.org.\n\nChange-Id: I650a80cb45febc457c42360061faf3a9799e6131\n(cherry picked from commit 5032b42b10548bda6d27dba81cfec2f95a67d91b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2800e8e97d65faf5b0d3881846049c08650bb82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2800e8e97d65faf5b0d3881846049c08650bb82"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
