)]}'
{"id":"openstack%2Fswift~878594","triplet_id":"openstack%2Fswift~stable%2Fwallaby~I650a80cb45febc457c42360061faf3a9799e6131","project":"openstack/swift","branch":"stable/wallaby","topic":"update-url-stable/yoga-stable/wallaby","attention_set":{},"removed_from_attention_set":{"33787":{"account":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"last_update":"2024-03-12 11:32:04.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I650a80cb45febc457c42360061faf3a9799e6131","subject":"Update url","status":"ABANDONED","created":"2023-03-27 02:08:05.000000000","updated":"2024-03-12 11:32:04.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":878593,"cherry_pick_of_patch_set":2,"meta_rev_id":"4bb5134c0c108eb145d5245af99ce9e96936be4f","_number":878594,"virtual_id_number":878594,"owner":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-03-27 03:16:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-27 03:16:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-12 11:32:04.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"c27f406eee6c967b088d9a941a38bd43cf9ee552","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"date":"2023-03-27 02:08:05.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"d0b71c59931ceced8c52cba1b7d5d82a9850a1f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 03:16:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e8114b1282e4fb7a436ff0f37c20df0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2c2ce7249cbb498c84939bf63d69c6fc : SUCCESS in 1m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c67c1b9eebb54a04a07329350d05311f : SUCCESS in 8m 29s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5742555cb37b471d8a6659ce03acfe56 : FAILURE in 4m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9ecc3c83f61e4ab8a368b29887907ca5 : FAILURE in 4m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/27b2b9a1d8df4228a8b5aa66423d8660 : SUCCESS in 31m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/255877b19af04428ab6ea4548aa739e6 : SUCCESS in 22m 24s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/404d0cdcb6134471b7881ebdbfb8e156 : SUCCESS in 15m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c3430501af23442fb225ed25c3886027 : SUCCESS in 14m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/debe01a1bece445f8f42708e0a7f702d : SUCCESS in 14m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bc88c5ca39a940999ff5a4c091b7e3e2 : SUCCESS in 5m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a38bb59416249df83319fe4c9aed3ab : SUCCESS in 14m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5d08ba93ba547c085a8868fefd53c8b : FAILURE in 4m 52s","accounts_in_message":[],"_revision_number":1},{"id":"3670026d207d2f5c1d30642c519814c78ffdac42","author":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"date":"2023-03-28 02:06:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7aca24aeada2b7b49ecb50734f4fd0a23021f65f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 03:20:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1de7ef3a73564d3c9fbc663141cef3f8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1e990bd8fee6482b811c08912f433df2 : SUCCESS in 2m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca964ced99164003902cdf63c4c905f3 : SUCCESS in 9m 43s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ed8a9276a69d4210b14b3d2e9782c141 : FAILURE in 4m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6009cd5b088042baae82fcd67f32d940 : FAILURE in 6m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/860c84fe8bb7462c8accc79be10729cf : SUCCESS in 12m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8845361b734546c28f7588c71faed3be : SUCCESS in 15m 12s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/897ffa187b1c4c75a5d9f491f9504fad : SUCCESS in 14m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5f6a8568f8864bc0ab8d0a257a60be71 : SUCCESS in 15m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bab343d740504c27a74b472807aa9dde : SUCCESS in 15m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2c992eeeffd64dc39433ae185982c003 : SUCCESS in 8m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16bca4239ae244faae7f01160181c305 : SUCCESS in 14m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf1e769b131e446ca3f6437f8e56e15f : FAILURE in 5m 04s","accounts_in_message":[],"_revision_number":1},{"id":"9042ac842b274145c2ffe1d8b1512d88c8c1c359","author":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"date":"2023-03-28 06:25:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2408aa1be4075575bf3004662399260111a3eb2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 06:47:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c32343a4bd49494db75ec0d17df2bf21\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/10eca763e7a44a2aa6fb3c5d696e4b68 : SUCCESS in 1m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c61fa753e4e44c3393246df97063eb8f : SUCCESS in 7m 33s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f1b9f0a0cb8b4c59b2ea51786a4f56fa : FAILURE in 3m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ac6079b736804250a0f35bafbd5855a8 : FAILURE in 3m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c160de23bce3426e9391cce074b8318f : SUCCESS in 16m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4084994143234a1d9356b2fc2d45d5ee : SUCCESS in 14m 29s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f37a6c27ea1749d187f2686d955bb79c : SUCCESS in 17m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/008f95ead827485a9f40564e9bc86322 : SUCCESS in 13m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9c661d9a46934c29a773a147f8ba9968 : SUCCESS in 14m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f2441a8bc07d4d138186a5b1083a6e54 : SUCCESS in 5m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95d7f20380b447efba594ab13c8fc66b : SUCCESS in 13m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7131177da082450d8ec6db81340e930b : FAILURE in 4m 29s","accounts_in_message":[],"_revision_number":1},{"id":"caca6116ac7f08d8fc37fbd073b1da0edb9f6871","author":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"date":"2023-04-11 05:29:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e04c20a2995729bf7331458c15c78dc2eb5652c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 06:01:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a42dffc66ca4afc8098d6b014e5f4db\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/26d7f08ef3aa40a4a5a8fbedcb9d1a3e : SUCCESS in 1m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9de41c7b0c3d42df9c481f327997f134 : SUCCESS in 7m 57s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cf2d72ba64ac4368be990fdeb4748b71 : FAILURE in 3m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/daf9ccc6443c441da5839d832ac07686 : FAILURE in 3m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/46646f69d09c49d28469dd69d15a80a9 : SUCCESS in 19m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7bc8cc96aa0f4d3396de0e88f4f56904 : SUCCESS in 20m 44s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e422acf3f6fe4a75b5e405b733c0dbd6 : SUCCESS in 14m 18s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9cc5519ec32e433a8a7574d8b705dc42 : SUCCESS in 13m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c8007b9d939346428c2d2aebffb9dc3a : SUCCESS in 18m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/95a88f7024f64b248fa54fdb329a263a : SUCCESS in 5m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/774f2e6156c04908b26b33aa041f0409 : SUCCESS in 13m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50e2e8cfd21841a6b6adde7b5d58b5ee : FAILURE in 4m 54s","accounts_in_message":[],"_revision_number":1},{"id":"4bb5134c0c108eb145d5245af99ce9e96936be4f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-12 11:32:04.000000000","message":"Abandoned\n\nstable/wallaby branch of openstack/swift is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/wallaby if you want to further work on this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e3a672e9a7bb330f25e0dddcf8319c4237a081f0","revisions":{"e3a672e9a7bb330f25e0dddcf8319c4237a081f0":{"kind":"REWORK","_number":1,"created":"2023-03-27 02:08:05.000000000","uploader":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"ref":"refs/changes/94/878594/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/878594/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/878594/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/878594/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/878594/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/878594/1"}}},"commit":{"parents":[{"commit":"ead53ba0c89ee9721981fcfc3784b335b4aef0ad","subject":"s3api: Prevent XXE injections","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ead53ba0c89ee9721981fcfc3784b335b4aef0ad"}]}],"author":{"name":"Wei LingFei","email":"weilingfei@uniontech.com","date":"2023-03-24 06:11:44.000000000","tz":480},"committer":{"name":"weilingfei","email":"weilingfei@uniontech.com","date":"2023-03-27 02:08:05.000000000","tz":0},"subject":"Update url","message":"Update url\n\nThe OpenStack project is currently maintained on opendev.org, with github.com serving as a mirror repository.\n\nReplace the source code repository address for the python-swiftclient project from github.com to opendev.org.\n\nChange-Id: I650a80cb45febc457c42360061faf3a9799e6131\n(cherry picked from commit 5032b42b10548bda6d27dba81cfec2f95a67d91b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e3a672e9a7bb330f25e0dddcf8319c4237a081f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e3a672e9a7bb330f25e0dddcf8319c4237a081f0"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
